)]}'
{"id":"openstack%2Fglance~948536","triplet_id":"openstack%2Fglance~master~I462195cef6cd5a13284de22e081652ef1c3bfeb6","project":"openstack/glance","branch":"master","topic":"eventlet-removal","attention_set":{},"removed_from_attention_set":{"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2025-06-16 19:15:09.000000000","reason":"Change was submitted"},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2025-06-16 18:08:36.000000000","reason":"\u003cGERRIT_ACCOUNT_8122\u003e replied on the change","reason_account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}}},"hashtags":[],"change_id":"I462195cef6cd5a13284de22e081652ef1c3bfeb6","subject":"Functional test migration test_wsgi.py","status":"MERGED","created":"2025-04-30 09:01:47.000000000","updated":"2025-06-16 19:16:29.000000000","submitted":"2025-06-16 19:15:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"948536-eventlet-removal","meta_rev_id":"2353fba8b2103bf560c97fa3cf5980f8bdd8e972","_number":948536,"virtual_id_number":948536,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-06-16 19:15:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":2,"date":"2025-06-16 18:08:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":1,"date":"2025-06-16 18:08:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-30 10:07:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-16 15:53:40.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"CC"},{"updated":"2025-06-16 18:08:36.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"079cbb62bb8d263636abb9f59870d6e99569bc3d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-04-30 09:01:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4f6e7855ae61037f061f42a1ce4fc1a56122c756","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-30 10:07:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3db57c615860423698deb78b5449344f\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/58cba0ff2af540c283423be4b4110e18 : SUCCESS in 51m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4dc3879f2e7045aaa6a0cd778e096179 : SUCCESS in 1h 00m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d452862b071d4ac2bf3e225d5f94332c : SUCCESS in 4m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3591d729d39d4349a801d592b4c37f86 : SUCCESS in 4m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/902ac1dbdc3149e09e00a73b158dbdb3 : SUCCESS in 5m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cc3e7cccc0834dc7bc79400af7785820 : SUCCESS in 8m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e069a0b2c2664ab4932ce645fd2529aa : SUCCESS in 4m 53s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6ddf5a66041e41b6bf5aa12b748abfd9 : SUCCESS in 7m 45s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/e4b8e492c1ed4dabbd7629bbf5aad139 : SUCCESS in 11m 03s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/929a90d8f5dc4c8e936d4bbc0a393c4a : SUCCESS in 31m 01s","accounts_in_message":[],"_revision_number":1},{"id":"1618fb5327c0e7a2b7b5d296379918da3e6a180c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-30 10:10:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/670c14876eeb480e93c1f5c192f084f2\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2eb676cbe6624fd4b8b51f9e3d7ff696 : SUCCESS in 6m 30s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/66d74785922a4652b09ab54090967c97 : SUCCESS in 9m 28s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f831feb340ff4cef8765ac01023aaff1 : SUCCESS in 15m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"01205e0a466a4fb191613c5d030ba27e1b46fc25","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-05-20 07:35:32.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"9556e700546ae7b3ca304afd06ffdf5d7f7d60c3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-05-20 07:44:17.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"531d2b50e7255f2f6a5d55d051be05562ade2bed","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 08:00:53.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/20af0587b89c43c8b377eb2939c50fad\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8e9026ba906e4ae3b9208cc6d026f962 : SUCCESS in 8m 15s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/cb508a2e14914e2a945de0219621ec44 : SUCCESS in 9m 59s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/05a5ec21f2074913bc29fb7ae7697074 : SUCCESS in 14m 19s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"aa33a468c9d06fa0ba0a796c885cc7e3af7bb379","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 08:49:03.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d26de02ff8fe44a68fca7c2c7947288f\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/33b19d17e1054699bb2a4238626fba0f : SUCCESS in 59m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4162d817e6984d07949dd39680cec572 : SUCCESS in 1h 00m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7208edce881f4ded86e891e5610f1cc8 : SUCCESS in 4m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b4061659d8514b7984aaade537a03592 : SUCCESS in 4m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f92d8aa01f1142c48e02bfd3e8ee5bac : SUCCESS in 5m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/73855b2233634d56bce8bc580b1a5e73 : SUCCESS in 8m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c7a210328944cf2b4b8563aa1a22ad2 : SUCCESS in 5m 51s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/fb63a4ffaa8343f9b99189bb7fbf1a05 : SUCCESS in 7m 07s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/9c540e4ae5ba4e14bf3c39b429a526be : SUCCESS in 9m 42s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/d9ce21b2422e444983a08346f472a960 : SUCCESS in 31m 26s","accounts_in_message":[],"_revision_number":3},{"id":"2d17abf47fd8703780bcef3004574698e11e20fa","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-06-16 15:53:40.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"6ab2c7f25686290e80cf18c6dc69e3af2bc49c35","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-06-16 17:12:17.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"b077f2dd8c573a9b6d693b107f07bc9ab094bfaa","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-06-16 17:13:18.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"10eacbf8d0d8cfcd0a9d5399de5dcabb5f0f837c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-16 17:31:08.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ec99cf34f83427988090f5ae123bf00\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/72abdf09aac5436fa62462814b3d8e4f : SUCCESS in 10m 23s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/75f2b55177a247868f4202abfa901552 : SUCCESS in 9m 02s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/5f14eb9ebb7f4b019e71196109f555d6 : SUCCESS in 16m 27s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"09c41cf0bd2f18664a2ec0897869d87868d5b311","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-06-16 18:08:36.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"55a560a8929a095215ccefc02d35cabd13bdbab5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-16 18:09:46.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad6887c14c2244a3a7763d0cf42ce9f0\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3c505ef373f1476da96fb71f877a93fc : SUCCESS in 48m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/195f4b6203a64d69b9fbf17e4749b8e8 : SUCCESS in 51m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/faefb8be020b4766b3d9fd9e8531c95d : SUCCESS in 4m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/790af3c6a9284037b6d0d52d5e582c4f : SUCCESS in 4m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8eee3ff227ca4c72af0c606b9181ff46 : SUCCESS in 5m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9e5fa25ebffb40a2930223d5bd286154 : SUCCESS in 6m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b61e1c218ab4d9098db66912b5ab9eb : SUCCESS in 4m 41s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/892ed93a75394a698eb24ca29dbf964b : SUCCESS in 6m 27s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/ccffb5c340404c8ea828daa2b4447e9a : SUCCESS in 10m 07s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c622ff822f084e71ba2317341ab91656 : SUCCESS in 33m 09s","accounts_in_message":[],"_revision_number":4},{"id":"ed0ece70f44ede276321319479ad61b05508fc19","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-16 18:10:12.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"b5bb1d7a709915e9dd730db257139fdfd9d53cb4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-16 19:15:09.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7221f1a84f004687900e3bea21eb3ad3\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/287127e7c63349e7b885ad75c153a46a : SUCCESS in 1h 03m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/afb5148007c34f3499861398cb1eb929 : SUCCESS in 54m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e733021b62bf4c329e915aabda7008fb : SUCCESS in 3m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ed5d27ab30174a1fbdc5bdf73e443b99 : SUCCESS in 4m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/dda1e1a029e54e92b23004ac59bb211b : SUCCESS in 5m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9047419398e64cfcb75daeef85c568fd : SUCCESS in 6m 01s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e31bffcb77eb49628fbde18a715ec5f1 : SUCCESS in 6m 04s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/949fe8b9ed224318b8ae11e59b93a0f8 : SUCCESS in 11m 27s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/4603071b698e4e13b2ca8aeb89b488c6 : SUCCESS in 31m 33s","accounts_in_message":[],"_revision_number":4},{"id":"0e172d9df1af5a1c6eaabc1905be8ade70805ec7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-16 19:15:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"2353fba8b2103bf560c97fa3cf5980f8bdd8e972","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-16 19:16:29.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dc645755a0dc45aba450d289d50eb0a9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14cc7ce63942497296e747692b580c6c : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"342ec4ed1b2403e158969411c271d9b2f0d6c239","revisions":{"f0c4582226610afb7762c8bf5e98a29b27cd7b47":{"kind":"REWORK","_number":1,"created":"2025-04-30 09:01:47.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/36/948536/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/36/948536/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/948536/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/948536/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/948536/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/36/948536/1"}}},"commit":{"parents":[{"commit":"6bca700b241e2a0032cd7de1661ab7e59a34a124","subject":"Functional tests migration of TestImageMultipleBackend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6bca700b241e2a0032cd7de1661ab7e59a34a124"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-04-30 08:40:22.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-04-30 08:40:22.000000000","tz":0},"subject":"Functional test migration test_wsgi.py","message":"Functional test migration test_wsgi.py\n\nAdded staging cleaner logic in start_server method to\nreplicate actual cleaning on glance service startup.\n\nChange-Id: I462195cef6cd5a13284de22e081652ef1c3bfeb6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f0c4582226610afb7762c8bf5e98a29b27cd7b47"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f0c4582226610afb7762c8bf5e98a29b27cd7b47"}]},"branch":"refs/heads/master"},"955a526b4f9bc7c7dece831ef3e891b8d4fc6ace":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-05-20 07:35:32.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/36/948536/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/36/948536/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/948536/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/948536/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/948536/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/36/948536/2"}}},"commit":{"parents":[{"commit":"8905e511268ed31a08a392d9c13867ca6c34eb75","subject":"Functional tests migration of TestImageMultipleBackend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8905e511268ed31a08a392d9c13867ca6c34eb75"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-04-30 08:40:22.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-05-20 07:35:32.000000000","tz":0},"subject":"Functional test migration test_wsgi.py","message":"Functional test migration test_wsgi.py\n\nAdded staging cleaner logic in start_server method to\nreplicate actual cleaning on glance service startup.\n\nChange-Id: I462195cef6cd5a13284de22e081652ef1c3bfeb6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/955a526b4f9bc7c7dece831ef3e891b8d4fc6ace"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/955a526b4f9bc7c7dece831ef3e891b8d4fc6ace"}]},"branch":"refs/heads/master","description":"Rebase"},"3a48741395a9a6c15002d6b491f5d033ea83066f":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2025-05-20 07:44:17.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/36/948536/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/36/948536/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/948536/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/948536/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/948536/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/36/948536/3"}}},"commit":{"parents":[{"commit":"7747b2d40a0b6cfac0dd9b9c5456d315814d2d90","subject":"Functional tests migration of TestImageMultipleBackend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7747b2d40a0b6cfac0dd9b9c5456d315814d2d90"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-04-30 08:40:22.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-05-20 07:44:17.000000000","tz":0},"subject":"Functional test migration test_wsgi.py","message":"Functional test migration test_wsgi.py\n\nAdded staging cleaner logic in start_server method to\nreplicate actual cleaning on glance service startup.\n\nChange-Id: I462195cef6cd5a13284de22e081652ef1c3bfeb6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3a48741395a9a6c15002d6b491f5d033ea83066f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3a48741395a9a6c15002d6b491f5d033ea83066f"}]},"branch":"refs/heads/master","description":"Rebase"},"342ec4ed1b2403e158969411c271d9b2f0d6c239":{"kind":"REWORK","_number":4,"created":"2025-06-16 17:12:17.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/36/948536/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/36/948536/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/948536/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/948536/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/948536/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/36/948536/4"}}},"commit":{"parents":[{"commit":"7747b2d40a0b6cfac0dd9b9c5456d315814d2d90","subject":"Functional tests migration of TestImageMultipleBackend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7747b2d40a0b6cfac0dd9b9c5456d315814d2d90"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-04-30 08:40:22.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-06-16 17:12:07.000000000","tz":0},"subject":"Functional test migration test_wsgi.py","message":"Functional test migration test_wsgi.py\n\nAdded staging cleaner logic in start_server method to\nreplicate actual cleaning on glance service startup.\n\nChange-Id: I462195cef6cd5a13284de22e081652ef1c3bfeb6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/342ec4ed1b2403e158969411c271d9b2f0d6c239"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/342ec4ed1b2403e158969411c271d9b2f0d6c239"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
