)]}'
{"id":"openstack%2Fglance~948563","triplet_id":"openstack%2Fglance~master~I3abd1e96eba499c220319bcb21abfa38b922d394","project":"openstack/glance","branch":"master","topic":"eventlet-removal","attention_set":{},"removed_from_attention_set":{"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2025-05-18 07:32:27.000000000","reason":"\u003cGERRIT_ACCOUNT_9303\u003e replied on the change","reason_account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}}},"hashtags":[],"change_id":"I3abd1e96eba499c220319bcb21abfa38b922d394","subject":"Functional tests migration test_images","status":"MERGED","created":"2025-04-30 17:31:02.000000000","updated":"2025-05-18 09:29:56.000000000","submitted":"2025-05-18 09:28:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"948563-eventlet-removal","meta_rev_id":"73cf1e9604c1241c3bb95ea97a8dfdfd14f1274e","_number":948563,"virtual_id_number":948563,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-18 09:28:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":2,"date":"2025-05-16 16:00:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":1,"date":"2025-05-16 16:00:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-30 18:26:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-04-30 18:45:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-16 16:00:45.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"b3d65ccf4dbb4dc011ca3a6717716c926bdbd2c1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-04-30 17:31:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aadd4e3aedda17a4274b93f270085611f1698f75","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-30 18:26:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/51060703bace428ebe6e066531356136\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3e0f63ef81de44198871ace0074a5cf3 : SUCCESS in 8m 45s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f527cb0a0be74902932cab36795677af : SUCCESS in 7m 45s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/66f0692ace8e40fcb8322ed141f3e79b : SUCCESS in 8m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a57f04e12558b2801354a56867f008b04ef3d459","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-30 18:45:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a3eb726dad24606acf15d8b3f236dc6\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e1768bf5b4ca4aae806ef5526ea52521 : SUCCESS in 53m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/224d9d86847c4bc0852a77318de5cfa2 : SUCCESS in 1h 07m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ee8269a5ee54efd9b05ca0fb7d4e23b : SUCCESS in 3m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/aba9d7325a914eeda89373849d400732 : SUCCESS in 4m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bd9a44779ec84235b7d8c02e597b754e : SUCCESS in 4m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/12ed38533735435e9edb197f3db379e7 : SUCCESS in 4m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69850ac2be214247a8b7737904508cd7 : SUCCESS in 5m 42s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7c45e9242f0d40cc99b695373f961016 : SUCCESS in 7m 57s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/414420a606234900974055fce18c5671 : SUCCESS in 13m 28s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/d1f83247de514b3ca2ec9c65ef662970 : SUCCESS in 28m 12s","accounts_in_message":[],"_revision_number":1},{"id":"eba6b182e7ad64e584bffa7e9e2c5d41413450f2","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-05-16 16:00:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"27910dbeec31b9ede889faa671feb68e86d75021","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-16 16:01:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6b22911f5cb4e88bf81defe1b4be467e8cb5902f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-16 17:23:09.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d7259f62d23a493695689fe94c563067\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2f4836a6495e40979992db1efe60e60d : SUCCESS in 35m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3e7849482e364051b0e9a664b2e41f3d : SUCCESS in 36m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0902d1488ca4756bffdbafe9ab77019 : SUCCESS in 4m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6613b10a9bf04c17b432c12aa7882f15 : SUCCESS in 4m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f15171cb4ca64014bd4d8aa7a36fb256 : SUCCESS in 4m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f0d69cc7def48a8b363c0bca0eb8ade : SUCCESS in 2m 58s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/dab872cb480b46ee809244954dfe7c4d : FAILURE in 6m 59s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/6c22f37db67f4daab87a1f8fdd53d99a : SUCCESS in 6m 42s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/74a8e75a4a3f45fe9152e957e171ec09 : SUCCESS in 26m 12s","accounts_in_message":[],"_revision_number":1},{"id":"42c32126137ead986caf7d81a9fd92f28ff47e87","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-05-16 17:47:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2183d94550b8c56ef263de5c2186e2b05dd7fa5d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-16 18:11:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4b04d974e86c4b8b94b5839347061c04\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/03a1d68d4ff344df9ea5b3bfd07dbfa6 : SUCCESS in 6m 02s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/082ac55e279c4c67938220fb22120677 : SUCCESS in 11m 49s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/5f82cf6625d949359d1b2405a916803e : SUCCESS in 8m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"556099e14b88ea151e3c7c9fa060a9e16af3b237","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-16 20:52:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e5f2e5a286da4ff3963e425e519248c4\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/81c0e7f410654b92974413837ee386a0 : TIMED_OUT in 3h 04m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cc1ac628177242e5b8260ecc884df99b : SUCCESS in 1h 07m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b451d9757a54216aad94f99ca0582c4 : SUCCESS in 2m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/91102343e59849e6879a260e994560f9 : SUCCESS in 2m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e0be846a9d1f46c3ac69c719d9b22295 : SUCCESS in 3m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/929634c237414dc5b4539103bd1805a2 : SUCCESS in 5m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5399e6e3097448a39b0d3236b7192561 : SUCCESS in 5m 24s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ee023023f3754c6ca086b3f749857223 : SUCCESS in 7m 51s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/ca4793b6e2274505ab7649105f12238a : SUCCESS in 11m 31s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/7f10b4d00aa54ff08ba98f5b8f28412f : SUCCESS in 25m 14s","accounts_in_message":[],"_revision_number":1},{"id":"25de432b2ef73a44653d485ac81858a69360f998","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-05-17 00:55:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4ac4f3bb76549e1477d5cc9a94ca06cd5b5fc367","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-17 01:06:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2a197a9b7e9141dba64c14faf593cbb3\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9fe3823d6ad947a396b2c7c3869e0036 : SUCCESS in 6m 56s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/42812b36de494af99f92551f96c03ac2 : SUCCESS in 6m 29s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/1092589eff4643dd80156ec09a2d6511 : SUCCESS in 9m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"626d38ad9d127c2ab4f74aa1eecc6fc645954a2a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-17 03:59:33.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cb69d2c7b122459e96f3141a74230e5f\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f6d76c461fae441a94533ffc997bc158 : TIMED_OUT in 3h 02m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7b65169954104cd09a441d6073ee2742 : SUCCESS in 31m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/726658eefcdd44eb909192aad2d7ac47 : SUCCESS in 3m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/544d1111bba844a19ab8ca957822ecdb : SUCCESS in 2m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/72482f428c6b481cabf58ed43aca2a64 : SUCCESS in 5m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bea7e6013761403da79e5e41da7555d6 : SUCCESS in 7m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df76bf0b37a14c918424cbc74684e880 : SUCCESS in 5m 35s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e33960e247d5468b990094bb5851b010 : SUCCESS in 7m 15s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/3a8508b879a343a9a9d95a50640ec1e2 : FAILURE in 9m 09s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f01af731f89d4ae8a9669f764bfd4c44 : SUCCESS in 22m 16s","accounts_in_message":[],"_revision_number":1},{"id":"40c33ebd59469f5e97a1b37ca7cf1648cf387463","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-05-17 04:32:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"99198c07483a1bf24cdf5c52e3c3093676a61f14","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-17 04:43:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/86656b463f21468084972a6a573d7760\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ee5aa87746b24fa08c107dd15395142a : SUCCESS in 5m 54s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/23edf9b4951d4de09edbf1bd65dd3620 : SUCCESS in 7m 38s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/06ff121725944475aa4ab26b3eef0dfc : SUCCESS in 8m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"83c7998c7590d9df4dc13bb8397804d366d68369","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-17 05:39:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/30c022b8d0504ea0a384fcc78db4044f\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/81e9fe38f0ba40e6bec0f7a1a746909b : SUCCESS in 1h 01m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dfc4bd5ee8214c5fbbcba0a6a45f1fbd : SUCCESS in 57m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a046429477a4dac86f7dbf9e7041324 : SUCCESS in 4m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2c21d96ad57341f5af8ac0bc51bde685 : SUCCESS in 4m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cfdc60b89f994435885561f9b97eaa09 : SUCCESS in 4m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fb4b43053c974ae78e0503b13ffe0812 : SUCCESS in 4m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16bf93fbd8934cf2aa38811c959c4df1 : SUCCESS in 5m 34s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/16072a2c6a9c42bf890c997b259e5af4 : SUCCESS in 6m 28s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/6da7d3172dcd4ea3809cd00c7674dc64 : SUCCESS in 8m 38s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/a9b1191692a74e48b686c6402d560e6b : SUCCESS in 22m 29s","accounts_in_message":[],"_revision_number":1},{"id":"fdef435b071567ee4f172ac0f20698932831f65c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-17 05:39:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4eaa96698c06c3875e65c510b980dfd9f598578a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-17 06:13:39.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b00e39ca395346a5b3225660528c735c\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3fcea3fe3915468b86c7846fd10d9d9c : SUCCESS in 32m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f79b8a39a1484a6a8dd9778629e1869e : SUCCESS in 33m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3ffb8b4ed68461f88cfba360ac4ba82 : SUCCESS in 2m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f98e2c7897bd48fea3a56ebdd59a63b5 : SUCCESS in 3m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/05a4a81d80be41a29e842873a239b982 : SUCCESS in 4m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9eee3525436d43adab0e244d8b849b6e : SUCCESS in 3m 06s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8b16924946ba4a6db495f63ad56ad7e1 : FAILURE in 6m 07s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/189e47262c874744bb167904660e8219 : FAILURE in 10m 44s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/4f1773e91c8442be82b7c271162cfb11 : SUCCESS in 14m 17s","accounts_in_message":[],"_revision_number":1},{"id":"a51043b165b99343c2952b274c0020546a9b7c01","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-05-17 06:41:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4c75c86a14a8c33c49e9335cb1aa3ba4a1ff8051","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-17 06:52:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/51763ffd193a481e9fa95f0a487dd689\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5d673ae17e884f7fbba2e81e561880d9 : SUCCESS in 6m 38s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/cf0d7826ce7d4a43838c69ea5e607965 : SUCCESS in 6m 41s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b60fed90dc054b59bb549a21efb671dc : SUCCESS in 8m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"05b192f5969e2477e011c042a345bd53cade3a37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-17 07:29:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3bf8c5c774a14af2b1f597ce1abea3c0\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/00624bc97e7041ac8f2345f172613523 : SUCCESS in 43m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/563d3a12911c476a8be1f3d44562da29 : SUCCESS in 47m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34becd5183874e3baeb3b74a6d057ddd : SUCCESS in 3m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d19939dfaea7432788775a7777e168a4 : SUCCESS in 3m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/70bc3ca338fc47f0acb5ae4dd5be9c65 : SUCCESS in 4m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d5dd34437fa84925baf3a4a821a1b9f4 : SUCCESS in 6m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6da85c687e1048949c564c16ad12fd37 : SUCCESS in 4m 22s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a91c140b48bf4fcaa7c47f8c45ec5445 : SUCCESS in 6m 49s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/09cf84f81b82465b86cfbbfa489bc11d : FAILURE in 10m 13s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/83a804e01a1e46a28a74f303c0d07708 : SUCCESS in 20m 39s","accounts_in_message":[],"_revision_number":1},{"id":"f783f966c5814906544eb1011af0dde003c92b34","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-05-17 08:33:59.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e41c2bf64442009ca0ee6621f3b4400adfa5b1fc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-17 08:48:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1cf683c2310d497eabc0fe00c49ea96e\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/727b512262af4e058270d403bf09ae1f : SUCCESS in 8m 02s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0106fae703fa4bf8b40bc62e854a9119 : SUCCESS in 9m 16s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a65dec3c29e04047b601ac57690c87a0 : SUCCESS in 8m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ca33c19216d0570d5cf5cc51570d49d37b579b45","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-17 11:37:57.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e51cd67abf16455f96a13e0e102e5bf3\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/dcc9fe06f91a4be982b9bc8b067608bb : TIMED_OUT in 3h 03m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1662743b7d3d4c19b635851cb1840e36 : SUCCESS in 31m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95da834514164476b39ddb4734e20d91 : SUCCESS in 5m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/12cc5270169b4cc1ab6ab3c73084c438 : SUCCESS in 2m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bd7e70ce4a2b441cbf63cb6e4cbc69cd : SUCCESS in 3m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e81b5e2e5ee34bbbb20ffa61093ed141 : SUCCESS in 7m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a7fb07e1aa247ce9c5a6dcf0b0ee692 : SUCCESS in 4m 36s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3bc748e9d5234a9ab8104eef7216bff6 : SUCCESS in 6m 51s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/bc3738f11f214c3da995ae16a052a5d5 : SUCCESS in 8m 08s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/8310eb197cf9403998726f7e12ae13cb : SUCCESS in 34m 08s","accounts_in_message":[],"_revision_number":1},{"id":"6f8eae39d233825a56815b5669e68c999f0ad868","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-05-18 07:32:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"301cabcf13c33b5e6fa9288496125be96033c1a4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-18 07:45:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/070ac0f296da42aaa7c357e4e4396777\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1bbaf9c0e4d741518b6691e861105f1b : SUCCESS in 8m 38s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d6dbbaa03df448259d5209933f6ffa96 : SUCCESS in 10m 00s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/dff60ebf06f547cb9eedb23c193eb092 : SUCCESS in 11m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8aa7d3b50cdb72a610c9a8b522f925ce537dae55","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-18 08:39:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0134270af6224b65a1a3583865f21b3d\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7a3228b7f1ff4ad785ac26030d0dc09b : SUCCESS in 1h 02m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/610a4e7dfe674b698795dd681ae65f32 : SUCCESS in 1h 00m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0cf51c30ec3a4f0fa425036897524809 : SUCCESS in 4m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5f342382e88d4bf3bf9571a66769558f : SUCCESS in 6m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cd39b271f71f4727a90f5f947df47c0f : SUCCESS in 4m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/99f4450fa1a84d8e833e9ec2f6f80083 : SUCCESS in 4m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14a6a65576ad475f8ed54effc7fe1a52 : SUCCESS in 4m 31s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/79dcdd7eac1944e2a9a734e804f69b97 : SUCCESS in 8m 11s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/bdd43017ae8f41abad67f50e15978925 : SUCCESS in 9m 44s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/9f601c09c1484073a6f4bfb0d3f816a1 : SUCCESS in 17m 12s","accounts_in_message":[],"_revision_number":1},{"id":"953b5f03760c5e47b29c43050baab05f8dc7b964","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-18 08:39:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9908c419be3a3814834c24b1edd0464e689ce360","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-18 09:28:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d3bed78a0dfc43f19cee2bd035de3f66\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f96f2444dea744ae94a3a3a1ed7f5c46 : SUCCESS in 48m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0e7170e86d0e4886b8fcf4bdc5e54355 : SUCCESS in 45m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/513b99f0b43447e9a7faa466bde8de6c : SUCCESS in 4m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b7a1d1821f95400ba5c946923cde41fe : SUCCESS in 3m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/92b16b71bfd14135a8b497b285888408 : SUCCESS in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d396ef9dce52495aad6a5b00934e2c5c : SUCCESS in 4m 42s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/33e07ce066b24857947e95c541371ff9 : SUCCESS in 5m 59s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/0251e9ff3f4d4e50aede3f00b0623618 : SUCCESS in 6m 55s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/382cd11bbd6c4f50b0773e3d88d4c994 : SUCCESS in 29m 42s","accounts_in_message":[],"_revision_number":1},{"id":"790d88515b72cec2ee1954afb55c35731172ff76","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-18 09:28:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"73cf1e9604c1241c3bb95ea97a8dfdfd14f1274e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-18 09:29:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d5944ffb4234345ac5455b1336a4b1e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b3880107430458ba81ac21cca04159e : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"35b44eaf83eadef6801a4ca1441ec447aa1e919f","revisions":{"35b44eaf83eadef6801a4ca1441ec447aa1e919f":{"kind":"REWORK","_number":1,"created":"2025-04-30 17:31:02.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/63/948563/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/63/948563/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/63/948563/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/63/948563/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/63/948563/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/63/948563/1"}}},"commit":{"parents":[{"commit":"e8da73973531d2a26afe3394faa8482a39787946","subject":"Functional tests migration test_images.TestImages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e8da73973531d2a26afe3394faa8482a39787946"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-04-30 16:25:11.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-04-30 16:25:11.000000000","tz":0},"subject":"Functional tests migration test_images","message":"Functional tests migration test_images\n\n- TestImageDirectURLVisibility\n- TestImageLocationSelectionStrategy\n- TestCopyImagePermissions\n\nChange-Id: I3abd1e96eba499c220319bcb21abfa38b922d394\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/35b44eaf83eadef6801a4ca1441ec447aa1e919f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/35b44eaf83eadef6801a4ca1441ec447aa1e919f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
