)]}'
{"id":"openstack%2Fglance~948564","triplet_id":"openstack%2Fglance~master~I7d6402175a18c846ddd435dd998a2b8b19b09de6","project":"openstack/glance","branch":"master","topic":"eventlet-removal","attention_set":{},"removed_from_attention_set":{"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2025-05-20 17:29:38.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7d6402175a18c846ddd435dd998a2b8b19b09de6","subject":"Functional tests migration of image members tests","status":"MERGED","created":"2025-04-30 17:31:02.000000000","updated":"2025-05-20 17:30:41.000000000","submitted":"2025-05-20 17:29:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"948564-eventlet-removal","meta_rev_id":"c2a7f37adb5b1b6700e061b70f21c472994efc33","_number":948564,"virtual_id_number":948564,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-20 17:29:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":2,"date":"2025-05-20 15:55:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":1,"date":"2025-05-20 15:55:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-30 18:31:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-04-30 18:43:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-20 15:55:30.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"1fd42074207e49c564ae0c220b4200dc351b7010","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-04-30 17:31:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f575553fe8c0c62e76f90f401d63028aab351129","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-30 18:31:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a04868239dfc4a0f99547cb1de117228\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/af6f475ef7504cbe91e63db4d9feca6f : SUCCESS in 10m 31s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3d028803d15d430685fb5339225d66de : SUCCESS in 7m 07s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/3eca5086a3a3472d8dbc7ce260720bba : SUCCESS in 10m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1d4a147baa693996d659516a9d9db15852b3906c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-30 18:43:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/89b59b6d9bb74abcbddf1f604f5f356c\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/904f568a2f7a482e866a6bd0a70d2bf3 : SUCCESS in 53m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7ed50d8316154483b8f302ade229228c : SUCCESS in 57m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0086add0439d4fe2a2c3cb54c0350174 : SUCCESS in 4m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/28d1eb66c5f04c5a8af26fae0434a5f1 : SUCCESS in 5m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6c10da160e62412cb510e5a371c3ede6 : SUCCESS in 5m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c7e55c449a8144b7afb0e73f7e10ab91 : SUCCESS in 8m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9923eb7046e452a9211b892955277ff : SUCCESS in 5m 17s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a53459e57b9240e7a028a6ee08903900 : SUCCESS in 7m 44s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/4d51e1b62c904c5882440cbdfc993c05 : SUCCESS in 10m 24s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/99159acbbf0b4fe2926ec78dc5d4ca64 : SUCCESS in 22m 59s","accounts_in_message":[],"_revision_number":1},{"id":"18ae3fbb05514838f8daf40feaaaccf73a5bde1a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-05-20 07:03:16.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"d1aeecd81dfb319b1b9a156c6d66192968044a56","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 07:18:53.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d29eede49dfd48fdb4415b0a8f7973fd\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c5b9965a79544a4faadd11fdb13b365a : SUCCESS in 7m 07s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/97ae97a5ee5a4e0c9b94d56651806649 : SUCCESS in 6m 58s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/7fb1dcbaced047378d91a66a7d8f4c8e : SUCCESS in 13m 32s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c7d65f60bde8535f39a1550d90831a1f8acb97aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 08:02:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7faa3bdb457148aa96e0f86414da0442\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9136930d48104c26a44af9e10ca37c1d : SUCCESS in 54m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/839ff41f229d451ca7ce3231c84a4344 : SUCCESS in 58m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a7fa19e03974f83a774299ac894871c : SUCCESS in 3m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/546754e51f774d9ca308aa0ec37f3e7d : SUCCESS in 4m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/28e209085c0544a9b34f50f91c995f0a : SUCCESS in 4m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bab2d4d9485f4bb8b849af806f1d7194 : SUCCESS in 6m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83fed98f388c4785bde86d7237eafe7b : SUCCESS in 5m 33s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/fb7a4fade0c74c26b76b1ba35d4231cb : SUCCESS in 6m 38s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/63384026e14340218f01212dade08b64 : SUCCESS in 6m 51s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/d1e410810c784cd1a3ba97121f746826 : SUCCESS in 32m 22s","accounts_in_message":[],"_revision_number":2},{"id":"9a375c980c5bbf9f3c7772d1863406e7740d8d4e","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-05-20 15:55:30.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9e770f69a46a192d836a61acdef876f8e6a27ebd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 15:56:00.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3a6218b75941caf252555d2f10fafb78935cb231","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 17:29:38.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f82941355564fd6a336589b7c4b603b\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2fbbd42ef4c448ada8c86ee844f94b71 : SUCCESS in 1h 01m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e008a738861e41a4b69250e710a6119d : SUCCESS in 1h 00m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27ed30da07634c00b6c3f6631d59442b : SUCCESS in 3m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0156180714ee460fa93ff1cb5a4f576e : SUCCESS in 4m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c9eaabe3ccd94d91b18080a074dd06d9 : SUCCESS in 6m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2992803e9f254992b7f012d567fa2380 : SUCCESS in 6m 12s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8fddf52ee9b8478a973131f1b6be0f3d : SUCCESS in 7m 33s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/0177bab92dd149beaca7036a589a857d : SUCCESS in 11m 40s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/cf9edba012f146428e3378cabe2d6e13 : SUCCESS in 35m 01s","accounts_in_message":[],"_revision_number":2},{"id":"4f0e07217ad196ec01a919910798ba28b52bcc32","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 17:29:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"c2a7f37adb5b1b6700e061b70f21c472994efc33","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 17:30:41.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/327b08b98892415aa1e776d4ba23b083\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4898f706438c4788920978e53bc8f885 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"faccbed1d9291808be563c22784467a65f1f8bd1","revisions":{"f1c07bbeebf86226c6186e85efb2db172a5cd20b":{"kind":"REWORK","_number":1,"created":"2025-04-30 17:31:02.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/64/948564/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/64/948564/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/64/948564/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/64/948564/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/64/948564/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/64/948564/1"}}},"commit":{"parents":[{"commit":"35b44eaf83eadef6801a4ca1441ec447aa1e919f","subject":"Functional tests migration test_images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/35b44eaf83eadef6801a4ca1441ec447aa1e919f"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-04-30 16:47:44.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-04-30 16:47:44.000000000","tz":0},"subject":"Functional tests migration of image members tests","message":"Functional tests migration of image members tests\n\n- Added new test file test_image_members.py\n- Move existing member tests from test_images.py to new file\n- Refactored TestMultipleStoreImageMembers to inherit from\n  TestImageMembers to avoid duplication\n- Still scope for refactoring which can be done in followup if anyone is\n  interested\n\nChange-Id: I7d6402175a18c846ddd435dd998a2b8b19b09de6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f1c07bbeebf86226c6186e85efb2db172a5cd20b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f1c07bbeebf86226c6186e85efb2db172a5cd20b"}]},"branch":"refs/heads/master"},"faccbed1d9291808be563c22784467a65f1f8bd1":{"kind":"REWORK","_number":2,"created":"2025-05-20 07:03:16.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/64/948564/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/64/948564/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/64/948564/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/64/948564/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/64/948564/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/64/948564/2"}}},"commit":{"parents":[{"commit":"7cf7a1445a222590027b3fca7c9ad026d3781083","subject":"Merge \"Revert \"Deprecate glance scrubber\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7cf7a1445a222590027b3fca7c9ad026d3781083"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-04-30 16:47:44.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-05-20 06:29:03.000000000","tz":0},"subject":"Functional tests migration of image members tests","message":"Functional tests migration of image members tests\n\n- Added new test file test_image_members.py\n- Move existing member tests from test_images.py to new file\n- Refactored TestMultipleStoreImageMembers to inherit from\n  TestImageMembers to avoid duplication\n- Still scope for refactoring which can be done in followup if anyone is\n  interested\n\nChange-Id: I7d6402175a18c846ddd435dd998a2b8b19b09de6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/faccbed1d9291808be563c22784467a65f1f8bd1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/faccbed1d9291808be563c22784467a65f1f8bd1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
