)]}'
{"id":"openstack%2Fglance~948565","triplet_id":"openstack%2Fglance~master~If73e3f70240fbda70b497353c16be19bcf76da0d","project":"openstack/glance","branch":"master","topic":"eventlet-removal","hashtags":[],"change_id":"If73e3f70240fbda70b497353c16be19bcf76da0d","subject":"Functional tests migration of TestImageMultipleBackend","status":"ABANDONED","created":"2025-04-30 17:31:02.000000000","updated":"2025-06-20 05:49:27.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"563bd0a77bcbc311f34c1c4ea2f1b1a30d8cc140","_number":948565,"virtual_id_number":948565,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-04-30 18:38:16.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-30 18:38:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"158a76a9741e36cc116174c6b237c39969c8b300","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-04-30 17:31:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c8f8ad7a101c11dd51a7c2ca9eef18c0cc3541c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-30 18:38:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/76cc71711d61492d9c69aa77a2d922f4\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a4bc030e19514163a28c704dc06e52be : SUCCESS in 55m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2a2323a039ae454aaf388363ffa60708 : SUCCESS in 1h 00m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6d1225fe716431a976ce00ce8da4dd0 : SUCCESS in 4m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/68efe4561d1441989b4efb0d4874d7e2 : SUCCESS in 5m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/632d130cd795488aa74d4b70e4275057 : SUCCESS in 5m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f683d0261e31473c919e4174159a1c20 : SUCCESS in 8m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c0f713613f9466d89cb36ac860b90c1 : SUCCESS in 5m 15s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/44b4825ce5e541ad9ce9a5fdf116f2c8 : SUCCESS in 7m 11s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/b663cd91e1724039bca26ec0592bdd64 : SUCCESS in 12m 08s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/606d1e674ba14b1f9f98da15906920d8 : SUCCESS in 34m 04s","accounts_in_message":[],"_revision_number":1},{"id":"86a43fa098f3e613735c3152777725fd458ece20","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-30 18:42:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c95a4c2c0faf456bab491850569e1e43\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e6c95ff464ab4cfaa01caa9a5b8ffc3c : SUCCESS in 8m 47s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c7d4f7ba2c2f402199eec096cb4c753b : SUCCESS in 9m 26s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/50a29440f6444e6db75bd1fa1b31fde1 : SUCCESS in 13m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"563bd0a77bcbc311f34c1c4ea2f1b1a30d8cc140","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-06-20 05:49:27.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ca1ef75607ca79a82f216734e63c1ca7bc365637","revisions":{"ca1ef75607ca79a82f216734e63c1ca7bc365637":{"kind":"REWORK","_number":1,"created":"2025-04-30 17:31:02.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/65/948565/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/65/948565/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/65/948565/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/65/948565/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/65/948565/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/65/948565/1"}}},"commit":{"parents":[{"commit":"f1c07bbeebf86226c6186e85efb2db172a5cd20b","subject":"Functional tests migration of image members tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f1c07bbeebf86226c6186e85efb2db172a5cd20b"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-04-30 17:00:30.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-04-30 17:00:30.000000000","tz":0},"subject":"Functional tests migration of TestImageMultipleBackend","message":"Functional tests migration of TestImageMultipleBackend\n\n- Moved all tests from above class to test_multi_store_images.py\n- Refactored tests to use common helper methods\n- Later we can move other multi store related tests here\n\nChange-Id: If73e3f70240fbda70b497353c16be19bcf76da0d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ca1ef75607ca79a82f216734e63c1ca7bc365637"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ca1ef75607ca79a82f216734e63c1ca7bc365637"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
