)]}'
{"id":"openstack%2Fglance~948566","triplet_id":"openstack%2Fglance~master~I092dac5b27488975dcc35ee8db4d97dc80e4e0d1","project":"openstack/glance","branch":"master","topic":"eventlet-removal","hashtags":[],"change_id":"I092dac5b27488975dcc35ee8db4d97dc80e4e0d1","subject":"Functional test migration test_wsgi.py","status":"ABANDONED","created":"2025-04-30 17:31:02.000000000","updated":"2025-06-20 05:47:50.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"52dcf1c61a4ab5979649d9e88ace854366b95dbf","_number":948566,"virtual_id_number":948566,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-04-30 18:22:17.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-30 18:22:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"77d9832d37c402d6a76444e62da839bfb5a889e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-04-30 17:31:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"647dd012309d3cedaf15fe61f732a90e2ab3a70e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-30 18:22:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f0a1136c43fb4da8868b5487fa44b14e\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/186f51b407c3498d8e1d521f0506f3ce : SUCCESS in 43m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d88361493c0344a7879df7a09f197453 : SUCCESS in 46m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49cacdd4af2646b9a4f1301e50bb4448 : SUCCESS in 4m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/caaee485fede4e5cb136be99e8e28e2b : SUCCESS in 5m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/da228fa8cc9440aebd377b24b807b6a1 : SUCCESS in 5m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/837e554e1ae3419a951be8bb58f52717 : SUCCESS in 7m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5069cad5e4b4034aa3fe8d7a1da7dd1 : SUCCESS in 5m 26s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ba4e7b1a2e344e758f99facc5f4661a3 : SUCCESS in 7m 12s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/9823373ae73c4818b6e0e5b7572d120f : SUCCESS in 10m 46s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/8abef37e95df49838b41487aeef8535d : SUCCESS in 35m 57s","accounts_in_message":[],"_revision_number":1},{"id":"38fa9051903625cdeeb27d0b49accd1eb5017c70","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-30 18:42:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb4d1d164b43461091b5d1269da9e5af\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/aa1f1ceba30f4ca0bf5a20fa36af4cbc : SUCCESS in 5m 20s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c7d426193ab548b4a925db6ff669b6ff : SUCCESS in 11m 22s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/568023b868904fdda9caf5129beed175 : SUCCESS in 11m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"52dcf1c61a4ab5979649d9e88ace854366b95dbf","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-06-20 05:47:50.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"04044a40857468beaa01d663e915e0ee205fc70d","revisions":{"04044a40857468beaa01d663e915e0ee205fc70d":{"kind":"REWORK","_number":1,"created":"2025-04-30 17:31:02.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/66/948566/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/66/948566/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/948566/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/948566/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/948566/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/66/948566/1"}}},"commit":{"parents":[{"commit":"ca1ef75607ca79a82f216734e63c1ca7bc365637","subject":"Functional tests migration of TestImageMultipleBackend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ca1ef75607ca79a82f216734e63c1ca7bc365637"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-04-30 17:10:21.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-04-30 17:10:21.000000000","tz":0},"subject":"Functional test migration test_wsgi.py","message":"Functional test migration test_wsgi.py\n\nAdded staging cleaner logic in start_server method to\nreplicate actual cleaning on glance service startup.\n\nChange-Id: I092dac5b27488975dcc35ee8db4d97dc80e4e0d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/04044a40857468beaa01d663e915e0ee205fc70d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/04044a40857468beaa01d663e915e0ee205fc70d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
