)]}'
{"id":"openstack%2Fglance~948665","triplet_id":"openstack%2Fglance~stable%2F2024.1~Ic56d897c441004d710ec89a536ad57b2fe480a41","project":"openstack/glance","branch":"stable/2024.1","attention_set":{},"removed_from_attention_set":{"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2025-05-05 15:41:41.000000000","reason":"Change was submitted"},"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2025-05-05 15:41:41.000000000","reason":"Change was submitted"},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2025-05-05 13:31:13.000000000","reason":"\u003cGERRIT_ACCOUNT_8122\u003e replied on the change","reason_account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}}},"hashtags":[],"change_id":"Ic56d897c441004d710ec89a536ad57b2fe480a41","subject":"Fix broken metadata function calls","status":"MERGED","created":"2025-05-01 18:39:21.000000000","updated":"2025-05-05 15:42:36.000000000","submitted":"2025-05-05 15:41:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"948665","cherry_pick_of_change":942411,"cherry_pick_of_patch_set":1,"meta_rev_id":"5a5f5a35f3134f6ceba000531f1df36b59b90f8d","_number":948665,"virtual_id_number":948665,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-05 15:41:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":2,"date":"2025-05-05 13:31:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-05-05 09:42:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":1,"date":"2025-05-05 13:31:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-01 18:49:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-05-01 20:43:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-05 09:42:19.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2025-05-05 09:42:19.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2025-05-05 09:42:19.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"2b9a78bd568ed97fdd392ada9f97ca74efa66dd5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-05-01 18:39:21.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2024.2.","accounts_in_message":[],"_revision_number":1},{"id":"12242ea1c0e7ea8c359c51865078a72f9e60a5ff","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-01 18:49:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/491f308f065846cc8699566846dfb569\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6c1b64e8e8c843efa8fb4f2693b3ce8d : SUCCESS in 8m 18s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5a99032237134d0bba28a688ddcbd261 : SUCCESS in 6m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"638cd5574b0e03a703bba5763dd954fe8443cee5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-01 20:43:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/80a91ff4202a48f286109fd63fe29aa2\n\n- grenade https://zuul.opendev.org/t/openstack/build/62eb48d0c6ee4d7f97b712f40fda80d4 : SUCCESS in 24m 52s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/63a51ba33ae644dfa40a6f8dd53d92a4 : SUCCESS in 23m 46s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ebd57e5479cb422c80e188f5f03f3171 : SUCCESS in 1h 02m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4e6376c23d3b4a4d8f01e2692763713e : SUCCESS in 25m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b229cf23e4b4cdf83dbd226181362be : SUCCESS in 3m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/403b4602bd7f4b3badccdaaf01cc01f0 : SUCCESS in 4m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/40f51fd5b00046a98b60ceed3f142e34 : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49e2dd58b9db42d68ec1e7529d54c3de : SUCCESS in 5m 06s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/13b75c00fe804aa6a7c657a59ff6a5ec : SUCCESS in 8m 30s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2196754a31a14846986d7ab8085f247d : SUCCESS in 8m 21s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/51d0019f147d42b9b248c8b97b355954 : SUCCESS in 7m 43s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/6c250b00387f4004aab77910507784f3 : SUCCESS in 1h 39m 34s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/1ccd960eecda4a63a7a30782ab9f0c78 : SUCCESS in 1h 58m 18s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/53aec5425c904139ae32ca0042f48cf7 : SUCCESS in 1h 17m 38s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/7a1fc695517c4b9bbd447b75957f5820 : SUCCESS in 1h 38m 43s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/7a16a03e31fa4cbea193309fda0388d4 : SUCCESS in 1h 37m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/302c2a1d9f054072b52f500340191174 : SUCCESS in 53m 36s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/7cd9af88004b455396ca10a05fd342bb : SUCCESS in 1h 48m 00s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/f8d6c32d8c8b4fb3a8b42cd2372ddbe6 : SUCCESS in 1h 42m 38s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/93e44ab42ca3499b945e12c4a31a6882 : SUCCESS in 1h 03m 26s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/376a70c489ec4d3b9a417c8bf426e91b : SUCCESS in 28m 01s","accounts_in_message":[],"_revision_number":1},{"id":"182d6350e3facc27647958b0c6efd8c274bdfacf","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-05-05 09:42:19.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"505d2501e296faedddef5f9b5c6111e71746a5f1","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-05-05 13:31:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c1d27fcde4f6a965d8e5046c8a51ed2ac496ddbb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-05 13:31:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d6f85f32cb7c5141b4305c7fab134bd3c7141adc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-05 15:41:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e5f5fe7d5719499697338670c2834d7a\n\n- grenade https://zuul.opendev.org/t/openstack/build/b301766261a940c5914dc812da67017e : SUCCESS in 1h 01m 55s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/f73c3885d3c644b8929d279322d48624 : SUCCESS in 1h 10m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4406ea886cd146a8bbffc415b90b3da0 : SUCCESS in 1h 35m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3ba3ad3b8d664c0da62a688026a9741d : SUCCESS in 59m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f484589d874641a59916ed9eba9f59e9 : SUCCESS in 4m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/822b090bb2504857adfbe533deefa08e : SUCCESS in 5m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/829c2594f6364cb192fdf8a6310839de : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9452ccdadd2446c09e8bc80b5ba4599f : SUCCESS in 5m 09s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/fd5f948c55eb423ab5b5ef2f35fef19b : SUCCESS in 8m 24s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4ce767da91e447128438ac0f59298a73 : SUCCESS in 8m 03s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/23f37dde1f8d44c6bc8e37e167d29f8b : SUCCESS in 1h 52m 25s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/3f819a4ce9fa4cdeaaff7315debde968 : SUCCESS in 2h 03m 37s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/5efb7258943f459a863d8c79fee58ed0 : SUCCESS in 1h 45m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4515c54e45124d44b035ddff5248f9c6 : SUCCESS in 57m 36s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d054df89d41f430da3c2e1293687548f : SUCCESS in 1h 32m 26s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f90ea5e29c4a4d1c998c6b21fc3dd079 : SUCCESS in 28m 45s","accounts_in_message":[],"_revision_number":1},{"id":"b35c6880a808f84186d6916e9e8942601d217ebc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-05 15:41:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"5a5f5a35f3134f6ceba000531f1df36b59b90f8d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-05 15:42:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8f729e1cc26048968a7b1adf8a13fd29\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9d5fc07717f4cc2ad6c7a8484e2f9d7 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"028de8e37e76db456de9a6de71258e60689adee1","revisions":{"028de8e37e76db456de9a6de71258e60689adee1":{"kind":"REWORK","_number":1,"created":"2025-05-01 18:39:21.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/65/948665/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/65/948665/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/65/948665/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/65/948665/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/65/948665/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/65/948665/1"}}},"commit":{"parents":[{"commit":"6925177fd74dbc4230a26b3a582747d19c455bbb","subject":"Merge \"Fix doc bug\" into stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6925177fd74dbc4230a26b3a582747d19c455bbb"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-01-14 15:43:16.000000000","tz":60},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-05-01 18:39:21.000000000","tz":0},"subject":"Fix broken metadata function calls","message":"Fix broken metadata function calls\n\nThe patch [1] cleanup so that the database\nconnection was re-used but forgot to update\nall function calls causing functionality to\nbe broken.\n\n[1] https://review.opendev.org/c/openstack/glance/+/889380\n\nChange-Id: Ic56d897c441004d710ec89a536ad57b2fe480a41\nCloses-Bug: #2094816\n(cherry picked from commit e791ed565e9b1a17743306caaf5c7a086fc47aee)\n(cherry picked from commit e74b91d24de43836fe1269fd8d3ea11afd0caa58)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/028de8e37e76db456de9a6de71258e60689adee1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/028de8e37e76db456de9a6de71258e60689adee1"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
