)]}'
{"id":"openstack%2Fglance~948812","triplet_id":"openstack%2Fglance~stable%2F2024.1~I69ce5e831fab6ebe061fed7ce32140591769a368","project":"openstack/glance","branch":"stable/2024.1","topic":"harden-copy-bust-lock-2024-1","attention_set":{},"removed_from_attention_set":{"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2025-05-16 19:45:36.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I69ce5e831fab6ebe061fed7ce32140591769a368","subject":"Harden copy bust lock test","status":"MERGED","created":"2025-05-05 15:02:09.000000000","updated":"2025-05-16 19:46:26.000000000","submitted":"2025-05-16 19:45:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"948812-harden-copy-bust-lock-2024-1","meta_rev_id":"f75270a2810d3fcff600e483f1bb9bdeb5dfbf3c","_number":948812,"virtual_id_number":948812,"owner":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-16 19:45:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2025-05-16 16:52:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2025-05-16 18:43:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}],"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-05 15:02:09.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"CC"},{"updated":"2025-05-05 17:01:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-16 16:52:46.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"9cac1b80896b724d675f64c1627f68e21964f566","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-05-05 15:02:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0e5aad5892846b9aa132b70b450dc65277bf974e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-05 17:01:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e7cc3c587ab04f98bb870e70921bc147\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/48b94f9797364246a22d4b8b8a4b7884 : SUCCESS in 47m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0dccfd7fe9e747449fb9b8c1bed8c2bc : SUCCESS in 53m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0cdadc39b83743d2ad4bdc61d818a8a7 : SUCCESS in 4m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0bad5084cb09444883b2caf70813d535 : SUCCESS in 5m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/304d418a3d21467ebb59ac9bba55707e : SUCCESS in 6m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ac5973939b24765aad767033a111c35 : SUCCESS in 5m 04s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/68c84cf6f5214c5c9cadedcbfcfcb98c : SUCCESS in 8m 26s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/85503dc82f084dc2bbf952c496f4e5ce : SUCCESS in 7m 21s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/eb22c81c23a943b3afaa3c92fcb05900 : SUCCESS in 7m 50s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/7a3287535935411b9ea3cb9282b70d13 : SUCCESS in 31m 10s","accounts_in_message":[],"_revision_number":1},{"id":"d36c46cfbe44fe0f8ad82aba79ee17aecbd86dd8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-05 17:46:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e9c47862f59b460a952b22ae8a066371\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f075074072384b68b38cdedc9a143105 : SUCCESS in 9m 56s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/aabfc75c98d040a18bbee53fce588915 : SUCCESS in 8m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d2b4aa697ab93818bf2e637844aed6d785c95e8a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-05-16 15:39:16.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"5c2ef82c60b5d7821a3bf8613dd534df321e03d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-16 16:32:15.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7fe1a3fc59664dc0ab39c278f105401c\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/929b57d9fb244428af9677e61c1f7d51 : SUCCESS in 46m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d756ddae46544cdd8eebf41d15ec3cfe : SUCCESS in 52m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db6e2415fd6040c1bc4706e296171bbe : SUCCESS in 4m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1342aef9219942e8b26fb4c31f02c148 : SUCCESS in 2m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/490ee12228d34f109d40541b0a598517 : SUCCESS in 5m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73ded131467e4d7788a62ebf911a3fee : SUCCESS in 6m 58s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c497d762b152423a8cb6ab1cbcfb0110 : SUCCESS in 8m 32s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7aa401554fd7427bb4677b2f3f9a61e9 : SUCCESS in 6m 20s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/dd998852d5054bfd918cb3266e25ba7b : SUCCESS in 5m 41s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/0f0ddac2566a4e9ba690c4f9442911b8 : SUCCESS in 29m 00s","accounts_in_message":[],"_revision_number":2},{"id":"c539b55e376febcea73e3630e0a3954bbefebe1e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-16 16:49:07.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/79da70b03cf74aaeb2c764efd1175c5a\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/142c3bd804664b4896473e3f9700e0f2 : SUCCESS in 6m 56s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/89717f9bfd8742118f153e6b06b9960b : SUCCESS in 9m 12s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"dcccc4409dd3da4d7afccda198c27f90474effe6","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-05-16 16:52:46.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"b8184bb6fd0450883d3f12ee5df9bcf5f72aeac6","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-05-16 18:43:23.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"af56b2a30c8838f35ab7eab592e65b8d70fe3fef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-16 18:43:45.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"420b6a942ffe97e75ad909115e1f8ec2a968d420","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-16 19:45:35.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c6d0d80c8b84411a143a073928d8385\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/36e83161a0be443492c6f1cb5e0f2111 : SUCCESS in 1h 01m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/958c6c342dea4449a67a22c8a2391fc6 : SUCCESS in 5m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c646b81c886b4c6dbeba901d13336826 : SUCCESS in 6m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ef3c403908b9403e908bcae49a4d96b1 : SUCCESS in 8m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac919c34781341c0a61fdfe72f4c78da : SUCCESS in 7m 10s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8a7786a3a3dd433aa9ef1cfc3e1b5cf5 : SUCCESS in 7m 45s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/afcdcf5daf444ff699e2b465f9a05335 : SUCCESS in 9m 28s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/ffd10b432f1a4407af706e5c8b75a88b : SUCCESS in 20m 47s","accounts_in_message":[],"_revision_number":2},{"id":"197311463b08d6ae110c74b1604e4ae12960dc12","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-16 19:45:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"f75270a2810d3fcff600e483f1bb9bdeb5dfbf3c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-16 19:46:26.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/534ae581d9524c1d926107179c455fbd\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9de50c637154f87994626f56bc9d633 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a514df4570b5b2a127037416aebdefb7aca7932f","revisions":{"32fcd84746970fd08f611b1018c9731431e7b7c5":{"kind":"REWORK","_number":1,"created":"2025-05-05 15:02:09.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/12/948812/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/12/948812/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/12/948812/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/12/948812/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/12/948812/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/12/948812/1"}}},"commit":{"parents":[{"commit":"6925177fd74dbc4230a26b3a582747d19c455bbb","subject":"Merge \"Fix doc bug\" into stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6925177fd74dbc4230a26b3a582747d19c455bbb"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2025-04-24 15:08:00.000000000","tz":-420},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2025-05-05 15:00:15.000000000","tz":120},"subject":"Harden copy bust lock test","message":"Harden copy bust lock test\n\nHopefully this resolves a race in the test_import_copy_bust_lock test,\nwhere we look for the task id before it may have actually started.\n\nChange-Id: I69ce5e831fab6ebe061fed7ce32140591769a368\n(cherry picked from commit caac5fc12c550a72848b0ded652d67311cb708f6)\n(cherry picked from commit 512f070cf7787c6dc6e779d4f5e69ad25c119916)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/32fcd84746970fd08f611b1018c9731431e7b7c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/32fcd84746970fd08f611b1018c9731431e7b7c5"}]},"branch":"refs/heads/stable/2024.1"},"a514df4570b5b2a127037416aebdefb7aca7932f":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":2,"created":"2025-05-16 15:39:16.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/12/948812/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/12/948812/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/12/948812/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/12/948812/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/12/948812/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/12/948812/2"}}},"commit":{"parents":[{"commit":"028de8e37e76db456de9a6de71258e60689adee1","subject":"Fix broken metadata function calls","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/028de8e37e76db456de9a6de71258e60689adee1"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2025-04-24 15:08:00.000000000","tz":-420},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2025-05-16 15:39:02.000000000","tz":120},"subject":"Harden copy bust lock test","message":"Harden copy bust lock test\n\nHopefully this resolves a race in the test_import_copy_bust_lock test,\nwhere we look for the task id before it may have actually started.\n\nChange-Id: I69ce5e831fab6ebe061fed7ce32140591769a368\n(cherry picked from commit caac5fc12c550a72848b0ded652d67311cb708f6)\n(cherry picked from commit eebe82f1a53131c076beaa230862d0eb8f1dc799)\n(cherry picked from commit 7af1ddb0db23f0091bee4ad0d34c217f70ccea51)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a514df4570b5b2a127037416aebdefb7aca7932f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a514df4570b5b2a127037416aebdefb7aca7932f"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
