)]}'
{"id":"openstack%2Fglance~948903","triplet_id":"openstack%2Fglance~master~Ib2c282bc19d48361b16fb1b2549cfcab80dea29c","project":"openstack/glance","branch":"master","topic":"spec/set-size-on-upload","attention_set":{},"removed_from_attention_set":{"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2025-08-06 20:02:22.000000000","reason":"Change was submitted"},"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2025-08-06 17:37:49.000000000","reason":"\u003cGERRIT_ACCOUNT_4393\u003e replied on the change","reason_account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2025-08-06 15:12:31.000000000","reason":"\u003cGERRIT_ACCOUNT_8122\u003e replied on the change","reason_account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}}},"hashtags":[],"change_id":"Ib2c282bc19d48361b16fb1b2549cfcab80dea29c","subject":"Read image size from request header","status":"MERGED","created":"2025-05-06 15:17:36.000000000","updated":"2025-08-06 20:03:14.000000000","submitted":"2025-08-06 20:02:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":19,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"948903-spec/set-size-on-upload","meta_rev_id":"41ebde93f49c8749b7d1b05bf5fe247822c97058","_number":948903,"virtual_id_number":948903,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-06 20:02:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"all":[{"value":2,"date":"2025-08-06 17:37:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":2,"date":"2025-08-06 15:12:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"all":[{"value":1,"date":"2025-08-06 17:37:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-06 15:52:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-05-06 17:34:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-03 08:33:37.000000000","updated_by":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"reviewer":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"state":"CC"},{"updated":"2025-07-03 14:24:19.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2025-08-04 13:17:04.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"e4ebbe5a4136ccd2ddb988b6d01d4a52a26c1427","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-05-06 15:17:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5e98e6e5972416e06be2d0f00032500fc0048c70","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-06 15:52:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc211dc605f94944a513f134bab31379\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e3dc72dd8ab5403796e2c6310ea20ac8 : SUCCESS in 6m 09s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/db9ffaee6f2d467ca65def9c2849e5df : SUCCESS in 7m 11s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/108803605530437a997402b27403e752 : SUCCESS in 9m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"82212d25bf69d3adf6b259644acfadc6fcd973d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-06 17:34:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/79f8b0f1c29d4e1dbdc78dcf2ffa4362\n\n- grenade https://zuul.opendev.org/t/openstack/build/c725910804054080a51f9a1b19d00557 : SUCCESS in 59m 47s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/21246d7c554d483fabf66925a4095629 : SUCCESS in 56m 26s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c77adcc00e464ea7aea427386ee53297 : SUCCESS in 1h 32m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a467176fd0104f4eb92bef7c8304b424 : SUCCESS in 54m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41a2115341c1431ba1792e811ae7d2f4 : SUCCESS in 7m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a302415a57b34b54abb9f3e28a40ddac : SUCCESS in 5m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5b0acc5a99d542228d65d8094cc9f32c : SUCCESS in 7m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5d2ba0f0e94e4f77bb12f1a1404e622b : SUCCESS in 9m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/222aa2fbccbb4a86b8f7c16e32937961 : SUCCESS in 7m 32s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/205caebe506948f28452b2b046ebc4c0 : SUCCESS in 9m 13s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/d413090cabc545ae85931c1e8590df29 : SUCCESS in 12m 35s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/50875aeeacfb40768809e2eb1d125828 : FAILURE in 1h 56m 57s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/13a8f9d123844968865d914b264aab48 : SUCCESS in 2h 05m 30s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/5564d468301c4cff9483c7e6925ff469 : SUCCESS in 1h 59m 36s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/2f949fcc6454439da2f655e82bc5ee4e : SUCCESS in 2h 00m 46s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/f65b2e97303c42739bcd8f4b6344a74c : SUCCESS in 2h 02m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b06ab1340e5d47fb85ea7d415baddf45 : SUCCESS in 55m 50s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/7c389370e2ab497c8a0a4e508838fa96 : SUCCESS in 2h 01m 57s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/4882a4dd7a1541f1b4d42edc6dc7a1b8 : SUCCESS in 1h 52m 42s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/19b8ca48c48e4d1ab039e3c8a86fd58c : SUCCESS in 1h 00m 57s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/83ae8b1a09b64211b1bb08e4a1c320dc : SUCCESS in 2h 02m 45s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/514e16963b03416b9eed46e79416c4a4 : SUCCESS in 29m 06s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/e96068421da14bbbb6769cb143babfd9 : SUCCESS in 56m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6403793c8c546d73bc738130015fbc71a900e064","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-05-06 18:33:17.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"eb8723c4a5b7de626e1fccea22e685c6ad90e840","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-06 18:46:05.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5cbb148b8ccf4edb9ff30824d2d0fc19\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9495dc99982446afa4757c2c6631cc56 : SUCCESS in 7m 37s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/571748a8b6a142738058e8865aa385ce : SUCCESS in 7m 58s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4911414a8af645fa83c0392395cb609f : SUCCESS in 11m 03s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"78cf4fd3a01ab40f88c924e28e6579c3052c646e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-06 20:38:09.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9f6bba338bb9433e9a59ce3aab8e533a\n\n- grenade https://zuul.opendev.org/t/openstack/build/fda72ecc1cf3438cb23b2d30aba0a9fa : SUCCESS in 39m 22s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3c60a9d29e1a4c618e91b8c9c510e843 : SUCCESS in 42m 33s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/fb89c24b1e2d49ac9b5c80545e2c27c8 : SUCCESS in 1h 14m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8ef0e2c93a144deb87508b67f06e8082 : SUCCESS in 1h 03m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96c6ddfc065d41d695c21fcdf77542ef : SUCCESS in 5m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/85f42937538549738e7b5ac42a4b902b : SUCCESS in 6m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/61393b5ab2fe4f67b139a0fd9fc2ad16 : SUCCESS in 8m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/40b1c9e0fc8d4ca5b16983e6b040816f : SUCCESS in 9m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29e10fad48d6462aa96df425b6a25f13 : SUCCESS in 6m 52s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0ef288bfb419472c8461a376f80be7da : SUCCESS in 9m 17s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/28d0b1c1002145d9b033b46e062fdccd : SUCCESS in 12m 30s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/7c63e20cbe0e41ccb7f46439329769ef : FAILURE in 2h 00m 58s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/ae3070a458c2436bb6af49ee08905af7 : SUCCESS in 1h 02m 36s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/c4ccc5312fd44c45bc4651da754b2d43 : SUCCESS in 1h 08m 49s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/b30e148eee8e48e8998a16497192a41a : SUCCESS in 1h 09m 47s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/88b23dcd288948be94894f6eeed9aceb : SUCCESS in 1h 11m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8f9612107d2e46c688c8490fcbff5fe0 : SUCCESS in 29m 24s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a18a63267864429396772204347fbac6 : FAILURE in 1h 10m 50s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/98be9a6a872c444ea18583100f09304d : SUCCESS in 1h 05m 26s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/f8ec87a31aa04cc7952f884305e7956a : SUCCESS in 1h 01m 01s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/9ef7c3da959948a89f9f151072182586 : SUCCESS in 1h 12m 23s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/48d685d273a84a059c1d086c179167ee : SUCCESS in 13m 46s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/64d87cfff46f47ccbf03b4bf8a4841be : SUCCESS in 1h 01m 44s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"be7e7851e4799b0e8c3f1963b0f16a414490cf10","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-05-15 18:50:26.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"20d475ecfdd8d806f6553c04326ca1168436f26e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-15 18:50:37.000000000","message":"Patch Set 3: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/glance_store for 949129,4","accounts_in_message":[],"_revision_number":3},{"id":"ebff1a392dfc271fdd274d55166889d75ae5d9f7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-15 18:50:37.000000000","message":"Patch Set 3:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/glance_store for 949129,4","accounts_in_message":[],"_revision_number":3},{"id":"f7a0d2c6e1022f1bb40dbaeb6a9a9380cb32a745","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-05-15 19:24:30.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f6c2b332de1763c74bdc88964430c969feb82a0a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-15 19:45:04.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1f4fa2b11e364c47acc2ebf431ba7a9a\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/78b59fdb22b544c5bd0d240bf55a5485 : SUCCESS in 6m 21s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ad49939749914b798aa45e81ee167284 : SUCCESS in 6m 50s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/0f503183e9c5439f9191af6daf97539c : SUCCESS in 18m 05s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"9747674c6923b42772436c71e8143bb58e03cb63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-15 22:27:49.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bc5909ec32444d579643459f4b125e44\n\n- grenade https://zuul.opendev.org/t/openstack/build/d67523bfded04f31a874c7478a88dd75 : SUCCESS in 56m 15s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b0077f993c5f4d84a2206fc06ecbbb73 : TIMED_OUT in 3h 02m 34s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/77a812ae1e624b86b60b4c441fddae55 : SUCCESS in 1h 29m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0797ba0d49c948ad9bafe3db8322876e : SUCCESS in 56m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dff539da0db74398a14c57de7f760a74 : SUCCESS in 2m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/230f737f8b6b4164a509ab49e27073cb : SUCCESS in 3m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b748a566f0314c1791a44e1d8e855ace : SUCCESS in 2m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7dff08987dc94bba88c6202c00cc865c : SUCCESS in 7m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7325dc9c08c54320b8f99b3028efaede : SUCCESS in 6m 21s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/390f3147db1a4893a34724ac0e4bcdfa : SUCCESS in 7m 49s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/d9ab93cfd13a49479e72b1c22dd719d8 : SUCCESS in 9m 56s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/e7e99b2621174424b467a48e57e55e4a : FAILURE in 1h 51m 43s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/d907e4b79e074f91b71241694795b23d : SUCCESS in 1h 41m 27s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/ac977c6fd5ca45fbb2b95d6c1b2484b5 : SUCCESS in 1h 50m 04s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/fea119d8b87d4fc1951ee10a317c6a64 : SUCCESS in 1h 41m 34s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/bc4404b78eb046139648601c3f14f080 : SUCCESS in 2h 16m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9eb7ec9a20774f9987c5b333d31dfec7 : SUCCESS in 54m 45s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/fa4d195fd6554e34a53d2febfe7982ef : FAILURE in 1h 35m 04s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/60fa9265e8cf45108f20f356462ce3e4 : SUCCESS in 1h 51m 33s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/8f5f7f9051d34b8ab128d505be36ad67 : FAILURE in 32m 07s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/a34dfa5d16a04524836bc3999b099818 : FAILURE in 1h 37m 51s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/9b4eb0f00bcd4912b489b10b52d16564 : SUCCESS in 29m 43s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/3b93bdecce0a4c95b608067deda7721b : SUCCESS in 1h 02m 30s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"61f449b6e964c3d5bc7507ea207ce5938a3d5918","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-05-16 07:01:48.000000000","message":"Topic spec/set-size-on-upload-1 removed","accounts_in_message":[],"_revision_number":3},{"id":"c5535986524e33a424a9e7f26932e13869e58694","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-05-16 07:01:53.000000000","message":"Topic set to spec/set-size-on-upload","accounts_in_message":[],"_revision_number":3},{"id":"adfdd698ea08aa814fe406b9ea160494cec22be1","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-05-16 14:01:52.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"230afb10ba858fb17fa26612013d22a62992896b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-16 14:15:45.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/77c24e39301d4e62b3d8aa076893d840\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/dfcbcdd512f745b484b572c5891d1f46 : SUCCESS in 7m 21s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e82b5ec910ca42f0a660b14bc142fa9c : SUCCESS in 8m 18s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/0527cc78c2054f2a9d8cc801491ac847 : SUCCESS in 12m 01s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"824d3b4719d44595d18fb3a8d8af2cd096cb2705","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-16 16:15:53.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/25164d993cba44769856767cdc89131a\n\n- grenade https://zuul.opendev.org/t/openstack/build/274cc00ef3c141f086fa149601c38dbf : SUCCESS in 28m 06s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4cd0621efd07445c89b2dfac915311cd : SUCCESS in 1h 04m 07s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/20241e125bfb441180d36bc3e5ecb720 : SUCCESS in 1h 28m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4b24973aca57431f9f6adcffa6050f0b : SUCCESS in 1h 05m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f11e9f8cabe43aca3f83b71b38fcbaa : SUCCESS in 5m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/eb51d43c151043e28465d6e144359ab3 : SUCCESS in 4m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/610b326baa684a2fb1a60a3ce9fd3e04 : SUCCESS in 5m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/428cf405a006442a94f4193a0776ad8c : SUCCESS in 7m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb63f0f4c4c9497fbb50d1e5b27b05d6 : SUCCESS in 6m 14s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/894ee26b447a44978efb0f20e5c6201a : SUCCESS in 8m 47s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/ddd8b7ad8a8c4156ab0858c3b3cd7d14 : SUCCESS in 12m 09s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/ffa8d92d4f5d499eb58fb27109c0439d : FAILURE in 1h 45m 10s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/1277decfe8764c97828b892dca0ebd20 : SUCCESS in 1h 50m 09s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/12d31749d4af4482ac87fc2d1430acb0 : SUCCESS in 1h 55m 45s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/eac7c77dbce5454196a500ac455b87d0 : SUCCESS in 1h 49m 17s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/165c3618359e417692fccb2b2730b98e : SUCCESS in 2h 07m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/df1fae270ac74e399cbc9a70fa827d01 : SUCCESS in 58m 24s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/93e196ae9a0a4676a6c2984f6e5dae60 : FAILURE in 1h 50m 22s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/e65b56c455244120b0922c9153cd8a03 : SUCCESS in 2h 12m 34s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/4f0f029c438e4fef921855342e889059 : FAILURE in 29m 15s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/c4845bc0bb694b319041993bcd4de1fe : SUCCESS in 1h 41m 27s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/4b92eb68a6ff4b3690cadf1f0c5a7ae5 : SUCCESS in 32m 24s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/aa1c36823cf345e3bfcf80acf2e3fdf8 : SUCCESS in 59m 19s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b57cf0619b99011e228114f3803c286d0be54725","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-06-03 05:59:08.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"07c8fd74daaa70bc7629e5ca81982f698759b7c0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-03 06:35:50.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/539e2b2fc3e449258825e235c28cd124\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/bc60505477144eb6b0fd54159e431256 : SUCCESS in 27m 50s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3c03c80cbd0a4b96b2130bf12c65a603 : SUCCESS in 19m 18s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/375235d702c84ed6add9794b59a0bcf9 : SUCCESS in 34m 13s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"6f90be2db57744371687cfe098b5133acd004a83","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-03 08:34:08.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/579e512a51204d44913280f62c9b059b\n\n- grenade https://zuul.opendev.org/t/openstack/build/00e6499041ca4b47bdfbdd9c6745fdf5 : SUCCESS in 55m 11s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/110ab1dfeb2d469a8c2364d7507cb3c1 : SUCCESS in 55m 50s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b8d83169429040029dd2d3b0917e9b16 : SUCCESS in 1h 19m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/06a4f0fb320e4af2bcb2a976dd348bc0 : SUCCESS in 1h 03m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ce13bd7161942a08c65e93c45fc455b : SUCCESS in 4m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/acea0b5665b9426ebd5b116b599b5cdc : SUCCESS in 5m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0b82ccf35678405a95d25e40abd7571a : SUCCESS in 4m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8e1343ced97b42e892bce9f0df6cb589 : SUCCESS in 6m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad34d0f8a2e64835b3f42dfdbd60651a : SUCCESS in 7m 59s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d0b3a2f7e82c4d39a0eb93fdc618892c : SUCCESS in 6m 24s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/2fea09eb5cdc44db9a4e29394e7fc796 : SUCCESS in 10m 31s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/03de2bda049b414b843370bd1b3153e8 : SUCCESS in 2h 15m 48s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/cbeb9b655c464f0481781b714490f332 : SUCCESS in 2h 03m 24s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/6cba26e5b89a44a089149062ffd2dd8c : SUCCESS in 2h 09m 26s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/2e0786571572406b8b37b27a75e0bb65 : SUCCESS in 2h 27m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4f2ba8bc091f4481b10490b05caee4b3 : SUCCESS in 55m 55s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ae6cbc35a39a4df58e1b6e6e30296619 : SUCCESS in 1h 33m 19s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/cb2c60bad014475dbd280e4f31e41896 : SUCCESS in 1h 58m 37s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/8be5cdac3578412fa21720cbc8b3a43d : FAILURE in 29m 51s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/48d02887fd964fa6871e529d557f6c43 : SUCCESS in 1h 58m 03s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/fad5aa9849b74528b12592d17fc9b2d6 : SUCCESS in 33m 50s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/57fb9f80639f4333ae992717c1a587b1 : SUCCESS in 1h 00m 27s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"0f31d686a0507489699c71fee4b5ae857e99549b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-06-27 10:04:26.000000000","message":"Patch Set 5: Patch Set 4 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"63824bdb5b3dcca6b0357e587bde68018e193725","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-27 10:30:39.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e78afc8ee854dde863b545298938bdf\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/84a71bd0ea174faf9b2180ebb7950935 : SUCCESS in 9m 39s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9eb0de5d29084e6f8089f854f3926dd2 : SUCCESS in 9m 35s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/32acc8abff724ee1a99c6ac65cf81d39 : SUCCESS in 10m 00s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"e0d2f61dbe2cbc0b3879355f06b513fa20d11e23","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-27 12:32:03.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/367d52bbd1b948bcb22529b83cb4d932\n\n- grenade https://zuul.opendev.org/t/openstack/build/40bb8fc2db8c4318991b71e535921778 : SUCCESS in 1h 00m 32s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/98628b8423c94fbe938997af856c048b : SUCCESS in 58m 15s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/02916824300e4a73a63f4c664e8186c6 : SUCCESS in 1h 24m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7f41316fffb7471087253bd0eca7b18c : SUCCESS in 1h 02m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf4c46f40db64d55889ae2aac480efe3 : SUCCESS in 4m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c75dd2ed3b484641998177d851b0b523 : SUCCESS in 4m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fbc2c7ed547b43618f55645094fd09e0 : SUCCESS in 6m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e1293f434afe4e968243736a06579658 : SUCCESS in 8m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f04af5478ab44faaf13837776014231 : SUCCESS in 3m 08s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9f8586e504e245ae8c2f8577b4792382 : SUCCESS in 8m 00s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/352d647c248d4413a27860ed64709e80 : SUCCESS in 10m 59s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/7c3173ac48a745dc938b7e416909bde4 : SUCCESS in 1h 03m 25s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/d2481d8336cc4cea9ecc19407615f2f8 : SUCCESS in 1h 59m 16s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/2dbceb3a6e2049dda1450ee2da18f709 : SUCCESS in 2h 21m 39s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/f1e16c05d331446eb8955a7cd51c6e1f : SUCCESS in 1h 12m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/18258201f86148609e11905488555563 : SUCCESS in 51m 16s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/7cf575c072bc4938ac8def29a1ad63d8 : SUCCESS in 1h 55m 21s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/fc969bfe1a7a4fccaa5867bab57fe8fd : FAILURE in 16m 47s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/56f5802e087d4600b395d64a87341404 : SUCCESS in 1h 45m 26s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/87b8369a39f145888236efad4e2a3c0f : SUCCESS in 27m 38s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/c8d16ef568fb4fea96a81f6b657b5bab : SUCCESS in 58m 50s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"31fd8e824a1f5705ffea6a1e57a63e46b032f29c","author":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"date":"2025-07-03 08:33:37.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"8da7ea20ae132bb58dba46489e49fe221780062a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-07-03 08:47:02.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"b8fdc8c017a06f58a77af2e9df13e14693d2e075","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-03 09:36:28.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0d58f82efd5a42e0aba4864af8fd37b2\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/18dcdbd39a2a4f4381fddd129e992da0 : SUCCESS in 9m 21s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/58b0f57cb6364325b933e366f295fdc2 : SUCCESS in 7m 28s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/7f27e2194d934eb19818ed94b58f56bc : SUCCESS in 15m 42s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"fb60bc95bc229c2e5fcb91524cd7af0657edd517","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-03 11:49:48.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e1fd7df32e224b748de6bc5dd82a1a40\n\n- grenade https://zuul.opendev.org/t/openstack/build/650285133e9745d98e32ccd2b4ec54ee : SUCCESS in 56m 37s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/843dd6708ba1420d9384433d1bf51ea6 : TIMED_OUT in 3h 01m 36s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/281fd06b936b441590f814eea7651b4c : SUCCESS in 1h 20m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8715da86870b4fa08bef3194e5408d7e : SUCCESS in 37m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75f3f850cd2b482ab2b16284b40bd591 : SUCCESS in 3m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/43fdb27e9e704948a5f3fdf4f81e38fc : SUCCESS in 4m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/420ba542ab1a4a44995608fe0a3161c6 : SUCCESS in 5m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/51f0cc39c92348c394b2a191704f4f82 : SUCCESS in 7m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c1364ea113242f09c51ab5cb4471210 : SUCCESS in 5m 21s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/12799db1ced64e009578af3f12b819bb : SUCCESS in 6m 51s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/80c1f65f403a4ca9981bc3964ca2f1e3 : SUCCESS in 10m 44s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/e9302c3dd15a42768f468cc3a038bfa9 : SUCCESS in 1h 27m 23s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/121adc541f924a51b04a5b6833337e63 : SUCCESS in 1h 56m 25s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/568098ee60174e66a0e602b76f04475a : SUCCESS in 2h 08m 33s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/adc47d2f3a4741168d40ee6621fa98e4 : SUCCESS in 2h 10m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c836b71006124e7492d4b45fa56067bb : SUCCESS in 59m 00s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/2bc71955895143dfb1198a5e53c8a5c7 : SUCCESS in 1h 51m 43s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/b5772689530448ef8a169bd66fab47c2 : FAILURE in 34m 03s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/ff66412a87804250b8c780c797073002 : SUCCESS in 1h 47m 51s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f9967ef208b6486c9e27533f59575e9f : SUCCESS in 35m 02s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/182863c349694cec8fe0192cddf17288 : SUCCESS in 1h 13m 24s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"706e5c53baed012f620b4935d23ab31f4e773a06","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-07-03 14:01:41.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"241c70e28d588862a2c0bdc328982374d45f0cc9","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2025-07-03 14:24:19.000000000","message":"Patch Set 6: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"52f1685fbbc2b2dfdae381f17cdbe6ce9f7142be","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-07-03 14:32:01.000000000","message":"Patch Set 6:\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"225455768ccecb73a9fc683d503f11e92df11ec9","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-07-03 14:45:38.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"69a0f5c67b2126f30d8a0b7543c72aa5c01ce893","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-07-03 14:51:53.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"69a2d227d4dbf1d63831381b6599bb99ea9a7044","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2025-07-03 15:08:42.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"7cc2ab2fe2ed43709132de5f9bca50ca92418cfb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-03 15:41:42.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/33447a8d1c8246a9bb5c5d91eb25579a\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/65e092e2470346a18fe66f3f44964b8d : SUCCESS in 9m 02s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8ff7921d1faf4978896dbecb7c1d9a58 : SUCCESS in 9m 07s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c4e58abf3e5a4f7ba7259357a5dbb564 : SUCCESS in 11m 30s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"83a0cd6bd1cd217299c5e83efd71d02a075d70ed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-07-03 15:54:43.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"bc65093f4eb551b40601f90c759f77db794be860","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-07-03 15:55:06.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"2805b16b0855094972dad89f45ad0bf82f9748d2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-03 17:34:49.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5dffaa2e5a9c4a51a7c75e07781af6c4\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/acb060e16a4f4e7c833333cb9a9bef2e : SUCCESS in 6m 20s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6779b6705efe4b8686eb6dba3fec48cb : SUCCESS in 7m 32s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/eafff91089414ce99051e8344ac73ba0 : SUCCESS in 7m 55s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"bab91d3ba111d1783bb777de0291e1162c382ee9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-03 18:13:41.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/56c002e657654c6689f1d61b87371112\n\n- grenade https://zuul.opendev.org/t/openstack/build/970ba1ffb6114299b121e98eaa88e2f0 : SUCCESS in 1h 03m 16s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e6fd1f3d13884de3886a4cf2dcd6630d : SUCCESS in 1h 04m 44s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2a795a6194fd4c55be31e1115690039e : SUCCESS in 1h 46m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1cb056d15cbb469989b7fd7f4b27374a : FAILURE in 1h 12m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/686b80f33e0f4ea9b9cc59cfff05842f : SUCCESS in 4m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/eeea627c77c54e66a9116f3e9785aad3 : SUCCESS in 2m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5cb738fecefe4143965bc63ae7efff77 : SUCCESS in 2m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b315c14a49f54477bea04298b71ab06d : SUCCESS in 7m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12e0e1cc209d4a40869d790df8985b3f : SUCCESS in 6m 58s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/06e2b852de29445db9167228a3ca68db : SUCCESS in 7m 38s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/b8abd450d71740df8fba9a19991393f8 : SUCCESS in 14m 11s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/7fc17201f16247ca862f3aca70a37da9 : SUCCESS in 1h 10m 30s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/02ef0c2fbfd546e49abcc3cf8a6ac0cc : SUCCESS in 2h 01m 26s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/d8a78eab52d349c5b958ab3f26e082de : SUCCESS in 1h 49m 51s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/4d3b55ec451644c187281eb0a4d88e85 : SUCCESS in 2h 07m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d2d5502897f649058e0db2f3342c0955 : SUCCESS in 57m 31s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/7fcc737f79004e67827c2d1988e57f22 : SUCCESS in 1h 43m 14s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/d61f8035fe7e4a34b8d9a734bdc9f78d : FAILURE in 34m 00s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/b7a4e5c3b0d0442fabec44090a306bed : SUCCESS in 1h 55m 19s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/ec93a596f02a441bab6b5174f6d6fe71 : SUCCESS in 38m 06s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/f7ab0a509af54fc980e71c99cfd2fbd3 : SUCCESS in 1h 05m 10s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"1d6768c6b789c10fe2a144376593c899b7e96093","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-07-04 14:07:08.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"643fbc21d0d00a192fb3da341db4017af7a209fa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-04 15:16:55.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b7b46645ea7485ab682b00ad1c2339d\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b470a10b0f634790b9f61666058d83ce : SUCCESS in 7m 31s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7b78830ca096435297161fc575fc8bb8 : SUCCESS in 9m 01s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f7aca9847dec47508532bf839d8ef1ea : SUCCESS in 8m 36s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"790f4dcd6ab0a2b6fad1c63caa49e151d141cb0e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-04 16:29:47.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/79684a412b274b8f822c8ad69047ddff\n\n- grenade https://zuul.opendev.org/t/openstack/build/480b371e31634b029ec575261711e8f3 : SUCCESS in 1h 02m 00s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2b2e3977c3034da1adc0e4353e1060f4 : SUCCESS in 1h 13m 42s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/cc2785037c554b7ab1728c6eb7f6e612 : SUCCESS in 1h 30m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/936c4c5918ee44ce86b1e9aa46bf1ed9 : SUCCESS in 1h 10m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c4e12974dca43b3913a716887a38428 : SUCCESS in 6m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d7e3932d6e29424dadb5369fcbc46b06 : SUCCESS in 4m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5f6b9180278d4ebfb7bb8b76bc07dd88 : SUCCESS in 6m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6364d520993444ac80916b832836d6fc : SUCCESS in 6m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9738764ba0314438888b6a98db50eda8 : SUCCESS in 4m 51s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/dc9c937074094b4396040626fb6902c6 : SUCCESS in 7m 26s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/522c6d04309d4bbe959648afe592a814 : SUCCESS in 9m 06s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/deaa5581938e42e9be7b4fad0b200ddd : SUCCESS in 1h 55m 31s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/9c6f02bb53fd4fd1a97f1d51acec4e08 : SUCCESS in 2h 01m 31s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/006b8da1db41410cb65c88e0ea02d204 : SUCCESS in 2h 13m 03s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/2d1c19b249f34298a255412a633b116b : SUCCESS in 2h 12m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d8d56802b687496dabd1913491245592 : SUCCESS in 51m 25s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9450cb985a7945fc8feef54a749ff0d5 : SUCCESS in 1h 43m 16s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/ba51b2b40db1495490a10da4230487c1 : FAILURE in 31m 23s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/1f899724dcc048ec909a77cf1cf0c219 : SUCCESS in 1h 59m 58s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/70a0d676214347e091e4a71fcdc40c51 : SUCCESS in 29m 16s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/3f9a5f9d37ad461bab36e9a39ad7b0b2 : SUCCESS in 1h 09m 06s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"f9a3d2cd8a0df69bd94fdeb3475c1abae8e59777","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-08-04 13:17:04.000000000","message":"Patch Set 7: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"8f84c27900cb8ecd280b583a3c82542baf4cff24","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-08-04 14:03:55.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"490da6e3614fef2255ff92bbe1df7a1e7b820b6b","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-08-04 14:04:11.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"c04e001514cb468fdadadd77f3d8be62869c9eed","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-04 14:19:09.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b4f9dbba4b354583a92403cc8661ab3e\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e2f32c76ba9049749256bc586d9fe77b : SUCCESS in 9m 49s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9089b82c5c834058913edf007a6c30fb : SUCCESS in 10m 32s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/7e7daea72ca24f779e1ae69435563a8b : SUCCESS in 13m 33s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"8827f25cf575f7ae03610397749bf36f5f38bed3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-04 16:15:21.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6a173b1a9d6e4a7c96f0a15cfc59a776\n\n- grenade https://zuul.opendev.org/t/openstack/build/cc2e43b040804406bb1e7fcf47075a5c : SUCCESS in 53m 56s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/93c7fe8b9f3a46a2b02d394c240271a1 : SUCCESS in 1h 05m 40s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5a28a2dd782b441195be7b5cbf96e6f7 : SUCCESS in 1h 38m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b7e97679ba2b4c40849e281387c393a3 : SUCCESS in 55m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa8f0c686fc946d381a96325f76c2556 : SUCCESS in 4m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7bc354693afd4bbba4bc44e3cfadb49a : SUCCESS in 4m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e6f6281431574e2daaa165fd488d33ab : SUCCESS in 4m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/52cd55e0bf5c45359f4720f96974c4f2 : SUCCESS in 9m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99f2c117bca0436b9ef0576bbde984f2 : SUCCESS in 5m 01s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/bac8b28c210d49fcafac55be39281421 : SUCCESS in 7m 03s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/da40e99fe00f43faa9b173f6c2b4ef00 : SUCCESS in 10m 15s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/41fdb7716c964cdf89444a7a5360b809 : SUCCESS in 1h 10m 05s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/e25897d8d4a94471b2206ee4f61218bd : SUCCESS in 1h 50m 35s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/b7e333da58fe4036b53bbbc7a247749c : SUCCESS in 1h 47m 46s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/7f8356cb66cd4ca083995080d134487e : SUCCESS in 2h 06m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/10cbbb2ee2b047a4be061523555820dd : SUCCESS in 42m 50s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/859d72fe51e445cbba86c46133be12cd : FAILURE in 58m 36s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/3325ac9c7c2a4424bbc11b79d84a87e8 : FAILURE in 15m 54s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/d34d67754c6341959335280718f063ac : SUCCESS in 2h 06m 03s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/51b88f9b116a4185806f12c32cced27c : SUCCESS in 31m 31s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/3ca58b000f714db2b4bc787988ddc0a4 : SUCCESS in 1h 42m 14s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"3b0601b960b9a2d01f7f8ff084405696f78cb951","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-08-04 16:18:01.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"78e4ceda50ba9c0f2a4f0242a2b073b694d1058d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-04 16:33:43.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/499a1a04319348bbae60166f9ca68348\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ef3e580f0d17475ab5a9fa0dd287bd0d : SUCCESS in 9m 21s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7f39f73bc5bf4f28810828cef7b88fb5 : SUCCESS in 11m 01s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b000d31d841b45d199905d8355f1d020 : SUCCESS in 14m 02s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"47d82cd86ae8fa074801f22002a5b2b4858876d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-04 18:23:09.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a1601c54bc8145ed9936ca8504e09e65\n\n- grenade https://zuul.opendev.org/t/openstack/build/97198f855eca4d5ebb0827a3a564506e : SUCCESS in 55m 29s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/be7c01023cad44799687ca1619a69173 : SUCCESS in 34m 00s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f29855372973451aaf8132c4bb0e6a2d : SUCCESS in 1h 24m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ef913f2954ef423f91a2dab3f361573c : SUCCESS in 33m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e674297740de4a8dbe339e4e105fdfa0 : SUCCESS in 2m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/55df4f4123404b1eb16a76e379b2a609 : SUCCESS in 2m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1238fcdbdb5b494d8116aaef1883f5d0 : SUCCESS in 2m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/53d6ae0dfc18409a90c99a16f0775a7e : SUCCESS in 4m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9de07f8efc074a2cb1921894c7720147 : SUCCESS in 3m 49s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7591cbf560b745ba86a1aeea4aebda7e : SUCCESS in 4m 42s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/adf7aaa640ef467890b4b1b7fa1eaf30 : SUCCESS in 6m 06s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/26dfb57b6c5f4559864e4906f19cda7b : SUCCESS in 1h 07m 26s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/df31171b011b4c969d2b2436c2ad13ff : SUCCESS in 1h 50m 24s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/29a20747a36a49a7891aeb783d1e4e57 : SUCCESS in 1h 45m 57s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/7b7f0d331b9b448681a8217f737b5baf : SUCCESS in 2h 03m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ad3b8a6c22854a93a21d70e180d7d080 : SUCCESS in 48m 55s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e397b266e9fb454ca3b992cef216957c : SUCCESS in 1h 31m 24s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/9c0e474cf56b483791825cd0ac7661b3 : FAILURE in 26m 00s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/67bf8983044e4c87a63699c8308e9aca : SUCCESS in 1h 39m 58s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/a451e3f8b1a44d5198ac22ef72410c7b : SUCCESS in 30m 50s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/15b8d187850d43cdb6cbbc797fadcfe7 : SUCCESS in 1h 08m 38s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"f0aa2bebdce03b2c8c58c417271f0fb0c1e68687","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-08-06 15:12:31.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"56d57112162448b6eab441bb8a37e4ea355356a2","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2025-08-06 17:37:49.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"ca91835acc7c686671ac8f2e7a63a40587c0c277","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-06 17:38:19.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"fc8a7d6133e568a37aaa1843972fcb93e2fd427a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-06 20:02:22.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/287d596a1d8047c1bc4f9056e1747efb\n\n- grenade https://zuul.opendev.org/t/openstack/build/6e64ce3be86c42bab08d88c926eff8e9 : SUCCESS in 32m 52s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3cf169c7a0014d39bcf0ae95c96faecd : SUCCESS in 34m 18s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7ec2b9a7a0874401b22c44b1b1b6f839 : SUCCESS in 1h 29m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9a0fe996443a4a1b8e68025e39b6380f : SUCCESS in 1h 02m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8480330f7714f9282c48f8995c4a7b6 : SUCCESS in 4m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/437ff2bfa446463eb8e77db85a931b6d : SUCCESS in 3m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/be79d2a58a664b5e930eb3394dd6a192 : SUCCESS in 7m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8bac670676af4fe8be0311540fc9ccf2 : SUCCESS in 5m 32s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b9dab8f236db4805926f6a69cfd2eb58 : SUCCESS in 8m 22s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/0922a720d0314008ac3ac2908e498d5b : SUCCESS in 11m 02s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/a285835726ac44a8a390c7b6e4f69326 : SUCCESS in 1h 53m 11s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/fbb7555e7ec14444952ddff755d42c66 : SUCCESS in 1h 59m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7a8d916c78904d1882b601078ced7840 : SUCCESS in 27m 39s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a529c1eeb6a74504be2d7470aade6fb0 : SUCCESS in 1h 01m 32s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/7aaa0f680f3f4f7da1a083367b97b975 : SUCCESS in 32m 01s","accounts_in_message":[],"_revision_number":8},{"id":"8cfe71df7bf3c4f4b0b4a931228a13f01e3362a5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-06 20:02:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8},{"id":"41ebde93f49c8749b7d1b05bf5fe247822c97058","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-06 20:03:14.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f204c6576011454d8d52737f51ff20f5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3dc585627bf4623aa2fca0fc65330a8 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"62a323aeddcc0d8895b99d111ffb89992687d340","revisions":{"644c0bed28fff5ba4cb5cc4c73b0a99039afbc3b":{"kind":"REWORK","_number":1,"created":"2025-05-06 15:17:36.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/03/948903/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/03/948903/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/03/948903/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/03/948903/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/03/948903/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/03/948903/1"}}},"commit":{"parents":[{"commit":"55158e2ac1d2954782396adda2ddb0c78c233e57","subject":"Merge \"Fix vTPM metadefs.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/55158e2ac1d2954782396adda2ddb0c78c233e57"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-05-06 06:22:48.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-05-06 14:25:08.000000000","tz":0},"subject":"DNM Test image size","message":"DNM Test image size\n\nChange-Id: Ib2c282bc19d48361b16fb1b2549cfcab80dea29c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/644c0bed28fff5ba4cb5cc4c73b0a99039afbc3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/644c0bed28fff5ba4cb5cc4c73b0a99039afbc3b"}]},"branch":"refs/heads/master"},"5980b3ed5585ed585ba8f50f943f8fd5b0b3b434":{"kind":"REWORK","_number":2,"created":"2025-05-06 18:33:17.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/03/948903/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/03/948903/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/03/948903/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/03/948903/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/03/948903/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/03/948903/2"}}},"commit":{"parents":[{"commit":"55158e2ac1d2954782396adda2ddb0c78c233e57","subject":"Merge \"Fix vTPM metadefs.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/55158e2ac1d2954782396adda2ddb0c78c233e57"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-05-06 06:22:48.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-05-06 18:11:35.000000000","tz":0},"subject":"DNM Test image size","message":"DNM Test image size\n\nChange-Id: Ib2c282bc19d48361b16fb1b2549cfcab80dea29c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5980b3ed5585ed585ba8f50f943f8fd5b0b3b434"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5980b3ed5585ed585ba8f50f943f8fd5b0b3b434"}]},"branch":"refs/heads/master"},"163297837c9ddcbeefa29293d0af0321e6f5ddc7":{"kind":"REWORK","_number":3,"created":"2025-05-15 18:50:26.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/03/948903/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/03/948903/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/03/948903/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/03/948903/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/03/948903/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/03/948903/3"}}},"commit":{"parents":[{"commit":"9e8bf37c850f19d3778194f3fbc4bb047ac1bf68","subject":"Merge \"Refactor scrubber to remove use of eventlet\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9e8bf37c850f19d3778194f3fbc4bb047ac1bf68"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-05-06 06:22:48.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-05-15 18:22:14.000000000","tz":0},"subject":"Read image size from request header","message":"Read image size from request header\n\nImage upload and image stage APIs now reads the request header\n`x-image-meta-size` and pass the image size to actual storage\nbakend while uploading the image data. This change now also\nset image size in the database even before image upload or\nstaging operation starts. As per backend implementation\nif actual image size mismatches with the size proivded in\nrequest header then the process will be aborted and HTTP 400\nresponse will be returned to the user.\n\nDepends-On: https://review.opendev.org/c/openstack/glance_store/+/949129\nDepneds-On: https://review.opendev.org/c/openstack/glance_store/+/949124\nDepends-On: https://review.opendev.org/c/openstack/glance_store/+/949626\nDepneds-On: https://review.opendev.org/c/openstack/glance_store/+/949362\nDepends-On: https://review.opendev.org/c/openstack/glance_store/+/949137\nDepends-On: https://review.opendev.org/c/openstack/python-glanceclient/+/949787\n\nImplementes: blueprint set-size-on-upload\n\nChange-Id: Ib2c282bc19d48361b16fb1b2549cfcab80dea29c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/163297837c9ddcbeefa29293d0af0321e6f5ddc7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/163297837c9ddcbeefa29293d0af0321e6f5ddc7"}]},"branch":"refs/heads/master"},"b562c73304e9e7d1e6929903dad57022e4c972d3":{"kind":"REWORK","_number":4,"created":"2025-06-03 05:59:08.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/03/948903/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/03/948903/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/03/948903/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/03/948903/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/03/948903/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/03/948903/4"}}},"commit":{"parents":[{"commit":"d12086ec6a3cf1a22a2766bd1ba9c5b85ffba8a4","subject":"Merge \"Functional tests migration of TestImageMultipleBackend\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d12086ec6a3cf1a22a2766bd1ba9c5b85ffba8a4"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-05-06 06:22:48.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-06-03 05:30:53.000000000","tz":0},"subject":"Read image size from request header","message":"Read image size from request header\n\nImage upload and image stage APIs now reads the request header\n`x-openstack-image-size` and pass the image size to actual storage\nbakend while uploading the image data. This change now also\nset image size in the database even before image upload or\nstaging operation starts. As per backend implementation\nif actual image size mismatches with the size proivded in\nrequest header then the process will be aborted and HTTP 400\nresponse will be returned to the user.\n\nDepends-On: https://review.opendev.org/c/openstack/glance_store/+/949129\nDepneds-On: https://review.opendev.org/c/openstack/glance_store/+/949124\nDepends-On: https://review.opendev.org/c/openstack/glance_store/+/949626\nDepneds-On: https://review.opendev.org/c/openstack/glance_store/+/949362\nDepends-On: https://review.opendev.org/c/openstack/glance_store/+/949137\nDepends-On: https://review.opendev.org/c/openstack/python-glanceclient/+/949787\n\nImplementes: blueprint set-size-on-upload\n\nChange-Id: Ib2c282bc19d48361b16fb1b2549cfcab80dea29c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b562c73304e9e7d1e6929903dad57022e4c972d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b562c73304e9e7d1e6929903dad57022e4c972d3"}]},"branch":"refs/heads/master"},"96d160e441677110c34651ae73096d398c666eb2":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2025-06-27 10:04:26.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/03/948903/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/03/948903/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/03/948903/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/03/948903/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/03/948903/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/03/948903/5"}}},"commit":{"parents":[{"commit":"2554cad01f4446488c132080830c286a7086dda7","subject":"Merge \"Remove glance-centralized-cache job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2554cad01f4446488c132080830c286a7086dda7"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-05-06 06:22:48.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-06-27 10:04:26.000000000","tz":0},"subject":"Read image size from request header","message":"Read image size from request header\n\nImage upload and image stage APIs now reads the request header\n`x-openstack-image-size` and pass the image size to actual storage\nbakend while uploading the image data. This change now also\nset image size in the database even before image upload or\nstaging operation starts. As per backend implementation\nif actual image size mismatches with the size proivded in\nrequest header then the process will be aborted and HTTP 400\nresponse will be returned to the user.\n\nDepends-On: https://review.opendev.org/c/openstack/glance_store/+/949129\nDepneds-On: https://review.opendev.org/c/openstack/glance_store/+/949124\nDepends-On: https://review.opendev.org/c/openstack/glance_store/+/949626\nDepneds-On: https://review.opendev.org/c/openstack/glance_store/+/949362\nDepends-On: https://review.opendev.org/c/openstack/glance_store/+/949137\nDepends-On: https://review.opendev.org/c/openstack/python-glanceclient/+/949787\n\nImplementes: blueprint set-size-on-upload\n\nChange-Id: Ib2c282bc19d48361b16fb1b2549cfcab80dea29c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/96d160e441677110c34651ae73096d398c666eb2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/96d160e441677110c34651ae73096d398c666eb2"}]},"branch":"refs/heads/master","description":"Rebase"},"1ea2795c0b1c8c1ea02edf8f67b85f94f7117db4":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":6,"created":"2025-07-03 08:47:02.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/03/948903/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/03/948903/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/03/948903/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/03/948903/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/03/948903/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/03/948903/6"}}},"commit":{"parents":[{"commit":"456886e0e3776ee88c7a6bbb81a7ca360df1971f","subject":"Fix py312 and py313 failing unit test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/456886e0e3776ee88c7a6bbb81a7ca360df1971f"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-05-06 06:22:48.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-07-03 08:23:04.000000000","tz":0},"subject":"Read image size from request header","message":"Read image size from request header\n\nImage upload and image stage APIs now reads the request header\n`x-openstack-image-size` and pass the image size to actual storage\nbakend while uploading the image data. This change now also\nset image size in the database even before image upload or\nstaging operation starts. As per backend implementation\nif actual image size mismatches with the size proivded in\nrequest header then the process will be aborted and HTTP 400\nresponse will be returned to the user.\n\nDepends-On: https://review.opendev.org/c/openstack/glance_store/+/949129\nDepends-On: https://review.opendev.org/c/openstack/glance_store/+/949124\nDepends-On: https://review.opendev.org/c/openstack/glance_store/+/949626\nDepends-On: https://review.opendev.org/c/openstack/glance_store/+/949362\nDepends-On: https://review.opendev.org/c/openstack/glance_store/+/949137\nDepends-On: https://review.opendev.org/c/openstack/python-glanceclient/+/949787\n\nImplements: blueprint set-size-on-upload\n\nChange-Id: Ib2c282bc19d48361b16fb1b2549cfcab80dea29c\nSigned-off-by: Abhishek Kekane \u003cakekane@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1ea2795c0b1c8c1ea02edf8f67b85f94f7117db4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1ea2795c0b1c8c1ea02edf8f67b85f94f7117db4"}]},"branch":"refs/heads/master"},"0996290c02d98e42b9b31109e7caf7ba7fe175ee":{"kind":"REWORK","_number":7,"created":"2025-07-03 15:54:43.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/03/948903/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/03/948903/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/03/948903/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/03/948903/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/03/948903/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/03/948903/7"}}},"commit":{"parents":[{"commit":"456886e0e3776ee88c7a6bbb81a7ca360df1971f","subject":"Fix py312 and py313 failing unit test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/456886e0e3776ee88c7a6bbb81a7ca360df1971f"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-05-06 06:22:48.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-07-03 15:29:58.000000000","tz":0},"subject":"Read image size from request header","message":"Read image size from request header\n\nImage upload and image stage APIs now reads the request header\n`x-openstack-image-size` and pass the image size to actual storage\nbakend while uploading the image data. This change now also\nset image size in the database even before image upload or\nstaging operation starts. As per backend implementation\nif actual image size mismatches with the size proivded in\nrequest header then the process will be aborted and HTTP 400\nresponse will be returned to the user.\n\nDepends-On: https://review.opendev.org/c/openstack/python-glanceclient/+/949787\n\nImplements: blueprint set-size-on-upload\n\nChange-Id: Ib2c282bc19d48361b16fb1b2549cfcab80dea29c\nSigned-off-by: Abhishek Kekane \u003cakekane@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0996290c02d98e42b9b31109e7caf7ba7fe175ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0996290c02d98e42b9b31109e7caf7ba7fe175ee"}]},"branch":"refs/heads/master"},"62a323aeddcc0d8895b99d111ffb89992687d340":{"kind":"REWORK","_number":8,"created":"2025-08-04 14:03:55.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/03/948903/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/03/948903/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/03/948903/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/03/948903/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/03/948903/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/03/948903/8"}}},"commit":{"parents":[{"commit":"17ad1aeaf86aea434952809b3ee727c4e6d487c2","subject":"Merge \"Replace oslo_utils.encodeutils.exception_to_unicode\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/17ad1aeaf86aea434952809b3ee727c4e6d487c2"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-05-06 06:22:48.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-08-04 13:37:58.000000000","tz":0},"subject":"Read image size from request header","message":"Read image size from request header\n\nImage upload and image stage APIs now reads the request header\n`x-openstack-image-size` and pass the image size to actual storage\nbakend while uploading the image data. This change now also\nset image size in the database even before image upload or\nstaging operation starts. As per backend implementation\nif actual image size mismatches with the size proivded in\nrequest header then the process will be aborted and HTTP 400\nresponse will be returned to the user.\n\nDepends-On: https://review.opendev.org/c/openstack/python-glanceclient/+/949787\n\nImplements: blueprint set-size-on-upload\n\nChange-Id: Ib2c282bc19d48361b16fb1b2549cfcab80dea29c\nSigned-off-by: Abhishek Kekane \u003cakekane@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/62a323aeddcc0d8895b99d111ffb89992687d340"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/62a323aeddcc0d8895b99d111ffb89992687d340"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
