)]}'
{"id":"openstack%2Fglance~948907","triplet_id":"openstack%2Fglance~master~I3736b9920506d32e1119b3f7475924e2b08d62f0","project":"openstack/glance","branch":"master","attention_set":{},"removed_from_attention_set":{"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2025-06-03 06:04:32.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I3736b9920506d32e1119b3f7475924e2b08d62f0","subject":"DNM Verify x-image-meta-size for all backends","status":"ABANDONED","created":"2025-05-06 15:21:13.000000000","updated":"2025-06-03 06:04:32.000000000","total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b266f451a28aa2bef4ff63906f64b22186cda76c","_number":948907,"virtual_id_number":948907,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-06-03 05:58:27.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-06 15:57:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-05-06 16:05:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f4eb5552b6f05f69aa0a3a87c6d0e70592eba2e3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-05-06 15:21:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b892affa287124c595b6a41e8996bf2aadefa6b1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-06 15:57:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8977ff350aa9484baa442c4514b9012f\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b26309c1f0204bdb893f24e8fb508578 : SUCCESS in 6m 42s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9f9cae196057477ab0bfe66ed982a6d3 : SUCCESS in 8m 37s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/3c21cbcb2d9842039abcc2e2c569f86f : SUCCESS in 10m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f49907c2012c0317c56b6dc973f339620cbd4e8f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-06 16:05:07.000000000","message":"Patch Set 1: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/948905 is needed.","accounts_in_message":[],"_revision_number":1},{"id":"48a681ae3e1a085aee86a32bee2f1b56c003680a","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-05-06 16:06:07.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"796851eb81880efe8d00c1b27dad8212360fdd71","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-06 16:20:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e4636bb1b50443d6a3c6a84d00b72b1b\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/dbcb87e693064381bfb1c0be0b28081a : SUCCESS in 7m 27s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/58e366ad5a3046c9813ebe89f0611922 : SUCCESS in 8m 14s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f5aaedd005e54256bc19d605a940a5a3 : SUCCESS in 12m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"47a737e9b5b306c06bbf44c7e89133a3dcc41258","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-06 18:11:33.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/03b74c2d0a4c42b18a2b77a6baef7110\n\n- grenade https://zuul.opendev.org/t/openstack/build/efb1a267f46c4261838ffcd330ead010 : FAILURE in 27m 39s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c0570760bcb54b21a269dc7cbcbefc82 : FAILURE in 26m 48s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f891b9c17d314ae5848f56bae669b2b4 : FAILURE in 1h 12m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/31bb5c5d65134e21b240e9bac68baf09 : SUCCESS in 37m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf1336004c1b4768baef394a85e5948f : SUCCESS in 5m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3d8af4b04f294814bf1da231734b14d7 : SUCCESS in 4m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/000d70ed3d384297a0f673eafe7f3a25 : SUCCESS in 7m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9b88b894d08b41748feef8cee4d46358 : SUCCESS in 7m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2114ba15db7649f381ce3f18f6826675 : SUCCESS in 5m 42s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/fd1d85370a00492fb38a2be35f9aca54 : SUCCESS in 9m 04s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/353a289836d844218acb01d5ad4c6c7e : SUCCESS in 12m 50s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/fea5e072396e456cbfceacf191948b4d : FAILURE in 1h 57m 24s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/c3ec5afffb31423d944b83b6e9d1246e : FAILURE in 1h 43m 51s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/bac8a4e0a3de4c9ca8234d94b7577a37 : FAILURE in 1h 03m 49s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/03e917898e994606b3f18232110363ff : FAILURE in 1h 54m 09s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/fb6ed4be98cf4028b70add6ef89c3d34 : FAILURE in 1h 49m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b2789e4469794a01a4e62ea4347196a8 : FAILURE in 56m 09s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/91a4b3b3b748472594348adbffde8ea3 : FAILURE in 1h 50m 16s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/13c1bf0654ef4725ae0d93606277241a : FAILURE in 1h 34m 26s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/6e5fd7879f3e4ac5a55c53384463e010 : FAILURE in 19m 45s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/a9ecfcbc49bd43a7b5c7b708511e40b5 : FAILURE in 1h 42m 13s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/035c7b568854498197f43a4c11471217 : SUCCESS in 27m 34s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/6ec8b1a991e14d1783dcd64e0b41c6c3 : SUCCESS in 48m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7cccdca245bba03a81a1161415680329ce63adeb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-05-06 18:33:17.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"67c1ce56531f2f3f1637a633f9f8023223d70824","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-06 18:43:10.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9501356171dc48d186357aeac5f48db2\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/854fc285eefe4e53aef46652a97afec0 : SUCCESS in 5m 30s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0e8aac6f686346208cad05f8ec663dc5 : SUCCESS in 6m 04s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/5f2271c889714cdb8fc4ebb9fcc8864e : SUCCESS in 7m 59s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"89cd9ec68463310f7e2385d4ba94b2609c7f080d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-06 20:49:27.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/08acd98e06ad472c843031c7c8d15fb3\n\n- grenade https://zuul.opendev.org/t/openstack/build/2bef34f195474e21acdeed04aaedf569 : SUCCESS in 57m 42s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0fa62828c9af4c08b94b187387091a95 : FAILURE in 1h 12m 15s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/dae6f19822574f93be6b7f9b3faf2122 : SUCCESS in 1h 32m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3a151dfac657419c9f4ba2823cdf4917 : SUCCESS in 1h 02m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86f10d75b4794b15837a25d415f9e406 : SUCCESS in 6m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/782e96ac315443d2bc6e8beaa0e6084d : SUCCESS in 4m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/274fa0271c694340991ecdf29cfd3446 : SUCCESS in 5m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/726812d20c0a4ac68ff903af8b813290 : SUCCESS in 7m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a314cc859ce430dac50e912f9f2dcec : SUCCESS in 6m 23s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/24a13ec2349641fcb3877d531c7f69a7 : SUCCESS in 8m 01s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/66e8daccdd634af5b07259c65e057e8a : SUCCESS in 11m 52s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/bbde10e1bcae49a18d549b0bb1762a4a : FAILURE in 1h 59m 14s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/ae08de56f6904b94a9989c2485101fca : SUCCESS in 1h 07m 41s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/8d71e84db8f04ecaa9db87215e7ddfcf : SUCCESS in 2h 06m 32s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/96cc77b11f9445cfb96857454cc8a261 : SUCCESS in 2h 10m 46s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/400a22349ab54f239deb9a9554f2a7f2 : SUCCESS in 1h 09m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0ee38065a57045029e301ee5d7c6c163 : SUCCESS in 33m 19s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d16110a093314a7989d781e39d971a00 : SUCCESS in 1h 52m 25s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/3ad5bd1874744ab8934e294de26c3eb0 : SUCCESS in 1h 06m 39s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/5e342b8a5c714070aeb07a87b35a74eb : SUCCESS in 58m 12s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/7f1a2fefad6d47a2b6681b22c7edbb72 : SUCCESS in 1h 09m 47s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/af87f09b3ca047498a126403a67b38ed : SUCCESS in 16m 36s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/4ee6d340d8d34a6ea1214e9dc676568c : SUCCESS in 48m 39s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"28a1b9e9ff33f480e38a0aeebd9ebb08184bb344","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-05-15 18:50:43.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"bbc7e3d582990864ddbafc5d3ee2398457697c10","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-15 18:50:54.000000000","message":"Patch Set 3: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/glance_store for 949129,4","accounts_in_message":[],"_revision_number":3},{"id":"d13f12da648c4e5baddc9f7caa192f5ede44ecaf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-15 18:50:54.000000000","message":"Patch Set 3:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/glance_store for 949129,4","accounts_in_message":[],"_revision_number":3},{"id":"e2687af0f8d0b3928b80b82f3f10b269f323b234","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-05-15 19:24:39.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"37fe45393992345db4fb26b2ca98fac96de90f3d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-15 19:44:20.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/537406cf648349d3926a9d4274908ff1\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/04bc2476abb6415693902dfdaabd30a3 : SUCCESS in 10m 48s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9fdca044cf714e409da2ad8314fe6c02 : SUCCESS in 12m 09s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f350edc12d4945378cfdf2be791e9b5b : SUCCESS in 16m 56s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0f0f185123694357bc9864c6fdbee3da115f5340","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-15 21:49:41.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a19950cbcef4638b119c898df359cd8\n\n- grenade https://zuul.opendev.org/t/openstack/build/454521f73ab9412b8cef9a0a4300a9f1 : SUCCESS in 57m 03s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/21a9069d1b1f4f54b9306b7d837c8366 : SUCCESS in 1h 07m 07s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9cc9bd66845e495595d0e14755e611fb : SUCCESS in 1h 29m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d619561581d44c95a66e17ac2397d6b7 : SUCCESS in 50m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25277a99d56140d182f3657817c794a3 : SUCCESS in 4m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/95c72330496b4345804a59102abbaad9 : SUCCESS in 5m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7c92b1487aba46e193e0a951978833cf : SUCCESS in 4m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/72a4bfa5b24b496685d11c05cfd46c74 : SUCCESS in 10m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/925f4c23cdcf4b339d5cdbbc6d36f2ac : SUCCESS in 6m 38s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1830717cceb64bcb96e0c067c90d1b8f : SUCCESS in 8m 19s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/1a1b28f24ae4439a93ab91201dc71013 : SUCCESS in 10m 42s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/f60c5ed38eab4bba8e384f990d26f2fb : FAILURE in 1h 47m 07s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/fe675b150d0f49e79bf333ad71d71b92 : SUCCESS in 1h 40m 34s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/41c6dd4f5f144f449c56de6a4811e102 : SUCCESS in 2h 06m 31s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/eec4257a6dfb48b08c41aee6ebab8d99 : SUCCESS in 2h 01m 02s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/3f09860c0e32475d8ca320de439af404 : SUCCESS in 2h 11m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/46b6ebf8438a44d380a6e48bff6c69bc : SUCCESS in 46m 11s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/89c2c42adfe74c72ad666896598dd72b : SUCCESS in 1h 33m 25s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/314406249ddc464fab2d54c59b2c994b : SUCCESS in 2h 14m 32s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/c3d453152db542d0a03618622427c68b : FAILURE in 25m 17s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/687f388799b24c879705a33bbd9cf282 : SUCCESS in 1h 38m 37s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/aeabf74df8d94125ab6784703d7cf439 : SUCCESS in 29m 11s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/817fe78ee6f34528aafebc7103b8e262 : SUCCESS in 59m 18s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"551d08243181e5e479a8c5b6fadf6be479b61c66","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-05-16 07:35:51.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"00d59097af5b4198782f30647c962694af5ddf85","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-16 07:47:14.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/85c37eaf04ad4383b926dc0c9cb70346\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0fb05d292ccc4c1f98567170e5d793f2 : SUCCESS in 5m 38s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0740ea3972404814b93f41e371faba48 : SUCCESS in 8m 03s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/77f50b69b9a74c118169f0129e8a48cf : SUCCESS in 8m 02s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"d4e8078a2bd78bdeb8170d7b78a2557c347c3fe3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-16 09:06:26.000000000","message":"Patch Set 4: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChanges https://review.opendev.org/948903, https://review.opendev.org/948905, https://review.opendev.org/949129, https://review.opendev.org/949626, https://review.opendev.org/949137 are needed.","accounts_in_message":[],"_revision_number":4},{"id":"759df903d8242528d7ecc105169b2f39c36452b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-05-16 09:07:51.000000000","message":"Patch Set 5: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"123680530bb1f08f93f5cc7b7869c7d9f75026de","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-16 09:24:26.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e792636e884c4b59951fe6ca58192341\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f7fa7f5729eb4d6ca3fe91da8cd677e3 : SUCCESS in 8m 18s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/395ab85ac5304e9ca85fa8f11caf425d : SUCCESS in 12m 46s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/08e4190476244ee190c97bd8dba4391e : SUCCESS in 9m 58s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"d5cf2ecc3eed2aa6c198a843db651b29eef7ba15","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-16 11:15:14.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/04b7a16ea9b44bf38c0b88a5b4d74ece\n\n- grenade https://zuul.opendev.org/t/openstack/build/3fb43d8c163b403b979efcd843dc883d : SUCCESS in 54m 43s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8326169afa124a0ab8302337a88b526a : FAILURE in 21m 49s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5f2ff609b74e4927bf0ba42a496e22b9 : SUCCESS in 1h 39m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2431cae6b50e49cd918b3915bd94b39b : SUCCESS in 1h 03m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9acba6c227904e91b666ab95ac6cfdb7 : SUCCESS in 4m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4994db24bcfc491aabf1d113e10e3709 : SUCCESS in 2m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/abdf801fcfa442619445bc01b3c8dbae : SUCCESS in 6m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/92f44ff30b64490fb691db9f91025909 : SUCCESS in 7m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ff1b5fab14f47b0b0a16528d507f3a4 : SUCCESS in 5m 04s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9a701ecf64d94e8480d31835a7151b3e : SUCCESS in 6m 53s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/5b7a4b0fdf234583a98b274505782de7 : FAILURE in 13m 26s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/4dfabfbe5b3c4ff3984c1721c5c85aae : FAILURE in 2h 00m 36s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/cdfc93c16f054320bc8af8ed8b09dff1 : SUCCESS in 1h 56m 13s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/e61adf335f424fa890a91149edebfa7b : SUCCESS in 1h 56m 30s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/aae0cb999aa242668dd4c2a86c3d1c78 : SUCCESS in 2h 05m 52s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/b7776e01a521461481c844fae36c9508 : SUCCESS in 1h 13m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2968f395ad004371ab9695626a48fad7 : SUCCESS in 57m 03s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/948dbca1467d4d65b69c02805edf6c9e : SUCCESS in 2h 00m 30s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/9484c25425cd430ab4e6a9720d8544b7 : SUCCESS in 1h 51m 41s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/dd5791b38ffd4e66b164c77c5c0a99e9 : FAILURE in 27m 01s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/6b43090cac4c49608e6dab271111d2e3 : SUCCESS in 1h 47m 39s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/432455a371734c65a6e2c145b75b4bc1 : SUCCESS in 29m 56s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/5e7fd6ef13774c619c1e07b139586f20 : SUCCESS in 29m 42s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"3ee7cda4983d7398a0ab754a61d0bf8066d66ea8","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-05-21 07:10:50.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"07d8ef910fabd2920dc0940f5a5b3558cfcd7283","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-05-21 07:11:01.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"95dc61015dc0fbf666c2266eaa40b22428e72b3e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-21 07:22:09.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/134527e2aa39467aa82a4f9101aa8a2e\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/97dfd17cb7b448c2a53c2ac3a0c08181 : SUCCESS in 5m 50s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f12d30175da04a75893502f9d471ac53 : SUCCESS in 6m 45s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ab92d2f59f394f889657eed2d780f115 : SUCCESS in 9m 12s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"969403b852355caa6e91857fd2ab0ad878bf03f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-21 09:55:06.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/09b0ac38c386414f980dd159d6cb9e06\n\n- grenade https://zuul.opendev.org/t/openstack/build/ecc0ce34212547eb8b898aad7402b05e : SUCCESS in 59m 31s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1abf36f63f544f6ea9e3ccf88bd112b6 : SUCCESS in 1h 04m 46s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2a9d6131b700466295ff737ee58acec2 : SUCCESS in 1h 19m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a604fc1a2ce64096a28e54633cf6b10e : SUCCESS in 1h 01m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0c3ab53a29c4f07a61579302809f1d9 : SUCCESS in 5m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/912758e332ed4415b184e0e63533b1af : SUCCESS in 4m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a26882cd8d994b76ad2dd44afed0312c : SUCCESS in 5m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a2f1f297da534c2fba7e891490756471 : SUCCESS in 7m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e563ad29c6b24f1f9f52d6e343ab2481 : SUCCESS in 6m 24s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/fc034b1511ea4bd4959d3577baa0221e : SUCCESS in 7m 28s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/ae31b3662022444aa79f52aeba7d0612 : SUCCESS in 10m 16s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/2aa2d8278cf34e14891128f746bc4030 : SUCCESS in 2h 02m 36s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/e2f44e64fa034e359178ec982cdca52b : SUCCESS in 1h 02m 06s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/1e11ed4bc78a48938acd47a60ecb9423 : SUCCESS in 2h 05m 02s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/b811800bdc0a4fd394f667c0d9079e49 : SUCCESS in 1h 56m 27s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/355b9a66bc004994beb5c9647b95ab05 : SUCCESS in 2h 29m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4741b1f984fc41a5b5113fbb1a2bcb0f : SUCCESS in 54m 22s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/59e6ac6af1c44bcaae705e4e285a3829 : SUCCESS in 1h 47m 34s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/c67432e06fe344deb649ae7d9539495b : SUCCESS in 1h 58m 49s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/dd3c5ef3792a4784b523ad08e857acf3 : FAILURE in 31m 23s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/772287508a3e47c6b245def059c533eb : SUCCESS in 1h 52m 05s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/61f0da90019a4a45bdd8946e24e91311 : SUCCESS in 29m 51s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/c910f81110024117a7055ece65ca54af : SUCCESS in 59m 51s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"fe4909d78abdf464a1d35bead82bb744fe92ff3c","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-06-03 05:58:04.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"92353c3c1891e87adabccf43dedaa00b59755573","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-03 05:58:27.000000000","message":"Patch Set 5: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/glance for 948907,5","accounts_in_message":[],"_revision_number":5},{"id":"780f3494f497e10e20531088add21594e3614495","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-03 05:58:28.000000000","message":"Patch Set 5:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/glance for 948907,5","accounts_in_message":[],"_revision_number":5},{"id":"1074b7c655e4cb3bd2d3e4d42b7fe8b9ed17d76d","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-06-03 06:03:46.000000000","message":"Topic spec/set-size-on-upload removed","accounts_in_message":[],"_revision_number":5},{"id":"b266f451a28aa2bef4ff63906f64b22186cda76c","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-06-03 06:04:32.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"d0f394f962547e3e62cb373fde8f11ecafffb2a6","revisions":{"473066c79c0bf04aafc43da52957380eb843ced6":{"kind":"REWORK","_number":1,"created":"2025-05-06 15:21:13.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/07/948907/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/07/948907/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/07/948907/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/07/948907/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/07/948907/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/07/948907/1"}}},"commit":{"parents":[{"commit":"644c0bed28fff5ba4cb5cc4c73b0a99039afbc3b","subject":"DNM Test image size","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/644c0bed28fff5ba4cb5cc4c73b0a99039afbc3b"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-05-06 14:59:41.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-05-06 14:59:41.000000000","tz":0},"subject":"DNM Verify x-image-meta-size for all backends","message":"DNM Verify x-image-meta-size for all backends\n\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/948905\nChange-Id: I3736b9920506d32e1119b3f7475924e2b08d62f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/473066c79c0bf04aafc43da52957380eb843ced6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/473066c79c0bf04aafc43da52957380eb843ced6"}]},"branch":"refs/heads/master"},"58e85b8174752e3f4b3505d64252d8120de7a857":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-05-06 18:33:17.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/07/948907/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/07/948907/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/07/948907/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/07/948907/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/07/948907/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/07/948907/2"}}},"commit":{"parents":[{"commit":"5980b3ed5585ed585ba8f50f943f8fd5b0b3b434","subject":"DNM Test image size","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5980b3ed5585ed585ba8f50f943f8fd5b0b3b434"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-05-06 14:59:41.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-05-06 18:12:17.000000000","tz":0},"subject":"DNM Verify x-image-meta-size for all backends","message":"DNM Verify x-image-meta-size for all backends\n\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/948905\nChange-Id: I3736b9920506d32e1119b3f7475924e2b08d62f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/58e85b8174752e3f4b3505d64252d8120de7a857"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/58e85b8174752e3f4b3505d64252d8120de7a857"}]},"branch":"refs/heads/master"},"0af80ebf31ef61fbd88a5316f9791c110ab6ace6":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2025-05-15 18:50:43.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/07/948907/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/07/948907/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/07/948907/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/07/948907/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/07/948907/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/07/948907/3"}}},"commit":{"parents":[{"commit":"163297837c9ddcbeefa29293d0af0321e6f5ddc7","subject":"Read image size from request header","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/163297837c9ddcbeefa29293d0af0321e6f5ddc7"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-05-06 14:59:41.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-05-15 18:50:43.000000000","tz":0},"subject":"DNM Verify x-image-meta-size for all backends","message":"DNM Verify x-image-meta-size for all backends\n\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/948905\nChange-Id: I3736b9920506d32e1119b3f7475924e2b08d62f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0af80ebf31ef61fbd88a5316f9791c110ab6ace6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0af80ebf31ef61fbd88a5316f9791c110ab6ace6"}]},"branch":"refs/heads/master","description":"Rebase"},"93f74967bd1ea14c4499bbaa2f99bb8b13185479":{"kind":"REWORK","_number":4,"created":"2025-05-16 07:35:51.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/07/948907/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/07/948907/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/07/948907/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/07/948907/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/07/948907/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/07/948907/4"}}},"commit":{"parents":[{"commit":"163297837c9ddcbeefa29293d0af0321e6f5ddc7","subject":"Read image size from request header","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/163297837c9ddcbeefa29293d0af0321e6f5ddc7"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-05-06 14:59:41.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-05-16 07:35:37.000000000","tz":0},"subject":"DNM Verify x-image-meta-size for all backends","message":"DNM Verify x-image-meta-size for all backends\n\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/948905\nDepends-On: https://review.opendev.org/c/openstack/glance_store/+/949129\nDepneds-On: https://review.opendev.org/c/openstack/glance_store/+/949124\nDepends-On: https://review.opendev.org/c/openstack/glance_store/+/949626\nDepneds-On: https://review.opendev.org/c/openstack/glance_store/+/949362\nDepends-On: https://review.opendev.org/c/openstack/glance_store/+/949137\nDepends-On: https://review.opendev.org/c/openstack/python-glanceclient/+/949787\n\nChange-Id: I3736b9920506d32e1119b3f7475924e2b08d62f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/93f74967bd1ea14c4499bbaa2f99bb8b13185479"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/93f74967bd1ea14c4499bbaa2f99bb8b13185479"}]},"branch":"refs/heads/master"},"d0f394f962547e3e62cb373fde8f11ecafffb2a6":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2025-05-16 09:07:51.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/07/948907/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/07/948907/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/07/948907/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/07/948907/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/07/948907/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/07/948907/5"}}},"commit":{"parents":[{"commit":"163297837c9ddcbeefa29293d0af0321e6f5ddc7","subject":"Read image size from request header","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/163297837c9ddcbeefa29293d0af0321e6f5ddc7"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-05-06 14:59:41.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-05-16 09:07:51.000000000","tz":0},"subject":"DNM Verify x-image-meta-size for all backends","message":"DNM Verify x-image-meta-size for all backends\n\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/948905\nDepends-On: https://review.opendev.org/c/openstack/glance_store/+/949973\nDepends-On: https://review.opendev.org/c/openstack/python-glanceclient/+/949787\n\nChange-Id: I3736b9920506d32e1119b3f7475924e2b08d62f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d0f394f962547e3e62cb373fde8f11ecafffb2a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d0f394f962547e3e62cb373fde8f11ecafffb2a6"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
