)]}'
{"id":"openstack%2Fglance~949206","triplet_id":"openstack%2Fglance~master~I52bc2d6cff5c5f073854e45350599b363af7e6f6","project":"openstack/glance","branch":"master","topic":"fix/import","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2026-02-13 08:23:27.000000000","reason":"Change was abandoned"},"35765":{"account":{"_account_id":35765,"name":"Felipe Figueroa Vergara","display_name":"Felipe Figueroa","email":"ffigueroa@whitestack.com","username":"felipeafv","status":"Whitestack"},"last_update":"2026-02-13 08:23:27.000000000","reason":"Change was abandoned"},"34437":{"account":{"_account_id":34437,"name":"Alejandro Garcia","email":"agarcia@whitestack.com","username":"agarciaws"},"last_update":"2025-10-02 11:57:09.000000000","reason":"\u003cGERRIT_ACCOUNT_34437\u003e replied on the change","reason_account":{"_account_id":34437,"name":"Alejandro Garcia","email":"agarcia@whitestack.com","username":"agarciaws"}},"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2025-05-08 14:56:25.000000000","reason":"Reviewer/Cc was removed"},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2026-02-13 08:23:27.000000000","reason":"Change was abandoned"},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2025-05-08 14:56:22.000000000","reason":"Reviewer/Cc was removed"}},"hashtags":[],"change_id":"I52bc2d6cff5c5f073854e45350599b363af7e6f6","subject":"Fixes image import to store for large images","status":"ABANDONED","created":"2025-05-08 14:36:48.000000000","updated":"2026-02-13 08:23:27.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1d02a618e8ab35e819b9dfbaff7bd01e69fbfb78","_number":949206,"virtual_id_number":949206,"owner":{"_account_id":35765,"name":"Felipe Figueroa Vergara","display_name":"Felipe Figueroa","email":"ffigueroa@whitestack.com","username":"felipeafv","status":"Whitestack"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":34437,"name":"Alejandro Garcia","email":"agarcia@whitestack.com","username":"agarciaws"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-09-20 16:11:56.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":34437,"name":"Alejandro Garcia","email":"agarcia@whitestack.com","username":"agarciaws"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2025-10-02 11:57:09.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":34437,"name":"Alejandro Garcia","email":"agarcia@whitestack.com","username":"agarciaws"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":34437,"name":"Alejandro Garcia","email":"agarcia@whitestack.com","username":"agarciaws"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":34437,"name":"Alejandro Garcia","email":"agarcia@whitestack.com","username":"agarciaws"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34437,"name":"Alejandro Garcia","email":"agarcia@whitestack.com","username":"agarciaws"}],"CC":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-08 14:47:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-05-08 14:56:06.000000000","updated_by":{"_account_id":35765,"name":"Felipe Figueroa Vergara","display_name":"Felipe Figueroa","email":"ffigueroa@whitestack.com","username":"felipeafv","status":"Whitestack"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2025-05-08 14:56:06.000000000","updated_by":{"_account_id":35765,"name":"Felipe Figueroa Vergara","display_name":"Felipe Figueroa","email":"ffigueroa@whitestack.com","username":"felipeafv","status":"Whitestack"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2025-05-08 14:56:06.000000000","updated_by":{"_account_id":35765,"name":"Felipe Figueroa Vergara","display_name":"Felipe Figueroa","email":"ffigueroa@whitestack.com","username":"felipeafv","status":"Whitestack"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2025-05-08 14:56:06.000000000","updated_by":{"_account_id":35765,"name":"Felipe Figueroa Vergara","display_name":"Felipe Figueroa","email":"ffigueroa@whitestack.com","username":"felipeafv","status":"Whitestack"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2025-05-08 14:56:22.000000000","updated_by":{"_account_id":35765,"name":"Felipe Figueroa Vergara","display_name":"Felipe Figueroa","email":"ffigueroa@whitestack.com","username":"felipeafv","status":"Whitestack"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REMOVED"},{"updated":"2025-05-08 14:56:25.000000000","updated_by":{"_account_id":35765,"name":"Felipe Figueroa Vergara","display_name":"Felipe Figueroa","email":"ffigueroa@whitestack.com","username":"felipeafv","status":"Whitestack"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REMOVED"},{"updated":"2025-05-08 17:32:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-08 18:03:08.000000000","updated_by":{"_account_id":34437,"name":"Alejandro Garcia","email":"agarcia@whitestack.com","username":"agarciaws"},"reviewer":{"_account_id":34437,"name":"Alejandro Garcia","email":"agarcia@whitestack.com","username":"agarciaws"},"state":"REVIEWER"},{"updated":"2026-02-11 06:45:03.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"CC"}],"messages":[{"id":"9c6f09c23f855657e8de301e75795a2719b345ce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35765,"name":"Felipe Figueroa Vergara","display_name":"Felipe Figueroa","email":"ffigueroa@whitestack.com","username":"felipeafv","status":"Whitestack"},"date":"2025-05-08 14:36:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2ca1945de0ee2fb0be9109825ae1c7a06a6704aa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-08 14:47:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e4ed6bee39db4204a017156630992176\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/fe7b77b106d6476ba4ab4c9978e290cb : SUCCESS in 7m 53s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/62c7e5d48cf746a09dbe6ab0c7823d77 : SUCCESS in 6m 05s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e4b503f4af3646658b049737caae6c5d : SUCCESS in 9m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9d61044e9b4d968f86860a09b9290b093c42ac8c","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":35765,"name":"Felipe Figueroa Vergara","display_name":"Felipe Figueroa","email":"ffigueroa@whitestack.com","username":"felipeafv","status":"Whitestack"},"date":"2025-05-08 14:56:22.000000000","message":"Removed reviewer \u003cGERRIT_ACCOUNT_37598\u003e.","accounts_in_message":[{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"_revision_number":1},{"id":"af17abb2d6c8665cd463f71e10d85cf3092dc87e","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":35765,"name":"Felipe Figueroa Vergara","display_name":"Felipe Figueroa","email":"ffigueroa@whitestack.com","username":"felipeafv","status":"Whitestack"},"date":"2025-05-08 14:56:25.000000000","message":"Removed reviewer \u003cGERRIT_ACCOUNT_4393\u003e.","accounts_in_message":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}],"_revision_number":1},{"id":"8a1f272c998c15afd940cb2a2676d0906f8e5a34","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-08 17:32:21.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/296e0c55c7d64c2084dccfe79d3ba1a5\n\n- grenade https://zuul.opendev.org/t/openstack/build/c3cff816c3b746b1a590238853fb6bd4 : SUCCESS in 48m 47s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/527af22dcfe14b39ab39af4c49aebc1e : SUCCESS in 1h 26m 54s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/258b7d7323594683976f3f4b9d5e844b : SUCCESS in 1h 38m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1a70d3f02b324b65bc4acef29720f677 : SUCCESS in 1h 01m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62ae885d10a64f6eb3fbf3e1cc1ad2e1 : FAILURE in 4m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/386163e852a245f0851c7a73b9db5cf7 : SUCCESS in 4m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/260f191f62284435a2093519ea3291a2 : SUCCESS in 4m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b939ebf4a5924025b0046030f7455f32 : SUCCESS in 7m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7bfca06a867241c88e0a0b62d79be5c9 : SUCCESS in 5m 59s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/87b3aaa13f2f4a99a5953fa42ffcf957 : SUCCESS in 8m 10s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/c7488e6d2d0047aa8b2fa272d82f72cf : SUCCESS in 11m 46s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/66091593aa204576816d14328f6a8ec9 : FAILURE in 1h 16m 12s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/6c27d49e62bb45328a99a2d65b0b6115 : SUCCESS in 1h 08m 15s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/7f556540ee404fe68b661e2c52142bd2 : SUCCESS in 1h 18m 27s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/439f4dc63b7440639332ae83874b0ac7 : SUCCESS in 1h 51m 11s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/cd393d49798942e9be3b2cf44b737e85 : SUCCESS in 2h 24m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6e1127f2dad542c39ba20bb99b420255 : SUCCESS in 56m 29s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/7fe79d3765b649a9ac5584ee919f2785 : TIMED_OUT in 2h 38m 02s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/5ab2b0e4dd5b47dead343a3ebf8a989f : SUCCESS in 2h 07m 43s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/7e1750ce5a2a4ac0b71f7659ba5cb962 : SUCCESS in 1h 03m 35s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/15166b83ad0c4bbd822e6d62d83c9636 : SUCCESS in 1h 45m 31s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/57055297f034449a9aedc9f23b0b81b7 : SUCCESS in 30m 41s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/1205a6de319f44c49f96b6057bc23bf2 : SUCCESS in 50m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d22b1708fd22ad1761cfff5ad15f074ca147cfab","author":{"_account_id":34437,"name":"Alejandro Garcia","email":"agarcia@whitestack.com","username":"agarciaws"},"date":"2025-05-08 18:03:08.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"33bdfc546bd95d4b2326e82fd4be1f881f4eb4c3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35765,"name":"Felipe Figueroa Vergara","display_name":"Felipe Figueroa","email":"ffigueroa@whitestack.com","username":"felipeafv","status":"Whitestack"},"date":"2025-05-08 18:05:44.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"125ee41f63d10631661d48633daf03e874e083df","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-08 18:20:31.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/90acd5a5c5bf4824999d7d51f7afcabd\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/415abc5a292745e6ad90eeff48ef8809 : SUCCESS in 4m 48s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7e37139ac945409d8469c93708252665 : SUCCESS in 5m 54s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/df33c413a08341b593496c09227bad38 : SUCCESS in 10m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"19da98eb06f6de6a86f19c47e8086b42ca797726","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-08 20:26:39.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/27a8f3b57ef249dc86753c02fe9fa424\n\n- grenade https://zuul.opendev.org/t/openstack/build/743309b16a744aac8e305c2663e83fd8 : SUCCESS in 55m 26s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e48e459fd06140858df76c0e5f36d76e : SUCCESS in 53m 46s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/47354135f28143548cc7c75d34082b01 : SUCCESS in 1h 21m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bf487e304ec94dab8cb4fda34b626442 : SUCCESS in 1h 05m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1bb61bc845a414c9c4f82445ffb52c2 : SUCCESS in 4m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/05f1a9f89044465f985eff887940d5db : SUCCESS in 3m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c0a644ce94c142bdb2d3ae40cc17f64f : SUCCESS in 2m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/dae67e55c82c442a8a3379c881342e83 : SUCCESS in 8m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0786f9b9a5d24613a196bc65108a41f2 : SUCCESS in 5m 04s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6fb0e01adc7a44c992f1319ace108881 : SUCCESS in 6m 55s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/27ec85f3ca5c4fcf9a260ae7c38aac19 : SUCCESS in 10m 27s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/8ee6615391f542bc9b8bca6270aee0a3 : FAILURE in 1h 51m 21s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/5e6d115adf2b411ea549d908867b1cea : SUCCESS in 1h 11m 24s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/9323e31848ec4e3996c8c048b0c6c876 : SUCCESS in 2h 15m 59s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/10064eee9c0847f9ad20b7fa3c0f222a : SUCCESS in 1h 17m 56s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/705a375f24c34fd59c2970cbc2962684 : SUCCESS in 2h 12m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8db20e2877ed469e92e069c8c0087e27 : SUCCESS in 1h 16m 37s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/8abf8d0feded49f7bb4798f9be777d0d : FAILURE in 23m 53s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/d79bf0efbceb4525924cf1eade655131 : SUCCESS in 2h 02m 20s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/ce3bbf35a6dc4a9ebc920446a7b55731 : SUCCESS in 55m 39s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/627beaf29ac141419a59efa14311177a : SUCCESS in 1h 33m 59s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/0b426d7d02964df0b9c0d14ec92dea36 : SUCCESS in 27m 11s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/ff669aa5612e4bfeb76ae507bb42b922 : SUCCESS in 1h 02m 03s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6b619f0a3bacacf345e695e26e7064e250dacead","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35765,"name":"Felipe Figueroa Vergara","display_name":"Felipe Figueroa","email":"ffigueroa@whitestack.com","username":"felipeafv","status":"Whitestack"},"real_author":{"_account_id":8108,"name":"Nobuto Murata","email":"nobuto.murata@canonical.com","username":"nobuto-m"},"date":"2025-09-20 14:08:12.000000000","message":"Patch Set 3: Patch Set 2 was rebased on behalf of \u003cGERRIT_ACCOUNT_35765\u003e\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":35765,"name":"Felipe Figueroa Vergara","display_name":"Felipe Figueroa","email":"ffigueroa@whitestack.com","username":"felipeafv","status":"Whitestack"}],"_revision_number":3},{"id":"38cc090643dd8d6d5f0ff55cec5346ce5cd0fda1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-20 14:22:25.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f4d3233e0c44f9394af852423c4b851\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9ff443c8bbad493d80b2b8bee2426924 : SUCCESS in 6m 57s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/44af6cb550c24790bef750041fd24452 : SUCCESS in 7m 43s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/3a2ec8889a12425f880c5f5e4965d26a : SUCCESS in 12m 36s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"cfeeb30f3d6744ba7449f5962ec96834ec53853b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-20 16:11:56.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad4586db86ab482ba44b7da5890a51d0\n\n- grenade https://zuul.opendev.org/t/openstack/build/434b41e4190644188c59226f014a5df0 : SUCCESS in 45m 28s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ffe27aa1c66b463f88951e05b98cb4bb : SUCCESS in 1h 02m 12s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d0ed75a11bc44475bd7bfcf6f4ec9d30 : SUCCESS in 1h 26m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/72e36c690f9641f4a449e1edd3100fcf : SUCCESS in 44m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ace5b6c64a348adbe74866599550a14 : SUCCESS in 5m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e18893512c1144218954529645f7d8f6 : SUCCESS in 5m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d1cc151253a84d4db6137444ae0f2e68 : SUCCESS in 6m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/68c944db27bf42a68a076ae22dc06540 : SUCCESS in 7m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74829f9cc3b549fba349871d1aed96b6 : SUCCESS in 6m 05s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/045469a5c974415f88967dc20fc8a02e : SUCCESS in 6m 14s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/a5bd40bd2d364be58a267982a7108ad3 : SUCCESS in 9m 22s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/e5f5ee74353b450ba43b382893593f92 : SUCCESS in 1h 46m 07s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/d54a73b44e1e4783a8880f01cf1aa18c : SUCCESS in 1h 58m 17s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/7f88238fd4994af994a6f42fe2465795 : SUCCESS in 1h 20m 03s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/c9c24a354feb40c79a64614c0e2866d5 : SUCCESS in 1h 45m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b57cb5c17aff43aeb3c61e7dcd38dd6c : SUCCESS in 57m 48s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e43ae9ecb6944f01be4bf9aa42f69450 : SUCCESS in 1h 16m 16s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/bc8c53ccc21f4a9d98ae8c90046d249a : FAILURE in 30m 32s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/3db79abbbec64228859e715a91642342 : SUCCESS in 1h 55m 33s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/fb7437c066c84a1aafb8efe3c39fa77d : SUCCESS in 30m 52s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/55f6c5c9d95b4bee94f126512b932e14 : SUCCESS in 1h 10m 28s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"970f65d121c883fa8e145c0b2cd25671c7e8e46a","author":{"_account_id":34437,"name":"Alejandro Garcia","email":"agarcia@whitestack.com","username":"agarciaws"},"date":"2025-10-02 11:57:09.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"2246d52d3c964583e7cd648b5eab86960ab0551b","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-02-11 06:45:03.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1d02a618e8ab35e819b9dfbaff7bd01e69fbfb78","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-02-13 08:23:27.000000000","message":"Abandoned\n\nThe fix i mentioned in my last comment is merged in master branch. We can abandon this patch. Sorry to hold you back for long","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"589bcb07dcc9b63ffd1b149bcc79bd4180d6e672","revisions":{"6a7e7d3f7f4a079a1b5f3c995356bfffff3ab341":{"kind":"REWORK","_number":1,"created":"2025-05-08 14:36:48.000000000","uploader":{"_account_id":35765,"name":"Felipe Figueroa Vergara","display_name":"Felipe Figueroa","email":"ffigueroa@whitestack.com","username":"felipeafv","status":"Whitestack"},"ref":"refs/changes/06/949206/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/06/949206/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/06/949206/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/06/949206/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/06/949206/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/06/949206/1"}}},"commit":{"parents":[{"commit":"1d8bc7d17ccf16bc0b66b14c54e60fd1b152d471","subject":"Merge \"Replace links to HypervisorSupportMatrix\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1d8bc7d17ccf16bc0b66b14c54e60fd1b152d471"}]}],"author":{"name":"FelipeAFV","email":"ffigueroa@whitestack.com","date":"2025-05-08 14:31:20.000000000","tz":-240},"committer":{"name":"FelipeAFV","email":"ffigueroa@whitestack.com","date":"2025-05-08 14:35:28.000000000","tz":-240},"subject":"Fixes image import to store for large images","message":"Fixes image import to store for large images\n\nRelated-Bug: #2017502\nRelated-Bug: #1885252\nChange-Id: I52bc2d6cff5c5f073854e45350599b363af7e6f6\nSigned-off-by: FelipeAFV \u003cffigueroa@whitestack.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6a7e7d3f7f4a079a1b5f3c995356bfffff3ab341"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6a7e7d3f7f4a079a1b5f3c995356bfffff3ab341"}]},"branch":"refs/heads/master"},"0d42143f037f9b58d7c0d9d2b51c708070287f1b":{"kind":"REWORK","_number":2,"created":"2025-05-08 18:05:44.000000000","uploader":{"_account_id":35765,"name":"Felipe Figueroa Vergara","display_name":"Felipe Figueroa","email":"ffigueroa@whitestack.com","username":"felipeafv","status":"Whitestack"},"ref":"refs/changes/06/949206/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/06/949206/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/06/949206/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/06/949206/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/06/949206/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/06/949206/2"}}},"commit":{"parents":[{"commit":"1d8bc7d17ccf16bc0b66b14c54e60fd1b152d471","subject":"Merge \"Replace links to HypervisorSupportMatrix\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1d8bc7d17ccf16bc0b66b14c54e60fd1b152d471"}]}],"author":{"name":"FelipeAFV","email":"ffigueroa@whitestack.com","date":"2025-05-08 14:31:20.000000000","tz":-240},"committer":{"name":"FelipeAFV","email":"ffigueroa@whitestack.com","date":"2025-05-08 18:05:17.000000000","tz":-240},"subject":"Fixes image import to store for large images","message":"Fixes image import to store for large images\n\nRelated-Bug: #2017502\nRelated-Bug: #1885252\nChange-Id: I52bc2d6cff5c5f073854e45350599b363af7e6f6\nSigned-off-by: FelipeAFV \u003cffigueroa@whitestack.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0d42143f037f9b58d7c0d9d2b51c708070287f1b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0d42143f037f9b58d7c0d9d2b51c708070287f1b"}]},"branch":"refs/heads/master"},"589bcb07dcc9b63ffd1b149bcc79bd4180d6e672":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2025-09-20 14:08:12.000000000","uploader":{"_account_id":35765,"name":"Felipe Figueroa Vergara","display_name":"Felipe Figueroa","email":"ffigueroa@whitestack.com","username":"felipeafv","status":"Whitestack"},"real_uploader":{"_account_id":8108,"name":"Nobuto Murata","email":"nobuto.murata@canonical.com","username":"nobuto-m"},"ref":"refs/changes/06/949206/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/06/949206/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/06/949206/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/06/949206/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/06/949206/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/06/949206/3"}}},"commit":{"parents":[{"commit":"bcfd2cac74618f9cf115963488e80cd6d576a5b2","subject":"Merge \"Remove the last reference to OsloDBDeprecationWarning\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bcfd2cac74618f9cf115963488e80cd6d576a5b2"}]}],"author":{"name":"FelipeAFV","email":"ffigueroa@whitestack.com","date":"2025-05-08 14:31:20.000000000","tz":-240},"committer":{"name":"Felipe Figueroa Vergara","email":"ffigueroa@whitestack.com","date":"2025-09-20 14:08:12.000000000","tz":0},"subject":"Fixes image import to store for large images","message":"Fixes image import to store for large images\n\nRelated-Bug: #2017502\nRelated-Bug: #1885252\nChange-Id: I52bc2d6cff5c5f073854e45350599b363af7e6f6\nSigned-off-by: FelipeAFV \u003cffigueroa@whitestack.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/589bcb07dcc9b63ffd1b149bcc79bd4180d6e672"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/589bcb07dcc9b63ffd1b149bcc79bd4180d6e672"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
