)]}'
{"id":"openstack%2Fglance~949614","triplet_id":"openstack%2Fglance~stable%2F2025.1~I9600d7bf45d48b83c8b066bda20680365bedade7","project":"openstack/glance","branch":"stable/2025.1","topic":"pip-23.1-support","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-09-05 14:26:27.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I9600d7bf45d48b83c8b066bda20680365bedade7","subject":"Add pyproject.toml to support pip 23.1","status":"ABANDONED","created":"2025-05-13 11:49:05.000000000","updated":"2025-09-05 14:26:27.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f1fcc75e8f703145a31c164851a314c6565194fa","_number":949614,"virtual_id_number":949614,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-05-13 14:28:25.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-13 11:49:05.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"CC"},{"updated":"2025-05-13 13:59:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-05-13 14:28:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"59fdbacc7cdaef33ce6765b9e66e651a7ef579ab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-05-13 11:49:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a8e37051e39fd484c4af250912e6719024806647","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-13 13:59:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/35241e6344a9443ebb0a777aa416a634\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3ae87b5afed341439b7fcc070fe7f313 : SUCCESS in 7m 34s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2667a6652be542f1a9adf599d68672c8 : SUCCESS in 7m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"40ae167c6c661023fd88c90c8d10f338ef8854f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-13 14:28:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4f2805ab8e524aa58758ac415886d5ee\n\n- grenade https://zuul.opendev.org/t/openstack/build/92e6b012f05b404aa55f45a889e06129 : SUCCESS in 54m 39s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d594d599922c4cceab8d8372ab35be63 : SUCCESS in 54m 46s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/751b849286344267b75d114c5d666d88 : SUCCESS in 1h 36m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cbbefe830e794a519d2ec707eaec5b0e : SUCCESS in 1h 06m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd13f5d2428f4b3da5d1379a93cde0bc : SUCCESS in 5m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f2f5fd4d95eb40c29be0df2a4d227545 : SUCCESS in 3m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/12b78a2615e140dd8f3655b82b3bf7f9 : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56a73463b2364a338a555149e257cac3 : SUCCESS in 5m 29s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c6eb57e85da34a208c76af63370ed9de : FAILURE in 6m 21s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8f1b517602aa49a4b2b82f8d71fe4a12 : SUCCESS in 8m 47s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/03d25377cfc6400da4b3fba8cacc1dac : SUCCESS in 11m 40s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/aae26caf963d4cfcb1889a0d14710b98 : SUCCESS in 1h 48m 26s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/6bf725d263494b0fb27d5abc66891e91 : SUCCESS in 1h 48m 03s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/910c86b9a0bf454884127e851ea78425 : SUCCESS in 1h 00m 20s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/ecf1911e0bae4f7fbf7ae2bcf1204bfc : SUCCESS in 2h 01m 39s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/5c1ec5a20a9a413181b81089be746d93 : SUCCESS in 1h 57m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4bcbe9a0fdca4d0fa5b2d489ccfc3ce5 : SUCCESS in 56m 46s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/81c805f879fd4bd28e9acd988ca529e0 : SUCCESS in 1h 44m 46s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/4cc839eb40c44b3ea3eadb01917fd677 : SUCCESS in 1h 49m 31s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/a6fa374514a146b9904e0b6ca5058579 : SUCCESS in 53m 13s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/9b8874f0455e43e58e1694c2563e4dc8 : SUCCESS in 1h 46m 15s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/2592a1e7600c4dd3a13ec955c6e50e9d : SUCCESS in 14m 02s","accounts_in_message":[],"_revision_number":1},{"id":"f1fcc75e8f703145a31c164851a314c6565194fa","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-09-05 14:26:27.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e64404dd3b207b1a357af9d4bbcace60a2772e3c","revisions":{"e64404dd3b207b1a357af9d4bbcace60a2772e3c":{"kind":"REWORK","_number":1,"created":"2025-05-13 11:49:05.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/14/949614/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/14/949614/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/949614/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/949614/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/949614/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/14/949614/1"}}},"commit":{"parents":[{"commit":"0e6443944e4594c5affe78633b449722a665c01b","subject":"Update TOX_CONSTRAINTS_FILE for stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0e6443944e4594c5affe78633b449722a665c01b"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2023-11-02 11:16:49.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-05-13 11:48:58.000000000","tz":60},"subject":"Add pyproject.toml to support pip 23.1","message":"Add pyproject.toml to support pip 23.1\n\npip 23.1 removed the \"setup.py install\" fallback for projects\nthat do not have pyproject.toml and now uses a pyproject.toml\nwhich is vendored in pip.\nTo address that, this change adds the minimal pyproject.toml\nto enable pbr to be properly used to build editable wheels.\n\nThis is required to support installing devstack on\ncentos stream 9 and related distros with GLOBAL_VENV\u003dTrue\nWithout this change the wsgi scripts are not generated in\neditable mode. i.e. pip install -e /opt/stack/keystone\n\nSee https://pip.pypa.io/en/stable/news/#v23-1\nand https://github.com/pypa/pip/issues/8368 for more\ndetails on the removal of the fallback support.\n\nsetuptools v64.0.0 is used to support editable installs\nvia its PEP-660 implmentation\nhttps://github.com/pypa/setuptools/pull/3488\n\nChange-Id: I9600d7bf45d48b83c8b066bda20680365bedade7\n(cherry picked from commit 95c5f281fd33f2dd71ad26b2258426cdf1112bbf)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e64404dd3b207b1a357af9d4bbcace60a2772e3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e64404dd3b207b1a357af9d4bbcace60a2772e3c"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
