)]}'
{"id":"openstack%2Fglance~949616","triplet_id":"openstack%2Fglance~stable%2F2024.1~I9600d7bf45d48b83c8b066bda20680365bedade7","project":"openstack/glance","branch":"stable/2024.1","topic":"pip-23.1-support","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-09-02 10:26:28.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I9600d7bf45d48b83c8b066bda20680365bedade7","subject":"Add pyproject.toml to support pip 23.1","status":"ABANDONED","created":"2025-05-13 11:50:34.000000000","updated":"2025-09-02 10:26:28.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"efb6544cd0343dc05312f1dd209c28ed79aa6270","_number":949616,"virtual_id_number":949616,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-05-13 13:43:07.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-13 11:50:34.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"CC"},{"updated":"2025-05-13 13:43:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"db38a963f364a42620e12702c73750abab128495","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-05-13 11:50:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c9e516f2f9d0d070297490d2d9ecb93514d76e1b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-13 13:43:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3a4bbbc1d27a4bf2a5bfc5261d8cb742\n\n- grenade https://zuul.opendev.org/t/openstack/build/a4e83aa8a690447a8063ba516168f39a : SUCCESS in 52m 09s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/55cb1a2e6c9a481190928775ae9a1788 : SUCCESS in 33m 10s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a3def8a846214c7588d7599661986117 : FAILURE in 19m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8d198dcf5e714947b77126941309097f : POST_FAILURE in 27m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c31f1d0248534c87b2fc07e1d2a9be38 : SUCCESS in 3m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3b2c292b77374643b3bcd843c6902342 : SUCCESS in 5m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9482b72865d042afbd3660be3bcce78e : SUCCESS in 5m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d6976f0e5ca4895955ab5c82ac805ba : SUCCESS in 4m 38s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8ccd0817a3eb4fa3aac9b174e597fcad : SUCCESS in 9m 08s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b04fd9663eea413ab8322fd12d6f579a : SUCCESS in 7m 23s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/180e19ba273b470394575cef2032dae6 : SUCCESS in 8m 31s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/b9fff5614bef4326bcf1156946450f94 : FAILURE in 25m 09s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/669f7da7f1534b168694bde81c2a108d : FAILURE in 20m 00s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/0d161e7dff2c41e4a0cc0749451f2c15 : FAILURE in 23m 53s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/03a9bdfd2610483e9e501d5b7aa01c3f : SUCCESS in 50m 51s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/4356b0e6dd514b559bbc2f122f99056a : FAILURE in 25m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dd7d57144ee54bcbb4ddc04d36db1262 : FAILURE in 22m 11s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/77f50c10573f4fb49a8bc218da91122f : SUCCESS in 1h 34m 45s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/182f9c3b59ee49059bcd7405c67f624f : FAILURE in 21m 46s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/29bde6b1796d4fe6b6143eecabfad486 : SUCCESS in 57m 48s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/74e070e8f1cf4f398fd736b51b907ff7 : FAILURE in 19m 13s","accounts_in_message":[],"_revision_number":1},{"id":"4eea0f996e1dda5bf24e286186dc8208a0c26dca","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-13 14:06:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b69dd2586aa34210be40958e67cfb847\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d3da7c4abf29492aa613002b8f92c4fe : SUCCESS in 6m 54s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/010ad8f5a6fa45c3886974aefcda4196 : SUCCESS in 10m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"efb6544cd0343dc05312f1dd209c28ed79aa6270","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-09-02 10:26:28.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"32d4c998926612425397f473d41dd5dd35c5dc43","revisions":{"32d4c998926612425397f473d41dd5dd35c5dc43":{"kind":"REWORK","_number":1,"created":"2025-05-13 11:50:34.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/16/949616/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/16/949616/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/16/949616/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/16/949616/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/16/949616/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/16/949616/1"}}},"commit":{"parents":[{"commit":"028de8e37e76db456de9a6de71258e60689adee1","subject":"Fix broken metadata function calls","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/028de8e37e76db456de9a6de71258e60689adee1"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2023-11-02 11:16:49.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-05-13 11:50:26.000000000","tz":60},"subject":"Add pyproject.toml to support pip 23.1","message":"Add pyproject.toml to support pip 23.1\n\npip 23.1 removed the \"setup.py install\" fallback for projects\nthat do not have pyproject.toml and now uses a pyproject.toml\nwhich is vendored in pip.\nTo address that, this change adds the minimal pyproject.toml\nto enable pbr to be properly used to build editable wheels.\n\nThis is required to support installing devstack on\ncentos stream 9 and related distros with GLOBAL_VENV\u003dTrue\nWithout this change the wsgi scripts are not generated in\neditable mode. i.e. pip install -e /opt/stack/keystone\n\nSee https://pip.pypa.io/en/stable/news/#v23-1\nand https://github.com/pypa/pip/issues/8368 for more\ndetails on the removal of the fallback support.\n\nsetuptools v64.0.0 is used to support editable installs\nvia its PEP-660 implmentation\nhttps://github.com/pypa/setuptools/pull/3488\n\nChange-Id: I9600d7bf45d48b83c8b066bda20680365bedade7\n(cherry picked from commit 95c5f281fd33f2dd71ad26b2258426cdf1112bbf)\n(cherry picked from commit e64404dd3b207b1a357af9d4bbcace60a2772e3c)\n(cherry picked from commit 9d8de8060c6bafd0229a6db596353f8b1af5ee66)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/32d4c998926612425397f473d41dd5dd35c5dc43"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/32d4c998926612425397f473d41dd5dd35c5dc43"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
