)]}'
{"id":"openstack%2Fglance~949668","triplet_id":"openstack%2Fglance~master~I21c04d73f58f45f74202da64ec0244fa0ced518b","project":"openstack/glance","branch":"master","topic":"fix-location-apis","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2025-05-20 16:11:54.000000000","reason":"Change was abandoned"},"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2025-05-20 16:11:54.000000000","reason":"Change was abandoned"},"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2025-05-20 16:11:54.000000000","reason":"Change was abandoned"},"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2025-05-20 16:11:54.000000000","reason":"Change was abandoned"},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2025-05-20 16:11:27.000000000","reason":"\u003cGERRIT_ACCOUNT_27615\u003e replied on the change","reason_account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}}},"hashtags":[],"change_id":"I21c04d73f58f45f74202da64ec0244fa0ced518b","subject":"Don\u0027t allow hash calculation by default","status":"ABANDONED","created":"2025-05-13 19:04:10.000000000","updated":"2025-05-20 16:11:54.000000000","total_comment_count":20,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"6cf00c95a60374e665856702b2186cb245bcb41d","_number":949668,"virtual_id_number":949668,"owner":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-05-14 08:21:12.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":-1,"date":"2025-05-19 15:51:57.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":-1,"date":"2025-05-19 15:22:02.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-13 19:14:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-05-13 21:13:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-14 07:35:34.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"CC"},{"updated":"2025-05-16 16:46:03.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"CC"},{"updated":"2025-05-19 15:22:02.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2025-05-19 15:22:02.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2025-05-19 15:22:02.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"}],"messages":[{"id":"273d7fbbe8ad384d4c564e6ec416d828c7ce3ec1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2025-05-13 19:04:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0b4dc920f271b78036342e08b7f9815a4adafe83","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-13 19:14:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f78ccd1ca7eb4af1becefca526c422e4\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4bf36303c59a40f0b6a6f1c5cf22d20b : SUCCESS in 6m 37s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d377f92961344c6992a3170ee029cde8 : SUCCESS in 6m 56s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/46f8d24abef047d4a0c634198a4bf297 : SUCCESS in 8m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f4ead8eb5bba790674c224e33acfde961545c39d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-13 21:13:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/680cd30978674ba3aac61fe028c7b3bc\n\n- grenade https://zuul.opendev.org/t/openstack/build/ad4acebbea894017abdbd8acdac65d8b : SUCCESS in 59m 49s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/14fa0006c09b4b3291dafb16e0fe8dae : SUCCESS in 55m 17s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f851cee038794a82896ca13699b74982 : SUCCESS in 1h 28m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ec7f033df55142d09b5d7b9750b362f2 : SUCCESS in 58m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/674449cdd426451e8e9fe49fe793d60d : SUCCESS in 5m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2929bbd30f1e4f858a7acc661c4cc786 : SUCCESS in 4m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b26c548192464a13aaf1dfff86ad6cde : SUCCESS in 5m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3760871e618d4014ba6752bd9c0fe674 : SUCCESS in 4m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a64a263cdf354bd7a59f6f89509947b5 : SUCCESS in 5m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/21b9f3c0b6dd4b88b9bbe487bdb1d531 : SUCCESS in 5m 09s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/37f7348cffd8491e970d1d9c4667c6aa : FAILURE in 7m 41s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/ce6ad217a0774ed1b02ce29d21056b6a : FAILURE in 11m 22s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/eabf6b3d303d4af5a41eb2ffba064792 : SUCCESS in 1h 46m 01s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/f7eeeae73472426183b5a56f7646d4df : SUCCESS in 1h 34m 20s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/8fda4c18d4a0493c8284aaaffd0c4d77 : SUCCESS in 1h 46m 38s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/1666d0b4c60f44529702717be0b679d7 : SUCCESS in 1h 57m 50s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/e8b30dea6e7a4f0bb863a8ac97a0bb3d : SUCCESS in 2h 06m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b4dcb46b8de240dcaec9662102f57e67 : SUCCESS in 55m 38s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b60d7b24515444f49f1333c73a143da3 : SUCCESS in 1h 53m 22s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/8b1d23d3286e4f8f8409ba178d72de22 : SUCCESS in 1h 49m 00s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/516fd0bf1d4a44e8baadeacd9351e50b : SUCCESS in 1h 02m 04s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/a27db7b9c2964aa8ab3ba694f36ccf50 : SUCCESS in 1h 55m 46s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c48249d9dd744abcb26532a5722054a8 : SUCCESS in 31m 00s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/bb9404628eaa4ca9ba58b407c245a3ba : SUCCESS in 59m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bd3e92733ec47737a22cb3471953a721d4f59f26","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2025-05-14 06:04:54.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"87c3053242328323e1718e8343586d7c4ddc7010","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-14 06:19:07.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fc1bd94419e544b190bacf41d0fba45f\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b7b79ab1cb294d61a1e90926e10d3ba7 : SUCCESS in 8m 52s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4a83b5cf8f90404c945bd06f05051319 : SUCCESS in 8m 44s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f251fd38067844ff8d49949db4dda3af : SUCCESS in 12m 12s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d74f1ef44a8a9cc5b1c22f7d974913e79466b70e","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-05-14 07:35:34.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7ab42ee8be52aa292f02e6ee451085ad6f680947","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-05-14 07:37:04.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b822a4bd986899ff2039b6019e62808d370fc30b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-14 08:21:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/75d64e11e2c54876834be962c1a66518\n\n- grenade https://zuul.opendev.org/t/openstack/build/013e1fb69f4441cab970a35b770353f3 : SUCCESS in 26m 55s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/99389e9775b347a8bad25ee87d5960b4 : SUCCESS in 26m 01s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4fe44e63972a474c8f9f5d8c7316e0ed : SUCCESS in 1h 26m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e900676647cb42659963f397b3bbbdda : SUCCESS in 59m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f2fa4daa3404ddc85def0df2393e143 : SUCCESS in 4m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c9598acbf2594c9f812762cf0a68c5e4 : SUCCESS in 2m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3e88fef60de14047bfb3716a8e78c4f7 : SUCCESS in 5m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9afd84467ebd457cba93a70aa6621b6d : SUCCESS in 8m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44e3e70d2f63422289f87aa0e700fb3d : SUCCESS in 5m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/981f9671bc144f4bb03eb36b0cb6c7df : SUCCESS in 5m 52s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/55eb7b35431d461e872b4e76bd164759 : SUCCESS in 8m 15s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/369f33c952c849a8801f83a20d2b3bfa : SUCCESS in 11m 43s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/4ceb3376f5744938948273f6bad22392 : SUCCESS in 1h 25m 37s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/da260d2a702046449f75e15522909c75 : SUCCESS in 1h 23m 26s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/09596f8492cf42b883ca8df6ed66bea7 : SUCCESS in 1h 44m 54s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/d1a4ba58bbfc4cadae2beefba40cc16f : SUCCESS in 1h 36m 45s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/14041304479e4f82b3e3965176c7a13a : SUCCESS in 2h 01m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5f1203edbf6749289a36fcb84478f1ef : SUCCESS in 43m 43s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/6846a60965c04e3684ff81e3cd34bd28 : SUCCESS in 1h 26m 42s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/f0bd9e47d2ee4867b76fa2bfeab7b2b5 : SUCCESS in 2h 15m 40s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/c9e6b702258041e0abddcfd70a53897d : SUCCESS in 55m 36s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/681650dc55ef4da4864ccadde41fe8c6 : SUCCESS in 1h 55m 48s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/a173933c4ab64cc3945a5edb7455cd1e : SUCCESS in 29m 34s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/9c4342572b4b45adadbd7fc5f6e6904f : SUCCESS in 1h 09m 17s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"86f6866c7f928e2b564adc68863e1fff0a2cd4af","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2025-05-16 16:46:03.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9035543c618f4e57a1c662f7d151ce670e9af894","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-05-16 17:03:28.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1e4e1128299003546ce18cb09f56d7381950dc9c","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2025-05-19 15:22:02.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ba3af1ae1c35504b03bb99957e5a2fe1bb1324bc","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-05-19 15:51:57.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bfcb51416029c11f1436c8f8d9c16d63563629b7","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2025-05-19 16:00:03.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"46357e66320f98ffeedd804bfcea1a556f9b89aa","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-05-19 16:01:45.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0cddaebe2e6f84f8792e3a614c7cfb686b9a5152","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2025-05-19 16:07:37.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b511c7a1d25626efae9817cc760cb62baa7935b9","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-05-20 15:10:05.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"93b3c882489c2e9f4216ac26fdae0859f38919cd","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-05-20 15:15:07.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1cf9df951ac1f8fd8bff9749d7ad6e1a92ec3504","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-05-20 15:24:12.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"37d46b6c880b78624233f3b99a61a6fea97e64f1","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-05-20 15:31:34.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ac4269dad8f8521dad1704ca173ed6fc4aa9c3e3","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-05-20 15:34:18.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"af2d86b714460fbdaf69cdc94ee1624b04ffbc34","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2025-05-20 15:36:10.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3e5b6db7bb5078b29f8c699bc110bf6647a5bcdb","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-05-20 15:41:02.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"decc581f12daa5bd01389729872a609c3c06a62d","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-05-20 15:42:50.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c24a3258bbc228733b1a8ab9f0ccd459b5bb8a7f","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2025-05-20 15:54:27.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"09e7caaca5642412f32018693a2927e51e99edb9","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2025-05-20 16:11:27.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6cf00c95a60374e665856702b2186cb245bcb41d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2025-05-20 16:11:54.000000000","message":"Abandoned\n\nbased on the discussion on this patch, this is no longer a viable solution","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"fb55edcfc77b5423f9b2a1925d727ed2c015617b","revisions":{"5088643d4a63f90884a68b960e8ffcc4202bd263":{"kind":"REWORK","_number":1,"created":"2025-05-13 19:04:10.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/68/949668/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/68/949668/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/949668/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/949668/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/949668/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/68/949668/1"}}},"commit":{"parents":[{"commit":"1d8bc7d17ccf16bc0b66b14c54e60fd1b152d471","subject":"Merge \"Replace links to HypervisorSupportMatrix\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1d8bc7d17ccf16bc0b66b14c54e60fd1b152d471"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2025-05-13 18:53:40.000000000","tz":330},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2025-05-13 18:53:40.000000000","tz":330},"subject":"Don\u0027t allow hash calculation by default","message":"Don\u0027t allow hash calculation by default\n\nIf a user creates a VM snapshot (in RBD) using the new location\nAPI workflow, a hash calculation task is started in the background.\nThis task reads from the image and calculates the checksum and hash\ndepending on the hash algorithm set.\n\nThe downside to this is the image is seen as \u0027active\u0027 and the user\ncan perform a delete operation on it which would fail later since\nthere is an open RBD handle on the image.\n\nThe new behavior can cause issues with existing scripts which rely\non the image status \u0027active\u0027 to perform tasks on it which might\nstart failing, example: breakage in tempest tests[1].\n\nThe best way to avoid breakage in deployments is to disable\nthe hash calculation by default which would skip the hash\ncalculation task[2].\n\n[1] https://review.opendev.org/c/openstack/tempest/+/949595\n[2] https://github.com/openstack/glance/blob/1d8bc7d17ccf16bc0b66b14c54e60fd1b152d471/glance/async_/flows/location_import.py#L318\n\nCloses-Bug: #949595\nChange-Id: I21c04d73f58f45f74202da64ec0244fa0ced518b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5088643d4a63f90884a68b960e8ffcc4202bd263"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5088643d4a63f90884a68b960e8ffcc4202bd263"}]},"branch":"refs/heads/master"},"fb55edcfc77b5423f9b2a1925d727ed2c015617b":{"kind":"REWORK","_number":2,"created":"2025-05-14 06:04:54.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/68/949668/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/68/949668/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/949668/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/949668/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/949668/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/68/949668/2"}}},"commit":{"parents":[{"commit":"1d8bc7d17ccf16bc0b66b14c54e60fd1b152d471","subject":"Merge \"Replace links to HypervisorSupportMatrix\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1d8bc7d17ccf16bc0b66b14c54e60fd1b152d471"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2025-05-13 18:53:40.000000000","tz":330},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2025-05-13 20:06:41.000000000","tz":330},"subject":"Don\u0027t allow hash calculation by default","message":"Don\u0027t allow hash calculation by default\n\nIf a user creates a VM snapshot (in RBD) using the new location\nAPI workflow, a hash calculation task is started in the background.\nThis task reads from the image and calculates the checksum and hash\ndepending on the hash algorithm set.\n\nThe downside to this is the image is seen as \u0027active\u0027 and the user\ncan perform a delete operation on it which would fail later since\nthere is an open RBD handle on the image.\n\nThe new behavior can cause issues with existing scripts which rely\non the image status \u0027active\u0027 to perform tasks on it which might\nstart failing, example: breakage in tempest tests[1].\n\nThe best way to avoid breakage in deployments is to disable\nthe hash calculation by default which would skip the hash\ncalculation task[2].\n\n[1] https://review.opendev.org/c/openstack/tempest/+/949595\n[2] https://github.com/openstack/glance/blob/1d8bc7d17ccf16bc0b66b14c54e60fd1b152d471/glance/async_/flows/location_import.py#L318\n\nCloses-Bug: #949595\nChange-Id: I21c04d73f58f45f74202da64ec0244fa0ced518b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fb55edcfc77b5423f9b2a1925d727ed2c015617b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fb55edcfc77b5423f9b2a1925d727ed2c015617b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
