)]}'
{"id":"openstack%2Fglance~949672","triplet_id":"openstack%2Fglance~master~I49f4c6337e7165268f41c6c37eb5f51b84956720","project":"openstack/glance","branch":"master","topic":"fix-location-apis","attention_set":{},"removed_from_attention_set":{"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2025-05-20 16:23:06.000000000","reason":"Change was abandoned"},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2025-05-14 08:39:12.000000000","reason":"\u003cGERRIT_ACCOUNT_27615\u003e replied on the change","reason_account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}}},"hashtags":[],"change_id":"I49f4c6337e7165268f41c6c37eb5f51b84956720","subject":"Docs: Add warning about ``do_secure_hash`` config","status":"ABANDONED","created":"2025-05-13 19:50:25.000000000","updated":"2025-05-20 16:23:06.000000000","total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"189b2e3cc134b1f8bcfd76d96ade2dec77b87b54","_number":949672,"virtual_id_number":949672,"owner":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-05-14 06:10:43.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-13 19:55:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-14 07:42:09.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"CC"}],"messages":[{"id":"41e9538b60cb3f50e8082bbfc8ad9a47e444af32","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2025-05-13 19:50:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6adb99a8e485e6587769df6b5b2f378da0d53f89","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2025-05-13 19:51:01.000000000","message":"Topic set to fix-location-apis","accounts_in_message":[],"_revision_number":1},{"id":"a7bf2d3f8209e11c8351ad320f9c1a08bd12b77b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-13 19:55:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a136aa6e20b49f8a60ce19bb70e3715\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2fcac01c9e394cd68e7cdebf131569a4 : SUCCESS in 4m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee3fa13bffab4334ae8f6c3da36d5cac : SUCCESS in 5m 15s","accounts_in_message":[],"_revision_number":1},{"id":"7323179a12d2e2630053c15beb1b8ac299653152","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2025-05-14 06:04:54.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"ebc8b01df211c0c03d6fd9510b1bd9613cf756a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-14 06:10:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fc1ecb9e5b5f4d2fb40cedcf7a39a942\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c774e769b9b9458398c038e1baf766af : SUCCESS in 4m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b423e9768ff4f5c8d24b1de6b76810e : SUCCESS in 4m 54s","accounts_in_message":[],"_revision_number":2},{"id":"f4a4ef1df36e0eb029173cdeb3590dd683e7ca7c","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-05-14 07:42:09.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"67f2523f803f0061264cec29e53445290d35eb9c","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-05-14 07:43:16.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1ba70f1dd07e0f8a192e3ed5b2d73ae2fd80f3fa","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2025-05-14 08:39:12.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"189b2e3cc134b1f8bcfd76d96ade2dec77b87b54","tag":"autogenerated:gerrit:abandon","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2025-05-20 16:23:06.000000000","message":"Abandoned\n\nSee discussions here for reason of abandonment https://review.opendev.org/c/openstack/glance/+/949668","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"fc889da7ea6913b40b2f933b8f6b9f2eddbd5fe5","revisions":{"f43702ea6326f40e39e07239072917d2795aaabd":{"kind":"REWORK","_number":1,"created":"2025-05-13 19:50:25.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/72/949672/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/72/949672/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/949672/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/949672/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/949672/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/72/949672/1"}}},"commit":{"parents":[{"commit":"5088643d4a63f90884a68b960e8ffcc4202bd263","subject":"Don\u0027t allow hash calculation by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5088643d4a63f90884a68b960e8ffcc4202bd263"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2025-05-13 19:47:02.000000000","tz":330},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2025-05-13 19:47:02.000000000","tz":330},"subject":"Docs: Add warning about ``do_secure_hash`` config","message":"Docs: Add warning about ``do_secure_hash`` config\n\nWhen ``do_secure_hash`` is enabled and a hash calculation task\nis going on in the background, we cannot delete the image.\nThis can break tooling in some deployments relying on the image\nstatus alone.\nThis patch adds a warning in the docs so operators are aware before\nsetting the ``do_secure_hash`` config option to ``True``.\n\nPartial-Bug: #2110581\nChange-Id: I49f4c6337e7165268f41c6c37eb5f51b84956720\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f43702ea6326f40e39e07239072917d2795aaabd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f43702ea6326f40e39e07239072917d2795aaabd"}]},"branch":"refs/heads/master"},"fc889da7ea6913b40b2f933b8f6b9f2eddbd5fe5":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-05-14 06:04:54.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/72/949672/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/72/949672/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/949672/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/949672/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/949672/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/72/949672/2"}}},"commit":{"parents":[{"commit":"fb55edcfc77b5423f9b2a1925d727ed2c015617b","subject":"Don\u0027t allow hash calculation by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fb55edcfc77b5423f9b2a1925d727ed2c015617b"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2025-05-13 19:47:02.000000000","tz":330},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2025-05-13 20:06:44.000000000","tz":330},"subject":"Docs: Add warning about ``do_secure_hash`` config","message":"Docs: Add warning about ``do_secure_hash`` config\n\nWhen ``do_secure_hash`` is enabled and a hash calculation task\nis going on in the background, we cannot delete the image.\nThis can break tooling in some deployments relying on the image\nstatus alone.\nThis patch adds a warning in the docs so operators are aware before\nsetting the ``do_secure_hash`` config option to ``True``.\n\nPartial-Bug: #2110581\nChange-Id: I49f4c6337e7165268f41c6c37eb5f51b84956720\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fc889da7ea6913b40b2f933b8f6b9f2eddbd5fe5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fc889da7ea6913b40b2f933b8f6b9f2eddbd5fe5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
