)]}'
{"id":"openstack%2Fglance~951031","triplet_id":"openstack%2Fglance~master~Ic58233302001a2396e84400ef7081afa51748a2a","project":"openstack/glance","branch":"master","topic":"bug/2045769","attention_set":{},"removed_from_attention_set":{"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2025-06-16 15:41:07.000000000","reason":"Change was submitted"},"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2025-06-16 15:41:07.000000000","reason":"Change was submitted"},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2025-06-16 13:36:35.000000000","reason":"\u003cGERRIT_ACCOUNT_8122\u003e replied on the change","reason_account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}}},"hashtags":[],"change_id":"Ic58233302001a2396e84400ef7081afa51748a2a","subject":"Add task cancellation tracker utility","status":"MERGED","created":"2025-05-27 20:24:11.000000000","updated":"2025-06-16 15:42:15.000000000","submitted":"2025-06-16 15:41:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":32,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"951031-bug/2045769","meta_rev_id":"67753b0b70d9b84f3221f7554cd7f970227dd258","_number":951031,"virtual_id_number":951031,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-06-16 15:41:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"all":[{"value":2,"date":"2025-06-13 15:09:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":2,"date":"2025-06-16 13:36:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":1,"date":"2025-06-16 13:36:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-27 20:26:18.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2025-05-27 20:26:18.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2025-05-27 20:37:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-05-27 22:51:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"86289c11883685390b2325260af44d0e26f02bb1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-05-27 20:24:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"16d10b63ad8a0df399822d9c4bd1069cd17ec540","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-27 20:37:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2255d943764249aaa64d07a19f5cae20\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/659cca7edbc94768b897ec7e5c14254c : SUCCESS in 8m 59s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/94975bab569b40c6acc8852c83190e8e : SUCCESS in 8m 21s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d77f7662fcef4c8d9db0a5beb748d42b : SUCCESS in 11m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"df4e7bfb4264fa1990a3871c4cac2158a208d84f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-27 22:51:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/25200a90c77b4125916886dc45f7a078\n\n- grenade https://zuul.opendev.org/t/openstack/build/716c1da69f674a5ab744eee61d29113e : SUCCESS in 1h 05m 11s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e87e3212f4124e16a3164376199abdc9 : SUCCESS in 53m 04s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0845a2d857bb42f19f9c18ef7d70e677 : SUCCESS in 1h 37m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a7fa5b4bd9be474bbd1e34dce370ee16 : SUCCESS in 1h 03m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/beaf8c56530045e998f31d279608abfd : SUCCESS in 4m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4178e9e7ced7444c8fbf226090240cb4 : SUCCESS in 4m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b581cf8dc2514752bcd782dccf840063 : SUCCESS in 5m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/79cc6a55cf584cedaa39b40eaeefc195 : SUCCESS in 7m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff09985c6b294ff4a5d0fd4114e4fe20 : SUCCESS in 5m 41s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6b82cc87367f4851b8978ed44ddc0748 : SUCCESS in 6m 22s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/f5334f44fb9241cea366b573c82212f8 : SUCCESS in 11m 02s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/76b5b8963e9b4f39933e2e62fea41a35 : SUCCESS in 1h 32m 24s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/671e81d92e544877a6a86b0a2feb4f9e : SUCCESS in 1h 52m 29s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/ea34192fdf8d414581c75845e700921d : SUCCESS in 2h 06m 40s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/eb80e15a0a5d4cb7ad7f1e50e3bfa7a5 : SUCCESS in 2h 11m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9164629a0ee644eda68ed6f4dcf04fe7 : SUCCESS in 53m 15s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/734041f2a5254324ad10cc1004da4bef : SUCCESS in 1h 28m 55s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/1e6d05be7beb430b84a701fc351932dc : SUCCESS in 1h 50m 40s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/30a3024b94b94e1e9e008c3a51932b26 : FAILURE in 23m 16s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/61b9b051bea64c5ebfcc73b105f17e6b : SUCCESS in 1h 51m 11s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/b059cb01f7b14e36be6ca191ed5c9abe : SUCCESS in 25m 05s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/7b71e67bde564287832ac10539d117d1 : SUCCESS in 1h 04m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e2219a58ef9f050ac193eb5ecff44ad737df9fc0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-05-28 11:20:49.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"c83f26fc0b4a955107d2b53e058f784012933c76","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-28 11:38:20.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/60d81ba192084531a8a0c97e88b435e0\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/423f55bf3f854899975ff7b9abfbf6df : SUCCESS in 10m 13s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/91bd4293c1b34370a565270456a179bf : SUCCESS in 10m 14s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/de5fbe6cf20942d7bbcb7a1b6ec867d5 : SUCCESS in 15m 22s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a3a04091c291298639fddea6732889a2022d4094","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2025-05-28 13:37:32.000000000","message":"Patch Set 2:\n\n(7 comments)","accounts_in_message":[],"_revision_number":2},{"id":"39428ce6a6588221a65ea664f179fadbae4a5ed0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-28 13:55:16.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e17f9892189b40bd9678d2a6a05bc981\n\n- grenade https://zuul.opendev.org/t/openstack/build/66355ad489c94fd0ab7fadbee6e28046 : SUCCESS in 52m 44s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c6ecc2eb7c5740d39ad51842c2acaf25 : SUCCESS in 54m 10s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/09d8609a4e4a487f810e2d7f7cb74744 : SUCCESS in 1h 24m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7ae54e6795e746c7acca6a21ed6d0737 : SUCCESS in 1h 05m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e65ca071472408baec74e900520d2e9 : SUCCESS in 4m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/592f343b19e74a148d306abb0bd7f0d9 : SUCCESS in 7m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4bfdad52025f44fe8e1bcbf93b5ce134 : SUCCESS in 5m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8020c5ed87424a2b87b2e28d143db353 : SUCCESS in 8m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29ec37de95374b46923751d7eb279d55 : SUCCESS in 7m 24s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/bd7acfd306e34d01a7e64bd3a306124d : SUCCESS in 7m 35s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/e2e2f5bd8b5e4f5c85d6fc4ce6f10f77 : SUCCESS in 9m 06s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/7ae54b0afcae49e5944d229c9d2a2f0b : SUCCESS in 2h 08m 30s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/38ca9b6fd35241b3a64b0490f439064f : SUCCESS in 2h 06m 20s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/78a028cf24b5430098a6344d77322ae3 : SUCCESS in 1h 55m 15s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/474c710353414b259697eba930397cf2 : FAILURE in 1h 50m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/022d05c838ea46b08e373662d963209e : SUCCESS in 1h 04m 06s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/5013ce76ce114bfd8f82596f4f3db655 : SUCCESS in 2h 01m 32s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/c21eeb1a81b3404499c4be7a62635e12 : SUCCESS in 1h 53m 11s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/4b4b11df09284b149a17f89c5d4ee583 : FAILURE in 28m 38s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/d010db8c98a64490b5b85f92af29fd9b : SUCCESS in 1h 45m 38s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c166cbe412014e83a41a98a0a69650c9 : SUCCESS in 28m 43s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/ccc961a0a2ff477d8b710636ac8c8380 : SUCCESS in 1h 02m 14s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7b8b2c59818b996fd735448f368dffe2a5ad8bd3","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-05-28 14:29:09.000000000","message":"Patch Set 2:\n\n(8 comments)","accounts_in_message":[],"_revision_number":2},{"id":"0eabb3dfee409b3ff44052d164ecc29fee12f62f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-05-28 16:19:37.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"9ba88eb78888d4400f9a7aeb589b97e61ed4bfb9","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-05-28 16:20:27.000000000","message":"Topic bug/2045769-locks removed","accounts_in_message":[],"_revision_number":3},{"id":"077528be8fb36bdd26bef80ed6d9bb391253db79","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-05-28 16:20:30.000000000","message":"Topic set to bug/2045769","accounts_in_message":[],"_revision_number":3},{"id":"b85491aae64a7b5edc27c075340ac1ff7bba3f1c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-28 16:28:33.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c764a1dfcd234bd580e6802a833766a1\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ffd499c5132645f1918422158894ea31 : SUCCESS in 5m 33s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5c0f43bd6f4c4e928704f3e0807e8e35 : SUCCESS in 5m 31s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/7257caaba05b4ba3b071d64a244a61ba : SUCCESS in 7m 17s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8017548d72c7b2b20910443c65c3162e3723c708","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-05-28 18:12:17.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"37f2e1ced9ba4f5e3ea15558a17e2b199e0f445b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-28 18:44:11.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/472e015897bf421d806539c4587df12e\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b6a8b68d7fae48b79f979d414655148e : SUCCESS in 5m 08s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/bb22bd559ee44dada0052cc958833cb5 : SUCCESS in 5m 07s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b756289532d74f4cbf12872f30cfd4ad : SUCCESS in 7m 39s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"c25bd814164313b40e16fbc4b0a1837a69683f27","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-28 20:36:06.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/64494e7bf90b42139024fb0f31cb01f9\n\n- grenade https://zuul.opendev.org/t/openstack/build/f7db69f6a3974e888fb8f92ffb8496bb : SUCCESS in 58m 47s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6beb517bda324787af74eb93c219b0c3 : SUCCESS in 53m 19s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e0a3fdde30c54f448c55fa50ae7cb234 : SUCCESS in 1h 41m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dab881aff41c4d569957e5570f17e5bb : SUCCESS in 1h 02m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/beef75e895f84fb88bcfe12dddfe6cb5 : SUCCESS in 4m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d54e17dd41ac40d4b88332893350dc09 : SUCCESS in 6m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/35946e0241ec4a35adcbcb3605c96f4e : SUCCESS in 5m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5eb90565442446cc8d6b8cd88121277d : SUCCESS in 9m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02cb662309074941b72e2150167ff8d3 : SUCCESS in 5m 20s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/82ff854596b44780a912609f50204742 : SUCCESS in 7m 52s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/26a55e141dfb4c41802983cf1493850e : SUCCESS in 9m 22s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/37cbe734af674918aec2068e32a0e871 : SUCCESS in 1h 26m 29s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/a5d1dc71858f49ba9730265261f6dc51 : SUCCESS in 1h 51m 30s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/96ab343bce1f43be8cc0ab50100e50b9 : SUCCESS in 1h 32m 16s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/7cbe20a236494c9ab488600b130451c1 : SUCCESS in 1h 49m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/117796de61364595afc3415904ad51ea : SUCCESS in 1h 00m 36s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/91b5fd38329743338b577b5573338d07 : SUCCESS in 1h 32m 55s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/76856dbe81b54750a840b84014d3a724 : SUCCESS in 1h 57m 41s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/6d29873cbbdd4c1e8a69ea35ac5bb0f8 : FAILURE in 30m 29s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/29924d32b2b84536afff45d5c645bb4a : SUCCESS in 2h 00m 12s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/b32a71c08c8b448d814a6f87680077b7 : SUCCESS in 32m 23s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/54f7217fc961489db2c64b5a49818423 : SUCCESS in 1h 06m 37s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"22959179bac85ad93cf8d7a65b8be8fa85a39c08","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2025-05-29 13:37:34.000000000","message":"Patch Set 4: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"f754b2d2174a829888d272721411539b174cdb0a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-05-29 14:17:18.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"b7aa0ae5063b23e332d0b232323be38808a0cd91","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-29 14:27:28.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/876b22e2f138447383342c7549cc0309\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d136b65c89044661906a6f7f196856bf : SUCCESS in 6m 03s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a156497091a64defb3033feac97c3d8a : SUCCESS in 6m 29s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f1ba493696184e0caa44bd1bfbc867d6 : SUCCESS in 8m 36s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"40f30c21c6ffba2821d581a683cc6b2b4208b68f","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-05-29 14:44:30.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"2e559009aeb1563887adbe51761ca7d4f2bfe435","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2025-05-29 14:57:23.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"1e88c2b1ef94b83cfaa792e4cf059fd25cab9e00","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-29 16:53:01.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e7a1518ea08245c8a4f3b55f0925ead5\n\n- grenade https://zuul.opendev.org/t/openstack/build/3e35a43cba504880825e746c7d9bbf04 : SUCCESS in 1h 01m 51s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/97094a9287944247af424df5ab5c94ee : SUCCESS in 1h 02m 36s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/acb53b4becdb45b18fc2897630c3cdbd : SUCCESS in 1h 43m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9eacb12f42d141d9bf1320b806ee42fe : SUCCESS in 55m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/176756bcd42c4386827e4765032f7fe7 : SUCCESS in 5m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5f0e71582bf24263804f157a0dd3aba0 : SUCCESS in 4m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b2f6678b625a4e008e658b516d9a5967 : SUCCESS in 5m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e960adf995244e49806d7507b79fe785 : SUCCESS in 7m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea02ac254f714a48910647f9bca376d0 : SUCCESS in 5m 14s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6b06751216e9496e902ecb147300da43 : SUCCESS in 6m 49s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/89ebe2fe863b4a23b0ae7a67bc062730 : SUCCESS in 12m 04s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/05f139274e7141a7a98ce684399c305c : SUCCESS in 1h 56m 09s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/6c3d0f7100e845a7ae1aafff452afd6d : SUCCESS in 1h 37m 54s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/0d7e9415df134b618c142e02bd440b0e : SUCCESS in 2h 03m 12s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/7818618a6408481a8747031225bb8a8e : SUCCESS in 2h 05m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6123f4122a054a0e9fac7537bce3bb03 : SUCCESS in 52m 24s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/39303e612a44453d85a63ef2bfb2a4cf : SUCCESS in 1h 37m 15s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/3784b382eba14a6f83ad86c99faa2c72 : SUCCESS in 1h 58m 51s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/13ba39be0ef447bc8b7ca01e7ada5ed3 : FAILURE in 33m 18s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/2a45a2f380084016860fa1a867821756 : SUCCESS in 1h 33m 43s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/42b25dd6d15d4b739c0418fca69dbbcf : SUCCESS in 30m 40s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/41c6a94f139f40e4aa4840323abf6c73 : SUCCESS in 1h 07m 07s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"bfdbcfa1f08b8dbe235c511dec13d5c383248068","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-05-29 17:07:39.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"4dbef5bec3d164eb0219356b31937e3e1a9f6b4f","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-05-29 17:08:05.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"ccf54a2cb0c8060bebd5f30a0937249a6e7ea4d4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-29 17:17:50.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2d44d9212f7944e0a8cd7a6506a14dd1\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5f10caddf3664213bc5f4b6fcc2e090b : SUCCESS in 5m 42s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/06981a9ea02b4e5c821cd5e590da187c : SUCCESS in 7m 30s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/8c3c0d0c20b148c8b137716cf762412c : SUCCESS in 8m 24s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"a91d448616e75d084aa140025be2046c831d50fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-29 20:13:08.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0ddd866fe6f041538a061e861ed04350\n\n- grenade https://zuul.opendev.org/t/openstack/build/fae12a1c85c544d6a778f242e3a1d209 : SUCCESS in 1h 01m 58s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/149ec44ad72a4a61b57250e08a798c03 : SUCCESS in 59m 20s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/320e79b769e4435e90772a03df1e82b0 : SUCCESS in 1h 31m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bef12a7223bd464c88081da10551daa3 : SUCCESS in 1h 01m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fdae1394f7464aaa95f53886733dc2ec : SUCCESS in 4m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2df7a438583e4b8da13169b56057ae4e : SUCCESS in 7m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4a88378d0b0b4e7a803678e06508ca45 : SUCCESS in 5m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e3d9171414a849feacc7e21de92914ef : SUCCESS in 8m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1bcc4d82c50041d8809137fc0e25023c : SUCCESS in 5m 11s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1710720ccf0841b9b10da29c9dddbb6c : SUCCESS in 8m 01s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/ca3a2f93d3f440c4bb2450693c5b5df3 : SUCCESS in 9m 53s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/db5e130175704155bb48be5e0721482c : SUCCESS in 2h 06m 35s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/d14c9a37e0954100bd8db8c042475c98 : SUCCESS in 1h 54m 23s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/08e77231d5ce4d1cbd42bd9c3f12ff1d : SUCCESS in 1h 59m 41s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/4a4643c071744b4f9e9d6ccd69c15354 : SUCCESS in 2h 20m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a874091a14b4454f89e6cbd137908b46 : SUCCESS in 1h 00m 01s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9cc30faab8c94a50892dc8e1acd2e705 : SUCCESS in 1h 58m 31s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/d637658e2d0f43ebb47a0137782a6682 : SUCCESS in 1h 52m 24s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/f587912d16e84fc6ae1359559f0609c0 : FAILURE in 27m 08s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/5c47a01c3abb47718e7b359a22347cbd : SUCCESS in 1h 55m 12s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/6d0a492a040c49bead57b8e6a020ab96 : SUCCESS in 27m 11s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/c92d020da123448d9a42c881f6f66092 : SUCCESS in 1h 06m 06s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"8a7f6ccd27abf768c5f92130150570a069dc8024","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2025-06-09 13:21:03.000000000","message":"Patch Set 6: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"d0cdb5353119ceae795df6cd4cdd5c1f37e174e8","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-06-09 13:45:33.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"ebe5cff01e9952d26da0109a7d2c776cf9ce970c","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2025-06-09 14:12:42.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"9eb3b10f1f761b812c0cfbd1ab6efc470ae9ab0d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-06-09 15:25:25.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"5d4a56253e20d884977142cf745e7097d17e6003","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-06-09 15:55:20.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"10161dca08a5a94f3de7f189f737ff985461d94f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-09 16:18:48.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8dbee6e863f74a24932bc059c3b1a65f\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b7f726d5232a4062aa3e4ffc6282d04e : SUCCESS in 10m 24s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e81f464e26174dfdacf324697c0cd62c : SUCCESS in 7m 19s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c958baca0952434bbdb60f7080e71ad6 : SUCCESS in 7m 59s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"0a0ae789c2218720e4dffe8511cd3a02de28ccf8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-09 17:42:02.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7edbeb22658847c79109565f32480984\n\n- grenade https://zuul.opendev.org/t/openstack/build/f7029207ccbc46908dfba9fac026291f : SUCCESS in 55m 13s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/75b371a0f993460b8e66740e49de4eff : SUCCESS in 1h 00m 48s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9ce6d0bbd1fb47df97229d1d25f5939e : SUCCESS in 1h 22m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a523c1e7e5174547828780b1ef628330 : SUCCESS in 1h 28m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc52a8babf764af0a0c126798f7e0ff7 : SUCCESS in 4m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5a0884524ca149d6b5723569914ffc4d : SUCCESS in 6m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8efff4ff73c640d8a505079ed8e0868e : SUCCESS in 5m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9936e09bb65947c0b5d53543f9383e5a : SUCCESS in 7m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e0ee0cf91574705b5725bff85a0114e : SUCCESS in 5m 47s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/91611eca2c2b402598a1523f48e0d060 : SUCCESS in 7m 12s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/a6f2a9817c9a4b6499551007a11abc80 : SUCCESS in 11m 15s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/776e8132da7049649d1c16348191db2b : SUCCESS in 2h 10m 52s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/61a6a64ad455462b9e9ef26ac4184731 : SUCCESS in 1h 55m 07s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/2ce2826e76e94f2b91498ab30c52a405 : SUCCESS in 1h 42m 49s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/8c7e169f11d449ce8929290fa9a827f2 : SUCCESS in 2h 06m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4092e5a516b94271b61703b53920de42 : SUCCESS in 46m 21s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/3a6418f3aae64b739bb88406c708918d : SUCCESS in 2h 02m 15s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/938eaa6cb08a409eb4c8a6af3a5e15f3 : SUCCESS in 1h 37m 03s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/e4743be43a0e4c5380952226550c3107 : FAILURE in 29m 17s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/01258bc0ab264db88775fa9b25012ae9 : SUCCESS in 1h 58m 26s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/18b896162dd643cea2956b666387d4dd : SUCCESS in 18m 41s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/f65e2b5aa9c0417f835aac86f37a432e : SUCCESS in 38m 15s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"de912ac8f5f89dd0357fbc841c948e772e472437","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-06-11 15:09:10.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"931cfbec8b4335c830923fb3e88325b04f7dbef0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-11 16:47:45.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/451ee2213ccd4155aa2cb2e6028c371a\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3a9a595c029648e68b4ab9f7bb7a473b : SUCCESS in 7m 30s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c58f2daf87284dd1aff00334dcee1b62 : SUCCESS in 8m 04s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/47f1c94a52bb4e479bc6bae0e54bf007 : SUCCESS in 17m 15s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"4955f6aa76a5326041f1cb7212ab987d2108ab21","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-11 17:20:11.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4aaddbb7c9564ef3ada18c3ff4c0d6c6\n\n- grenade https://zuul.opendev.org/t/openstack/build/667a57acf84548c8b15fefcdf5f097a4 : SUCCESS in 56m 50s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/bdd15891dcaa4184ac4ce78e89749270 : SUCCESS in 52m 48s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d95802042be04a289031d0a6587e87ec : SUCCESS in 1h 46m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5333ed5f8efe4d07a45438cd24a3d993 : SUCCESS in 53m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6c4c97d308f40bd89a03b0439560112 : SUCCESS in 4m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0d2f3437322b40d990a5bb320f71f826 : SUCCESS in 4m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0161c610ac814e24bf7ef4faf1c5643b : SUCCESS in 5m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b354b6c2bd7e47c2906698eb7b1308f0 : SUCCESS in 7m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7282dbdd096949f9bec5e5803a67b148 : SUCCESS in 5m 36s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/50d5c8c60ae9494daba7c4b354ae6a35 : SUCCESS in 7m 23s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/47e431cab8d84bfcbe20e774333be921 : SUCCESS in 11m 28s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/5b29b4f3e1c1447aa3f26a8effb0791d : SUCCESS in 1h 40m 06s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/7d4c7a26ba3249b98847e4f85bc5dfca : SUCCESS in 1h 45m 28s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/d74f49d3bbd94a4ea1a9cd1e0a072e53 : SUCCESS in 2h 05m 45s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/21cf249310794e8e9ef3f04da845551a : SUCCESS in 2h 03m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d70027703cb44c6c9f1cd68adabe8e52 : SUCCESS in 49m 12s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/6e4685c386e04a2b8d0eee907cc9524f : SUCCESS in 1h 50m 15s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/d07793e32e25459fad8e81cb5bfe8d8c : SUCCESS in 1h 17m 59s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/08c81f690b4b4f31bc66bb9fc3fa01f2 : FAILURE in 35m 42s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/80417dc58e364971bd0afdde76634b75 : SUCCESS in 1h 45m 24s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/5d0f178d4b0f41f4b71811a0c0fd6d35 : SUCCESS in 27m 10s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/7d87e7b6b43a436abadcb623d1e28d0d : SUCCESS in 1h 14m 56s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"3e7b895df067de4101de041a4c70fcc1b18e70af","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2025-06-13 15:09:39.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"0933b8395bae2e376169abba0eea2e5d5138d605","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-06-13 18:23:47.000000000","message":"Patch Set 8: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"fe1a819bd07a1e517af355d44b6502a7e18d8d05","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2025-06-13 19:53:32.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"9d59a1d16c2bcbe02bc49f0da1f95a97ed49897c","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-06-13 21:20:53.000000000","message":"Patch Set 8: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"987f12eb194dcbe35f2f1680339e0efc26531ccc","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-06-16 06:24:58.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"9c59ed9e672217e2147d1c0bdad11385039c531d","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-06-16 13:36:35.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"e0b190f233ce27c8ff4b4ab72aca9cfcac01bb51","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-16 13:37:06.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"241ed56f1fb350c55c2c26257f62b6cd389df301","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-16 15:41:06.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b74e3abffc744ca8897b5e353d92a458\n\n- grenade https://zuul.opendev.org/t/openstack/build/aa35833f50e94306bdf07b97f5a43009 : SUCCESS in 58m 18s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9f1ddf137ba24fa6a6ec31a54b0d0425 : SUCCESS in 1h 00m 47s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a7839564dd0a4e19a2bbbb402e2b8ba9 : SUCCESS in 1h 25m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/42ff0f35aebd4608a833daeebb4327f0 : SUCCESS in 1h 00m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e9713e4d74d478999dcbb45367613ef : SUCCESS in 6m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6a22866f7eb040ee8f157b625baf892f : SUCCESS in 5m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/27e331d960dd4913b429eeb503ba35b5 : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/045b1b54c41f4027bd6190904ab37d0d : SUCCESS in 5m 28s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/34e01ecfb5414ab38961e8e3280037fe : SUCCESS in 8m 49s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/fb752db58c2b4c178e48cd6ba2cb9ceb : SUCCESS in 11m 43s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/ab50d827f55e4699b8f976ee5b6fbd38 : SUCCESS in 1h 42m 14s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/9762d27e38c348f9ba79113f400a4dfe : SUCCESS in 2h 03m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b903f400c8d546bd87d084f0e9ab3f2f : SUCCESS in 1h 00m 25s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b9458eb7e23549c0988371ae5eed7365 : SUCCESS in 1h 15m 26s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/7851fce0f96e438988c033661f9288ff : SUCCESS in 29m 03s","accounts_in_message":[],"_revision_number":8},{"id":"ff866ebad2e12243a48ff7944b57b2f6cbcfc7ca","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-16 15:41:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8},{"id":"67753b0b70d9b84f3221f7554cd7f970227dd258","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-16 15:42:15.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5eb7b38f04bc4f1da149c3f19e050485\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5519de5ba84470fac216dea1750fcc1 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"1311d5fe2f1e308e1abd930ab17a8fb368b670fe","revisions":{"06bb868025883502f509003116d2d004293581e8":{"kind":"REWORK","_number":1,"created":"2025-05-27 20:24:11.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/31/951031/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/31/951031/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/31/951031/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/31/951031/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/31/951031/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/31/951031/1"}}},"commit":{"parents":[{"commit":"d12086ec6a3cf1a22a2766bd1ba9c5b85ffba8a4","subject":"Merge \"Functional tests migration of TestImageMultipleBackend\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d12086ec6a3cf1a22a2766bd1ba9c5b85ffba8a4"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-05-27 20:00:14.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-05-27 20:00:14.000000000","tz":0},"subject":"Add task cancellation tracker utility","message":"Add task cancellation tracker utility\n\nThis module provides functions to register, cancel, and signal cancellation\nof long-running tasks using lock files and external locking. It ensures\natomic operation registration, safe cancellation, and proper cleanup,\nsupporting concurrent task management.\n\nChange-Id: Ic58233302001a2396e84400ef7081afa51748a2a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/06bb868025883502f509003116d2d004293581e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/06bb868025883502f509003116d2d004293581e8"}]},"branch":"refs/heads/master"},"ab71623087941c03359cc03ec06db6f4993c3154":{"kind":"REWORK","_number":2,"created":"2025-05-28 11:20:49.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/31/951031/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/31/951031/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/31/951031/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/31/951031/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/31/951031/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/31/951031/2"}}},"commit":{"parents":[{"commit":"d12086ec6a3cf1a22a2766bd1ba9c5b85ffba8a4","subject":"Merge \"Functional tests migration of TestImageMultipleBackend\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d12086ec6a3cf1a22a2766bd1ba9c5b85ffba8a4"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-05-27 20:00:14.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-05-28 10:57:59.000000000","tz":0},"subject":"Add task cancellation tracker utility","message":"Add task cancellation tracker utility\n\nThis module provides functions to register, cancel, and signal cancellation\nof long-running tasks using lock files and external locking. It ensures\natomic operation registration, safe cancellation, and proper cleanup,\nsupporting concurrent task management.\n\nChange-Id: Ic58233302001a2396e84400ef7081afa51748a2a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ab71623087941c03359cc03ec06db6f4993c3154"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ab71623087941c03359cc03ec06db6f4993c3154"}]},"branch":"refs/heads/master"},"88d2f289e6573bbfa07370445f1fd006ff8dc730":{"kind":"REWORK","_number":3,"created":"2025-05-28 16:19:37.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/31/951031/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/31/951031/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/31/951031/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/31/951031/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/31/951031/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/31/951031/3"}}},"commit":{"parents":[{"commit":"d12086ec6a3cf1a22a2766bd1ba9c5b85ffba8a4","subject":"Merge \"Functional tests migration of TestImageMultipleBackend\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d12086ec6a3cf1a22a2766bd1ba9c5b85ffba8a4"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-05-27 20:00:14.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-05-28 15:57:35.000000000","tz":0},"subject":"Add task cancellation tracker utility","message":"Add task cancellation tracker utility\n\nThis module provides functions to register, cancel, and signal cancellation\nof long-running tasks using lock files and external locking. It ensures\natomic operation registration, safe cancellation, and proper cleanup,\nsupporting concurrent task management.\n\nChange-Id: Ic58233302001a2396e84400ef7081afa51748a2a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/88d2f289e6573bbfa07370445f1fd006ff8dc730"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/88d2f289e6573bbfa07370445f1fd006ff8dc730"}]},"branch":"refs/heads/master"},"75c96949a09ed87e8b3b0604b56519cd7139de71":{"kind":"REWORK","_number":4,"created":"2025-05-28 18:12:17.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/31/951031/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/31/951031/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/31/951031/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/31/951031/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/31/951031/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/31/951031/4"}}},"commit":{"parents":[{"commit":"d12086ec6a3cf1a22a2766bd1ba9c5b85ffba8a4","subject":"Merge \"Functional tests migration of TestImageMultipleBackend\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d12086ec6a3cf1a22a2766bd1ba9c5b85ffba8a4"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-05-27 20:00:14.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-05-28 17:30:10.000000000","tz":0},"subject":"Add task cancellation tracker utility","message":"Add task cancellation tracker utility\n\nThis module provides functions to register, cancel, and signal cancellation\nof long-running tasks using lock files and external locking. It ensures\natomic operation registration, safe cancellation, and proper cleanup,\nsupporting concurrent task management.\n\nChange-Id: Ic58233302001a2396e84400ef7081afa51748a2a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/75c96949a09ed87e8b3b0604b56519cd7139de71"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/75c96949a09ed87e8b3b0604b56519cd7139de71"}]},"branch":"refs/heads/master"},"dc2af9bd9d45877fdeaae452e01fa68e52ffd31f":{"kind":"REWORK","_number":5,"created":"2025-05-29 14:17:18.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/31/951031/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/31/951031/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/31/951031/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/31/951031/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/31/951031/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/31/951031/5"}}},"commit":{"parents":[{"commit":"d12086ec6a3cf1a22a2766bd1ba9c5b85ffba8a4","subject":"Merge \"Functional tests migration of TestImageMultipleBackend\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d12086ec6a3cf1a22a2766bd1ba9c5b85ffba8a4"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-05-27 20:00:14.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-05-29 13:55:28.000000000","tz":0},"subject":"Add task cancellation tracker utility","message":"Add task cancellation tracker utility\n\nThis module provides functions to register, cancel, and signal cancellation\nof long-running tasks using lock files and external locking. It ensures\natomic operation registration, safe cancellation, and proper cleanup,\nsupporting concurrent task management.\n\nChange-Id: Ic58233302001a2396e84400ef7081afa51748a2a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/dc2af9bd9d45877fdeaae452e01fa68e52ffd31f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/dc2af9bd9d45877fdeaae452e01fa68e52ffd31f"}]},"branch":"refs/heads/master"},"26ec8b0d013a87824f8975414a4f4e121b0068ee":{"kind":"REWORK","_number":6,"created":"2025-05-29 17:07:39.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/31/951031/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/31/951031/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/31/951031/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/31/951031/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/31/951031/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/31/951031/6"}}},"commit":{"parents":[{"commit":"d12086ec6a3cf1a22a2766bd1ba9c5b85ffba8a4","subject":"Merge \"Functional tests migration of TestImageMultipleBackend\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d12086ec6a3cf1a22a2766bd1ba9c5b85ffba8a4"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-05-27 20:00:14.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-05-29 15:08:26.000000000","tz":0},"subject":"Add task cancellation tracker utility","message":"Add task cancellation tracker utility\n\nThis module provides functions to register, cancel, and signal cancellation\nof long-running tasks using lock files and external locking. It ensures\natomic operation registration, safe cancellation, and proper cleanup,\nsupporting concurrent task management.\n\nChange-Id: Ic58233302001a2396e84400ef7081afa51748a2a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/26ec8b0d013a87824f8975414a4f4e121b0068ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/26ec8b0d013a87824f8975414a4f4e121b0068ee"}]},"branch":"refs/heads/master"},"928790da456f5efd3a48830528b4e688d6f0b23f":{"kind":"REWORK","_number":7,"created":"2025-06-09 15:25:25.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/31/951031/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/31/951031/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/31/951031/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/31/951031/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/31/951031/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/31/951031/7"}}},"commit":{"parents":[{"commit":"e055873be4079bc9d37165cf75de5495603133b0","subject":"Get ready for oslo.limit 2.7.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e055873be4079bc9d37165cf75de5495603133b0"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-05-27 20:00:14.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-06-09 14:12:29.000000000","tz":0},"subject":"Add task cancellation tracker utility","message":"Add task cancellation tracker utility\n\nThis module provides functions to register, cancel, and signal cancellation\nof long-running tasks using lock files and external locking. It ensures\natomic operation registration, safe cancellation, and proper cleanup,\nsupporting concurrent task management.\n\nChange-Id: Ic58233302001a2396e84400ef7081afa51748a2a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/928790da456f5efd3a48830528b4e688d6f0b23f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/928790da456f5efd3a48830528b4e688d6f0b23f"}]},"branch":"refs/heads/master"},"1311d5fe2f1e308e1abd930ab17a8fb368b670fe":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2025-06-11 15:09:10.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/31/951031/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/31/951031/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/31/951031/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/31/951031/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/31/951031/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/31/951031/8"}}},"commit":{"parents":[{"commit":"dac9741c2ecb62a79d9eaf3ed5bffa60f4f8bf69","subject":"Remove Python 3.9 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/dac9741c2ecb62a79d9eaf3ed5bffa60f4f8bf69"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-05-27 20:00:14.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-06-11 13:22:09.000000000","tz":0},"subject":"Add task cancellation tracker utility","message":"Add task cancellation tracker utility\n\nThis module provides functions to register, cancel, and signal cancellation\nof long-running tasks using lock files and external locking. It ensures\natomic operation registration, safe cancellation, and proper cleanup,\nsupporting concurrent task management.\n\nChange-Id: Ic58233302001a2396e84400ef7081afa51748a2a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1311d5fe2f1e308e1abd930ab17a8fb368b670fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1311d5fe2f1e308e1abd930ab17a8fb368b670fe"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
