)]}'
{"id":"openstack%2Fglance~951611","triplet_id":"openstack%2Fglance~master~I082c806fb492a770b374b91fabc7937608a40de7","project":"openstack/glance","branch":"master","topic":"spec/set-size-on-upload","attention_set":{},"removed_from_attention_set":{"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2026-06-24 06:30:51.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I082c806fb492a770b374b91fabc7937608a40de7","subject":"DNM Verify x-image-meta-size for all backends","status":"ABANDONED","created":"2025-06-03 06:12:52.000000000","updated":"2026-06-24 06:30:51.000000000","total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f36e3611e6fbddc0cd0d6aeb30cbbbb30a9cace3","_number":951611,"virtual_id_number":951611,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-07-06 17:07:49.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-03 06:58:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-06-03 09:21:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"86d3be738bcb03b91bf23050836dd1fd764c2ac1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-06-03 06:12:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3ff1e2b69275c25444957e860a162250f3fee5d6","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-06-03 06:13:18.000000000","message":"Topic test-dnm-size-pathc removed","accounts_in_message":[],"_revision_number":1},{"id":"e581e450d5dd526eac4f2a6c9d21025bfad5c44f","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-06-03 06:13:22.000000000","message":"Topic set to spec/set-size-on-upload","accounts_in_message":[],"_revision_number":1},{"id":"c03f85349cdcf586a4b5bc52e36bd4e275a7e24f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-03 06:58:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f6d681c5fb364288ac87029d81775562\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/bc9a8ae3ff2447ba8c432cb1a6199a5d : SUCCESS in 16m 58s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8e7129f89e734e34872341ef84b29fd6 : SUCCESS in 12m 36s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/765780fc69064d0d92478ed154ade28c : SUCCESS in 24m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9907fc750b72e1fa31f58909b39cb06808a40dcc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-03 09:21:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9aeeeadcd0f1427b9021b37a72a4bebe\n\n- grenade https://zuul.opendev.org/t/openstack/build/143e2c9fbe9b48e0870d063761f8f219 : SUCCESS in 53m 46s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/436f4e322ba74feebdcd2831697d340b : TIMED_OUT in 3h 02m 05s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e8301d2f94de4888828ef476f65b3f62 : SUCCESS in 1h 41m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9a560ebdf299492ba9fc8fcad2208809 : SUCCESS in 1h 01m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1d8c148ee4e479d98fffb24030659b3 : SUCCESS in 5m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9f87abee67244fe0a1772bd3708d5b5b : SUCCESS in 4m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/75a45c827c15470fb117116799e4109c : SUCCESS in 6m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a5145a91f84440fcb0c0fb4c61058cbc : SUCCESS in 8m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ca50eb567cf4ab7afb771e82d6358fb : SUCCESS in 5m 57s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/76c81718334f435987ba70382c3a4024 : SUCCESS in 6m 59s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/1e17aede9af84711b2e621682f1b6cbb : SUCCESS in 9m 48s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/4f9548ace454436eb3e1107e97701ed0 : SUCCESS in 1h 43m 58s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/06087cdc1eed4293b35adbd89346380e : SUCCESS in 1h 48m 01s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/3d84ca80c8aa4c269713c3e61bfda4f3 : SUCCESS in 1h 56m 52s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/c9289f7e3fed4d4db95bfe084d5d7bd0 : SUCCESS in 2h 11m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cdcdec215e6441639d0a9649536cf6fa : SUCCESS in 50m 58s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/bef39601342c487687dcb77d08ff5168 : SUCCESS in 1h 43m 33s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/23c7c0b2b49c4b5296ac184ec15c8c65 : SUCCESS in 2h 13m 08s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/91c91987f28c487fa7bdc34754ecc52b : FAILURE in 29m 49s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/1f52073ec1c5464794ef6d6c3b49862e : SUCCESS in 1h 50m 02s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/68f161b887d44647b162183feb1cc80f : SUCCESS in 31m 55s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/01ed0385dc3d438f9a0fb2efe657ab2c : SUCCESS in 1h 04m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"57def4306c08632fab34b928894ce2b21617391e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-06-27 09:59:58.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"827f0ca6267236fa60a9fe4c8aba1d6ebd873d38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-27 10:00:20.000000000","message":"Patch Set 2: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/glance for 951611,2","accounts_in_message":[],"_revision_number":2},{"id":"11436b4a8ad4c38e735c7bdfe3cc4474a9e8b9db","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-27 10:00:21.000000000","message":"Patch Set 2:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/glance for 951611,2","accounts_in_message":[],"_revision_number":2},{"id":"9a030c4cfede8f1d054c975c22ba09d40e0e9524","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-06-27 10:05:09.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nThe following files contain Git conflicts:\n* .zuul.yaml\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"996b282f871722418ec9ee998854a706cb11fa98","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-27 10:05:45.000000000","message":"Patch Set 3:\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/glance on branch master.  The\nproblem was:\n\n  while scanning a simple key\n    in \"openstack/glance/.zuul.yaml@master\", line 256, column 1\n  could not find expected \u0027:\u0027\n    in \"openstack/glance/.zuul.yaml@master\", line 257, column 9","accounts_in_message":[],"_revision_number":3},{"id":"75640b0e772a47ac995d42a97a08c96b9affe5c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-27 10:05:48.000000000","message":"Patch Set 3: Verified-1\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/glance on branch master.  The\nproblem was:\n\n  while scanning a simple key\n    in \"openstack/glance/.zuul.yaml@master\", line 256, column 1\n  could not find expected \u0027:\u0027\n    in \"openstack/glance/.zuul.yaml@master\", line 257, column 9","accounts_in_message":[],"_revision_number":3},{"id":"f06b723ae7e25f1069771d6bb5f2fea4b2ce568e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-06-27 10:06:00.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"741bd767cdac6d8d9e30832527c5620a841a2528","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-06-27 10:06:24.000000000","message":"Patch Set 4:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":4},{"id":"e688d25cd49a42d35a846d9dc1cc0254d19487c7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-27 10:41:11.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b066cbfc80674b1f87b7a9314c4c97fb\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/50d235a955e44e5783092994b6452ea0 : SUCCESS in 9m 05s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e691231d489546fa8810aac1f9b95884 : SUCCESS in 7m 44s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/2a70e9b25100434a9744ec552c766e21 : SUCCESS in 10m 46s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"bf4861a8ad2226a3f24accd9b4a335e94484c7c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-27 12:22:29.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/403d2d0bd390445a9f9d3770c7258fc8\n\n- grenade https://zuul.opendev.org/t/openstack/build/2b3e36ea80d344dc8157e57f7026d606 : SUCCESS in 54m 18s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/18d6da73975c4afe937cf0e14ec4fddc : SUCCESS in 1h 03m 00s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5f97ef9987f446a4b7664db4ef73ee0e : SUCCESS in 1h 27m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/52ad71ca9b4d4ea88efe355ee258f99d : SUCCESS in 1h 03m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef8a847463c54e1bbc0f98ff264f6574 : SUCCESS in 4m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9fc78afa4b524a3284a3bed176592fe1 : SUCCESS in 4m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d8563f37f25c40dfac29216d27d64ba9 : SUCCESS in 7m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7f33efefb3a44a88ab884b08d052a4bb : SUCCESS in 8m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e59eb29bb414451094ffbe726ef5ae84 : SUCCESS in 7m 57s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/561460e899f3474c8e53c32b6a9110dd : SUCCESS in 7m 56s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/b04c6f6e140343b898c58aac7839b65c : SUCCESS in 11m 03s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/282a9cd50aaa4fbf8d7c6926c1548b90 : SUCCESS in 2h 10m 12s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/c1ee2854ee334bf581e68ed873a1e083 : SUCCESS in 1h 55m 30s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/4afb80bce0744beda3d77461fb5b1c18 : SUCCESS in 1h 51m 26s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/c2fe1d55b24941c1a04d546ce216d55a : SUCCESS in 1h 46m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3e8392169c9344e5a3a276630a4fe354 : SUCCESS in 51m 09s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/946e8ac5724c49a69deed89c6969d790 : SUCCESS in 1h 58m 47s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/0504d7ade2154cdc8dbda2a1ca8796f6 : FAILURE in 31m 14s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/873545de7bb648d6ac1f1f994eb36e43 : SUCCESS in 1h 49m 04s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/e3b9e7f8be0546399eb7628d04b888f4 : SUCCESS in 13m 52s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/613b0586a47f45c9af95fe7e282d0535 : SUCCESS in 1h 05m 25s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"0ebe10c1f6ba2fbd1aa154c47873486f84f0c4bb","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-06-27 14:50:28.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"6e4df2ec3c589cfa3b2a8dae4dd6b68699b26122","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-27 15:05:40.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4276db44d2bf4abb8e5f16407d81bcf6\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e36f6981d1f44e09bedefb76a9ea81a0 : SUCCESS in 8m 25s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0e7172e2a7594ec996b82d103eb74b3d : SUCCESS in 9m 29s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/3c3e3b3a01024cd7a8aa6049e86b2bf6 : SUCCESS in 12m 55s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"1fa85424fd5faea9a9d111a6e73fb02c5797644d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-27 17:01:23.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9c667d0c854843af84d8c36eef597558\n\n- grenade https://zuul.opendev.org/t/openstack/build/89196b6a29a342d19204eda6c3ead9d1 : FAILURE in 35m 58s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b9deea979b2a4364846cbc80bfc71a61 : FAILURE in 43m 52s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c4753a19fa5c4b67bd75a3dbdd8b098e : FAILURE in 1h 26m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e8b6f718d0e84f41bec6cdcace880995 : SUCCESS in 32m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0cd5866a40e49c6a96c10adf5dc826a : SUCCESS in 4m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4e44ca3e58fb43c48c9da3affa79fd35 : SUCCESS in 4m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d63a76e1fa5b4b5988db99c3a1d67e7e : SUCCESS in 3m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/12ac7553cd114ce095774474950c3b1a : SUCCESS in 7m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2673e0a18e424a1480ab27641c5f25cd : SUCCESS in 4m 47s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/65bfdaea77304377961ef2a5cebc3b40 : SUCCESS in 8m 03s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/57a77a214818487dbcd70858df273855 : SUCCESS in 7m 50s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/72b021e76a984bca8883985df7fe9786 : SUCCESS in 1h 45m 38s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/3d8779762fc44a0b90d0fce8f0e78712 : FAILURE in 2h 09m 20s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/bb7044be7a23482980e07f382ed95416 : FAILURE in 1h 35m 28s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/886984d68e2d438993f7422d27a2f7eb : SUCCESS in 1h 52m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8fb2e25347c04ddfa24cf2633e3d557f : FAILURE in 54m 18s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/2716280d3fe94f698afe39f5eb1b1d38 : FAILURE in 1h 58m 00s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/6873d1b0e046427abb40607f06286243 : FAILURE in 29m 05s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/65f0684199cc47b997a89515362048c7 : SUCCESS in 1h 56m 00s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/577f2f629bf14e3d97cb75a60b1d2053 : SUCCESS in 32m 03s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/c811c22eff164437812f5ae8a50731e7 : SUCCESS in 1h 03m 02s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"e3dc82f9f354087f94c6d9a66179b82c183e9057","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-06-27 17:54:01.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a01a2cbf1dc68a73478850ff24279af5318107cb","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-07-01 06:43:28.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"7ea59723d3c5f9450e4ea2c55eec391b8fb9853b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-01 08:35:25.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/823f935da2fa45158d1b4670a62298b7\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5758a3df4787483397cb90021e7b0b9c : SUCCESS in 10m 17s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/95d9c4eaf1764d37b99929a893b2e7f5 : SUCCESS in 9m 16s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d7e4259055614acb862ba3e14560e8d8 : SUCCESS in 11m 12s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"58686c0d2252a78d0075dc235b0ef85ba7f4743b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-01 09:24:32.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/695e8af93442436685255ff064cb97ec\n\n- grenade https://zuul.opendev.org/t/openstack/build/08ad8e728bfc41a0819cd860a4557376 : FAILURE in 36m 06s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3b9f68966b7041889d46364ff0180db8 : FAILURE in 35m 14s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/089211515c21429c82c5633581a3cd2b : FAILURE in 1h 04m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a98f1b6674df4601bab216dfb79485c2 : SUCCESS in 1h 02m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6cbc4091f9940dcb8ec5a0d4578ff63 : SUCCESS in 4m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1842472651df4a96920a0d72349ffd06 : SUCCESS in 4m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d40cea5f62c64eee91f1bdc419a58eda : SUCCESS in 5m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/059bce078cc0455e93cf0281e74e5e6e : SUCCESS in 7m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8769367cc95c4787a3a6687850b6890d : SUCCESS in 5m 36s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d5addf90d26746fa8798b2d81ae67ace : SUCCESS in 6m 59s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/bd8941b291fb4099b0a3dcc76ed41509 : SUCCESS in 9m 52s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/b1fecbdae8cf447a8cfa9442a0343671 : FAILURE in 6m 40s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/09a69fa5c91f4d47bdc9795f23ae0618 : FAILURE in 1h 51m 01s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/46a3650adffb46fc8379a759beb3a50e : FAILURE in 1h 31m 45s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/6c4d60999fa24e27900a85dba396872e : SUCCESS in 2h 35m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a3171b8bb43541e4836346059fdcbfd1 : FAILURE in 48m 40s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/8855665562e24b9dbdad4e24f0e4efce : FAILURE in 4m 28s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/363f1ce2c9604568bad8a994985b9525 : FAILURE in 31m 52s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/b0c6e852de93425c83e66976d4dddec0 : SUCCESS in 1h 05m 47s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/1c7c582098d343dcaadee0efcc46992b : SUCCESS in 24m 52s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/5ab9482f47224c079564e4f3086c115f : SUCCESS in 42m 55s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"4bee615c97179659cecc2e56896168d1829fedb3","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-07-02 07:36:17.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"5a8c6df80de19d8e0f8031ca2e38b9ee9f4b5906","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-02 08:01:58.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a892623de75f47d9b134073de3e3fa70\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/000d214172e44f1eadbbbc42c0358377 : SUCCESS in 8m 28s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2f604770e52d46da82755f27f2a62b20 : SUCCESS in 9m 31s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/5af624671dfa4b26ae966e53c6a8a096 : SUCCESS in 14m 03s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"76100179f37ac8dfc437e57d6dbe82dd4a1bb8a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-02 09:55:07.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/88511d1d10bd4555b5ef082f6916c0ef\n\n- grenade https://zuul.opendev.org/t/openstack/build/5733fc00f87846d082a1f39ef440dbd6 : FAILURE in 25m 13s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8f7c2df289bb49499fd630806fb20c56 : FAILURE in 30m 28s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/22aeb1d7b3924990976742324ea63cef : FAILURE in 1h 24m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/89ec5e06bad54af38add9d4ef51d27c8 : SUCCESS in 52m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e7119731e1644a1b0a7d96eaa4cdc0b : SUCCESS in 5m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d68b1c4fb7a14e16bb798abb9d5c5155 : SUCCESS in 4m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/18deffd1b2444d9abfd8afd8a02e91a4 : SUCCESS in 5m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5241f3d3bae94c0b93b38cbe64f22b1e : SUCCESS in 8m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d9b5d42f7d648188a730182be7378a0 : SUCCESS in 5m 33s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/887098477f23425f8bb0c0aa0cec3f2f : SUCCESS in 6m 51s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/366ae22f27c243e6a39183eded09519a : SUCCESS in 9m 38s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/f9e474fbe8a14f46818f16e1eb517ce2 : SUCCESS in 1h 39m 39s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/2c486401c2484c8a8175c3866f7d5e4d : FAILURE in 1h 43m 06s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/e379e91147a14da8bc0b806c7396945f : FAILURE in 1h 06m 00s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/85f3eada67a84bb5b1dd5e1c00d8fdb5 : SUCCESS in 1h 44m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/927f63314077413d946ff5e4e44cbbef : FAILURE in 35m 26s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/0c14338ab3a340ab89fa28140d69139d : FAILURE in 2h 12m 49s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/953fa1faf9194f77aa06c0e22a965709 : FAILURE in 30m 37s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/8113a135da344234832b51789f74f0f9 : SUCCESS in 2h 01m 23s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/7efa746eef974324895ee88c9d862a1e : SUCCESS in 26m 51s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/7ffdf16812a74df798df6793ff248da0 : SUCCESS in 1h 08m 46s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"3a2918f0386a11a22b598f0af05fd8cab70c5c30","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-07-03 08:47:42.000000000","message":"Patch Set 5: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"d18093695489d1bba3de86b8fbaa76ecd087cb10","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-03 09:28:34.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b55d41c7008f4444999ae4a7e7a9f366\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9e53cacab9094bb6a4735229f8ea9915 : SUCCESS in 9m 06s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3a7b930eaf75435897d09671eccd487f : SUCCESS in 11m 39s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/59fb5a1bbc0d4d2ebf7ec607061fa30c : SUCCESS in 11m 24s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"2df807cb80dba769174c3104b65ce8f89252fe9f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-03 11:15:00.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/961589678e0347139ae3ca4720e4a90a\n\n- grenade https://zuul.opendev.org/t/openstack/build/741639e8ad5f480bbbeee208ab4039b6 : FAILURE in 34m 41s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7ae950154b61427ca8e8df839ec11f9d : FAILURE in 40m 29s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d4fb702394dd4616b94e7bbcd5a365e2 : FAILURE in 1h 20m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/60bd94a2fcfd41d8bec6d9ede26ac7aa : SUCCESS in 1h 00m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/542debfde1454c2ba125018a5ebceaed : SUCCESS in 5m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6a929b30b453480696d43b9b090dfce3 : SUCCESS in 4m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c662f8856a48417ab2da44c401b96367 : SUCCESS in 4m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b81f1854fd6e4e4a830261298f142342 : SUCCESS in 7m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a33e02aa50cd47f2bae59182b8d1f016 : SUCCESS in 7m 09s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b802eaac6f5149f7bb435891a4f87f52 : SUCCESS in 6m 24s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/dbc4f2e684594dff9e9f938586228a9d : SUCCESS in 12m 18s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/fae947b833db463f9b4a21bfd5be5e69 : SUCCESS in 1h 49m 46s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/55519423fd094238a11d0af490b28cc6 : FAILURE in 1h 42m 39s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/855ca9ed92894743aa6ece453771b7d9 : FAILURE in 1h 59m 14s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/618fb5e15bed4415b6fe4e7dc992ef23 : SUCCESS in 2h 21m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5ed62f184c424651908eb87174fe40f8 : FAILURE in 51m 19s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9e43878a3ca141b089becbb185fc5a05 : FAILURE in 1h 55m 59s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/c9cc5d9b9d724e9ca0d16febabe73f3d : FAILURE in 32m 54s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/c5b2f4439db644c5bdf0eff0bc42837d : SUCCESS in 1h 54m 03s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/94b39e3d39154e6cadd7764705123618 : SUCCESS in 28m 51s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/63a1b6d867ed4249ae53fbf557ba2dbe : SUCCESS in 1h 04m 31s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"c99a64ade7e4168e76fbe822e8174c1fe1600818","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-07-03 14:16:36.000000000","message":"Patch Set 6: Patch Set 5 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"bd314aa17bf57e293fdcab95368965b7dcddd904","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-03 15:54:50.000000000","message":"Patch Set 6:\n\nThis change depends on a change that failed to merge.\n\nChanges https://review.opendev.org/948905, https://review.opendev.org/948903 are needed.","accounts_in_message":[],"_revision_number":6},{"id":"b0632d2faa1bbc722866ecad880fad6550014def","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-03 15:55:40.000000000","message":"Patch Set 6: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChanges https://review.opendev.org/948905, https://review.opendev.org/948903 are needed.","accounts_in_message":[],"_revision_number":6},{"id":"b000c77c5eb9eed0a4d8666855d6a639fc80502b","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-07-06 14:39:53.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"296911391a7167e5f314f75f78950eee49692153","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-07-06 14:40:55.000000000","message":"Patch Set 7: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"5fdbd4adb152e7dc3f4c7a32f14885a792684762","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-07-06 14:57:39.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"2720b8ad11821710d602d26e9d475c565f67c74b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-06 15:13:48.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e6ef655ea9841d38f3499df436aad83\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8154e70c017143dea08c143e9e6266d7 : SUCCESS in 9m 45s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/513a2b91182a4d5cbace59d8fb0f7b88 : SUCCESS in 11m 30s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/cb808c4f692f4fb68b853133704bd591 : SUCCESS in 14m 03s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"7d153ed53e653aa9fa0fb71d65d40b7284457822","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-06 17:07:49.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bb37eb5b254f46708d333f6bb356f87f\n\n- grenade https://zuul.opendev.org/t/openstack/build/d134cf3fa7d94cefb9e72302831839a3 : FAILURE in 32m 59s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/84ee3ae84cb2457fa645c7dee65f6042 : FAILURE in 47m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c47eb1b6870943fea56ffa28e88edb88 : FAILURE in 1h 17m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/58d12e218e1f4d5e9e9c58c9c6841b5d : SUCCESS in 1h 03m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9bd16d429ae640b0ab78ee8daa83a4ac : SUCCESS in 5m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e65407a4d1ed4db4b886bff715a83df9 : SUCCESS in 4m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/322998dd56fa4b54aa70dac0a0b4e336 : SUCCESS in 6m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c6682add1974463d88c4dd09ac88d280 : SUCCESS in 10m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cef43841a34f4196bcfaf3d5d639135d : SUCCESS in 5m 42s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/11dc8f1a853e4290980e0e81d341ac96 : SUCCESS in 6m 28s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/dc889181d9ef4aba94c86d246f303681 : SUCCESS in 8m 49s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/cd2f34038ddc45da91c9b36483cd792f : SUCCESS in 1h 52m 14s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/6ca57a39b76e4cb086f56db5d866cf15 : FAILURE in 1h 44m 17s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/7f49a727734c437aa441479187bf6855 : FAILURE in 1h 44m 54s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/5181f2213d3f489c9ed98b2b312679a2 : SUCCESS in 2h 07m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1fa9fa9cbe454a05bee9d7d1e54b3fa6 : FAILURE in 49m 28s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/dd029c42653b46e0bd0fdee5eae890df : FAILURE in 1h 51m 25s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/8bac5204939747569dc589cedc55691a : FAILURE in 40m 07s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/9b5dc23a298b48ce830248514a3a3a12 : SUCCESS in 2h 03m 25s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/e55b8c6efbb545d384709796520dd043 : SUCCESS in 27m 02s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/9e9873e58fe845f0aa692de10dc233f8 : SUCCESS in 58m 13s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"fb51fd86899678b6d0d052d264fa5bc2fcc6f37f","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-07-10 10:20:44.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"ab61ceab5c4145f5fb9e47f73b2737b28ff327b7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 11:03:42.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f2e4cd81ad814236ad55204fb1af3c56\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ac1836606dac4702bf8776cb2f1b4633 : SUCCESS in 6m 54s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/490d5020b9dc4b85b0d0b96f724d724f : SUCCESS in 11m 27s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e07f605be1ce415695c77941c1c543bf : SUCCESS in 15m 45s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"187db92b5e7e1f7c0a6255e335ae6c3feb576f9b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 12:40:24.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2147b96e990542a9b8a2900f6f6778b5\n\n- grenade https://zuul.opendev.org/t/openstack/build/2c80c1e44ab24d028d5a884a51b28474 : FAILURE in 22m 20s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/bf98ca7bc1014f73b3ad192e670dd44e : FAILURE in 24m 40s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/476e2f62c99f463f8fac599466d39f6d : FAILURE in 1h 32m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fedca3d9e8ef41e3be03127c015acaa9 : SUCCESS in 39m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/270fb54662994590b1338b074f6e9cc0 : SUCCESS in 3m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0108f234f36f4e109ff2b94b822f6d44 : SUCCESS in 3m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/03da4a195ad04d4ca9984b257bf0057c : SUCCESS in 3m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/abd6e8986a704928afa4665da57e5d31 : SUCCESS in 5m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7cffc1c44854e2fa9a7ae2c01ca9bca : SUCCESS in 3m 51s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/429e8215ba234fb3958f85d7bbdfabeb : SUCCESS in 8m 48s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/8e3074b38cc042c1872703b17129d466 : SUCCESS in 7m 47s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/f1d2c91d43784ed08380031c27ad440a : SUCCESS in 1h 07m 30s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/464ad7e4ce704a40bef245ef56b658a9 : FAILURE in 1h 05m 39s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/99c6e7fe6ee74cd18836896c8414a06a : FAILURE in 1h 15m 24s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/2c99e9f05f27478cbb37054bdafb8805 : SUCCESS in 2h 17m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bea289612caf4d06ab9dbfcb22f6cdf0 : FAILURE in 38m 47s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9ae8f77c42324445a156f5b544f993ff : FAILURE in 1h 13m 11s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/5b4e3a6d738d4c45a47cc2c62b52d3e9 : FAILURE in 21m 10s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/bd164f1b270443a5973070cce8045a63 : SUCCESS in 1h 55m 47s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/66a5992e49d64426b2ffd5c4add726dc : SUCCESS in 20m 03s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/fca179eb2a1e453ca629300f5eca2b8f : SUCCESS in 1h 14m 29s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"f36e3611e6fbddc0cd0d6aeb30cbbbb30a9cace3","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-06-24 06:30:51.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"a2a546b484f7c0b64ef3cfc0d6c48664e1c813e8","revisions":{"6147aec47f81eb285716c5fe04eb2719eb1b7257":{"kind":"REWORK","_number":1,"created":"2025-06-03 06:12:52.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/11/951611/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/11/951611/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/951611/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/951611/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/951611/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/11/951611/1"}}},"commit":{"parents":[{"commit":"d12086ec6a3cf1a22a2766bd1ba9c5b85ffba8a4","subject":"Merge \"Functional tests migration of TestImageMultipleBackend\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d12086ec6a3cf1a22a2766bd1ba9c5b85ffba8a4"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-06-03 05:49:36.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-06-03 05:49:36.000000000","tz":0},"subject":"DNM Verify x-image-meta-size for all backends","message":"DNM Verify x-image-meta-size for all backends\n\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/948905\nDepends-On: https://review.opendev.org/c/openstack/glance_store/+/949973\nDepends-On: https://review.opendev.org/c/openstack/python-glanceclient/+/949787\nChange-Id: I082c806fb492a770b374b91fabc7937608a40de7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6147aec47f81eb285716c5fe04eb2719eb1b7257"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6147aec47f81eb285716c5fe04eb2719eb1b7257"}]},"branch":"refs/heads/master"},"fc1ada7c94b8676914bbf2a9ff7df28953d9f220":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-06-27 09:59:58.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/11/951611/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/11/951611/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/951611/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/951611/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/951611/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/11/951611/2"}}},"commit":{"parents":[{"commit":"d12086ec6a3cf1a22a2766bd1ba9c5b85ffba8a4","subject":"Merge \"Functional tests migration of TestImageMultipleBackend\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d12086ec6a3cf1a22a2766bd1ba9c5b85ffba8a4"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-06-03 05:49:36.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-06-27 09:59:58.000000000","tz":0},"subject":"DNM Verify x-image-meta-size for all backends","message":"DNM Verify x-image-meta-size for all backends\n\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/948905\nDepends-On: https://review.opendev.org/c/openstack/glance_store/+/949137\nDepends-On: https://review.opendev.org/c/openstack/python-glanceclient/+/949787\nChange-Id: I082c806fb492a770b374b91fabc7937608a40de7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fc1ada7c94b8676914bbf2a9ff7df28953d9f220"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fc1ada7c94b8676914bbf2a9ff7df28953d9f220"}]},"branch":"refs/heads/master","description":"Edit commit message"},"93644b566854690ce61441c49c670db5e6a53b78":{"kind":"REWORK","_number":3,"created":"2025-06-27 10:05:09.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/11/951611/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/11/951611/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/951611/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/951611/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/951611/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/11/951611/3"}}},"commit":{"parents":[{"commit":"2554cad01f4446488c132080830c286a7086dda7","subject":"Merge \"Remove glance-centralized-cache job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2554cad01f4446488c132080830c286a7086dda7"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-06-03 05:49:36.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-06-27 10:05:09.000000000","tz":0},"subject":"DNM Verify x-image-meta-size for all backends","message":"DNM Verify x-image-meta-size for all backends\n\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/948905\nDepends-On: https://review.opendev.org/c/openstack/glance_store/+/949137\nDepends-On: https://review.opendev.org/c/openstack/python-glanceclient/+/949787\nChange-Id: I082c806fb492a770b374b91fabc7937608a40de7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/93644b566854690ce61441c49c670db5e6a53b78"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/93644b566854690ce61441c49c670db5e6a53b78"}]},"branch":"refs/heads/master","description":"Rebase"},"b98de96ffdffc13586bf49a0bd1b69734a38134e":{"kind":"REWORK","_number":4,"created":"2025-06-27 10:06:00.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/11/951611/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/11/951611/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/951611/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/951611/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/951611/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/11/951611/4"}}},"commit":{"parents":[{"commit":"2554cad01f4446488c132080830c286a7086dda7","subject":"Merge \"Remove glance-centralized-cache job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2554cad01f4446488c132080830c286a7086dda7"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-06-03 05:49:36.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-06-27 10:06:00.000000000","tz":0},"subject":"DNM Verify x-image-meta-size for all backends","message":"DNM Verify x-image-meta-size for all backends\n\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/948905\nDepends-On: https://review.opendev.org/c/openstack/glance_store/+/949137\nDepends-On: https://review.opendev.org/c/openstack/python-glanceclient/+/949787\nChange-Id: I082c806fb492a770b374b91fabc7937608a40de7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b98de96ffdffc13586bf49a0bd1b69734a38134e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b98de96ffdffc13586bf49a0bd1b69734a38134e"}]},"branch":"refs/heads/master"},"62220b23ce527a1b1b829ec3bd7575895a0a4555":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2025-07-03 08:47:42.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/11/951611/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/11/951611/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/951611/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/951611/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/951611/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/11/951611/5"}}},"commit":{"parents":[{"commit":"2554cad01f4446488c132080830c286a7086dda7","subject":"Merge \"Remove glance-centralized-cache job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2554cad01f4446488c132080830c286a7086dda7"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-06-03 05:49:36.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-07-03 08:47:42.000000000","tz":0},"subject":"DNM Verify x-image-meta-size for all backends","message":"DNM Verify x-image-meta-size for all backends\n\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/948905\nDepends-On: https://review.opendev.org/c/openstack/glance_store/+/949137\nDepends-On: https://review.opendev.org/c/openstack/python-glanceclient/+/949787\nDepends-On: https://review.opendev.org/c/openstack/glance/+/948903\nChange-Id: I082c806fb492a770b374b91fabc7937608a40de7\nSigned-off-by: Abhishek Kekane \u003cakekane@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/62220b23ce527a1b1b829ec3bd7575895a0a4555"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/62220b23ce527a1b1b829ec3bd7575895a0a4555"}]},"branch":"refs/heads/master","description":"Edit commit message"},"f28dc4c333dd9a9f26c6685bbefebeeac60fa243":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2025-07-03 14:16:36.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/11/951611/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/11/951611/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/951611/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/951611/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/951611/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/11/951611/6"}}},"commit":{"parents":[{"commit":"1ea2795c0b1c8c1ea02edf8f67b85f94f7117db4","subject":"Read image size from request header","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1ea2795c0b1c8c1ea02edf8f67b85f94f7117db4"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-06-03 05:49:36.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-07-03 14:16:36.000000000","tz":0},"subject":"DNM Verify x-image-meta-size for all backends","message":"DNM Verify x-image-meta-size for all backends\n\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/948905\nDepends-On: https://review.opendev.org/c/openstack/glance_store/+/949137\nDepends-On: https://review.opendev.org/c/openstack/python-glanceclient/+/949787\nDepends-On: https://review.opendev.org/c/openstack/glance/+/948903\nChange-Id: I082c806fb492a770b374b91fabc7937608a40de7\nSigned-off-by: Abhishek Kekane \u003cakekane@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f28dc4c333dd9a9f26c6685bbefebeeac60fa243"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f28dc4c333dd9a9f26c6685bbefebeeac60fa243"}]},"branch":"refs/heads/master","description":"Rebase"},"64d840bfd373d925eec9c6f41ff2734e3ea8a378":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2025-07-06 14:40:55.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/11/951611/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/11/951611/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/951611/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/951611/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/951611/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/11/951611/7"}}},"commit":{"parents":[{"commit":"1ea2795c0b1c8c1ea02edf8f67b85f94f7117db4","subject":"Read image size from request header","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1ea2795c0b1c8c1ea02edf8f67b85f94f7117db4"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-06-03 05:49:36.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-07-06 14:40:55.000000000","tz":0},"subject":"DNM Verify x-image-meta-size for all backends","message":"DNM Verify x-image-meta-size for all backends\n\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/948905\nDepends-On: https://review.opendev.org/c/openstack/python-glanceclient/+/949787\nDepends-On: https://review.opendev.org/c/openstack/glance/+/948903\nChange-Id: I082c806fb492a770b374b91fabc7937608a40de7\nSigned-off-by: Abhishek Kekane \u003cakekane@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/64d840bfd373d925eec9c6f41ff2734e3ea8a378"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/64d840bfd373d925eec9c6f41ff2734e3ea8a378"}]},"branch":"refs/heads/master","description":"Edit commit message"},"a2a546b484f7c0b64ef3cfc0d6c48664e1c813e8":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2025-07-06 14:57:39.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/11/951611/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/11/951611/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/951611/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/951611/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/951611/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/11/951611/8"}}},"commit":{"parents":[{"commit":"0996290c02d98e42b9b31109e7caf7ba7fe175ee","subject":"Read image size from request header","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0996290c02d98e42b9b31109e7caf7ba7fe175ee"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-06-03 05:49:36.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-07-06 14:57:39.000000000","tz":0},"subject":"DNM Verify x-image-meta-size for all backends","message":"DNM Verify x-image-meta-size for all backends\n\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/948905\nDepends-On: https://review.opendev.org/c/openstack/python-glanceclient/+/949787\nDepends-On: https://review.opendev.org/c/openstack/glance/+/948903\nChange-Id: I082c806fb492a770b374b91fabc7937608a40de7\nSigned-off-by: Abhishek Kekane \u003cakekane@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a2a546b484f7c0b64ef3cfc0d6c48664e1c813e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a2a546b484f7c0b64ef3cfc0d6c48664e1c813e8"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
