)]}'
{"id":"openstack%2Fglance~951635","triplet_id":"openstack%2Fglance~master~I3f8197ff90d36f933a5b0580be533176c3357e3e","project":"openstack/glance","branch":"master","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2025-08-04 16:07:01.000000000","reason":"Change was submitted"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-08-04 16:07:01.000000000","reason":"Change was submitted"},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2025-08-04 16:07:01.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3f8197ff90d36f933a5b0580be533176c3357e3e","subject":"Get rid of thin wrapper of LimitFixture","status":"MERGED","created":"2025-06-03 09:11:41.000000000","updated":"2025-08-04 16:08:20.000000000","submitted":"2025-08-04 16:07:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"951635","meta_rev_id":"641cc5077b8820ba5c7f3fa87fc7b13eadc2cdc7","_number":951635,"virtual_id_number":951635,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-04 16:07:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"all":[{"value":2,"date":"2025-08-04 14:21:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"all":[{"value":1,"date":"2025-08-04 14:21:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}],"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-03 09:30:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-06-03 10:12:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-03 12:14:23.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"CC"},{"updated":"2025-06-10 14:12:07.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"CC"},{"updated":"2025-06-10 14:12:07.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2025-08-04 14:21:43.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"}],"messages":[{"id":"6590b17a0c22c7d97ee322ad6e70343cf01e6e9d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-06-03 09:11:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ea07f7f81ea9d073063008fa4891a909294234a7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-03 09:30:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/31e774b1a3684707856a9ac45689be38\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d08d0839ed7b4f53a5985c74dadd4147 : SUCCESS in 17m 13s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/de457799277948e5aa6ab8b9ba3b5aa4 : SUCCESS in 13m 47s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/9fba6dadb16b415eaebedecb1d375d87 : SUCCESS in 16m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4977b740c1598b8b01bbc1dd1d4dddb0e032a30b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-03 10:12:09.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d647f03f943f4a27bb9cba8b8aca2d9f\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e65961b14ebf4be586e0097b2678803c : SUCCESS in 59m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c331e324ab17421f882ed6cb6578e590 : SUCCESS in 54m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b3709036b5d4a069a393cab8fa6b758 : FAILURE in 4m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f7262f68476945458a2f830bf7893741 : SUCCESS in 4m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/73fe479c981a408db866236231a7b057 : SUCCESS in 6m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1d36235cf09e4136acc64363f02d2fa8 : SUCCESS in 8m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/259b5046ee86400f9c33fa9575b5139f : SUCCESS in 6m 48s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7865781e327f4c5ea6dd59fb74db1366 : SUCCESS in 6m 57s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/ed911be8aa92495b8b0ae417df7fbbbc : SUCCESS in 8m 42s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/01a3aeb697794649854f33018f88b4de : SUCCESS in 32m 13s","accounts_in_message":[],"_revision_number":1},{"id":"78958b38490bd04c5faad58e0611450dfec4de86","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-06-03 12:14:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f9c71bfa861b74cdd589883e289a40d0263d30be","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-06-03 14:09:15.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"a39f2c4af7217b6ae332efea9cf01a8f5d34d8cc","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-06-03 14:15:59.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ba0ad5e924439b429da703a37fe084f8ed3c82a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-03 15:20:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/52e7f34a96c044d4a9cfd445ecddb402\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d6ba21e32af04052964e841c52be4430 : SUCCESS in 1h 10m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/344b14ad4a8646c0b2e6a22dc5719547 : SUCCESS in 57m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/287e61e5de6d4b249e762eaf6c48a93f : SUCCESS in 4m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e3db72e83cba4dcbb6c7c233ea80e0e5 : SUCCESS in 4m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8213e3f9a2a24a8588cd2f5368bd85b5 : SUCCESS in 6m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fcd9aa3146844584949ba88c52fd4d7b : SUCCESS in 9m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efd7462461594d7e9ff9b197ba35c482 : SUCCESS in 5m 52s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2e953c05021c405f9ae3ea53fb0e9fe6 : SUCCESS in 6m 37s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/6ad719b1dbd647fabc8560452c195ed2 : SUCCESS in 10m 03s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/51162516329e4ee2bfd010180d589bf4 : SUCCESS in 27m 46s","accounts_in_message":[],"_revision_number":2},{"id":"9b54eaebdd127dc14acf138d87859de60cba0094","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-03 21:03:34.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a1562353a9914658982bcf2ad2aa7cd3\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/109ad22f48c143fba73ef3629bee4c38 : SUCCESS in 25m 48s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a62cec31f1e74e938929eeab8d01cf29 : SUCCESS in 13m 06s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/88a893e7ce2445c6b1c133918ddf4b5e : SUCCESS in 46m 56s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"455be4864250282f8315f4b7edc57414f26c245e","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-06-10 14:12:07.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"372d89bdd176b4d01250d5a393e4c56ec193d0a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-02 18:03:05.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"d50bd62da86105bf4ae9c2c3668ba71d5850d6b8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-02 18:14:45.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9dd042276dda4619991dcea2c45e8063\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f6c72be67597482499c110264ece270f : SUCCESS in 6m 23s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/200d7165177b460989394fbf2e75bff3 : SUCCESS in 7m 08s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b24b8b2db3d44556a8dee93c08785574 : SUCCESS in 9m 39s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"037debe9af0a3467a77a6585c895dba7531081d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-02 18:36:24.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a2567124567a46c08656725aa45ba2ca\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6644b1e044b340308630f7279f160f24 : FAILURE in 24m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/087c56efe42d43348a614932d491bccc : SUCCESS in 32m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3385fc0a0cb41c7a7b0dfb2ae94ef17 : SUCCESS in 4m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a28c81f294ab4ff2a20cce37d85c6583 : SUCCESS in 5m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/75767fd25dd140799c1752885e36737a : SUCCESS in 5m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8ce5ebcb0eef4472a15cfd965b6fb5eb : SUCCESS in 9m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7984b50d047642178278baeb89c5225b : SUCCESS in 2m 56s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/62c28ce9b2c340c4a497bdb52c7bbb23 : SUCCESS in 7m 13s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/057f5e1c0df84d55ae3a7a631719e8a2 : SUCCESS in 10m 20s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/41e00727862f4bfda33f44688800cf09 : SUCCESS in 14m 30s","accounts_in_message":[],"_revision_number":3},{"id":"1f65c7707a90432de331b25b28c54376e1fea0c5","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-04 05:50:49.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e8f0c055c91d2ec765ca4d4bf480266c953efd85","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-04 06:05:12.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/90e522756a0f47a28639d37547440b7c\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e8a278bbf8764269bc8c573a6094c9bb : SUCCESS in 9m 31s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/cd8180e3d0554d1494d1d1c655ecad6c : SUCCESS in 7m 47s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/0f151dbef003436982c229c61433a97d : SUCCESS in 12m 55s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"13e43500f703c4f8b5fe6f7013928ab2a93ff24c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-04 06:41:56.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/59b1d3dd458444c48fcfa493f689cab5\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ce17bf61d09d44878c22e50cc419a8c8 : SUCCESS in 32m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5ccc8f8144904ec8b7213f045aae8102 : SUCCESS in 49m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e5ba7326d6c4d6b88def617a37f247e : SUCCESS in 4m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/98e02e3fbca847209f097b61c45c3af9 : SUCCESS in 5m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4ace0e8acc95467f9d54bf32b5eb64d0 : SUCCESS in 7m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9b04624d1689485f920e12d487bc694b : SUCCESS in 7m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0ccbe5d3f1e4082ba47488e34edfef1 : SUCCESS in 3m 52s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8e8a3005905b4abe93e71f283154104d : SUCCESS in 6m 43s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/5c070dd0b7c240a6a5ac0da19daf8e8d : SUCCESS in 8m 35s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/d606ccf0cb044977b7b547569d337d2c : SUCCESS in 32m 11s","accounts_in_message":[],"_revision_number":3},{"id":"76aedb803daaf4207d2dea898c3e8c8a8d61932c","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2025-08-04 14:21:43.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"5faf4b1a26929276b76362b299b8be4fc070ef22","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-04 14:22:08.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"9878f3542666e881a99cbddfc58dc03926c40267","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-04 16:07:01.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/64932111a09a433cb02a2f99a898597e\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/20271d9cb435450696619fa8d42efc2b : SUCCESS in 1h 00m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9bcd7de4d0dd4afe95390c6c43c0f516 : SUCCESS in 55m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4c213fb03584109a6482e094eaa974d : SUCCESS in 5m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7eb7edeceec54d3680c12ec746c4d020 : SUCCESS in 5m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b513443498064b0fab9aca54b0643fa8 : SUCCESS in 6m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/025ff950596243adb5fa2ecc0e8e4172 : SUCCESS in 7m 41s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a09f063cf75b4f35965c4a5a5c6319e6 : SUCCESS in 7m 48s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/c62469237527443f93a92e1531599c47 : SUCCESS in 9m 33s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/88c83c735ec2428f9ef3d38490365809 : SUCCESS in 24m 52s","accounts_in_message":[],"_revision_number":3},{"id":"23ef4e183cb73634cf1cf3bd4ea43e54c3e1b9a8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-04 16:07:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"641cc5077b8820ba5c7f3fa87fc7b13eadc2cdc7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-04 16:08:20.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/442102b7771b4c3fa3e37de52637fb1e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5dca6c9e14147e584928f21badd12fc : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"27bfd6f2d27b3e9e6af0878b2ea428f76ca2a9d4","revisions":{"7298e16b1c45f0e874ecd3106efcf537ab43d4e3":{"kind":"REWORK","_number":1,"created":"2025-06-03 09:11:41.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/35/951635/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/35/951635/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/35/951635/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/35/951635/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/35/951635/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/35/951635/1"}}},"commit":{"parents":[{"commit":"e055873be4079bc9d37165cf75de5495603133b0","subject":"Get ready for oslo.limit 2.7.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e055873be4079bc9d37165cf75de5495603133b0"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-06-03 09:10:40.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-06-03 09:10:40.000000000","tz":540},"subject":"Get rid of thin wrapper of LimitFixture","message":"Get rid of thin wrapper of LimitFixture\n\n... to leverage the common implementation and reduce amount of logic\nmaintained specifically in this repository.\n\nChange-Id: I3f8197ff90d36f933a5b0580be533176c3357e3e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7298e16b1c45f0e874ecd3106efcf537ab43d4e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7298e16b1c45f0e874ecd3106efcf537ab43d4e3"}]},"branch":"refs/heads/master"},"2d8337d1cef159a523a4e105431b407f818de0e6":{"kind":"REWORK","_number":2,"created":"2025-06-03 14:09:15.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/35/951635/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/35/951635/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/35/951635/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/35/951635/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/35/951635/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/35/951635/2"}}},"commit":{"parents":[{"commit":"e055873be4079bc9d37165cf75de5495603133b0","subject":"Get ready for oslo.limit 2.7.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e055873be4079bc9d37165cf75de5495603133b0"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-06-03 09:10:40.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-06-03 14:09:15.000000000","tz":0},"subject":"Get rid of thin wrapper of LimitFixture","message":"Get rid of thin wrapper of LimitFixture\n\n... to leverage the common implementation and reduce amount of logic\nmaintained specifically in this repository.\n\nChange-Id: I3f8197ff90d36f933a5b0580be533176c3357e3e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2d8337d1cef159a523a4e105431b407f818de0e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2d8337d1cef159a523a4e105431b407f818de0e6"}]},"branch":"refs/heads/master"},"27bfd6f2d27b3e9e6af0878b2ea428f76ca2a9d4":{"kind":"REWORK","_number":3,"created":"2025-08-02 18:03:05.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/35/951635/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/35/951635/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/35/951635/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/35/951635/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/35/951635/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/35/951635/3"}}},"commit":{"parents":[{"commit":"17ad1aeaf86aea434952809b3ee727c4e6d487c2","subject":"Merge \"Replace oslo_utils.encodeutils.exception_to_unicode\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/17ad1aeaf86aea434952809b3ee727c4e6d487c2"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-06-03 09:10:40.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-02 18:02:55.000000000","tz":540},"subject":"Get rid of thin wrapper of LimitFixture","message":"Get rid of thin wrapper of LimitFixture\n\n... to leverage the common implementation and reduce amount of logic\nmaintained specifically in this repository.\n\nChange-Id: I3f8197ff90d36f933a5b0580be533176c3357e3e\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/27bfd6f2d27b3e9e6af0878b2ea428f76ca2a9d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/27bfd6f2d27b3e9e6af0878b2ea428f76ca2a9d4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
