)]}'
{"id":"openstack%2Fglance~952361","triplet_id":"openstack%2Fglance~master~Ifdcb4567cb0e3f6d0e004fd52930c1f888a3f0fc","project":"openstack/glance","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-07-10 15:27:15.000000000","reason":"Change was submitted"},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2025-07-10 15:27:15.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ifdcb4567cb0e3f6d0e004fd52930c1f888a3f0fc","subject":"Drop redundant check of sqlite3 module availability","status":"MERGED","created":"2025-06-11 12:42:31.000000000","updated":"2025-07-10 15:28:19.000000000","submitted":"2025-07-10 15:27:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"952361","meta_rev_id":"5d040419d3c7f0b2239f91bab70cb1def32ca0c7","_number":952361,"virtual_id_number":952361,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-10 15:27:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2025-07-10 13:58:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2025-07-10 01:59:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2025-07-10 13:58:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-11 13:37:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-06-11 13:52:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-10 01:57:39.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2025-07-10 13:58:44.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"d547ff0aa322a3837e2f0e0fbcd80b91942910c5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-06-11 12:42:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2c4ca1f71549afda8d564566a3039be69e1c75ad","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-11 13:37:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6fb2f11dd6834ded8f9a9bc88873e477\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/db6f9790ca1848828480912d7639b1ba : SUCCESS in 11m 10s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a9ac8cb759e642889e3fbc6b0b6ef54c : SUCCESS in 13m 05s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d9b84841e4d443b98982fa07794bed68 : SUCCESS in 10m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d65be07cb42dd7f6a6b92d88b41b89af7afbf7a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-11 13:52:02.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a1407bdf1e484cf69a3fb3b2793c0b37\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7c3a466ac50e4a078e1619c86293ea31 : SUCCESS in 50m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/55d9e0d88f7b45329a4867baa708487b : SUCCESS in 31m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/143def4f51c347a9b4f03fef43f1961a : FAILURE in 4m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b962e5b6adbd4db7bc8e090a839185d2 : SUCCESS in 4m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d9e6571935f147038de5d8774400c0ae : SUCCESS in 2m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bc039f85acf54a879fc904ed4a3bc4c2 : SUCCESS in 5m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42fb7dcd5aa44462aa924c6ecc336172 : SUCCESS in 6m 57s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/24ea9f8bbe464daab4ce2365ab309f3d : SUCCESS in 6m 47s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/1108908fa71c44a6b49540c8e8a1978f : SUCCESS in 11m 23s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/d4b98c9ee6e9475b8db4626f127164f2 : SUCCESS in 34m 56s","accounts_in_message":[],"_revision_number":1},{"id":"816cf635113251c5064d32ce1fc73276130c5537","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-06-11 13:52:28.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"acaf60a54d5b9ac1e0886af776907d083f88073a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-11 14:52:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/513f7b1ae95d4b11a78f7e8aecf211f1\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/35adf56b4422479aa5c20744badd213c : SUCCESS in 54m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/193a500b160645b7b23b7ccaae7d633b : SUCCESS in 57m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6621f9050f634daebbf2da16f1a9411d : SUCCESS in 5m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ab701c642ca64a9abc67d86c68d78cb8 : SUCCESS in 4m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c8fbe9c7288b4c4c9f807297649a36b2 : SUCCESS in 5m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5c0169922f9b4d01a4d9c623314401fb : SUCCESS in 8m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9fa6713cce874014b055ee78af51b9f1 : SUCCESS in 5m 31s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3c58fb48f7c445a680f43bc5bb887ad0 : SUCCESS in 5m 43s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/cd8feaab648a429f8abbe11cfe2a8e16 : SUCCESS in 9m 52s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/12f7a37569564c959d7e3142d0d2790f : SUCCESS in 18m 10s","accounts_in_message":[],"_revision_number":2},{"id":"5654f7d97b22e48c10ba3350c5fed2ce4b15fe97","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-11 15:16:58.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2633c92501c44e33b87a484aa654a7c0\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d825e114a51f4f2999624f297d85e918 : SUCCESS in 8m 08s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6f8346e830494a0a9619ca9898f685c9 : SUCCESS in 13m 40s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e7fa8e01fc2242c5a5f04e3b16d5278f : SUCCESS in 8m 29s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6549edfb75fb011e493b6213a5ffca0247291270","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-07-10 01:57:39.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"1faa3f704a8e10f09b23b6604b9d8506fdacc599","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-07-10 01:59:13.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"77fc451822b809a633789086623dda23b4b558e1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 02:16:36.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a0d70be7533483f98231a829f34a925\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/36e003906c6b44d98f7baf4475e54793 : SUCCESS in 11m 20s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6d651f10058d467eba895df7f208cfda : SUCCESS in 9m 29s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e0c284851d21452e9be8f208b5f96299 : SUCCESS in 16m 57s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"9d3a12d34c76b9b201a7fa2a049b6d9cebb0168e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 03:08:09.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2929c1facd404e8bbad22be8f26a2c05\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ab29cabd4d9047faa2b9e9eb6a0c9916 : SUCCESS in 34m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/19028306c43943c89e4457ece30edfec : SUCCESS in 1h 04m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf9ee492d4b2413c8c467d1d8c566038 : SUCCESS in 8m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f7e03ebd702e44119e5c1ae0ef7cb97a : SUCCESS in 4m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/75ecc939a2124e07b5272e2bed432874 : SUCCESS in 2m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/15837b700792481ea08e21660c04b302 : SUCCESS in 11m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91dfc55844b245c9980048ba701c31ed : SUCCESS in 4m 59s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1d7afc532132400d91481a1e01d065d0 : SUCCESS in 6m 56s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/c9da449167684a2aac70626b9c33d886 : SUCCESS in 10m 30s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/2bbf61b8375746f7817071e3128cc11a : SUCCESS in 33m 38s","accounts_in_message":[],"_revision_number":3},{"id":"6db8611aa85af1d8f30ee1b2a57c0f042bd38c4b","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-07-10 13:58:44.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"ff3bc8fc126aa57ce01443f70fec19abe7782e5e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 13:59:19.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"1a7c175e2b3b343c47f8a7d6eb695228f1ffcd34","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 15:27:15.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/112c7b04cd064696bd497bebbf4c37b8\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2dac232101254761a59414aedf096530 : SUCCESS in 1h 18m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1628093a8a704e2fa3bb0018999ec573 : SUCCESS in 58m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b2edce03bba4e17b598a5c596b01ae2 : SUCCESS in 6m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6cdf5090867b4f289df749f367808f37 : SUCCESS in 5m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9dcff0ae2246469f968f14f11f8382fd : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea7f1323f61f486e886f348dadadb6ee : SUCCESS in 5m 35s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/cc57735cbd204390a172c1d35cd91837 : SUCCESS in 7m 06s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/558c4ef73b8e4c20a0980271d0a4bbb5 : SUCCESS in 11m 14s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/cee7b5dd478a4148aaf743ef930be0b7 : SUCCESS in 33m 22s","accounts_in_message":[],"_revision_number":3},{"id":"bd9df8ead5bc0fb6c7b0ff820efe1c64e6109b35","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 15:27:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"5d040419d3c7f0b2239f91bab70cb1def32ca0c7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 15:28:19.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/faf1d8888d6147668af7637fcdbfbd6e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd9f69bd152a4138bf9b7fa8c4f0e7c0 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"63a9bd3dbc421aa8f0bd2505bd69e9eec5be26b6","revisions":{"c7f12345d9aa1937be4e79ea89be4ed2adb86823":{"kind":"REWORK","_number":1,"created":"2025-06-11 12:42:31.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/61/952361/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/61/952361/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/61/952361/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/61/952361/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/61/952361/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/61/952361/1"}}},"commit":{"parents":[{"commit":"dac9741c2ecb62a79d9eaf3ed5bffa60f4f8bf69","subject":"Remove Python 3.9 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/dac9741c2ecb62a79d9eaf3ed5bffa60f4f8bf69"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-06-11 12:39:23.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-06-11 12:42:21.000000000","tz":540},"subject":"Drop redundant check of sqlite3 module availability","message":"Drop redundant check of sqlite3 module availability\n\nThe sqlite3 module was merged into the core python in Python 2.5, and\nwe can safely assume it is always available.\n\nChange-Id: Ifdcb4567cb0e3f6d0e004fd52930c1f888a3f0fc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c7f12345d9aa1937be4e79ea89be4ed2adb86823"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c7f12345d9aa1937be4e79ea89be4ed2adb86823"}]},"branch":"refs/heads/master"},"1a3467271feda63ec321c57f35241cc09754de4a":{"kind":"REWORK","_number":2,"created":"2025-06-11 13:52:28.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/61/952361/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/61/952361/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/61/952361/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/61/952361/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/61/952361/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/61/952361/2"}}},"commit":{"parents":[{"commit":"dac9741c2ecb62a79d9eaf3ed5bffa60f4f8bf69","subject":"Remove Python 3.9 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/dac9741c2ecb62a79d9eaf3ed5bffa60f4f8bf69"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-06-11 12:39:23.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-06-11 13:52:28.000000000","tz":0},"subject":"Drop redundant check of sqlite3 module availability","message":"Drop redundant check of sqlite3 module availability\n\nThe sqlite3 module was merged into the core python in Python 2.5, and\nwe can safely assume it is always available.\n\nChange-Id: Ifdcb4567cb0e3f6d0e004fd52930c1f888a3f0fc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1a3467271feda63ec321c57f35241cc09754de4a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1a3467271feda63ec321c57f35241cc09754de4a"}]},"branch":"refs/heads/master"},"63a9bd3dbc421aa8f0bd2505bd69e9eec5be26b6":{"kind":"REWORK","_number":3,"created":"2025-07-10 01:57:39.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/61/952361/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/61/952361/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/61/952361/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/61/952361/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/61/952361/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/61/952361/3"}}},"commit":{"parents":[{"commit":"dac9741c2ecb62a79d9eaf3ed5bffa60f4f8bf69","subject":"Remove Python 3.9 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/dac9741c2ecb62a79d9eaf3ed5bffa60f4f8bf69"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-06-11 12:39:23.000000000","tz":540},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2025-07-10 01:57:30.000000000","tz":120},"subject":"Drop redundant check of sqlite3 module availability","message":"Drop redundant check of sqlite3 module availability\n\nThe sqlite3 module was merged into the core python in Python 2.5, and\nwe can safely assume it is always available.\n\nChange-Id: Ifdcb4567cb0e3f6d0e004fd52930c1f888a3f0fc\nSigned-off-by: Cyril Roelandt \u003ccyril@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/63a9bd3dbc421aa8f0bd2505bd69e9eec5be26b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/63a9bd3dbc421aa8f0bd2505bd69e9eec5be26b6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
