)]}'
{"id":"openstack%2Fglance~952468","triplet_id":"openstack%2Fglance~master~I1878543382b7ca4deac125a97648ea7067d76e18","project":"openstack/glance","branch":"master","hashtags":[],"change_id":"I1878543382b7ca4deac125a97648ea7067d76e18","subject":"WIP: Run a job with cinder-nfs backend","status":"NEW","created":"2025-06-12 15:03:29.000000000","updated":"2025-06-12 17:11:42.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8fe07eb01a973017ad8438adbf066ca8ab22d9e4","_number":952468,"virtual_id_number":952468,"owner":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-06-12 17:11:42.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-12 15:33:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-06-12 17:11:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b6bb06352c733342dddad68be7a111ab18e68b9c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2025-06-12 15:03:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bd841ab6adb1f4e8f9db74bea3cf11a82d76cec2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-12 15:33:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/391c699ada93431abbd09a8a1642313b\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/434a26f5e96d40c8b9fefdea508245a0 : SUCCESS in 5m 46s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/00fea195021349318ce19e4e69f99205 : SUCCESS in 6m 37s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/7380f928c9eb44b69ee9237d57585019 : SUCCESS in 7m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8fe07eb01a973017ad8438adbf066ca8ab22d9e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-12 17:11:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b6f2eeb59e564897acdc7294bb453970\n\n- grenade https://zuul.opendev.org/t/openstack/build/af0143dddc7a4b24a77a7e79105fc026 : SUCCESS in 57m 49s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f3a56b1e6f174527b9d2b6a782be9773 : SUCCESS in 51m 17s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1a91b35276b94a0eaacbfe6175905a20 : SUCCESS in 1h 16m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ea13341a665f4bd8babff38073b083dc : SUCCESS in 51m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7ebf785298c4058914201ffee131843 : SUCCESS in 4m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f930914980eb44aaba67afa414133c23 : SUCCESS in 5m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ac0c5e8243904b958b9d2eafb64ec89c : SUCCESS in 5m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d7796d7552644a93bb73c31f287e0852 : SUCCESS in 7m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9a017ba89664a9594f814de95fc84b4 : SUCCESS in 5m 47s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/33ceca065b9c46c9b0173618b7b6474a : SUCCESS in 6m 56s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/a5d56ef9a1334f24ad871abd173e62e8 : SUCCESS in 9m 07s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/500cba89188e45a6836dc9a3373d4d33 : SUCCESS in 1h 49m 59s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/fa72960c62534a17a1d33fd90e3adec3 : SUCCESS in 1h 49m 59s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/97d0197132c849a1aa7a2d1044a464f3 : SUCCESS in 1h 51m 02s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/26c5ac7de6714619a95a1194815bf1e5 : SUCCESS in 2h 04m 45s\n- glance-multistore-cinder-nfs-import https://zuul.opendev.org/t/openstack/build/7d3c876f06c74d10818a2e384a61e01a : FAILURE in 29m 59s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/42c1ac9ecaf24619b59546118435a03d : SUCCESS in 51m 11s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/83ff8b5dc16b497685f3a21729342fb6 : SUCCESS in 1h 10m 12s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/054f99ce72f0456290314cc62a48746f : SUCCESS in 1h 51m 48s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/fc0aef8d5290407ba8c86d95b3e86918 : FAILURE in 32m 45s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/3f82e69e0ab84bf7848272baf906946f : SUCCESS in 59m 09s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/65074f9e970747de84404d4496dd7b69 : SUCCESS in 14m 37s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/fbf691c2b9cb4162a653f46c847f710b : SUCCESS in 1h 02m 38s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"dbf0845c4584bb64836503831b7f5dad8976a549","revisions":{"dbf0845c4584bb64836503831b7f5dad8976a549":{"kind":"REWORK","_number":1,"created":"2025-06-12 15:03:29.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/68/952468/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/68/952468/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/952468/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/952468/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/952468/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/68/952468/1"}}},"commit":{"parents":[{"commit":"dac9741c2ecb62a79d9eaf3ed5bffa60f4f8bf69","subject":"Remove Python 3.9 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/dac9741c2ecb62a79d9eaf3ed5bffa60f4f8bf69"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2025-06-12 15:02:30.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2025-06-12 15:03:12.000000000","tz":-240},"subject":"WIP: Run a job with cinder-nfs backend","message":"WIP: Run a job with cinder-nfs backend\n\nChange-Id: I1878543382b7ca4deac125a97648ea7067d76e18\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/dbf0845c4584bb64836503831b7f5dad8976a549"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/dbf0845c4584bb64836503831b7f5dad8976a549"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
