)]}'
{"id":"openstack%2Fglance~952738","triplet_id":"openstack%2Fglance~stable%2F2024.1~I62d336565eaf6a50e6bdc7aa87946023fdcd7c54","project":"openstack/glance","branch":"stable/2024.1","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2025-07-15 15:38:15.000000000","reason":"Change was abandoned"},"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2025-07-15 15:38:15.000000000","reason":"Change was abandoned"},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2025-07-15 15:38:15.000000000","reason":"Change was abandoned"},"38109":{"account":{"_account_id":38109,"name":"Darshan Malagimani","display_name":"Darshan S M","email":"darshan.malagimani@outlook.com","username":"darshanmalagimani"},"last_update":"2025-06-18 16:10:14.000000000","reason":"\u003cGERRIT_ACCOUNT_38109\u003e replied on the change","reason_account":{"_account_id":38109,"name":"Darshan Malagimani","display_name":"Darshan S M","email":"darshan.malagimani@outlook.com","username":"darshanmalagimani"}}},"hashtags":[],"change_id":"I62d336565eaf6a50e6bdc7aa87946023fdcd7c54","subject":"docs: Fix incorrect oslo_limit endpoint_id in Glance install guide","status":"ABANDONED","created":"2025-06-17 07:56:18.000000000","updated":"2025-07-15 15:38:15.000000000","total_comment_count":11,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"55aa6f8d94ca494626be0c632030cf939db9a9d5","_number":952738,"virtual_id_number":952738,"owner":{"_account_id":38109,"name":"Darshan Malagimani","display_name":"Darshan S M","email":"darshan.malagimani@outlook.com","username":"darshanmalagimani"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-06-17 08:27:30.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":-2,"date":"2025-06-18 15:48:19.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-17 08:04:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-18 04:12:07.000000000","updated_by":{"_account_id":38109,"name":"Darshan Malagimani","display_name":"Darshan S M","email":"darshan.malagimani@outlook.com","username":"darshanmalagimani"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2025-06-18 04:12:07.000000000","updated_by":{"_account_id":38109,"name":"Darshan Malagimani","display_name":"Darshan S M","email":"darshan.malagimani@outlook.com","username":"darshanmalagimani"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2025-06-18 04:12:07.000000000","updated_by":{"_account_id":38109,"name":"Darshan Malagimani","display_name":"Darshan S M","email":"darshan.malagimani@outlook.com","username":"darshanmalagimani"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"0f55ce724fb4f75fc6519c9b013f8d9e66877aed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38109,"name":"Darshan Malagimani","display_name":"Darshan S M","email":"darshan.malagimani@outlook.com","username":"darshanmalagimani"},"date":"2025-06-17 07:56:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3e428866a5336a364df005028d3908468afb0761","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38109,"name":"Darshan Malagimani","display_name":"Darshan S M","email":"darshan.malagimani@outlook.com","username":"darshanmalagimani"},"date":"2025-06-17 07:58:00.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"128714c3d22fed52eeb36f661ae6c749263451d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-17 08:04:27.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f917f0ad9a4a4d36b1de8e27311d3835\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f18ff4efcbb4b50a364849492b4dc68 : FAILURE in 5m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e95e8f7cf85249c6b45b16562c5ec2d2 : SUCCESS in 2m 41s","accounts_in_message":[],"_revision_number":2},{"id":"22bdd542712cecb51ecac15084da1171926fcc1b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38109,"name":"Darshan Malagimani","display_name":"Darshan S M","email":"darshan.malagimani@outlook.com","username":"darshanmalagimani"},"date":"2025-06-17 08:10:58.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"dc4838c60d3e3cee9a9093caff97946344f4afff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-17 08:15:36.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d271a70529fe4799b99bae48b4ed58ad\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de2c8c5de9f5471b93e546fd4dc240f9 : FAILURE in 2m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f374c55b3b140bea5b554b7802b75b1 : FAILURE in 4m 20s","accounts_in_message":[],"_revision_number":3},{"id":"cfa964af9440040cbaa3736a9f630698754e8f3e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38109,"name":"Darshan Malagimani","display_name":"Darshan S M","email":"darshan.malagimani@outlook.com","username":"darshanmalagimani"},"date":"2025-06-17 08:22:40.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"1b9590f48fa00f16d7bfc4230f8b56e943eee257","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-17 08:27:30.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/03d292a9bf95424896ac55785d1ad9f4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2bd2a1012a5b4245affe2ca0b84f5a00 : SUCCESS in 3m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca5c3fdf2a19498c9cea53eff9f8b443 : SUCCESS in 4m 15s","accounts_in_message":[],"_revision_number":4},{"id":"ba2ef9f109c518d78a4f29925e285f7357e27a27","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-06-18 15:48:19.000000000","message":"Patch Set 4: Code-Review-2\n\n(5 comments)","accounts_in_message":[],"_revision_number":4},{"id":"33af9c1b050d8d99bfb659c5f29254e981680d8b","author":{"_account_id":38109,"name":"Darshan Malagimani","display_name":"Darshan S M","email":"darshan.malagimani@outlook.com","username":"darshanmalagimani"},"date":"2025-06-18 16:10:14.000000000","message":"Patch Set 4:\n\n(6 comments)","accounts_in_message":[],"_revision_number":4},{"id":"55aa6f8d94ca494626be0c632030cf939db9a9d5","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-07-15 15:38:15.000000000","message":"Abandoned\n\nMerged in https://review.opendev.org/c/openstack/glance/+/952847","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"8c675f64d875d9c18a397605f58f9797454bcae8","revisions":{"c92bf7a5d7e584f3f4b49075d5d8cf0bf30ff9dc":{"kind":"REWORK","_number":1,"created":"2025-06-17 07:56:18.000000000","uploader":{"_account_id":38109,"name":"Darshan Malagimani","display_name":"Darshan S M","email":"darshan.malagimani@outlook.com","username":"darshanmalagimani"},"ref":"refs/changes/38/952738/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/38/952738/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/952738/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/952738/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/952738/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/38/952738/1"}}},"commit":{"parents":[{"commit":"a514df4570b5b2a127037416aebdefb7aca7932f","subject":"Harden copy bust lock test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a514df4570b5b2a127037416aebdefb7aca7932f"}]}],"author":{"name":"Darshan Malagimani","email":"darshan.malagimani@outlook.com","date":"2025-06-17 07:51:17.000000000","tz":330},"committer":{"name":"Darshan Malagimani","email":"darshan.malagimani@outlook.com","date":"2025-06-17 07:51:17.000000000","tz":330},"subject":"docs: Fix incorrect oslo_limit endpoint_id in Glance install guide","message":"docs: Fix incorrect oslo_limit endpoint_id in Glance install guide\n\nThe Caracal 28.1.1 Glance documentation incorrectly hardcodes an\nendpoint_id in the  section of glance-api.conf, leading\nto HTTP 500 Internal Server Error when uploading an image.\n\nThis patch clarifies that users must replace  with the\nactual ID of their image endpoint using:\n\n    openstack endpoint list --service glance --region RegionOne\n\nThis change aligns the 2024.1 docs with the correct instructions\nalready provided in the Epoxy 2025.1 release.\n\nChange-Id: I62d336565eaf6a50e6bdc7aa87946023fdcd7c54\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c92bf7a5d7e584f3f4b49075d5d8cf0bf30ff9dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c92bf7a5d7e584f3f4b49075d5d8cf0bf30ff9dc"}]},"branch":"refs/heads/stable/2024.1"},"9d09ec3a83e5022fc7037d8aab8c581396c71e45":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-06-17 07:58:00.000000000","uploader":{"_account_id":38109,"name":"Darshan Malagimani","display_name":"Darshan S M","email":"darshan.malagimani@outlook.com","username":"darshanmalagimani"},"ref":"refs/changes/38/952738/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/38/952738/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/952738/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/952738/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/952738/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/38/952738/2"}}},"commit":{"parents":[{"commit":"a514df4570b5b2a127037416aebdefb7aca7932f","subject":"Harden copy bust lock test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a514df4570b5b2a127037416aebdefb7aca7932f"}]}],"author":{"name":"Darshan Malagimani","email":"darshan.malagimani@outlook.com","date":"2025-06-17 07:51:17.000000000","tz":330},"committer":{"name":"Darshan S M","email":"darshan.malagimani@outlook.com","date":"2025-06-17 07:58:00.000000000","tz":0},"subject":"docs: Fix incorrect [oslo_limit] endpoint_id in Glance install guide","message":"docs: Fix incorrect [oslo_limit] endpoint_id in Glance install guide\n\nThe Caracal 28.1.1 Glance documentation incorrectly hardcodes an\nendpoint_id in the  section of glance-api.conf, leading\nto HTTP 500 Internal Server Error when uploading an image.\n\nThis patch clarifies that users must replace ENDPOINT_ID with the\nactual ID of their image endpoint using:\n\n    openstack endpoint list --service glance --region RegionOne\n\nThis change aligns the 2024.1 docs with the correct instructions\nalready provided in the Epoxy 2025.1 release.\n\nChange-Id: I62d336565eaf6a50e6bdc7aa87946023fdcd7c54\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9d09ec3a83e5022fc7037d8aab8c581396c71e45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9d09ec3a83e5022fc7037d8aab8c581396c71e45"}]},"branch":"refs/heads/stable/2024.1","description":"Edit commit message"},"03fa1c99d4f9e4de6f9b97fa70031cf7aa4dd604":{"kind":"REWORK","_number":3,"created":"2025-06-17 08:10:58.000000000","uploader":{"_account_id":38109,"name":"Darshan Malagimani","display_name":"Darshan S M","email":"darshan.malagimani@outlook.com","username":"darshanmalagimani"},"ref":"refs/changes/38/952738/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/38/952738/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/952738/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/952738/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/952738/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/38/952738/3"}}},"commit":{"parents":[{"commit":"a514df4570b5b2a127037416aebdefb7aca7932f","subject":"Harden copy bust lock test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a514df4570b5b2a127037416aebdefb7aca7932f"}]}],"author":{"name":"Darshan Malagimani","email":"darshan.malagimani@outlook.com","date":"2025-06-17 07:51:17.000000000","tz":330},"committer":{"name":"Darshan Malagimani","email":"darshan.malagimani@outlook.com","date":"2025-06-17 08:10:40.000000000","tz":330},"subject":"docs: Fix incorrect oslo_limit endpoint_id in Glance install guide","message":"docs: Fix incorrect oslo_limit endpoint_id in Glance install guide\n\nThe Caracal 28.1.1 Glance documentation incorrectly hardcodes an\nendpoint_id in the  section of glance-api.conf, leading\nto HTTP 500 Internal Server Error when uploading an image.\n\nThis patch clarifies that users must replace  with the\nactual ID of their image endpoint using:\n\n    openstack endpoint list --service glance --region RegionOne\n\nThis change aligns the 2024.1 docs with the correct instructions\nalready provided in the Epoxy 2025.1 release.\n\nChange-Id: I62d336565eaf6a50e6bdc7aa87946023fdcd7c54\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/03fa1c99d4f9e4de6f9b97fa70031cf7aa4dd604"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/03fa1c99d4f9e4de6f9b97fa70031cf7aa4dd604"}]},"branch":"refs/heads/stable/2024.1"},"8c675f64d875d9c18a397605f58f9797454bcae8":{"kind":"REWORK","_number":4,"created":"2025-06-17 08:22:40.000000000","uploader":{"_account_id":38109,"name":"Darshan Malagimani","display_name":"Darshan S M","email":"darshan.malagimani@outlook.com","username":"darshanmalagimani"},"ref":"refs/changes/38/952738/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/38/952738/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/952738/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/952738/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/952738/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/38/952738/4"}}},"commit":{"parents":[{"commit":"a514df4570b5b2a127037416aebdefb7aca7932f","subject":"Harden copy bust lock test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a514df4570b5b2a127037416aebdefb7aca7932f"}]}],"author":{"name":"Darshan Malagimani","email":"darshan.malagimani@outlook.com","date":"2025-06-17 07:51:17.000000000","tz":330},"committer":{"name":"Darshan Malagimani","email":"darshan.malagimani@outlook.com","date":"2025-06-17 08:22:17.000000000","tz":330},"subject":"docs: Fix incorrect oslo_limit endpoint_id in Glance install guide","message":"docs: Fix incorrect oslo_limit endpoint_id in Glance install guide\n\nThe Caracal 28.1.1 Glance documentation incorrectly hardcodes an\nendpoint_id in the  section of glance-api.conf, leading\nto HTTP 500 Internal Server Error when uploading an image.\n\nThis patch clarifies that users must replace  with the\nactual ID of their image endpoint using:\n\n    openstack endpoint list --service glance --region RegionOne\n\nThis change aligns the 2024.1 docs with the correct instructions\nalready provided in the Epoxy 2025.1 release.\n\nChange-Id: I62d336565eaf6a50e6bdc7aa87946023fdcd7c54\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8c675f64d875d9c18a397605f58f9797454bcae8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8c675f64d875d9c18a397605f58f9797454bcae8"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Code-Review\u003dMIN"],"failing_atoms":["label:Code-Review\u003dMAX"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
