)]}'
{"id":"openstack%2Fglance~952847","triplet_id":"openstack%2Fglance~stable%2F2024.1~I2b6425f8a2bce4e47ebb15f4161fa238b3b6c6f9","project":"openstack/glance","branch":"stable/2024.1","attention_set":{},"removed_from_attention_set":{"38109":{"account":{"_account_id":38109,"name":"Darshan Malagimani","display_name":"Darshan S M","email":"darshan.malagimani@outlook.com","username":"darshanmalagimani"},"last_update":"2025-07-07 15:46:44.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2b6425f8a2bce4e47ebb15f4161fa238b3b6c6f9","subject":"Docs: be more explicit regarding oslo_limit.endpoint_id","status":"MERGED","created":"2025-06-18 15:57:39.000000000","updated":"2025-07-07 15:47:42.000000000","submitted":"2025-07-07 15:46:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"952847","meta_rev_id":"56d8bb106102ebafc426bb81d60f344a3d5a1faf","_number":952847,"virtual_id_number":952847,"owner":{"_account_id":38109,"name":"Darshan Malagimani","display_name":"Darshan S M","email":"darshan.malagimani@outlook.com","username":"darshanmalagimani"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-07 15:46:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":2,"date":"2025-07-07 14:42:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":1,"date":"2025-07-07 14:42:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-18 15:57:39.000000000","updated_by":{"_account_id":38109,"name":"Darshan Malagimani","display_name":"Darshan S M","email":"darshan.malagimani@outlook.com","username":"darshanmalagimani"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"CC"},{"updated":"2025-06-18 16:06:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-07 14:42:53.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"d7f0fb95054e3c314d7707efcb37193c3e63c900","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38109,"name":"Darshan Malagimani","display_name":"Darshan S M","email":"darshan.malagimani@outlook.com","username":"darshanmalagimani"},"date":"2025-06-18 15:57:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f5878107400d24a5833ea597fd8692cd223a85c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-18 16:06:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0e53aa693dbd4f88bc7b6e2677b6a845\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f72fe31c2054ebe871a88e62fdf5fbd : SUCCESS in 6m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8245b6bf63a3435f8ab62e237a3b6927 : SUCCESS in 6m 00s","accounts_in_message":[],"_revision_number":1},{"id":"69801ded2f188598dffc87fd8a9f4b36cab74fe9","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-07-07 14:42:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ba5bcf4c6f71525db22bf3f257c7c0b6ca6a8628","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-07 14:44:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7bbf03ef60c691de158c68aca1dca69584e2b222","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-07 15:46:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2613964413764e15b94e3b88ad07a95a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f59f576ca9c14203af54192b4a7526e5 : SUCCESS in 3m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a31b8b02a80249aa98d99a1afa9265d9 : SUCCESS in 6m 06s","accounts_in_message":[],"_revision_number":1},{"id":"694332690295ecbea8a4084e681bbca2647b0e0c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-07 15:46:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"56d8bb106102ebafc426bb81d60f344a3d5a1faf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-07 15:47:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ccfe7277195642659bcfb63872673cbb\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bba75b4cdcf542309c1668bcc0dfdfde : SUCCESS in 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4914bb9db5453f0721e37a631bd87f24135d5e19","revisions":{"4914bb9db5453f0721e37a631bd87f24135d5e19":{"kind":"REWORK","_number":1,"created":"2025-06-18 15:57:39.000000000","uploader":{"_account_id":38109,"name":"Darshan Malagimani","display_name":"Darshan S M","email":"darshan.malagimani@outlook.com","username":"darshanmalagimani"},"ref":"refs/changes/47/952847/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/47/952847/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/952847/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/952847/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/952847/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/47/952847/1"}}},"commit":{"parents":[{"commit":"a514df4570b5b2a127037416aebdefb7aca7932f","subject":"Harden copy bust lock test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a514df4570b5b2a127037416aebdefb7aca7932f"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2024-04-03 00:39:21.000000000","tz":120},"committer":{"name":"Darshan Malagimani","email":"darshan.malagimani@outlook.com","date":"2025-06-18 15:57:29.000000000","tz":330},"subject":"Docs: be more explicit regarding oslo_limit.endpoint_id","message":"Docs: be more explicit regarding oslo_limit.endpoint_id\n\nMake sure readers understand where this id comes from.\n\nChange-Id: I2b6425f8a2bce4e47ebb15f4161fa238b3b6c6f9\nCloses-Bug: #2059829\n(cherry picked from commit 70d18c868a82f347a1ec3247084b330148be9c3a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4914bb9db5453f0721e37a631bd87f24135d5e19"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4914bb9db5453f0721e37a631bd87f24135d5e19"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
