)]}'
{"id":"openstack%2Fglance~953008","triplet_id":"openstack%2Fglance~stable%2F2024.1~I8f7835449a35583972e1d995202161fe22f0b52c","project":"openstack/glance","branch":"stable/2024.1","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2025-07-06 18:16:44.000000000","reason":"\u003cGERRIT_ACCOUNT_38109\u003e replied on the change","reason_account":{"_account_id":38109,"name":"Darshan Malagimani","display_name":"Darshan S M","email":"darshan.malagimani@outlook.com","username":"darshanmalagimani"}},"16708":{"account":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"last_update":"2025-07-06 18:16:44.000000000","reason":"\u003cGERRIT_ACCOUNT_38109\u003e replied on the change","reason_account":{"_account_id":38109,"name":"Darshan Malagimani","display_name":"Darshan S M","email":"darshan.malagimani@outlook.com","username":"darshanmalagimani"}},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-07-06 18:16:44.000000000","reason":"\u003cGERRIT_ACCOUNT_38109\u003e replied on the change","reason_account":{"_account_id":38109,"name":"Darshan Malagimani","display_name":"Darshan S M","email":"darshan.malagimani@outlook.com","username":"darshanmalagimani"}},"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2025-06-28 20:37:05.000000000","reason":"\u003cGERRIT_ACCOUNT_38109\u003e replied on the change","reason_account":{"_account_id":38109,"name":"Darshan Malagimani","display_name":"Darshan S M","email":"darshan.malagimani@outlook.com","username":"darshanmalagimani"}},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2025-07-07 14:39:42.000000000","reason":"\u003cGERRIT_ACCOUNT_8122\u003e replied on the change","reason_account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},"27900":{"account":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"last_update":"2025-07-06 18:16:44.000000000","reason":"\u003cGERRIT_ACCOUNT_38109\u003e replied on the change","reason_account":{"_account_id":38109,"name":"Darshan Malagimani","display_name":"Darshan S M","email":"darshan.malagimani@outlook.com","username":"darshanmalagimani"}},"38109":{"account":{"_account_id":38109,"name":"Darshan Malagimani","display_name":"Darshan S M","email":"darshan.malagimani@outlook.com","username":"darshanmalagimani"},"last_update":"2025-07-07 14:39:52.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I8f7835449a35583972e1d995202161fe22f0b52c","subject":"Docs: clarify usage of `endpoint_id` in oslo_limit config","status":"ABANDONED","created":"2025-06-20 18:20:13.000000000","updated":"2025-07-07 14:39:52.000000000","total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e50981199e827074640bb4f746d199bf92d3d83a","_number":953008,"virtual_id_number":953008,"owner":{"_account_id":38109,"name":"Darshan Malagimani","display_name":"Darshan S M","email":"darshan.malagimani@outlook.com","username":"darshanmalagimani"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-06-20 20:40:18.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-20 18:27:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-20 18:40:05.000000000","updated_by":{"_account_id":38109,"name":"Darshan Malagimani","display_name":"Darshan S M","email":"darshan.malagimani@outlook.com","username":"darshanmalagimani"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2025-06-24 04:15:41.000000000","updated_by":{"_account_id":38109,"name":"Darshan Malagimani","display_name":"Darshan S M","email":"darshan.malagimani@outlook.com","username":"darshanmalagimani"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2025-06-28 20:37:05.000000000","updated_by":{"_account_id":38109,"name":"Darshan Malagimani","display_name":"Darshan S M","email":"darshan.malagimani@outlook.com","username":"darshanmalagimani"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REMOVED"},{"updated":"2025-06-28 20:37:05.000000000","updated_by":{"_account_id":38109,"name":"Darshan Malagimani","display_name":"Darshan S M","email":"darshan.malagimani@outlook.com","username":"darshanmalagimani"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REMOVED"},{"updated":"2025-06-28 20:37:05.000000000","updated_by":{"_account_id":38109,"name":"Darshan Malagimani","display_name":"Darshan S M","email":"darshan.malagimani@outlook.com","username":"darshanmalagimani"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"REVIEWER"},{"updated":"2025-06-28 20:37:05.000000000","updated_by":{"_account_id":38109,"name":"Darshan Malagimani","display_name":"Darshan S M","email":"darshan.malagimani@outlook.com","username":"darshanmalagimani"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2025-06-28 20:37:05.000000000","updated_by":{"_account_id":38109,"name":"Darshan Malagimani","display_name":"Darshan S M","email":"darshan.malagimani@outlook.com","username":"darshanmalagimani"},"reviewer":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"state":"REVIEWER"},{"updated":"2025-06-28 20:37:05.000000000","updated_by":{"_account_id":38109,"name":"Darshan Malagimani","display_name":"Darshan S M","email":"darshan.malagimani@outlook.com","username":"darshanmalagimani"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2025-07-06 18:16:44.000000000","updated_by":{"_account_id":38109,"name":"Darshan Malagimani","display_name":"Darshan S M","email":"darshan.malagimani@outlook.com","username":"darshanmalagimani"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"REMOVED"},{"updated":"2025-07-06 18:16:44.000000000","updated_by":{"_account_id":38109,"name":"Darshan Malagimani","display_name":"Darshan S M","email":"darshan.malagimani@outlook.com","username":"darshanmalagimani"},"reviewer":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"state":"REMOVED"},{"updated":"2025-07-06 18:16:44.000000000","updated_by":{"_account_id":38109,"name":"Darshan Malagimani","display_name":"Darshan S M","email":"darshan.malagimani@outlook.com","username":"darshanmalagimani"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REMOVED"},{"updated":"2025-07-06 18:16:44.000000000","updated_by":{"_account_id":38109,"name":"Darshan Malagimani","display_name":"Darshan S M","email":"darshan.malagimani@outlook.com","username":"darshanmalagimani"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REMOVED"},{"updated":"2025-07-06 18:16:44.000000000","updated_by":{"_account_id":38109,"name":"Darshan Malagimani","display_name":"Darshan S M","email":"darshan.malagimani@outlook.com","username":"darshanmalagimani"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"d214fe734b8857963474d16404cf28558a45a43b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38109,"name":"Darshan Malagimani","display_name":"Darshan S M","email":"darshan.malagimani@outlook.com","username":"darshanmalagimani"},"date":"2025-06-20 18:20:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8a3f1a75c7a15d3ec1a6e1c3fccac89d15443848","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-20 18:27:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a133fe58d2e4ca0b49f268ef1c0afc4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d6037b51061486f8d15f09940aa4735 : SUCCESS in 4m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d5b7078bc404134998a3e25039d19c5 : SUCCESS in 4m 48s","accounts_in_message":[],"_revision_number":1},{"id":"76e1c115f85bed9334b58a9383da7aa64f73ce6a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38109,"name":"Darshan Malagimani","display_name":"Darshan S M","email":"darshan.malagimani@outlook.com","username":"darshanmalagimani"},"date":"2025-06-20 18:32:43.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"4b5299e633ac975328723f72e90dc220d5c2eb87","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38109,"name":"Darshan Malagimani","display_name":"Darshan S M","email":"darshan.malagimani@outlook.com","username":"darshanmalagimani"},"date":"2025-06-20 18:33:27.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"af4cb54eb2b55351136ecc1f70631e6c750d52bf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38109,"name":"Darshan Malagimani","display_name":"Darshan S M","email":"darshan.malagimani@outlook.com","username":"darshanmalagimani"},"date":"2025-06-20 18:33:36.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"d1c6ba3f5d741e9da4096b2e4e5b22a4aab05b27","author":{"_account_id":38109,"name":"Darshan Malagimani","display_name":"Darshan S M","email":"darshan.malagimani@outlook.com","username":"darshanmalagimani"},"date":"2025-06-20 18:40:05.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"cdde6865d25abbadf856a7a7a493f26abfb402ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-20 18:42:29.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/510c47cfc94a4b39914944da946b0c3c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2edc392cef0d493f9c269a447c655df3 : SUCCESS in 3m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c21505e86dde4bcb8d0f53a893608759 : SUCCESS in 7m 19s","accounts_in_message":[],"_revision_number":4},{"id":"131b7d44c8ece9eacb37f7a0d14c331b343581a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38109,"name":"Darshan Malagimani","display_name":"Darshan S M","email":"darshan.malagimani@outlook.com","username":"darshanmalagimani"},"date":"2025-06-20 20:29:37.000000000","message":"Patch Set 5: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"d3463934a9019b1492474ebce041a385b41c8403","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-20 20:40:18.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/563f72b0687949bc9bad5680aca89bf8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2701a5405bf34e9e92f7ed19fe429adc : SUCCESS in 4m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf90927c6c09486185d4a32854c0c4ac : SUCCESS in 5m 18s","accounts_in_message":[],"_revision_number":5},{"id":"0ffedf251a4f3314a3959f0b6e13fe140a4ff120","author":{"_account_id":38109,"name":"Darshan Malagimani","display_name":"Darshan S M","email":"darshan.malagimani@outlook.com","username":"darshanmalagimani"},"date":"2025-06-28 20:37:05.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"651aa1b13170819f80a7c98c5cc775055565ddd3","author":{"_account_id":38109,"name":"Darshan Malagimani","display_name":"Darshan S M","email":"darshan.malagimani@outlook.com","username":"darshanmalagimani"},"date":"2025-07-06 18:16:44.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"8a298e7b63742240a6049ec820a9fc3c627b3423","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-07-07 14:39:42.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"e50981199e827074640bb4f746d199bf92d3d83a","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-07-07 14:39:52.000000000","message":"Abandoned\n\nIn favor of https://review.opendev.org/c/openstack/glance/+/952847","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"d98656a8d895c4c9119095e1ec1d79cb5846f769","revisions":{"23afaa480f0b0e16cf6f733d4829b8dce62c35d3":{"kind":"REWORK","_number":1,"created":"2025-06-20 18:20:13.000000000","uploader":{"_account_id":38109,"name":"Darshan Malagimani","display_name":"Darshan S M","email":"darshan.malagimani@outlook.com","username":"darshanmalagimani"},"ref":"refs/changes/08/953008/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/08/953008/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/08/953008/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/08/953008/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/08/953008/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/08/953008/1"}}},"commit":{"parents":[{"commit":"a514df4570b5b2a127037416aebdefb7aca7932f","subject":"Harden copy bust lock test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a514df4570b5b2a127037416aebdefb7aca7932f"}]}],"author":{"name":"Darshan Malagimani","email":"darshan.malagimani@outlook.com","date":"2025-06-20 18:19:17.000000000","tz":330},"committer":{"name":"Darshan Malagimani","email":"darshan.malagimani@outlook.com","date":"2025-06-20 18:19:17.000000000","tz":330},"subject":"Docs: Clarify quota configuration using oslo.limit","message":"Docs: Clarify quota configuration using oslo.limit\n\nThis patch improves the Glance installation docs to clarify how to\nconfigure quotas using Keystone unified limits via oslo.limit.\n\nChanges include:\n- Setting `endpoint_id` and `region_name` in the `[oslo_limit]` section,\n  which are required for proper endpoint targeting.\n- Assigning the `reader` system role to the `glance` user, so it can\n  retrieve system-scope registered limits from Keystone.\n- Enabling `use_keystone_limits \u003d True` in `[DEFAULT]` to allow Glance\n  to use the centralized quota management system.\n\nThis is a fresh patch in `stable/2024.1` to replace an earlier patch\nthat included a full file rewrite.\n\nThanks for reviewing.\n\nChange-Id: I8f7835449a35583972e1d995202161fe22f0b52c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/23afaa480f0b0e16cf6f733d4829b8dce62c35d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/23afaa480f0b0e16cf6f733d4829b8dce62c35d3"}]},"branch":"refs/heads/stable/2024.1"},"090f20bb5725d3fe766766b708a1c93ece73c67a":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-06-20 18:32:43.000000000","uploader":{"_account_id":38109,"name":"Darshan Malagimani","display_name":"Darshan S M","email":"darshan.malagimani@outlook.com","username":"darshanmalagimani"},"ref":"refs/changes/08/953008/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/08/953008/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/08/953008/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/08/953008/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/08/953008/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/08/953008/2"}}},"commit":{"parents":[{"commit":"a514df4570b5b2a127037416aebdefb7aca7932f","subject":"Harden copy bust lock test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a514df4570b5b2a127037416aebdefb7aca7932f"}]}],"author":{"name":"Darshan Malagimani","email":"darshan.malagimani@outlook.com","date":"2025-06-20 18:19:17.000000000","tz":330},"committer":{"name":"Darshan S M","email":"darshan.malagimani@outlook.com","date":"2025-06-20 18:32:43.000000000","tz":0},"subject":"Docs: clarify usage of `endpoint_id` in oslo_limit config","message":"Docs: clarify usage of `endpoint_id` in oslo_limit config\n\nThis patch updates the quota configuration docs to clarify how to\nset the `endpoint_id` value in the `[oslo_limit]` section for Glance.\n\nSpecifically, it adds instructions on how to find the image endpoint ID\nusing the command:\n\n  $ openstack endpoint list --service image --interface public --region RegionOne\n\nThis is important because users need to provide the correct `endpoint_id`\nto enable Glance to connect properly to Keystone\u0027s unified limits system.\n\nWithout this clarification, users might be confused about what value to\nuse for `endpoint_id`, potentially causing misconfiguration or quota\nenforcement failures.\n\nOnly documentation change in stable/2024.1.\n\nThanks for reviewing.\n\n\nChange-Id: I8f7835449a35583972e1d995202161fe22f0b52c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/090f20bb5725d3fe766766b708a1c93ece73c67a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/090f20bb5725d3fe766766b708a1c93ece73c67a"}]},"branch":"refs/heads/stable/2024.1","description":"Edit commit message"},"69d0d418e22dcea6a82efcfbeb72eb256698c510":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2025-06-20 18:33:27.000000000","uploader":{"_account_id":38109,"name":"Darshan Malagimani","display_name":"Darshan S M","email":"darshan.malagimani@outlook.com","username":"darshanmalagimani"},"ref":"refs/changes/08/953008/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/08/953008/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/08/953008/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/08/953008/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/08/953008/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/08/953008/3"}}},"commit":{"parents":[{"commit":"a514df4570b5b2a127037416aebdefb7aca7932f","subject":"Harden copy bust lock test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a514df4570b5b2a127037416aebdefb7aca7932f"}]}],"author":{"name":"Darshan Malagimani","email":"darshan.malagimani@outlook.com","date":"2025-06-20 18:19:17.000000000","tz":330},"committer":{"name":"Darshan S M","email":"darshan.malagimani@outlook.com","date":"2025-06-20 18:33:27.000000000","tz":0},"subject":"Docs: clarify usage of `endpoint_id` in oslo_limit config","message":"Docs: clarify usage of `endpoint_id` in oslo_limit config\n\nThis patch updates the quota configuration docs to clarify how to\nset the `endpoint_id` value in the `[oslo_limit]` section for Glance.\n\nSpecifically, it adds instructions on how to find the image endpoint ID\nusing the command:\n\n  $ openstack endpoint list --service image --interface public --region RegionOne\n\nThis is important because users need to provide the correct `endpoint_id`to enable Glance to connect properly to Keystone\u0027s unified limits system.\n\nWithout this clarification, users might be confused about what value to\nuse for `endpoint_id`, potentially causing misconfiguration or quota\nenforcement failures.\n\nOnly documentation change in stable/2024.1.\n\nThanks for reviewing.\n\n\nChange-Id: I8f7835449a35583972e1d995202161fe22f0b52c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/69d0d418e22dcea6a82efcfbeb72eb256698c510"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/69d0d418e22dcea6a82efcfbeb72eb256698c510"}]},"branch":"refs/heads/stable/2024.1","description":"Edit commit message"},"ba5aea6d4587db85bc0165cd983aa05899d442e9":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2025-06-20 18:33:36.000000000","uploader":{"_account_id":38109,"name":"Darshan Malagimani","display_name":"Darshan S M","email":"darshan.malagimani@outlook.com","username":"darshanmalagimani"},"ref":"refs/changes/08/953008/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/08/953008/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/08/953008/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/08/953008/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/08/953008/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/08/953008/4"}}},"commit":{"parents":[{"commit":"a514df4570b5b2a127037416aebdefb7aca7932f","subject":"Harden copy bust lock test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a514df4570b5b2a127037416aebdefb7aca7932f"}]}],"author":{"name":"Darshan Malagimani","email":"darshan.malagimani@outlook.com","date":"2025-06-20 18:19:17.000000000","tz":330},"committer":{"name":"Darshan S M","email":"darshan.malagimani@outlook.com","date":"2025-06-20 18:33:36.000000000","tz":0},"subject":"Docs: clarify usage of `endpoint_id` in oslo_limit config","message":"Docs: clarify usage of `endpoint_id` in oslo_limit config\n\nThis patch updates the quota configuration docs to clarify how to\nset the `endpoint_id` value in the `[oslo_limit]` section for Glance.\n\nSpecifically, it adds instructions on how to find the image endpoint ID\nusing the command:\n\n  $ openstack endpoint list --service image --interface public --region RegionOne\n\nThis is important because users need to provide the correct `endpoint_id`to enable Glance to connect properly to Keystone\u0027s unified limits system.\n\nWithout this clarification, users might be confused about what value to\nuse for `endpoint_id`, potentially causing misconfiguration or quota\nenforcement failures.\n\nOnly documentation change in stable/2024.1.\n\nThanks for reviewing.\n\nChange-Id: I8f7835449a35583972e1d995202161fe22f0b52c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ba5aea6d4587db85bc0165cd983aa05899d442e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ba5aea6d4587db85bc0165cd983aa05899d442e9"}]},"branch":"refs/heads/stable/2024.1","description":"Edit commit message"},"d98656a8d895c4c9119095e1ec1d79cb5846f769":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2025-06-20 20:29:37.000000000","uploader":{"_account_id":38109,"name":"Darshan Malagimani","display_name":"Darshan S M","email":"darshan.malagimani@outlook.com","username":"darshanmalagimani"},"ref":"refs/changes/08/953008/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/08/953008/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/08/953008/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/08/953008/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/08/953008/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/08/953008/5"}}},"commit":{"parents":[{"commit":"a514df4570b5b2a127037416aebdefb7aca7932f","subject":"Harden copy bust lock test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a514df4570b5b2a127037416aebdefb7aca7932f"}]}],"author":{"name":"Darshan Malagimani","email":"darshan.malagimani@outlook.com","date":"2025-06-20 18:19:17.000000000","tz":330},"committer":{"name":"Darshan S M","email":"darshan.malagimani@outlook.com","date":"2025-06-20 20:29:37.000000000","tz":0},"subject":"Docs: clarify usage of `endpoint_id` in oslo_limit config","message":"Docs: clarify usage of `endpoint_id` in oslo_limit config\n\nThis patch updates the quota configuration docs to clarify how to\nset the `endpoint_id` value in the `[oslo_limit]` section for Glance.\n\nSpecifically, it adds instructions on how to find the image endpoint ID\nusing the command:\n\n  $ openstack endpoint list --service image --interface public --region RegionOne\n\nThis is important because users need to provide the correct `endpoint_id`to enable Glance to connect properly to Keystone\u0027s unified limits system.\n\noslo.limit uses the endpoint_id (along with region, auth, etc.) to query Keystone for registered limits for a specific service (in this case, Glance).\n\nIf the wrong or missing endpoint_id is set, oslo.limit may fail to resolve the limit from Keystone.\n\nSo, setting the correct endpoint_id is critical for quota enforcement to work as expected with unified limits.\n\nWithout this clarification, users might be confused about what value to\nuse for `endpoint_id`, potentially causing misconfiguration or quota\nenforcement failures.\n\nOnly documentation change in stable/2024.1.\n\nThanks for reviewing.\n\nChange-Id: I8f7835449a35583972e1d995202161fe22f0b52c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d98656a8d895c4c9119095e1ec1d79cb5846f769"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d98656a8d895c4c9119095e1ec1d79cb5846f769"}]},"branch":"refs/heads/stable/2024.1","description":"Edit commit message"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
