)]}'
{"id":"openstack%2Fglance~953161","triplet_id":"openstack%2Fglance~stable%2F2025.1~I2ebbeee9cb700e4a1ea9b669744ff49f8ccc9dfc","project":"openstack/glance","branch":"stable/2025.1","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-08-11 15:17:09.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2ebbeee9cb700e4a1ea9b669744ff49f8ccc9dfc","subject":"Fix incomplete migration to datetime.fromtimestamp","status":"MERGED","created":"2025-06-24 01:21:10.000000000","updated":"2025-08-11 15:18:06.000000000","submitted":"2025-08-11 15:17:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"953161","cherry_pick_of_change":952682,"cherry_pick_of_patch_set":2,"meta_rev_id":"e9bcdaf45e07f05b3ed5ad6a498246073ca16389","_number":953161,"virtual_id_number":953161,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-11 15:17:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2025-08-11 12:46:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2025-08-01 15:12:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-07-02 10:08:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2025-08-11 12:46:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-24 01:35:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-06-24 03:37:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-02 10:08:45.000000000","updated_by":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"reviewer":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"state":"REVIEWER"},{"updated":"2025-07-07 15:53:11.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"CC"},{"updated":"2025-08-01 15:12:28.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2025-08-11 12:46:19.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"}],"messages":[{"id":"dd31b68ea5095bcd59cc26523e15739deb3c796a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-06-24 01:21:10.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"0bbabeabfedcdf8f940d0f590e6006d4651fc237","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-24 01:35:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7dc2a3e692c844698b1981c1dec3d4ee\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/502ce40fc5954b54ba51aff2a89ee7f9 : SUCCESS in 11m 20s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b59c3dce33d74309b0ce9d1fce6cc688 : SUCCESS in 12m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"85826019024441601d6ca6daa8dce6ae1307bdb1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-24 03:37:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4aa4abfb2bc849dd88440ac33b4fd59e\n\n- grenade https://zuul.opendev.org/t/openstack/build/814a91860a0f4e2482d6c995bffebf73 : SUCCESS in 27m 02s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/51bf004d43b647289cb454c7f9af4088 : SUCCESS in 46m 54s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/759f9435482d4c198db8ad7f078704eb : SUCCESS in 1h 26m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2a749c202e6345a8aaf8eb71a414c5ad : SUCCESS in 51m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ccd5be62d2d24023a26c2057abcaa4f3 : SUCCESS in 2m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6d512534e82b400d9f848beea8a7b783 : SUCCESS in 5m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/494cef8daef3419fbf9f7073b61b8c9e : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/075a33e8178d45f495f4537e40b3e710 : SUCCESS in 6m 13s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ddfcd35a16304eadb7bafdc3077183ec : SUCCESS in 6m 00s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/48f2387cdd1e4c96b198fbc1448c5f87 : SUCCESS in 8m 22s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/0a0d98fd529b4e3791e9cbc9603b8738 : SUCCESS in 9m 25s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/cca6a84ac17641519337787339dbdb79 : FAILURE in 10m 05s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/4167fa70f95249068e7f974450dd394e : SUCCESS in 1h 50m 13s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/37e528b4097d4bc39a9bf790ecc24685 : FAILURE in 9m 58s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/aab0d2101bac49c9bbe8ca4b15f7bb6c : SUCCESS in 1h 04m 56s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/c653f50b1b2a403cafa2d6a0ae130171 : SUCCESS in 2h 15m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/80d109202c7546a9bbf37ed3b6fe18b7 : SUCCESS in 41m 50s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/59d3e91affaa4f6cb022e933f37f3110 : SUCCESS in 1h 47m 45s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/65fc317abd6e41fe84ce7011ed7d3322 : SUCCESS in 1h 05m 11s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/4004f3aa725641ccb6698076357c2eec : FAILURE in 16m 08s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/0c79335ba94e4568828750ffba57d6f0 : FAILURE in 1h 38m 57s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/e650936ae4ef45f88b5b0a8314f8eaf3 : SUCCESS in 24m 58s","accounts_in_message":[],"_revision_number":1},{"id":"7cb8c67e713962cf4e08fc4c275d714808326e59","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-06-24 15:02:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"94143caddac739e74536c6493c8ec2ee280e1cac","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-24 15:31:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9521027e7e1e4150a3a21421b5426fb2\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a67c61cc799e4b00a7dbcf21bc242e4d : SUCCESS in 11m 47s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5d94b696e4fa49558e31eeb1f3b5704a : SUCCESS in 11m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"df56eb9d3290fc8a863c786059303e323039d7c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-24 17:26:59.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d38e811acc974417ad8a5035487247eb\n\n- grenade https://zuul.opendev.org/t/openstack/build/019812e1c82f43739d5bc720e0abb05c : SUCCESS in 56m 38s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e3853955f8d84780bfe29b5c0554514b : SUCCESS in 43m 32s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/dc6485f90d634f7fb5ee953d3b4457b3 : SUCCESS in 1h 35m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/781746d187624b1f8bc48ff1e80f55e3 : SUCCESS in 1h 09m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a851104189d4f5e83e98f0cf8bcc54f : SUCCESS in 2m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a8174ae1cbb64410bbb262f1e24ecdaf : SUCCESS in 5m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ea77e05051be4da0b2e02bb4e59e0735 : SUCCESS in 5m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5bcda7c797f4e089819d06e266babeb : SUCCESS in 5m 06s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5d8bd5ec3ab344fd958ca97594c93699 : SUCCESS in 10m 58s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c8e311fc9eb44fd694fea7e96f4ec440 : SUCCESS in 6m 07s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/291db071ea584caf99e760efb535b250 : SUCCESS in 11m 46s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/158fc4fd49fc4606b0d019cb773e0bbb : SUCCESS in 1h 48m 27s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/125a4e36579e470d9098281055164e10 : SUCCESS in 2h 05m 44s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/ca0b809b108f4960ab2ffdfb83b0be91 : SUCCESS in 1h 57m 01s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/822f57139be2417392dbddc7069140f0 : FAILURE in 11m 03s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/ce8d317bdbbd47a0a674af089613a6c2 : SUCCESS in 1h 18m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8b1652cd6f59413bb56da14d838df8b9 : SUCCESS in 55m 57s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/0a95e7b954174a5999d0c8106cde10d1 : SUCCESS in 1h 40m 42s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/56c8af1513a94d84b97f0d0441e8b5df : SUCCESS in 1h 57m 33s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/dbb4c474a6214566adb241487d76f741 : SUCCESS in 57m 06s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/a273f6942987424f932b5a89fe0f0fc4 : FAILURE in 10m 31s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/1f17f68b799a4fc7b1d3fdae3dc4ffcc : SUCCESS in 15m 41s","accounts_in_message":[],"_revision_number":1},{"id":"4e103dcfeabb968da9cf3fb3f82554b0728452b5","author":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"date":"2025-07-02 10:08:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e11aeab627674e54579b5fc7254992ed0e6222e4","author":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"date":"2025-07-02 10:08:56.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"995d1e657bdcdb38ca8b70e95cea460f0d68b2d1","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-07-07 15:53:11.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"45158afcc90de8e18a2b520a2605c438a3fbcf73","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-07 16:28:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c4a27ffc993d469d878349b993fab817\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/39608f8c238c461eb1e9db417eaf914b : NODE_FAILURE Node(set) request 300-0027361639 failed in 0s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f1036b95b9a1485294e8ac0c0375cbf8 : SUCCESS in 8m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f0f45eba602b449e05bd98f5f6802bad39c7dd90","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-07 18:12:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/77915fdd9d8442a88abe3e72058bc2a7\n\n- grenade https://zuul.opendev.org/t/openstack/build/f028abb2b40b4e69b0ced308571ee615 : SUCCESS in 56m 05s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ca6fda10fcaa439f89bf1a736d5842fb : SUCCESS in 26m 44s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7098c32ce9ab42059e746931bae2c2d4 : SUCCESS in 1h 25m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dff1b27979e64825903ecd90a698f00f : SUCCESS in 1h 02m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50920ab5c95a40078cf310e4cbafeaec : SUCCESS in 6m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/679476f30e53429a8149a380d154456f : SUCCESS in 4m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/372ddf6c321d4594a25308bf51b6cf24 : SUCCESS in 5m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cea83710222c419bb0cda95806f3c9db : SUCCESS in 5m 07s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c343cba7cd104dfd937c9c0c7efcca29 : SUCCESS in 7m 10s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ae25517b1e2f439ea03faa3e25e5a52a : SUCCESS in 10m 53s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/fa92efa277b54c14abc9879b9695817c : SUCCESS in 13m 15s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/bea6d2317fa5485fae2b36ebe80ed40c : SUCCESS in 1h 50m 58s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/8a9a7bac67a3485bb840879256058415 : SUCCESS in 2h 04m 25s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/8e12c27c8cd147338eb0d50f481b1dfe : SUCCESS in 1h 39m 41s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/de145b59288f4c12aec95b72856b3b2c : SUCCESS in 1h 40m 15s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/b4fa33573daa4a379114800c206be9ef : SUCCESS in 2h 14m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ffb0e18f20ce4b27a275475611c53825 : SUCCESS in 54m 05s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/75a28a6208944a5e908e243bbfa0de41 : SUCCESS in 1h 38m 49s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/9c243ae50b4244ec92d6e5d0e15eff92 : SUCCESS in 1h 54m 03s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/03f350d975a84d5d8979ec80b84dba62 : SUCCESS in 53m 39s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/65a7fc0bacf54a1d9879b21664ab24fa : SUCCESS in 1h 53m 32s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/34357ae19a774059b62d9c5acf33b3c4 : SUCCESS in 36m 25s","accounts_in_message":[],"_revision_number":1},{"id":"b49d5932e4a8934fc92fffcb6cba4263692744b2","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-08-01 15:12:28.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"238eaed8aee94ac595c51afbd785c9d1ae96e62e","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-08-11 12:46:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"631bf499e6ae77b20c26c7266938d3113e073144","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-11 12:46:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c923c37246c5ec17e3d069f8c573223426a4d9d3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-11 15:17:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8aa92b8b816a46bbb7447d865445732a\n\n- grenade https://zuul.opendev.org/t/openstack/build/4a80f055ffad432faf20d4fb15ba418f : SUCCESS in 58m 42s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ed7a8b6ada904278a83c72080c572092 : SUCCESS in 44m 33s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1981c829b8024fe599cf76c8632ddad8 : SUCCESS in 1h 35m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0aed07c1476d41009badfe71f5e95bfc : SUCCESS in 52m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54a1ac9575cf46ce8567bcca05097902 : SUCCESS in 6m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/39bb1293861c46c5bfa47994e682c1bc : SUCCESS in 7m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b1279d11a03c45938ad3d2d0f7a8e990 : SUCCESS in 5m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3899d5e31504105b30a88ae2375ee38 : SUCCESS in 5m 54s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d81b31cdb04d4384b555ca6c5e6d42e4 : SUCCESS in 9m 16s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/627e86bf0f7448db9bd03a868510ea68 : SUCCESS in 10m 06s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/4e8241096d4043639e3b7defac21d87d : SUCCESS in 12m 22s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/c64b22561889484a933634c25a8fa1a0 : SUCCESS in 1h 36m 56s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/3ab7a0dccd8b4c069a320e01062c6fdc : SUCCESS in 2h 02m 31s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/e22bc3ac1ff84709bbff2beddac0059d : SUCCESS in 2h 25m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1cf99f12c54640388f6142c0fe1fcf12 : SUCCESS in 1h 00m 12s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/4fdf4ebe0bce4ceca191a9495f667301 : SUCCESS in 1h 42m 29s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/a410d2ba108043348f859c1bd10e047e : SUCCESS in 33m 08s","accounts_in_message":[],"_revision_number":1},{"id":"573f09fce8e79ceaefc0fdf5629cc7cff4bf1a8b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-11 15:17:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e9bcdaf45e07f05b3ed5ad6a498246073ca16389","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-11 15:18:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8c0032860db0463892ffe78859c90509\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6fb9b5a0d114ba0ad316ee0f8c42905 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2596f7253c36aeb99d53e00eca7ac47a18e7f30f","revisions":{"2596f7253c36aeb99d53e00eca7ac47a18e7f30f":{"kind":"REWORK","_number":1,"created":"2025-06-24 01:21:10.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/61/953161/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/61/953161/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/61/953161/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/61/953161/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/61/953161/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/61/953161/1"}}},"commit":{"parents":[{"commit":"99b54964031408555917539b73eb5c18e463f89a","subject":"Merge \"Fix vTPM metadefs.\" into stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/99b54964031408555917539b73eb5c18e463f89a"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-06-16 15:06:28.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-06-24 01:21:10.000000000","tz":0},"subject":"Fix incomplete migration to datetime.fromtimestamp","message":"Fix incomplete migration to datetime.fromtimestamp\n\nCommit 1f8a84cdccec4292f6fa8d9ca4f013fe8ebc4735 replaced\nutcfromtimestamp by fromtimestamp, due to deprecation of\nutcfromtimestamp, but it updated one usage incompletely.\n\nChange-Id: I2ebbeee9cb700e4a1ea9b669744ff49f8ccc9dfc\n(cherry picked from commit 9522939920b4fda6319acd6db7dcad5a6d880445)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2596f7253c36aeb99d53e00eca7ac47a18e7f30f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2596f7253c36aeb99d53e00eca7ac47a18e7f30f"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
