)]}'
{"id":"openstack%2Fglance~954190","triplet_id":"openstack%2Fglance~master~I19ff02322cad117b7d21d7ccf1709b1f259fd383","project":"openstack/glance","branch":"master","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2026-02-06 10:55:41.000000000","reason":"Change was submitted"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-02-06 10:55:41.000000000","reason":"Change was submitted"},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2026-01-05 15:15:26.000000000","reason":"\u003cGERRIT_ACCOUNT_8122\u003e replied on the change","reason_account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}}},"hashtags":[],"change_id":"I19ff02322cad117b7d21d7ccf1709b1f259fd383","subject":"Drop legacy arguments from RequestContext","status":"MERGED","created":"2025-07-06 14:49:24.000000000","updated":"2026-02-06 10:56:49.000000000","submitted":"2026-02-06 10:55:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"954190","meta_rev_id":"e113c5f29b058e4ae35f9d89f945ad0f1159efb0","_number":954190,"virtual_id_number":954190,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-06 10:55:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2026-02-05 15:32:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2026-02-05 14:00:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-02-05 14:00:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2026-02-05 15:32:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-06 15:02:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-07-06 16:59:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-17 14:58:28.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2026-01-05 15:38:16.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2026-01-05 15:38:16.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2026-02-05 15:32:44.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"a488b51316bf2b0dd6c73a6a7daf17c71a94f582","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-07-06 14:49:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"42225a84e1b3034bc39682650464b9601fbced82","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-06 15:02:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/edd5ea4b2cef4304ad97dde2d3ee4576\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/125e92409a764b25b52b545b4f4b0313 : SUCCESS in 9m 54s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/eb3e36db45c6407291e4081a789077c7 : SUCCESS in 7m 43s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/114da84d9fb543c7b8e64b1e517930db : SUCCESS in 10m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6921fe5253b0607411fc32003a35e1290f9ceba5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-06 16:59:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/808960ec8fbc4d848244aa1e00a2a83b\n\n- grenade https://zuul.opendev.org/t/openstack/build/024355f0e6224485add58a70ce424d7f : SUCCESS in 59m 52s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c959a14fdbda43c090526f8d0ef337cd : SUCCESS in 33m 12s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ee424c57162f49aca5453ea43b1dc7d4 : SUCCESS in 1h 30m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6808a592d88b4a45b06c65f964d09988 : SUCCESS in 31m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e6ac69ef7264d19a3b8824b33bf59ca : SUCCESS in 6m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/42844274a46643d181cb68e89380a8d8 : SUCCESS in 3m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2f52e96d54994b3494265140eebbe8d6 : SUCCESS in 5m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/10b67a31aff2484488f903f08a1bf447 : SUCCESS in 3m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4f4e49fa7e846369b6344e54a6b111c : SUCCESS in 5m 40s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d49ffe2de6534760924ff5eca2df44f7 : SUCCESS in 8m 04s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/ee71c455aab745ae80fe71a795108413 : SUCCESS in 6m 22s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/9abeaad22e214e558fa710b42b8af3f4 : SUCCESS in 1h 06m 05s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/2b4d5e982cce42a29b4eede7f78ae037 : SUCCESS in 1h 36m 07s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/d9fc0b061a4f487aaba9f2479258c2c9 : SUCCESS in 2h 01m 17s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/a17f0cee33d944389fde8a8cabb171ea : SUCCESS in 2h 05m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2433f4b85934473781f12bc8b9b39729 : SUCCESS in 33m 49s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b9048d50e7a440af9fe67752deabee6a : SUCCESS in 1h 56m 41s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/e25a2473bb394dd5a5d35f5ded2178a1 : FAILURE in 19m 15s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/9f17b6d3caf24d33b45574572f9496ec : SUCCESS in 1h 28m 41s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/8d0fcab503f84763a9128fcdd218b540 : SUCCESS in 20m 27s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/6c04976825ff4764ad2624640cc5e4a5 : SUCCESS in 58m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8bda2e3392557043bcfbaa7407f68c399dad0873","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-07-17 14:58:28.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"082fbe180702a4f27f7f84d46b3f046e02fa8c05","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-12-13 19:22:51.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"66764d1ad4ccd3e42f66fae4aa3a8dfd5e841dba","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-13 19:33:38.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/11b1ff2732004b94a079689fecf6ee14\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/45c0f103407c4452a4f7b2d581c51b8f : FAILURE in 8m 31s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/07b6d69f62bb408380052732b8ce6c34 : FAILURE in 6m 20s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e58ffdb5d8854fcd98260463caddba96 : FAILURE in 8m 52s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b7430ef8521b1f1804e442a1a723b899f46ccbc0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-13 21:59:34.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/845f9047ea8d45808ebddf97cc03b6a4\n\n- grenade https://zuul.opendev.org/t/openstack/build/58690182ce0a43fe9525738eced4b1c1 : SUCCESS in 59m 59s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/937f6924c46d4a88bb0ab0651b45e791 : SUCCESS in 54m 49s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/65512fd18d6a450ba84bce0423f6f60a : SUCCESS in 1h 37m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/60291b41cb90472d86b9a605ffc9d54e : SUCCESS in 35m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f8e981f677440baa6be1f39ff9331ed : SUCCESS in 3m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b36215a6d2314029a35ea7e53059b605 : FAILURE in 5m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/734ccda48877408dbbff4160285fa8bb : FAILURE in 7m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b28577ef45834dfeb1e69c4c3241cc08 : SUCCESS in 6m 23s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a0b36bccab2e48e3bed67e5aa81e9de9 : SUCCESS in 6m 45s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/a8382c77b16a4ebab18ea9d630759c9c : SUCCESS in 6m 42s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/6dd2e5e7cab04e0f93600565d06f06dd : SUCCESS in 1h 59m 51s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/6002f15451bf4a61ae0645d5709e27ba : SUCCESS in 1h 08m 57s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/a6d809f4256e44a9a6deee0691c69d70 : FAILURE in 1h 47m 44s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/d676fee49dae4a689c482f508fcc3021 : SUCCESS in 2h 31m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0da32cdd31ce4cb9b8b4ea3fcb92f81f : SUCCESS in 1h 01m 10s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/457cecf4b5294773ba29f422d0a82fea : SUCCESS in 1h 01m 06s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/e0dd1971a42c4a00b9f9006fbce4965d : FAILURE in 21m 35s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/51dc96780d4a476eb2191ac35a30bd3a : FAILURE in 11m 51s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/6d29649b958349ebb541793de3879496 : SUCCESS in 18m 10s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/b849975ea1814651898773e3ce334a7c : SUCCESS in 1h 18m 22s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"73904fe8fbc28391143a7f70d5bc8a3ef9bfa876","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-12-14 05:07:34.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"0fc24549ebb25175f7fe4c209af8eff5c2b15631","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-12-14 05:08:07.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"db49cb4bffe3a5edb2566f5ee06d0da0b0a2b86c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-14 05:17:34.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cece48558daa48bf817b6a8d899f97d6\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b65f5c0b0d3740658c734d9ad292c7eb : SUCCESS in 6m 32s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c33b9f4f82f14cb69d5916cd7c586d1f : SUCCESS in 6m 51s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/07c8f8f4c1b64ce8b39bbd0200048b41 : SUCCESS in 8m 19s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"498f4f840b623bfccc773476ab40ea91522f212a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-14 07:15:38.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc38232e7396485194dbb6c4018b299a\n\n- grenade https://zuul.opendev.org/t/openstack/build/8bc6367a104240a2a1b270e7261db555 : SUCCESS in 57m 09s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/282e9a84f9da406db0ab50f44ed3c8f2 : SUCCESS in 58m 37s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8563c5e1b1f74b66a84e31cb35b1ac2f : SUCCESS in 2h 02m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/07b7b14d1dad4f7a9ba6b764af0fb121 : SUCCESS in 1h 02m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3eaa3643e04e464eac0fbde4e9824e8d : SUCCESS in 5m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/58a65c25fcbe429d83d7a5a366ddda57 : SUCCESS in 4m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/73aa8bb9801d41758cfecaf9cba7d566 : SUCCESS in 5m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b58dd4fb864a4abe8cf5eb018c0f4405 : SUCCESS in 5m 13s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b772ced5fa494a1abed52f07afff62c8 : SUCCESS in 6m 04s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/d45cb955f6724133bd35df575e39a4e1 : SUCCESS in 10m 41s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/9087e0f3c7ea4a55b943d55652743ffa : SUCCESS in 1h 52m 34s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/5942bfbf0fbb4921817e743110312ab1 : SUCCESS in 1h 22m 44s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/1ae56b7c3cc245188bf3fb7e85f40e0b : SUCCESS in 1h 58m 29s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/6ce44d633bd04f71a9a35ed14465c7e5 : SUCCESS in 1h 44m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9bbedc5d7fe34b3ca2c37103556cdda5 : SUCCESS in 28m 44s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/16e78158623e4ac28c483911fd3f1313 : SUCCESS in 56m 28s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/998a1eaff0ea438990d300debe6d77d7 : FAILURE in 21m 55s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/dff79c77a736474b968e83f25e1ccb64 : FAILURE in 25m 59s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/2c3b9a1769d64876b5a1a115a6eadf72 : SUCCESS in 23m 05s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/facefd7cc9024355b9d8b06b3a4b25b0 : SUCCESS in 1h 13m 32s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ef5f4bc1bc169c24c4cc9e566d071400c250dcd6","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-01-05 15:15:26.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"825fb0237a171b2ca5f1631ee972db3dbccd4b1d","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-05 15:38:16.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"0d94e99c53f9ff86f0b589a434aca80db2931e5d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-02-05 14:00:42.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"aafbac73f6b554a28ff1e21b26519d7e2e2a6e28","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-05 15:30:52.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/23958441ce3d4852b22b0f6460bd2980\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ad44faeea3c140e28b7724a8c7b4a839 : SUCCESS in 8m 55s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e34212f6c95b43fe81e325f92c55d2d7 : SUCCESS in 10m 11s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/fd3b3819e8e94adbad091e4aaf1fe5ce : RETRY_LIMIT in 4m 21s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"e3d5bcf8e866a42cf3f1727d2b0ca7a928c96ba7","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-02-05 15:32:44.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"d55adb6442eaa26041c4eac87b11e0795a07bd89","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-05 16:31:03.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/99eeb69e37fd43a38d41723e6400c8b6\n\n- grenade https://zuul.opendev.org/t/openstack/build/8bb0c300c3d5462cb18abc32f1405843 : SUCCESS in 58m 49s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/19087dc156d6499ba7273fb541b9f8ea : SUCCESS in 1h 07m 23s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a4af40440bdf4d13a396591f56f35de1 : SUCCESS in 1h 33m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5307cefeec5a414388ab3de62df74d2b : SUCCESS in 36m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03e6ac897b8e4261a4e6ae6c7db0812a : SUCCESS in 4m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a684b1f021764633843250b434b4da7a : SUCCESS in 3m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8a80faeb8a8749029b6eb6ad6fa1f853 : SUCCESS in 3m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e7a0d275cc24c5c91186f9134851dbc : SUCCESS in 7m 09s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e2bce8193dab4c4496ce922704c4dd90 : SUCCESS in 5m 46s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/828546d91b884fe2aa7b4e89e9e72cc3 : SUCCESS in 8m 19s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/707b930681c64ae78a4ee2706f36a12d : SUCCESS in 2h 01m 48s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/2875e737ece24b0ea48e8adbd2dbc2e8 : SUCCESS in 2h 02m 49s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/ab8d952fb2ba41caad190b62f0d961ad : SUCCESS in 1h 45m 33s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/7d27aabe6c174c9282bc69af61a644f8 : SUCCESS in 2h 20m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9314ee29eda2418fa97136de06729524 : SUCCESS in 56m 20s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/8cc66de5838846999258387646176792 : FAILURE in 1h 46m 14s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/1d819247be69449ca55cbbb535219af0 : FAILURE in 21m 09s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/a2d0dfe41db347cfb3e081822af4e9b6 : SUCCESS in 1h 44m 06s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/b6ea719d58bc474980eb5cc83c7590df : SUCCESS in 25m 02s\n- ironic-tempest-bios-ipmi-direct https://zuul.opendev.org/t/openstack/build/46b15687a9e4444aab8c6d3829f40973 : SUCCESS in 1h 12m 06s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"d5dba763df9d02bea7e4aae09f7f18f0981ce06c","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-02-06 06:19:45.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"cac1cc3d1a8e117324ebb51fc49ad3146881489e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-06 06:37:23.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0c5b26bfc98c4bafb03f70b78f73cc5e\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1d34c1f8f61f483385f417f6bc68dfe8 : SUCCESS in 6m 54s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f8c6ce2e22c24b2b9312db5a7b210377 : SUCCESS in 7m 32s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/286561298044428e96c20c8512d001b1 : RETRY_LIMIT in 3m 35s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"530eb66d5555b6187e1c1ae9938550ff671f7876","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-06 08:45:00.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/73c61f9715f44ed4817aafcacf0d9960\n\n- grenade https://zuul.opendev.org/t/openstack/build/cfd7e2a096734ddbb9974f696c89e17a : SUCCESS in 57m 07s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/12d2572364e74af2bf080b90e312aecc : SUCCESS in 1h 03m 51s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e432805e239f46a0b4cbcb79501422df : SUCCESS in 1h 19m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3b171d0d3d25420ebc18ee0d033582b5 : SUCCESS in 1h 05m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76ceb1dc41684dbba9d014d7f8471831 : SUCCESS in 4m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f1e6deb67bcf48ad9b30bfb44b93b73f : SUCCESS in 4m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4a24e422eb474063819b47c66618982b : SUCCESS in 5m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80b00c33b39444f4b223318e724faf0d : SUCCESS in 5m 15s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/01292098eebb4290bfdeac50748cc30f : SUCCESS in 4m 21s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/f5ec9cc3e10e47f0bec19ac65be155d7 : SUCCESS in 6m 31s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/939c8064d6054310aef3ce1be9f46d17 : SUCCESS in 1h 02m 58s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/6354e886aa0e457eb494c72fca346b16 : SUCCESS in 1h 28m 07s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/0871dbb25ac54ad1b4f298c6a16d1a82 : SUCCESS in 2h 16m 24s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/fbd8b730cab14636a154f6d204862f92 : SUCCESS in 2h 17m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/65ba2c40b05040efb04c22347b1a6b3c : SUCCESS in 1h 06m 15s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/adbdcf8b9ac344ac8b94acf4d80b0085 : SUCCESS in 2h 07m 36s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/390cc921ef3d47d68325014e9e0cb113 : FAILURE in 17m 13s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/a6aed8e0c0a144aa8f0cc48158d2b6a5 : SUCCESS in 1h 51m 43s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/7e50f32811a64e929e41370871e320e0 : SUCCESS in 38m 40s\n- ironic-tempest-bios-ipmi-direct https://zuul.opendev.org/t/openstack/build/b70d6871f22d4a4882ae4fb8a8c6bdc7 : SUCCESS in 1h 15m 43s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"7c0058f033670c9cb5929f81083936c653c4b056","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-06 08:45:27.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"e3e34b0746b42f57d3841439d2ee2390c3c69a20","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-06 10:55:41.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0790a574414e4726a1f5a80cab997edf\n\n- grenade https://zuul.opendev.org/t/openstack/build/c4e019187b9648cfa67e69e7c2c4bd29 : SUCCESS in 1h 01m 35s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/620ec179efbd4e70ac9ff621640d5ebb : SUCCESS in 57m 29s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/77eafbf4043b495ebd113213ab4177a7 : SUCCESS in 1h 13m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/10764804199945b696a0af672606efa4 : SUCCESS in 1h 01m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/453aee1fd03c4c9ab0d8835e204551b9 : SUCCESS in 4m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/44fa5bfce1ae4fff8b49e20413ec3892 : SUCCESS in 4m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5a42162141804ad19135a76a1ad674b5 : SUCCESS in 5m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/063394e6eb31443e93c99178e47f7008 : SUCCESS in 5m 14s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5471486fd07b4c788ea5f72ed8cda179 : SUCCESS in 6m 25s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/26f92f561bf740688659ec7f0bc66c78 : SUCCESS in 6m 57s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/daa45dfe6fc9424a8e04366f870a511b : SUCCESS in 1h 28m 58s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/bbcbbf70257343b0b34faff2ff3203ff : SUCCESS in 1h 58m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b76daf8fa2a74b3094fcbc0f3a8fd98f : SUCCESS in 46m 54s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/8b813f445a5745dfa1841de758dbe83c : SUCCESS in 2h 05m 01s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/22a16fe34a064a3b8f496fd496e53e1c : SUCCESS in 15m 28s","accounts_in_message":[],"_revision_number":4},{"id":"89e601d7584080c01546a2252d7b59e1127cc737","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-06 10:55:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"e113c5f29b058e4ae35f9d89f945ad0f1159efb0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-06 10:56:49.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a9ba3ca6b95347648d4de63913ef8b2a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/709bbb281c2c41faa3174822c00470df : SUCCESS in 47s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"f06d130ff1849ee706b8fb56c359a7c7147a3bc0","revisions":{"7a108e8dbdf90806da5276853f9235c0c8f8a404":{"kind":"REWORK","_number":1,"created":"2025-07-06 14:49:24.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/90/954190/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/90/954190/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/90/954190/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/90/954190/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/90/954190/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/90/954190/1"}}},"commit":{"parents":[{"commit":"dc19f3817226017e88f9f139d024957b0e70ddeb","subject":"Merge \"Fix incomplete migration to datetime.fromtimestamp\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/dc19f3817226017e88f9f139d024957b0e70ddeb"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-07-06 14:44:07.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-07-06 14:49:11.000000000","tz":540},"subject":"Drop legacy arguments from RequestContext","message":"Drop legacy arguments from RequestContext\n\nThese were kept mainly for unit tests, but can be now completely\nreplaced by the native arguments.\n\nChange-Id: I19ff02322cad117b7d21d7ccf1709b1f259fd383\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7a108e8dbdf90806da5276853f9235c0c8f8a404"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7a108e8dbdf90806da5276853f9235c0c8f8a404"}]},"branch":"refs/heads/master"},"180aa6818ebdaf9dd18cf52270905d9e7a243a99":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-12-13 19:22:51.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/90/954190/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/90/954190/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/90/954190/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/90/954190/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/90/954190/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/90/954190/2"}}},"commit":{"parents":[{"commit":"06b80afa970510bb314aebb01657efef1844ded7","subject":"Merge \"Document Uwsgi support for production\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/06b80afa970510bb314aebb01657efef1844ded7"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-07-06 14:44:07.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-12-13 19:22:51.000000000","tz":0},"subject":"Drop legacy arguments from RequestContext","message":"Drop legacy arguments from RequestContext\n\nThese were kept mainly for unit tests, but can be now completely\nreplaced by the native arguments.\n\nChange-Id: I19ff02322cad117b7d21d7ccf1709b1f259fd383\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/180aa6818ebdaf9dd18cf52270905d9e7a243a99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/180aa6818ebdaf9dd18cf52270905d9e7a243a99"}]},"branch":"refs/heads/master","description":"Rebase"},"436a9b3fdc865423d6d2e66af702f7c8e7f389a4":{"kind":"REWORK","_number":3,"created":"2025-12-14 05:07:34.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/90/954190/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/90/954190/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/90/954190/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/90/954190/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/90/954190/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/90/954190/3"}}},"commit":{"parents":[{"commit":"06b80afa970510bb314aebb01657efef1844ded7","subject":"Merge \"Document Uwsgi support for production\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/06b80afa970510bb314aebb01657efef1844ded7"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-07-06 14:44:07.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-12-14 05:07:25.000000000","tz":540},"subject":"Drop legacy arguments from RequestContext","message":"Drop legacy arguments from RequestContext\n\nThese were kept mainly for unit tests, but can be now completely\nreplaced by the native arguments.\n\nChange-Id: I19ff02322cad117b7d21d7ccf1709b1f259fd383\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/436a9b3fdc865423d6d2e66af702f7c8e7f389a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/436a9b3fdc865423d6d2e66af702f7c8e7f389a4"}]},"branch":"refs/heads/master"},"f06d130ff1849ee706b8fb56c359a7c7147a3bc0":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-02-05 14:00:42.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/90/954190/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/90/954190/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/90/954190/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/90/954190/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/90/954190/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/90/954190/4"}}},"commit":{"parents":[{"commit":"ff1a8dd978334389915f472631a42415e9059d2e","subject":"Merge \"Raised NotImplementedError instead of returning\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ff1a8dd978334389915f472631a42415e9059d2e"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-07-06 14:44:07.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-02-05 14:00:42.000000000","tz":0},"subject":"Drop legacy arguments from RequestContext","message":"Drop legacy arguments from RequestContext\n\nThese were kept mainly for unit tests, but can be now completely\nreplaced by the native arguments.\n\nChange-Id: I19ff02322cad117b7d21d7ccf1709b1f259fd383\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f06d130ff1849ee706b8fb56c359a7c7147a3bc0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f06d130ff1849ee706b8fb56c359a7c7147a3bc0"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
