)]}'
{"id":"openstack%2Fglance~954191","triplet_id":"openstack%2Fglance~master~I8a59d609fadf5da5d549e1fc8008e44314c0177d","project":"openstack/glance","branch":"master","topic":"bug/2116041","attention_set":{},"removed_from_attention_set":{"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2026-03-17 16:36:59.000000000","reason":"Change was submitted"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-03-17 16:36:59.000000000","reason":"Change was submitted"},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2025-09-03 15:23:22.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I8a59d609fadf5da5d549e1fc8008e44314c0177d","subject":"Remove non-direct dependency","status":"MERGED","created":"2025-07-06 15:01:01.000000000","updated":"2026-03-17 16:38:05.000000000","submitted":"2026-03-17 16:36:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"954191-bug/2116041","meta_rev_id":"d887b1ecd41e8d4aaaca5c04f8b5af6f75ccfd48","_number":954191,"virtual_id_number":954191,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-17 16:36:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2026-03-17 15:23:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2026-03-17 15:23:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-06 15:04:20.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2025-07-06 15:22:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-07-06 17:27:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-07 15:49:38.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"635c40b30e419cf7b04581e2ebb0cd75ef2e743f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-07-06 15:01:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0a4d44046a04c7374d9530f26e99e782ec38f208","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-07-06 15:04:20.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a384261121cb3524ae8610e8319291aadfdd8344","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-07-06 15:09:52.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":2},{"id":"b1cca6d27a199640d10e79392fd8263c2399bc18","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-07-06 15:13:04.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b1255217ad107238d813c0a912aa45a1a8abd05f","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-07-06 15:15:09.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"c45bb13aebd2fac8f209981ac6e6909139d33fba","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-06 15:22:36.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/80444cb2117d4391b74f1508aa27f04c\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0458ea5047bf470a85f88bc427d39f58 : SUCCESS in 7m 47s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b6645fae636b4256a534c98d2dd2f936 : SUCCESS in 6m 19s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/741fc98982b04817be60b8fbbd467fb6 : SUCCESS in 10m 59s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d6ea2a2f5944efc36b8ff37904113f30bf79b4b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-06 17:27:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a81cf6e538a84ab09fefd3606701d7a3\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/f4c7bb1f6d434e46aa8f50a1d32c55de : SUCCESS in 2m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/17349ef9b82d49639cdb5148363e4730 : SUCCESS in 45m 52s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/992cd14f945e4685a29c179c9ba1f4b9 : SUCCESS in 47m 41s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b09927d6a7c442da9efac1e0901705fd : SUCCESS in 1h 09m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c4df52a0d158453ab5d9952ce6da3235 : SUCCESS in 46m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/086e9f98f8404d7ab3cb2813350e0d0d : SUCCESS in 4m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ab990fa1a62e4b6d9a6ecf2995f50de5 : SUCCESS in 4m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4f465611a1064685938adbe3f85d7e4f : SUCCESS in 4m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3f1fe4dd5b8940b58bdcf8ec82577e33 : SUCCESS in 6m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/acefb344182447d8aca13251e2fbdf19 : SUCCESS in 6m 34s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/25a76dd336c2498da522619538c2bf33 : SUCCESS in 5m 27s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/dea734c3b464429c985f2dd9f1736be7 : SUCCESS in 11m 28s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/46c2f3b09e2c49228f9bdae677f9170b : SUCCESS in 1h 52m 02s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/2e756eff5c61484aa47f6477d8557142 : SUCCESS in 1h 38m 41s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/e48c51a511514308a66c622319552a6c : SUCCESS in 1h 07m 13s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/473ec8d3589f4af8824a09ead7b1a749 : SUCCESS in 2h 14m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b290a6eb03d048b79c50c6a5ff81455c : SUCCESS in 47m 04s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/58ac1fdecdd7460fb89668ef4a2afc2a : SUCCESS in 1h 43m 00s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/fbf9b4acdc3e4fa4b238cf334da70cf6 : FAILURE in 19m 18s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/413f3d0eb0d44b2c84eb72f031a76e13 : SUCCESS in 1h 49m 59s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/040c8ea84c49490aa60d3c617f4baf37 : SUCCESS in 31m 53s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/987dd36741a2479bb009a1985d312b25 : SUCCESS in 45m 15s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"554371e2f59e4a37e60e6b4f45a3a7fe4e8d60b2","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-07-07 15:49:38.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c1d1e9aa0ed8d6e84761041e18acb0ebe293789c","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-07-08 06:56:44.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8104c01acf25df22eea23f049e4f75f9a87e4b6c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-08 16:03:34.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"d8d965146605c09391f4c5663bc4a5b7c1f7095e","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-08 16:04:23.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a3f4983db39c05e6ab3609453e060c3dd0a7eca3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-08 16:05:16.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"bf82ca026506511bb1dc306c051f4b19e519d411","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-08 16:05:20.000000000","message":"Topic set to bug/2116041","accounts_in_message":[],"_revision_number":4},{"id":"cac52cfdd31f4bf5378dcd4ae43ac6fdea7a7b80","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-08 17:09:15.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/162cf3f1807c4447a387d47827d5a69d\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/3deee5229e6f421391ac70febd7cb8f5 : SUCCESS in 3m 09s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e9715b7cde1e4fa08c548df849a3ffeb : SUCCESS in 1h 00m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8c1e42db5dd14aabb50d7635bf7e62cc : SUCCESS in 46m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7445ef803a414ef897398b19d3d95b73 : SUCCESS in 2m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4e99a3a0893a4df386acdfa69f86dd33 : FAILURE in 4m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e08905595a984d629ecb779a15728b95 : FAILURE in 5m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6a1dbabc545c41cda6974e293ce7cb1a : FAILURE in 6m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b9ea01e2b804bda9223013f0204034c : SUCCESS in 2m 59s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5463effd32a049789f4b14e406373399 : FAILURE in 3m 05s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/96f2f21af80f4616b444adf10f8aee0c : FAILURE in 6m 27s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/9f2a815d705444ab8b1b50c1c3b79004 : SUCCESS in 23m 27s","accounts_in_message":[],"_revision_number":4},{"id":"a0e6db4fbe4bd6e61de033dac8ab3fd1ced725ac","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-08 19:45:10.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb61aec994ed445e8f24231c8484daf5\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/934576397bd74dbd846526a79362d0bd : FAILURE in 9m 28s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/fc962dc9cb2e4af18bae80644c726edc : FAILURE in 11m 27s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/50c848e4fd524a6a8da31f7a9a31b9e4 : FAILURE in 9m 44s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"6613389022624a38dbfa2e84d2b1eb5428a1707e","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-09-03 14:45:23.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"882cc2ff8f905d8bb6c85df8fdd88cc57e4e1fa2","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-09-03 15:23:22.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"9db94e12689d16008ca835381fe35e06b78bb779","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-03 16:25:53.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/03f9080ce10a4d3fb197d0d3d62b7070\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/fb91b84aedda48e18755a0e7759f94f3 : SUCCESS in 4m 13s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/fe932f5a8f494ab78d36b22d7de3432b : SUCCESS in 57m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/83566a84a14c4706b247d7db3076e56d : SUCCESS in 56m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36f782ab5141470f992cd61b0bdbaa01 : SUCCESS in 5m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/65a3c82a465c470eabc8d8c2fefbfc3c : FAILURE in 4m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ada162a0c5d94c95af2feaba23602223 : FAILURE in 6m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/67d363e5826243deb0a428effa63bf32 : FAILURE in 10m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b0a6cc5e80d4b36a96d739d854f60e4 : SUCCESS in 6m 01s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/592de1fca83a4063b154bd0542ce2e96 : FAILURE in 5m 04s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/adaafd1aa4154bf0874c43fd2e050a21 : FAILURE in 7m 08s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/d00be92ef87642229494c679dffa8ba2 : SUCCESS in 33m 29s","accounts_in_message":[],"_revision_number":4},{"id":"7f55571e50453266a8c471f1caa84989009faa1d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-03 18:53:59.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d01c91c4a86e45b69f73d48ffa12576c\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0c18483a01434776aebb5a18ad5432fd : FAILURE in 8m 12s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4075bdcbe5ed4a80b66b3ba2c9dbf7fc : FAILURE in 11m 39s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4b0f0f1cc0bd466eb3069f08c2b77708 : FAILURE in 19m 41s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"a95d78c37d29031ee36f0c0219e652219f2d497b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-03-08 07:07:17.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"2661600261234cfe4d37b2ca286f6d4dd9103cb8","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-03-08 07:07:35.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"1b1f69b4731490fce9a1e8a8bfad44977d07a687","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-03-08 07:08:20.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"e67b9ca20290b0a74064baa0255c2d510d715bf5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-08 07:19:56.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9cafdf0810c34af49dc5e5d35f45644f\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2e794b04bb9f4b368b48d6bd38e1271b : SUCCESS in 6m 17s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a7e414e81bb34d0a92eadfdc10d8f1cf : SUCCESS in 6m 52s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/5b61ed06cd1845ada367a0ada96c85ff : SUCCESS in 10m 51s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"bd8bfd612378988f2401e7f1e6bffe2ee05b447e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-08 08:11:53.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8aed76389dde4364ab82a9d70e982014\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/a95cde3fc9a94ba4921be14953c96a6b : SUCCESS in 2m 51s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0bf71d8c11c341bca664d89ac4ad8985 : SUCCESS in 57m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/71cb0cc706b54c5bbd0223a6c692b87d : SUCCESS in 59m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef3c3c71cf084f4e93e81e17937d0903 : SUCCESS in 6m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f4fdb9ea64c4432b894845bf0cd863db : SUCCESS in 2m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/651965f3ed994dfbb6cef11dfd531bc4 : SUCCESS in 4m 58s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/da0797a716d7478c92970b47900c6963 : SUCCESS in 7m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30eed10be60741c79c0da372413e43ea : SUCCESS in 5m 37s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/728742dcdd854a0f8ea0abfa245daf46 : SUCCESS in 7m 26s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/ba59a466a10a4d8bb1cc28ad834f6a54 : SUCCESS in 6m 11s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/958b8923de884cb7b588ea8d8184f5c8 : SUCCESS in 19m 36s","accounts_in_message":[],"_revision_number":6},{"id":"1c2fc665d3bc372d4bf98b7fbe140486cb49de51","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-03-17 15:23:32.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"2dc6a3d5b74b415b745e846d4a06efb9fe644d10","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-17 15:24:15.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"da5fd5eda56eab323fff96fe1d25df359a3211c8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-17 16:36:59.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f866623ade0a42858af4ace498049970\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/10fae54832a744879ddb1ae511eedcc2 : SUCCESS in 4m 30s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2da8aa50910743f1be605a78dc2f6b4c : SUCCESS in 1h 07m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b1c4dc4ce7294ea390775c21ffa876d7 : SUCCESS in 54m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a198bb7b3f84cc8a98704d44512cca0 : SUCCESS in 5m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/899d14e6ce7f47f8817c55d536099a3a : SUCCESS in 6m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3e02a2be81eb415bae17c4220b240095 : SUCCESS in 5m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7cdd2af075a48dc94515b23268ea4af : SUCCESS in 6m 28s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/18550c72c7d54d44acdc3131858aecc4 : SUCCESS in 7m 04s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/0a258443eee34c54a1fc9f0c86a9c1b4 : SUCCESS in 7m 06s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/073edc3051524abca93bd160a65b8027 : SUCCESS in 36m 30s","accounts_in_message":[],"_revision_number":6},{"id":"a171262e6245f496c494337615c9d25918a07082","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-17 16:36:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"d887b1ecd41e8d4aaaca5c04f8b5af6f75ccfd48","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-17 16:38:05.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/15c8b043abf641a4968c9fda71300c54\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9855a5f62ca49bfb46cef0d901f2219 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"744ffe624dac5b37296f18b8188d585ebebad24f","revisions":{"35cd524a3b58a8a0258ef4ae3e476bef6c05593d":{"kind":"REWORK","_number":1,"created":"2025-07-06 15:01:01.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/91/954191/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/91/954191/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/91/954191/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/91/954191/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/91/954191/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/91/954191/1"}}},"commit":{"parents":[{"commit":"dc19f3817226017e88f9f139d024957b0e70ddeb","subject":"Merge \"Fix incomplete migration to datetime.fromtimestamp\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/dc19f3817226017e88f9f139d024957b0e70ddeb"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-07-06 14:57:30.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-07-06 15:00:52.000000000","tz":540},"subject":"Use common wrapper for retry upon db dealock","message":"Use common wrapper for retry upon db dealock\n\nThe oslo.db library provides its built-in wrapper to implement retry\nwhen db deadlock is detected.\n\nChange-Id: I8a59d609fadf5da5d549e1fc8008e44314c0177d\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/35cd524a3b58a8a0258ef4ae3e476bef6c05593d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/35cd524a3b58a8a0258ef4ae3e476bef6c05593d"}]},"branch":"refs/heads/master"},"5d4140384721a00a3932f8e8057d4475be7adc8c":{"kind":"REWORK","_number":2,"created":"2025-07-06 15:09:52.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/91/954191/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/91/954191/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/91/954191/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/91/954191/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/91/954191/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/91/954191/2"}}},"commit":{"parents":[{"commit":"dc19f3817226017e88f9f139d024957b0e70ddeb","subject":"Merge \"Fix incomplete migration to datetime.fromtimestamp\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/dc19f3817226017e88f9f139d024957b0e70ddeb"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-07-06 14:57:30.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-07-06 15:09:40.000000000","tz":540},"subject":"Use common wrapper for retry upon db dealock","message":"Use common wrapper for retry upon db dealock\n\nThe oslo.db library provides its built-in wrapper to implement retry\nwhen db deadlock is detected.\n\nA similar migration was done in the other projects. (For example\nfc851521bf407eb1a16f542c468003173435c1ee was done in nova.)\n\nChange-Id: I8a59d609fadf5da5d549e1fc8008e44314c0177d\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5d4140384721a00a3932f8e8057d4475be7adc8c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5d4140384721a00a3932f8e8057d4475be7adc8c"}]},"branch":"refs/heads/master"},"3ade1fba0f3b9630cc7cf2d14339f6aee5420900":{"kind":"REWORK","_number":3,"created":"2025-08-08 16:03:34.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/91/954191/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/91/954191/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/91/954191/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/91/954191/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/91/954191/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/91/954191/3"}}},"commit":{"parents":[{"commit":"4bf2ab7c3ed66c2f6b616c7a9b1e207e98a3ba71","subject":"Merge \"Remove LegacyEngineFacade\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4bf2ab7c3ed66c2f6b616c7a9b1e207e98a3ba71"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-08 16:02:15.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-08 16:02:15.000000000","tz":540},"subject":"Remove non-direct dependency","message":"Remove non-direct dependency\n\nThe retrying library is no longer a direct dependency since\nthe commit 4653c3e338bb626cc5b50ed25e33a8968379cd0b was merged.\n\nDepends-on: https://review.opendev.org/c/openstack/glance_store/+/954192\nChange-Id: I8a59d609fadf5da5d549e1fc8008e44314c0177d\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3ade1fba0f3b9630cc7cf2d14339f6aee5420900"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3ade1fba0f3b9630cc7cf2d14339f6aee5420900"}]},"branch":"refs/heads/master"},"3aa91ba8c75ec055d64310a5d91f18c58ed345d3":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2025-08-08 16:05:16.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/91/954191/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/91/954191/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/91/954191/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/91/954191/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/91/954191/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/91/954191/4"}}},"commit":{"parents":[{"commit":"4bf2ab7c3ed66c2f6b616c7a9b1e207e98a3ba71","subject":"Merge \"Remove LegacyEngineFacade\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4bf2ab7c3ed66c2f6b616c7a9b1e207e98a3ba71"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-08 16:02:15.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-08 16:05:16.000000000","tz":0},"subject":"Remove non-direct dependency","message":"Remove non-direct dependency\n\nThe retrying library is no longer a direct dependency since\nthe commit 4653c3e338bb626cc5b50ed25e33a8968379cd0b was merged.\n\nRelated-Bug: #2116041\nDepends-on: https://review.opendev.org/c/openstack/glance_store/+/954192\nChange-Id: I8a59d609fadf5da5d549e1fc8008e44314c0177d\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3aa91ba8c75ec055d64310a5d91f18c58ed345d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3aa91ba8c75ec055d64310a5d91f18c58ed345d3"}]},"branch":"refs/heads/master","description":"Edit commit message"},"4896ffa5d8958d7819b3de5183ba767d6c096db3":{"kind":"REWORK","_number":5,"created":"2026-03-08 07:07:17.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/91/954191/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/91/954191/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/91/954191/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/91/954191/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/91/954191/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/91/954191/5"}}},"commit":{"parents":[{"commit":"4bf2ab7c3ed66c2f6b616c7a9b1e207e98a3ba71","subject":"Merge \"Remove LegacyEngineFacade\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4bf2ab7c3ed66c2f6b616c7a9b1e207e98a3ba71"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-08 16:02:15.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-08 07:06:59.000000000","tz":540},"subject":"Remove non-direct dependency","message":"Remove non-direct dependency\n\nThe retrying library is no longer a direct dependency since\nthe commit 4653c3e338bb626cc5b50ed25e33a8968379cd0b was merged.\n\nRelated-Bug: #2116041\nDepends-on: https://review.opendev.org/c/openstack/glance_store/+/954192\nChange-Id: I8a59d609fadf5da5d549e1fc8008e44314c0177d\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4896ffa5d8958d7819b3de5183ba767d6c096db3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4896ffa5d8958d7819b3de5183ba767d6c096db3"}]},"branch":"refs/heads/master"},"744ffe624dac5b37296f18b8188d585ebebad24f":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2026-03-08 07:08:20.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/91/954191/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/91/954191/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/91/954191/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/91/954191/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/91/954191/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/91/954191/6"}}},"commit":{"parents":[{"commit":"8070fae771718eda7f626844ca84a949f77b2dd5","subject":"Merge \"feat: return size in get_image_data_iter\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8070fae771718eda7f626844ca84a949f77b2dd5"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-08 16:02:15.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-08 07:08:09.000000000","tz":540},"subject":"Remove non-direct dependency","message":"Remove non-direct dependency\n\nThe retrying library is no longer a direct dependency since\nthe commit 4653c3e338bb626cc5b50ed25e33a8968379cd0b was merged.\n\nRelated-Bug: #2116041\nDepends-on: https://review.opendev.org/c/openstack/glance_store/+/954192\nChange-Id: I8a59d609fadf5da5d549e1fc8008e44314c0177d\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/744ffe624dac5b37296f18b8188d585ebebad24f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/744ffe624dac5b37296f18b8188d585ebebad24f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
