)]}'
{"id":"openstack%2Fglance~955249","triplet_id":"openstack%2Fglance~master~I03da75b899b5da03c22ad3195a61343386ef9585","project":"openstack/glance","branch":"master","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2025-08-26 14:18:30.000000000","reason":"\u003cGERRIT_ACCOUNT_5314\u003e replied on the change","reason_account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2025-08-26 16:13:44.000000000","reason":"Change was submitted"},"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2025-08-26 16:13:44.000000000","reason":"Change was submitted"},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2025-08-01 15:08:13.000000000","reason":"removed on reply"},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2025-08-26 16:13:44.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I03da75b899b5da03c22ad3195a61343386ef9585","subject":"Replaced usage outdate egrep to grep","status":"MERGED","created":"2025-07-16 22:06:05.000000000","updated":"2025-08-26 16:17:06.000000000","submitted":"2025-08-26 16:13:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":13,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"955249","meta_rev_id":"38c9bed46a87cedd14d846e1b8a86fe790956236","_number":955249,"virtual_id_number":955249,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-26 16:13:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2025-08-26 14:18:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2025-08-01 15:08:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-08-03 11:54:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2025-08-26 14:18:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-16 22:09:30.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2025-07-16 22:09:51.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2025-07-16 23:00:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-07-17 00:22:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-03 11:54:18.000000000","updated_by":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"reviewer":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"state":"REVIEWER"},{"updated":"2025-08-03 12:09:26.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2025-08-08 17:06:12.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"}],"messages":[{"id":"a1aa450a426128683aa1d5e648e1811e48a31b6f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-07-16 22:06:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"96a4e6973fdc3e4d064dbc03d650397c2b55c3bb","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-07-16 22:08:56.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"3bff0bfd1eb9e5f7e57114aad2af8843f9c102b4","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-07-16 22:09:01.000000000","message":"Patch Set 2:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"083fa2e6a6f814d6b6ea6e2971bbb74d3df8dfec","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-16 23:00:19.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/297c47c2e19446378641ef5f8dccd639\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5cd195c6368341c098a1e94d54db915c : SUCCESS in 10m 31s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/28adaa22a953440f96982d362582eaa4 : SUCCESS in 8m 07s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/fb6130453c8b49098ed3a57390bdefe6 : SUCCESS in 11m 55s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"95e59235a36a5656448b7576304e5fc3db849f51","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-17 00:22:54.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/25bd7c83ce954107b8a8d3af2d8c9104\n\n- grenade https://zuul.opendev.org/t/openstack/build/45c2dfb38f494e12a9e57bb0ce0e124a : SUCCESS in 58m 41s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b2beb3a530d242fba89c07b93f359f31 : SUCCESS in 1h 08m 04s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f52cda5b050245a88bb220fe95b17b6d : SUCCESS in 1h 36m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8624425333874264882972644a718eb9 : SUCCESS in 57m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8fd2304ee72a4e0d899ec0d0553c13d1 : SUCCESS in 4m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c083f95ce43f4536952c359f4ee53ed4 : SUCCESS in 4m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/27dd1a6e757149908f1f5f2be4c5cd2d : SUCCESS in 2m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/34c0d33d81a44d2d86aeb7ee0a20a95a : SUCCESS in 7m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92fc3721cd744df580c6dd699f0db07e : SUCCESS in 3m 04s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b12827d5f0d64df78f0126ea080ec43f : SUCCESS in 8m 40s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/bcbaf119822547009f9eb6a0b26380bb : SUCCESS in 10m 06s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/e01b2eb4f3b9412486b2637bd50f61a9 : SUCCESS in 1h 43m 15s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/6a49adafb65045ffa38b067938f17398 : SUCCESS in 1h 55m 22s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f8624ac8d9c74f2aba1d9ff5f04258dc : SUCCESS in 1h 34m 43s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/1329b7762ef34acbb2425b2827127759 : SUCCESS in 2h 08m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0bd77bb6a1ca49d5a2f1d714e6ab8d2e : SUCCESS in 51m 55s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/46923eecd99c43049a8c247b07d9c092 : SUCCESS in 1h 27m 03s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/de0f0171f0b94b7fbfde13509dafeb7b : FAILURE in 15m 39s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/1398ae38e1f94258a34acc81eb711b00 : SUCCESS in 1h 43m 41s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/82e476311389440ebae2549250216319 : FAILURE in 32m 42s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/3cc1babc90f7469485d59d98be2e8db1 : SUCCESS in 1h 12m 19s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c24720030f415cd07b18a59e398987ddb50ba1c2","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-07-17 08:39:42.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a8dacea738c03612942a4cf920cb1fb6c7eac9c6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-17 08:56:31.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d8714657a5df41f0926ba50c72cb2b5a\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/928d9e5d29b749cb82952a188d832b8a : SUCCESS in 9m 11s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/970642d902ef422690b65993463e1177 : SUCCESS in 7m 22s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f0d7993ff9924f68b425e539df3ffcdd : SUCCESS in 14m 55s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1edbf9997641328fff6f26129668bb053846809a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-17 10:38:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/edc4400650374c688fc94024074e316a\n\n- grenade https://zuul.opendev.org/t/openstack/build/28303a5b7b4f44c5a24779e8358b6b14 : SUCCESS in 48m 37s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a5f26de9e6ab426ba564ca40bdc2edab : SUCCESS in 48m 06s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8e5afc5f8c2f4d74b2f73fe089223b0d : SUCCESS in 1h 19m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d83c1a21273c48b48dd385629f63a31f : SUCCESS in 56m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ead4a9895f17427aaaa3fa99f868f1b8 : SUCCESS in 4m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/013c06c57aa84c6aab90fa7b90a5b5c0 : SUCCESS in 5m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/31688e41c9c74d32a49d62a85824e352 : SUCCESS in 4m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/767e8b9b51364583b875ad0d9a37d28b : SUCCESS in 8m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ae586d497e6496594780a4d636685d1 : SUCCESS in 5m 13s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0a89ab5b907042a98e5c599ff893dbb3 : SUCCESS in 7m 20s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/482f53a086034ec3b6e6d94c52e8bddc : SUCCESS in 8m 01s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/625b2ef2ee554ef5a85d243ad2612b14 : SUCCESS in 1h 41m 31s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/1eab63f0627543339dc77676798f3014 : SUCCESS in 1h 56m 39s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/22a246003e7a43f897df1dee490db8c5 : SUCCESS in 1h 41m 07s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/e0200fc966004c03be5f1b383c016857 : SUCCESS in 1h 57m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f8fc7e3635ce43b690a8aa4de3d0cb26 : SUCCESS in 51m 40s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ed04ad51131841099f4a15afcdbeb883 : SUCCESS in 1h 33m 31s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/18de5036edf5499f9c05a8617e75a87f : FAILURE in 29m 50s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/254f25b75c884914806eb07773fd48d2 : SUCCESS in 1h 44m 19s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/e380023ec3ac4364aff06dbece9bb4bb : SUCCESS in 23m 29s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/9d5e60ffdb1c4232b4dc4178346a6862 : SUCCESS in 31m 54s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3300ce8e84f5e636ebcce584233b3c896eb775b8","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-07-17 14:43:44.000000000","message":"Patch Set 2: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"06c5e1c6fbca8334a692c3cc139a615a40255d60","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-07-17 15:09:26.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"12cc880f8edc23e9ac20004d5f70d93eaa2b5289","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-07-17 15:10:43.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"cd4e3dae3c27ed1c3b6fcba988eb1df740074284","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-07-17 15:10:50.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"06a8e44630c5cd401ef51246843ca654b489be2d","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-07-17 16:23:26.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"c56ed8f6aa4a358be1041d6f96335f2dc7419ee6","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-07-17 16:27:44.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ea487a5bd4f7d01757f1753e796800477ab68d39","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-17 17:16:59.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c51db6bb7da4ec488a9b5bddd1e4d11\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8ebe5974200a4cf2816d14d7db7102a2 : SUCCESS in 9m 12s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ad617156548340bc9aa16edcfc528a29 : SUCCESS in 9m 43s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b38337ef9aef42e58ecf8104fb9ffb68 : SUCCESS in 14m 57s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f7287b67dbe7e11ff906df450a4f09364f64f827","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-17 18:17:28.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a663cecd2e8e4c49b718825de0fa6897\n\n- grenade https://zuul.opendev.org/t/openstack/build/0cfc30e713c94019877a5cad06a6508b : SUCCESS in 53m 22s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/78099f6ccff04bf4b46569087a106e8d : SUCCESS in 58m 51s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ec6cb4f770b3483da5bd4e99e082782b : SUCCESS in 2h 04m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f1e8932d93464d7bb17aa07d2b024d25 : SUCCESS in 53m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d90e7cb8b9c24bfabc40a7e683d74b7c : SUCCESS in 4m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0c07f7605b1a47aaa3973d6f80f580f2 : SUCCESS in 5m 12s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/27544ff63ebc42769aca15aa8b90d518 : SUCCESS in 4m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c1ec8e43c01c4e49a0d332ffc9d29d71 : SUCCESS in 9m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3092991f10134d02a5a47871db106448 : SUCCESS in 5m 33s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1b5b1fa2957f416cb7cd7cfa3f6d9c7e : SUCCESS in 6m 13s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/d2811cc2120b4e48ac78af72f9b8c80b : SUCCESS in 10m 28s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/aaacc034e2934ec8a99c1051b04a71e8 : SUCCESS in 1h 33m 08s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/3a8cbecf97b54d56b68149e3aa893df7 : SUCCESS in 1h 51m 35s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/33debfa764684d809ad33f622a9d2aa8 : SUCCESS in 2h 22m 57s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/7a4e2c5f2d964c5eabf44f678845424b : SUCCESS in 2h 52m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f85e6b3a73ff43fd90a26e1e6b284e49 : SUCCESS in 1h 02m 45s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e4000d9b4b51468c8458707007d74943 : SUCCESS in 2h 17m 40s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/450e11ca781b45e3bc85035c55b40337 : FAILURE in 35m 00s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/fa7fa185f0e640c1b5d12fb9a0b4d175 : SUCCESS in 1h 51m 46s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/a76aebba72844b149b6e908a0e8f7eed : SUCCESS in 31m 56s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/3263418b94b04bae83d4e47259ff1463 : FAILURE in 33m 36s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"0a1a4c65ed3721177c3cccf8bf4a8695f14b21f8","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-08-01 14:54:04.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d4b349a55a84c6c60f715e3c31432a0e91acb86c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-08-01 15:02:11.000000000","message":"Patch Set 5: Published edit on patch set 4.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"98a9bf877b7dd304e063db4019607963bfb22d73","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-08-01 15:02:23.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"0155ecf71711c6b95998fdf917e278b26fe4c2c8","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-08-01 15:08:13.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"07ebb03392921ffc0d32aa982f864b58ebf37260","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-01 16:27:05.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8d1bb7889b9c49599698de32b5805fdb\n\n- grenade https://zuul.opendev.org/t/openstack/build/39114072c40f41a08384c44adc818f33 : SUCCESS in 32m 53s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d85c919da1734135b7b500c3a6190646 : SUCCESS in 40m 53s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7c2cb6452d6d40e0a875077ce02a3f39 : SUCCESS in 1h 23m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0db07df8383e4ef2b05456a278a88f2d : FAILURE in 36m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c796bbc1065479e893063e20fd529a6 : SUCCESS in 5m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/292c863ff5c34e909dbb19ca32b45d06 : SUCCESS in 4m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9aae2cded98f4a3a9765a856dd8ad393 : SUCCESS in 3m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3d2583aac28c42bd920bb5fa42b75f55 : SUCCESS in 8m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ce7269e53824446a1abc67b2b4c9b7e : SUCCESS in 5m 37s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d02940fc02094d2aa9f0a611ef5cdd2f : SUCCESS in 6m 34s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/16bf1b1ef3fc426790e4a9a19380ba30 : SUCCESS in 6m 58s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/900677a5de084020ae90128bde50045b : SUCCESS in 1h 06m 57s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/322c6d7e149e414594cb7a77f1f8624c : SUCCESS in 1h 12m 03s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/d8b638aba24349949c7ede8f9d794a76 : SUCCESS in 1h 12m 07s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/50f6c07a715644278331901d05a590e4 : SUCCESS in 1h 17m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7d60403a1ed049e2903af832a51b66dd : SUCCESS in 35m 25s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/1aeaf56e765541f5a02754874ddf03de : FAILURE in 1h 01m 05s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/bd6e40944c5b4ee382cc320948a2738d : FAILURE in 30m 57s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/9934cb01a6764c56a6c1bf71add67bae : SUCCESS in 1h 07m 12s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c61b844977b84bde9dff5789a727627b : SUCCESS in 20m 26s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/0d49bf53ee2a492aa7062a78e2a80ca8 : SUCCESS in 32m 14s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"dea6791d9d228eabd8dc9a67e09d4abc3c24a323","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-01 19:13:34.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dc60e5054b96450aaeec88b8562fb8d0\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/50e24a72c0c24a3698f19bdba05dcc32 : SUCCESS in 13m 07s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/36b645425f114cb0bc345400399bae67 : SUCCESS in 9m 19s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a3eca360543e45b3923cf83c77064354 : SUCCESS in 18m 12s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"82bc57f3f5205ecde7a237a26d07a2f6653590ac","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-08-02 00:53:08.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"fe35f9f65470b1c2ac66e208f5767d8e1cc2a3f4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-02 01:06:45.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d43c50b22174dec92fd0bd9da04945b\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/92b7fe061c7148fb9d08d7694c60a970 : SUCCESS in 10m 56s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/bd5ccf6b566f43eeadcf39024af0333b : SUCCESS in 11m 11s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b077936abcfc48fc82f7ef0667aabfbe : SUCCESS in 11m 32s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"7a5182bc6149569aba220994bffc7f44458377a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-02 03:08:56.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2d10d4e7d6304239881b7b3a366d4dbd\n\n- grenade https://zuul.opendev.org/t/openstack/build/00d594e487344ea7a4a8a8c1725c1c9f : SUCCESS in 55m 55s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5b3ab36175fa4043a56109663c1319b1 : SUCCESS in 49m 10s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/83201ac1b839461d8ab9754b624167b7 : FAILURE in 1h 24m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6c15a0d7af14457ea4c41ca6b30b052b : SUCCESS in 37m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c31b2e21fe5d41fd9ffeb75e8055db96 : SUCCESS in 3m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9fae5160a28c4af1b205f4d9446111c4 : SUCCESS in 6m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1c8f1089cfdb4f53b2200dbe325e7d4e : SUCCESS in 5m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/17619d9fb3f84d19909c260b6e52b03e : SUCCESS in 9m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/020d486950d1428984239f617af86582 : SUCCESS in 7m 27s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b84112fefb40468889a864bb72182cf2 : SUCCESS in 4m 27s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/35311142f4bc46b491b0f60214a604d9 : SUCCESS in 10m 52s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/bf047211d7b148e29f1285d2f4b63ef8 : SUCCESS in 1h 21m 58s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/3d7ed29f01ff4f82ae86ad975829a12b : SUCCESS in 2h 04m 47s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/480b91575d134f478aeba8ebee534687 : FAILURE in 1h 55m 22s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/4c6242b205b4430f9cfb16b08aff2d75 : SUCCESS in 1h 52m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a6a046a020f24298bf9ec9b06e650bb0 : SUCCESS in 38m 54s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a095b177284344cda322fc0fe81472bd : SUCCESS in 1h 51m 14s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/e8c2bd08267b47f1bf8eca5a1b029a21 : FAILURE in 16m 24s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/2ba07a9d6310437d8be06935b1d5f0ac : SUCCESS in 1h 29m 33s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/7c53851b158c4d9e98e7a0f588c6c676 : SUCCESS in 31m 53s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/f999b504963147608654566c64c77055 : SUCCESS in 27m 03s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"6f0028addffc69b22e39a50a9a3ce2b43e50aaf5","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-08-02 10:17:01.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"72f76abf82f46244dfb42a3ade2ec34cad4ee5a6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-02 10:34:01.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b1a8bfbd21044829d4a91a012ede01e\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1b4c1ae1a05b46c19de459339f77693e : SUCCESS in 9m 30s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/fe6f68715d09413dae284231ad674acd : SUCCESS in 7m 29s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/92edf067ec3a4fe0aaaf8e872471f997 : SUCCESS in 14m 50s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"28175efed2f3fb4b86f72075df800b0cd3fa5eca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-02 12:57:41.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b46554d147de46b8ae4f91e6f8229425\n\n- grenade https://zuul.opendev.org/t/openstack/build/b627bf48b0954aa18eda4f179f013af3 : SUCCESS in 59m 44s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d116ac4926af44e88facd37c1614d0b3 : SUCCESS in 1h 01m 34s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2f4eca2c91c441449d2c8c8fa8bc3bb3 : SUCCESS in 1h 26m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/826965c2ab51456ea559eac02c942b3d : SUCCESS in 1h 00m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/74c10b7e7dcf4bc68aa7a2c8b9db0d89 : SUCCESS in 4m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8b595970b8a440cb88cfe13667381964 : SUCCESS in 4m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3d46716c435a4421aa596774f5f5d516 : SUCCESS in 5m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/22698258cd804336b74aa8890cfdbb36 : SUCCESS in 7m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3583314d6db24b0b97ab73f9239416a2 : SUCCESS in 5m 28s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2102792f591c4b9bad0dbad925b59a2d : SUCCESS in 9m 49s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/9063ff949fd44a94a47675a200c0477d : SUCCESS in 11m 29s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/b97ff92d54ac4df2b0882de4cd38bb13 : SUCCESS in 1h 50m 16s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/74e463a97ed84f22a7a975106df46d0a : SUCCESS in 1h 49m 50s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/a7f3962521b44d3ba081c86438adbed7 : SUCCESS in 2h 06m 05s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/ab339b007c7d47139b1c144685f24ea5 : SUCCESS in 2h 32m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/93f4b18b126d42039ae6fa668b48c1f0 : SUCCESS in 56m 08s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b0b5c11216634812a5b1dccd5f01a33a : SUCCESS in 1h 55m 58s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/fd0f76c51eae498ab553196cc66d7380 : FAILURE in 21m 54s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/c46af3470cf444bfa1bcc990fb9c0a5b : SUCCESS in 1h 57m 05s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/97de44e5c2714e1388bb2c98b1add451 : SUCCESS in 29m 55s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/7e856dc7d40b4f3baad6af84c390e480 : SUCCESS in 47m 27s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"324b2e011cb08181fc6b53b8f6352027f6f658e5","author":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"date":"2025-08-03 11:54:18.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"1b829df3f7ed9917d2def9c9983a86c2d41ad385","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-08-26 14:18:30.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"b7ad38e56a5b003c8d25300e0036c8c40c45950f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-26 14:19:06.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"a9d07474f71158e43078a8f59f174804d5cef4bb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-26 16:13:43.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/58eaec54a0f74072956527fecedcfe45\n\n- grenade https://zuul.opendev.org/t/openstack/build/31b674564abd4b9888d61992b530035e : SUCCESS in 33m 23s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d23d1c78e03d419cab9964d05f694ff7 : SUCCESS in 1h 03m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0989553d7a3f461489bd3bce05440485 : SUCCESS in 1h 27m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/29480e54de2f460c9881b48d2a81f69d : SUCCESS in 1h 09m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ffd10e427114e339584a8ecab55de6a : SUCCESS in 5m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d3365eacf0e74ac08c61543a34fbd944 : SUCCESS in 4m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9ed2885b4aa441169a22d5e4579dfd21 : SUCCESS in 5m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e6023d59b8347c88bb9236edddd635b : SUCCESS in 6m 16s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f9f1c210c1634a27b61590ce0c8891fe : SUCCESS in 7m 43s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/af73e4584bc549309c53252e0c7e4322 : SUCCESS in 7m 53s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/b38aca61d1dd4fdf9f645744ccd57dde : SUCCESS in 1h 53m 40s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/dc24e5ef163c41a7965a902275f6ab7f : SUCCESS in 1h 23m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/348d1707ae704dc580478b010fb26b0a : SUCCESS in 43m 23s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f5dc506b6b6b4676b930c2dad75b8a4e : SUCCESS in 1h 49m 46s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/276be6b1dd0f4e34b4d06feec28b3999 : SUCCESS in 31m 12s","accounts_in_message":[],"_revision_number":5},{"id":"a85f360e8f4965fe579fadaa232898902b05f524","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-26 16:13:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"38c9bed46a87cedd14d846e1b8a86fe790956236","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-26 16:17:06.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5dc125629bea4bc990dca06f005fc6bc\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ff72fa71b094765b754caae937eb14e : SUCCESS in 3m 13s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"4bc50f7a2d7343b2c69da80dd60d6a4286c4a88b","revisions":{"24693f6d57c1abf7f1767fd6e3e59dcc50c6760b":{"kind":"REWORK","_number":1,"created":"2025-07-16 22:06:05.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/49/955249/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/49/955249/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/49/955249/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/49/955249/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/49/955249/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/49/955249/1"}}},"commit":{"parents":[{"commit":"56100a6113d6a592fbb54a6d8aa08715a22c67bb","subject":"Merge \"Drop redundant check of sqlite3 module availability\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/56100a6113d6a592fbb54a6d8aa08715a22c67bb"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-07-16 22:06:05.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-07-16 22:06:05.000000000","tz":0},"subject":"Replaced usage outdate egrep to grep","message":"Replaced usage outdate egrep to grep\n\nChange-Id: I03da75b899b5da03c22ad3195a61343386ef9585\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/24693f6d57c1abf7f1767fd6e3e59dcc50c6760b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/24693f6d57c1abf7f1767fd6e3e59dcc50c6760b"}]},"branch":"refs/heads/master"},"27e9c160748619f2a512d70e0d6f4f9929f2abbf":{"kind":"REWORK","_number":2,"created":"2025-07-16 22:08:56.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/49/955249/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/49/955249/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/49/955249/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/49/955249/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/49/955249/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/49/955249/2"}}},"commit":{"parents":[{"commit":"56100a6113d6a592fbb54a6d8aa08715a22c67bb","subject":"Merge \"Drop redundant check of sqlite3 module availability\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/56100a6113d6a592fbb54a6d8aa08715a22c67bb"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-07-16 22:06:05.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-07-16 22:08:56.000000000","tz":0},"subject":"Replaced usage outdate egrep to grep","message":"Replaced usage outdate egrep to grep\n\nChange-Id: I03da75b899b5da03c22ad3195a61343386ef9585\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/27e9c160748619f2a512d70e0d6f4f9929f2abbf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/27e9c160748619f2a512d70e0d6f4f9929f2abbf"}]},"branch":"refs/heads/master"},"83dd7f7820c3ef0b43a4b5c66d56efd069aa751f":{"kind":"REWORK","_number":3,"created":"2025-07-17 15:09:26.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/49/955249/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/49/955249/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/49/955249/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/49/955249/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/49/955249/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/49/955249/3"}}},"commit":{"parents":[{"commit":"56100a6113d6a592fbb54a6d8aa08715a22c67bb","subject":"Merge \"Drop redundant check of sqlite3 module availability\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/56100a6113d6a592fbb54a6d8aa08715a22c67bb"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-07-16 22:06:05.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-07-17 15:08:50.000000000","tz":0},"subject":"Replaced usage outdate egrep to grep","message":"Replaced usage outdate egrep to grep\n\nChange-Id: I03da75b899b5da03c22ad3195a61343386ef9585\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/83dd7f7820c3ef0b43a4b5c66d56efd069aa751f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/83dd7f7820c3ef0b43a4b5c66d56efd069aa751f"}]},"branch":"refs/heads/master"},"a6b7569479f3522a7971fa823e08f7004ad330be":{"kind":"REWORK","_number":4,"created":"2025-07-17 15:10:43.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/49/955249/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/49/955249/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/49/955249/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/49/955249/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/49/955249/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/49/955249/4"}}},"commit":{"parents":[{"commit":"56100a6113d6a592fbb54a6d8aa08715a22c67bb","subject":"Merge \"Drop redundant check of sqlite3 module availability\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/56100a6113d6a592fbb54a6d8aa08715a22c67bb"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-07-16 22:06:05.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-07-17 15:10:42.000000000","tz":0},"subject":"Replaced usage outdate egrep to grep","message":"Replaced usage outdate egrep to grep\n\nChange-Id: I03da75b899b5da03c22ad3195a61343386ef9585\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a6b7569479f3522a7971fa823e08f7004ad330be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a6b7569479f3522a7971fa823e08f7004ad330be"}]},"branch":"refs/heads/master"},"4bc50f7a2d7343b2c69da80dd60d6a4286c4a88b":{"kind":"REWORK","_number":5,"created":"2025-08-01 15:02:11.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/49/955249/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/49/955249/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/49/955249/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/49/955249/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/49/955249/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/49/955249/5"}}},"commit":{"parents":[{"commit":"56100a6113d6a592fbb54a6d8aa08715a22c67bb","subject":"Merge \"Drop redundant check of sqlite3 module availability\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/56100a6113d6a592fbb54a6d8aa08715a22c67bb"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-07-16 22:06:05.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-08-01 15:02:11.000000000","tz":0},"subject":"Replaced usage outdate egrep to grep","message":"Replaced usage outdate egrep to grep\n\nChange-Id: I03da75b899b5da03c22ad3195a61343386ef9585\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4bc50f7a2d7343b2c69da80dd60d6a4286c4a88b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4bc50f7a2d7343b2c69da80dd60d6a4286c4a88b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
