)]}'
{"id":"openstack%2Fglance~956086","triplet_id":"openstack%2Fglance~master~I491c486a0b967baba9bbdca4cd241ba691a94435","project":"openstack/glance","branch":"master","topic":"bug/2118756","attention_set":{},"removed_from_attention_set":{"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2025-08-29 05:34:17.000000000","reason":"\u003cGERRIT_ACCOUNT_9303\u003e replied on the change","reason_account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},"34510":{"account":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"last_update":"2025-08-29 09:53:20.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I491c486a0b967baba9bbdca4cd241ba691a94435","subject":"Fix inaccurate usage reporting when delayed_delete is enabled","status":"MERGED","created":"2025-07-29 14:59:38.000000000","updated":"2025-08-29 09:57:03.000000000","submitted":"2025-08-29 09:53:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"956086-bug/2118756","meta_rev_id":"c2b83e1c669376c9f4fdccffa764178490e8bf23","_number":956086,"virtual_id_number":956086,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-29 09:53:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":2,"date":"2025-08-25 12:36:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-08-25 14:57:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":1,"date":"2025-08-28 15:16:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-29 17:31:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-25 12:36:25.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2025-08-25 13:32:02.000000000","updated_by":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"reviewer":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"state":"REVIEWER"}],"messages":[{"id":"cbd06e1f7f0755ed7a33ba9c36d0b0c33b6c8b0d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-07-29 14:59:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7a0cfdac94c032a1573a87fc7988a46534810073","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-29 17:31:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca9de9c4ea4a41a0bf27028968b71697\n\n- grenade https://zuul.opendev.org/t/openstack/build/2fd87f9184134f2384488099f87f8e2e : SUCCESS in 1h 02m 21s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/49ee405dee7d4b4aa0af783c52a36a7d : SUCCESS in 51m 04s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/fbcf17f172704383ae9f195326394e70 : SUCCESS in 1h 21m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0d00f99ef0164a8aa32ae48445eae146 : SUCCESS in 58m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad4b196cb0dd49539fe933135c14ce89 : SUCCESS in 2m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/420b2d212d734050ab2abc3fed575d39 : SUCCESS in 4m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/028be4e6b84247809fb9edf772691e58 : SUCCESS in 7m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f76e8da8da644f5183f64557874b3df7 : SUCCESS in 7m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f703a63808514460a6bf2f4e814e89a0 : SUCCESS in 7m 48s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8e25c8b44827472ca49052957ddf8ba0 : SUCCESS in 7m 52s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/e7ebdb48bf79443c84165ace3adf087b : SUCCESS in 10m 23s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/cc7ca59c0b7f4be8a381a45bbefdec5a : SUCCESS in 2h 08m 26s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/1254634dffb84fd891caf480e4366b90 : SUCCESS in 1h 58m 28s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/958c8221ae7f41b8a352c40fddaa0172 : SUCCESS in 1h 41m 14s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/4c7d00b2792a425788dc0d74240d4cdf : SUCCESS in 2h 10m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/03cf74d6095846c9b59d8370665f9e7f : SUCCESS in 28m 47s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/6c94ff517f5a4ddeae8f6c3b3de08488 : SUCCESS in 2h 03m 50s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/c731a01a342342a0b915acf22d4d6f14 : FAILURE in 32m 49s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/a0c603b141714d97b8ed5335d731dd34 : SUCCESS in 2h 04m 20s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/b22877f324ef46ab93c611548514bd0d : SUCCESS in 26m 39s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/b33befcdd1d84836aa60c5c4d4ca0768 : SUCCESS in 1h 04m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"66cf99f6e7cbad01f0713bc26c556939eada4916","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-29 17:41:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a1f5a36a404e43d5bff837e79417003f\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/536a9149ec384358aed53e1563e4aa97 : SUCCESS in 10m 14s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/dc3f20c72a2e427883488447a83f2e09 : SUCCESS in 9m 36s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/8fb5287556804d04b34e27d88aa93a1c : SUCCESS in 16m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5b92beed2c327809c8712af601ef08eaa44a0221","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-08-25 10:36:44.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"7473d46b34523a95a6e3c94b60f103f68bf0cdb0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-25 12:12:11.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/92812686b7054ccfa56e2629452ba53a\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/724ef23d81ea4de69f7d0da357500300 : SUCCESS in 11m 33s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/65019a129da04e848891012ced867efd : SUCCESS in 13m 43s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/8be493d132914e95b72af049ecd38b18 : SUCCESS in 21m 01s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2691f85a6a2b68d74f9f9315c75aae6c703d4935","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-08-25 12:36:25.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"a141bb12908c00e244585d88297d11e0ffdc8298","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-25 12:38:26.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8c0a09ab16fb40619e3ffeb847647989\n\n- grenade https://zuul.opendev.org/t/openstack/build/1b425867c3f64a80955b222e0d4e3d03 : SUCCESS in 57m 16s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1d6c06314d8d4679bb8ce794a7992097 : SUCCESS in 1h 02m 32s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5ef3e5f9174a4cc1ace568f90a1746b4 : SUCCESS in 1h 19m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7476620e3a86409ea70f1a2a13433e4d : SUCCESS in 37m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8096e33915cf414eb03fbb2cc4037f77 : SUCCESS in 4m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/328fcceae4264ef68fc755e7d7ea97c3 : SUCCESS in 3m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/40fae46d4df14e9bac12841afe987bf8 : SUCCESS in 4m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a6b162a7852140a9ba646b726a245771 : SUCCESS in 8m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0316dcb6b65948fcb57fb361c3130c8c : SUCCESS in 5m 37s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/57024418fa384b388f016770756623ee : SUCCESS in 6m 43s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/e82c0a16279349998e870d34cf920b7e : SUCCESS in 10m 24s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/6da4d9a35ae74f71a3bf58600024b5ca : SUCCESS in 1h 32m 52s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/05c696bfb8954ddeaa73868bda954617 : SUCCESS in 1h 46m 28s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/3449b2b531774d4f9d7f71668c6e403a : SUCCESS in 1h 55m 32s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/93ed7355b0f34a258424b91b31333393 : SUCCESS in 1h 13m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dc896e51c1f244039bd211412244b86d : SUCCESS in 1h 00m 41s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/4302b5254bf64a28b3f7e4d42685504a : FAILURE in 1h 14m 06s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/ea8365de23b4419bbafd149032ca034e : FAILURE in 31m 32s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/ae91ebb93cdb4454a25f81f161ec74f9 : SUCCESS in 1h 52m 05s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/63015b64ef374f4e9efac6ac30e1583b : SUCCESS in 30m 04s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/9ca82115cd394195bd2623470a5980a2 : SUCCESS in 1h 05m 03s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7df2eee8f300b79a018d75cd4854990681da9645","author":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"date":"2025-08-25 13:51:36.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"13caacc856b3e43b8da212dbc5f8d7df14a8156c","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-08-25 14:02:43.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7a4dbec2c4545c5f78227872f4e6bd87c2c7a80f","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-08-25 14:03:44.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b47bb86fed1398fa77c5db87c6771a42aa3f4e56","author":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"date":"2025-08-25 14:55:54.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a6107bf98690e6158a03a36d6f291725b1a181ce","author":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"date":"2025-08-25 14:57:01.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"b3506b02532c878d9bb8f9b6142f3f99196534b6","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-08-25 14:57:23.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4bd57a60dd2f72bc85ed90e1d2d77d1847f9e0c2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-25 15:19:11.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b105a4aba1c4d35965268712ad62f92\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8a7efba338254748994e15e5464c3aa5 : SUCCESS in 8m 03s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/79a27e08f101414781730f7e27838185 : SUCCESS in 10m 56s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e5a3ca584b6f45dfbd4f9198ad700295 : SUCCESS in 14m 33s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"fab1d1ad008df622b50ce3609f846869a7d739b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-25 16:11:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/78d1afad43434796a851ad83059e84c5\n\n- grenade https://zuul.opendev.org/t/openstack/build/a79a35587890464ea9d5a04d23bf528b : SUCCESS in 1h 07m 25s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c2b78b6c93944fa39b5c00c2af530026 : SUCCESS in 55m 34s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2884096af7ed446789dc0b04176aa271 : SUCCESS in 1h 34m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fe9ee433c23d4fe69e061a6ddfce757a : SUCCESS in 1h 00m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53637ac24582466cbc01c63aaf7a4577 : SUCCESS in 4m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c07145aebd0b45b5ae8e67e713b86511 : SUCCESS in 5m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bb1b17177aee46fea36491cb2185dab8 : SUCCESS in 4m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/faa158751f8f42a2b380940d0ee6ed82 : SUCCESS in 8m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e253b45ca8f451b80f20d660e644ff8 : SUCCESS in 5m 32s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/62a911f3dd714a0bb5ad11d7feeb057e : SUCCESS in 6m 33s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/bbd65e8a1e7449dc98961234c690e55a : SUCCESS in 10m 37s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/6af98ecd7f404c9492413eed52977042 : SUCCESS in 1h 55m 35s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/e2cf0ce0c1404725bf0dec6dce7cb2ab : SUCCESS in 2h 01m 02s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/32899e66eea14808a4d59ef6ce6b47c5 : SUCCESS in 1h 36m 11s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/d857f6ce843f49cca8d45583c8e79e38 : SUCCESS in 1h 12m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/30a577116c5948bf8ed8a1c88203e846 : SUCCESS in 58m 41s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/921189d7a0b547ea8a99a56a19f83b5b : SUCCESS in 1h 51m 07s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/c3ef3a7c13f44986ab5d8f28619c4478 : FAILURE in 35m 11s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/51ec73da14d140428bdfcf4278498f4a : SUCCESS in 1h 44m 36s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/3efc91d5e4924704ae0f650c7ce5611f : SUCCESS in 17m 31s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/4229be7b1afd44c2abd0ec0016a916fc : SUCCESS in 32m 12s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"bfb9be89d6421aeabeace010e35a500c4fb5aac9","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-08-28 15:16:23.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b00754df3684a04509674e547ab82ea49e1121f4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-28 15:17:42.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"07a8682aec3f1fd891cd4cf9611977758f4b1e52","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-28 19:29:53.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/025cd44495b84c93beea3a84e17f99e9\n\n- grenade https://zuul.opendev.org/t/openstack/build/b6881e5dc4b84dde868a047d52f3df7a : SUCCESS in 28m 30s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b1fe7875f65e429a911ad5a6bd016d7f : SUCCESS in 43m 52s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1a5c00437dc74af2837cb8c5aaa8d635 : SUCCESS in 1h 20m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/25f2e1c4e55944d08d2dfb0ce9a518bf : SUCCESS in 32m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f680a79e03a473199d884d515370813 : SUCCESS in 5m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/01eb6af96af349ed963d712617519886 : SUCCESS in 4m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bcf2bccdb2a44b9da15c904f4b6a5b90 : SUCCESS in 6m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/154b8c4d820341719ed046d613a6a7ad : SUCCESS in 5m 59s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/64d3fd33690c4b65b9461bf20efc4598 : SUCCESS in 7m 19s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/d1ef41f2b0eb4c029985d0f643a879ca : SUCCESS in 12m 04s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/00e5702a6efc47148330ae7394d53dcc : SUCCESS in 1h 48m 08s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/2e3f80f442914415aa95a59547e4701b : SUCCESS in 1h 33m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c37975988b784a8d901de117962dd60c : SUCCESS in 57m 46s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/7cf2575bd5ab47d78f779da13fdba7ec : SUCCESS in 2h 20m 15s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c791024589974252b9cda46db7703510 : FAILURE in 31m 39s","accounts_in_message":[],"_revision_number":2},{"id":"018c8fa196adebae63eecbede266df67c17ebb17","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-08-29 05:34:17.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d70f120ed2d9dc6e43327a108c010dbfc59e88be","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-29 05:49:53.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd4991967e974dfe9fb308a8b82cc4f2\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/edda4caac6924069b62fec6bc3a805f3 : SUCCESS in 9m 24s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c704a6f29c364838b14865fa2f515802 : SUCCESS in 9m 32s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/bc5a023b155b4258987af5697eee3239 : SUCCESS in 12m 59s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8a249bb8a1c77575d69af16c3fe336212da9d464","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-29 08:04:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/52c0b54f93cb425394ce1d50d1d66ec3\n\n- grenade https://zuul.opendev.org/t/openstack/build/02f03f22deb64faaadbe0a46695e331a : SUCCESS in 33m 47s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2f1c1a233de6454e880561d57acea93f : SUCCESS in 38m 23s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0220dc4e04254cf59ac5b8255f5415e0 : SUCCESS in 1h 21m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2c2a6cdecd4b4f88b3c01b585f20899a : SUCCESS in 45m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0ebd7b8a25d406dabcf6bc145d1e178 : SUCCESS in 18m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/44046f85eee04eb4a921a28a14fbb3d2 : SUCCESS in 4m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d180eaee24b04b3ba3d6d40345ada972 : SUCCESS in 6m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fe8c1db0cbe4477d8f397749e619922e : SUCCESS in 5m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f180039e86d8433a8c8e30e10b0ecd1d : SUCCESS in 5m 37s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/33a0e8661a61467192064cb350dbbc6d : SUCCESS in 6m 37s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/9f8d3da75ca5423dac20786c34a94186 : SUCCESS in 9m 56s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/b7d3d25d3a6d4bd6b91f4185927993c5 : SUCCESS in 1h 23m 36s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/5cfc792b0d8741f1ba988bc98b4dfb71 : SUCCESS in 1h 55m 04s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/6fbfbff3ca8746328709d9fdfaa9e6ca : SUCCESS in 2h 04m 07s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/1127c1f97ab54e7bb2580ce6993528fa : SUCCESS in 2h 14m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3411daf0e186477face8449e51b5772a : SUCCESS in 58m 16s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/1dc893721cd34570a1b2939eccac0fa2 : SUCCESS in 1h 37m 05s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/4c4449f494804b3b95c7539ab854bb6c : FAILURE in 33m 53s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/c31576e42c044529b0e9b2f6799cc71b : SUCCESS in 2h 07m 10s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/570ede76c5bd4c86a12178eb3c7762a1 : SUCCESS in 14m 40s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/c6b56ff1205d44d3a116be5d922707df : SUCCESS in 1h 08m 54s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"578d74805600f3794ef5246887449305cad0dcf4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-29 08:05:38.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"91cf68a3c9431410e6424f85ad6745d7e0b96734","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-29 09:53:20.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c3ac04768b004c78b706478c7a5a3c58\n\n- grenade https://zuul.opendev.org/t/openstack/build/ddc5b5ce8f5c43ee80f4c914a26c7b0a : SUCCESS in 49m 34s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/765180e9a515409bbb086ffbea2ed0ce : SUCCESS in 1h 01m 12s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/045c471272334435aa51925226ee2879 : SUCCESS in 1h 20m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a4d8dc9de5344837ad0d6281d3fdbd5c : SUCCESS in 54m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2eacfd9dec73431a961864f0d27deb01 : SUCCESS in 5m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e7fcc92595544b599ceb1da5d303c272 : SUCCESS in 3m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c2a06651f1a94597a7c77d92f3ea37d1 : SUCCESS in 3m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99ae61f001b7418a8fdc1bff20af08a9 : SUCCESS in 6m 59s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/061fc21f2f1944bdaec580ce0f6a0f61 : SUCCESS in 6m 19s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/7af23c46e3dd49ebb8ec8711d59d32a4 : SUCCESS in 6m 40s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/fb7b9b8755aa46cf8d2997cea5834be1 : SUCCESS in 1h 45m 38s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/9891bf27457d4a0585e2bbf2ea12c14d : SUCCESS in 1h 43m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f117299b680f4316b1c6e1f24c9c7eac : SUCCESS in 49m 49s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/766cb747f30e4a26895242c1efd501d1 : SUCCESS in 1h 30m 30s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/b24907d489c64aae8ac5aef4338052a7 : SUCCESS in 28m 30s","accounts_in_message":[],"_revision_number":2},{"id":"94bb696d35db84e9addce117310aca178eac78aa","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-29 09:53:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"c2b83e1c669376c9f4fdccffa764178490e8bf23","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-29 09:57:03.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9409101cc52e44ccab6b1ae6379d4794\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa4a939d01e24721bd4108e40a5dce31 : SUCCESS in 3m 26s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"13b3c7283fd085131a3dd73f2b7300b91ca97007","revisions":{"5a89d3ec06424555bed49be1d76f690c3edb320b":{"kind":"REWORK","_number":1,"created":"2025-07-29 14:59:38.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/86/956086/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/86/956086/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/86/956086/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/86/956086/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/86/956086/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/86/956086/1"}}},"commit":{"parents":[{"commit":"0989e4d336658ff5c183a8d676b0a62190bc61ac","subject":"Merge \"Enable Hash calculation for new location API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0989e4d336658ff5c183a8d676b0a62190bc61ac"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-07-29 14:02:39.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-07-29 14:34:10.000000000","tz":0},"subject":"Fix inaccurate usage reporting when delayed_delete is enabled","message":"Fix inaccurate usage reporting when delayed_delete is enabled\n\nThis change addresses issues where usage reports include images in\npending_delete, leading to potential user blocking and limit\nenforcement errors.\n\nChange-Id: I491c486a0b967baba9bbdca4cd241ba691a94435\nSigned-off-by: Abhishek Kekane \u003cakekane@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5a89d3ec06424555bed49be1d76f690c3edb320b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5a89d3ec06424555bed49be1d76f690c3edb320b"}]},"branch":"refs/heads/master"},"13b3c7283fd085131a3dd73f2b7300b91ca97007":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-08-25 10:36:44.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/86/956086/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/86/956086/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/86/956086/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/86/956086/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/86/956086/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/86/956086/2"}}},"commit":{"parents":[{"commit":"0989e4d336658ff5c183a8d676b0a62190bc61ac","subject":"Merge \"Enable Hash calculation for new location API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0989e4d336658ff5c183a8d676b0a62190bc61ac"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-07-29 14:02:39.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-08-25 10:36:44.000000000","tz":0},"subject":"Fix inaccurate usage reporting when delayed_delete is enabled","message":"Fix inaccurate usage reporting when delayed_delete is enabled\n\nThis change addresses issues where usage reports include images in\npending_delete, leading to potential user blocking and limit\nenforcement errors.\n\nCloses-Bug: #2118756\nChange-Id: I491c486a0b967baba9bbdca4cd241ba691a94435\nSigned-off-by: Abhishek Kekane \u003cakekane@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/13b3c7283fd085131a3dd73f2b7300b91ca97007"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/13b3c7283fd085131a3dd73f2b7300b91ca97007"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
