)]}'
{"id":"openstack%2Fglance~956434","triplet_id":"openstack%2Fglance~master~Ic6888fb0b362ba46c869517b1c4030aca1e18f77","project":"openstack/glance","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-01-15 16:25:42.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic6888fb0b362ba46c869517b1c4030aca1e18f77","subject":"Simplify import of sqlalchemy","status":"MERGED","created":"2025-08-02 18:36:56.000000000","updated":"2026-01-15 16:28:00.000000000","submitted":"2026-01-15 16:25:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"956434","meta_rev_id":"bfcee43af0bf1f137686cbf500cc72b93fa9097a","_number":956434,"virtual_id_number":956434,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-15 16:25:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2025-09-12 15:55:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2025-09-15 13:58:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-01-11 20:39:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2026-01-06 15:06:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-02 18:46:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-08-02 20:53:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-19 10:20:37.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2025-09-15 13:58:17.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2026-01-11 20:39:38.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"}],"messages":[{"id":"126856fc4bf7365a98c21a6ca222f34a1d7a0a9d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-02 18:36:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"407b5c4d59397d3d0682a2eef8b6e3243c42b7ef","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-02 18:46:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f1f72f0cee2d45c4baba0c1d8e669e9e\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/eed30e7d10a848f4b2b13687a8c23fdc : SUCCESS in 6m 54s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/474dcd85f7394b7ab34473db3d70c52a : SUCCESS in 6m 40s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/64e9916848324a6583f120da4ff6b5b6 : SUCCESS in 8m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a8e2cd0ff7942b08d6dc1767d7392a14bbf87511","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-02 20:53:53.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/36239fafabf243f693347f7e26a14997\n\n- grenade https://zuul.opendev.org/t/openstack/build/31dd0c9046684028bf338505fc1992a8 : SUCCESS in 1h 00m 19s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/38e7b32730634173ada75733926f1996 : SUCCESS in 1h 04m 18s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/560ccfbd9022490baa1f429d18d84039 : SUCCESS in 1h 23m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/24e0f8d3496340e7b3d5f288f3102063 : SUCCESS in 1h 01m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e14891d919ec4bd38fd2eb6fd3aaca08 : FAILURE in 4m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/96fa59a9191b4b80a32dcccc795cbdfb : SUCCESS in 4m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0f8ea589050240b2ad011707e796e330 : SUCCESS in 7m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/097d32bd875d4c92955e278e2fbd8289 : SUCCESS in 9m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6998197a1957492e97fb8f99000bfe56 : SUCCESS in 5m 08s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/70c60f61f5bc4d67a9da436cf1cef8b7 : FAILURE in 6m 25s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/1c644535c4a44e3ab4546d331404ffd6 : FAILURE in 9m 55s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/066d0d1f71a94af290684a42dd447715 : SUCCESS in 1h 42m 37s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/324fd79f427343b9ab9bbdf94b8cf371 : SUCCESS in 2h 03m 55s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/52427a85b55446cc8f9206e2479bd962 : SUCCESS in 1h 52m 20s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/a3db798a0bdc4c1a84c34b10598e7f42 : SUCCESS in 2h 10m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bb8ebedcf3ac49838fd950f2c5a21e2b : SUCCESS in 56m 56s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f733165979ff4135b948a4c11158d490 : SUCCESS in 58m 42s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/7727e2b4bee14256bfb86ce6e411894c : FAILURE in 30m 37s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/e04c062da27e4911ac601d9747dc484f : SUCCESS in 2h 07m 23s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/a7bee3b72ad14bf0a6098174dfd2c844 : SUCCESS in 42m 00s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/a0ecd1dec96a4e1bb5defaf7b347508b : SUCCESS in 27m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"46760f04c5855374764c697d6c424afde3f3e242","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-03 02:29:59.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"4b4dbc13e0044a7f30104b0c00aac23f889b587b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-03 03:34:10.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9820b1b2b48242d3824d9a5bf1e90a89\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d11e5105e2da480ba7164ab6e5cdc48c : SUCCESS in 13m 44s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/496fb8651b144c3bbff4da3130f0a276 : SUCCESS in 12m 34s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/0fdf750472fa4bf3942de6db614baa69 : SUCCESS in 17m 18s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e512d5fbd864aa8412c4442ba189708e490b6919","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-03 04:51:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/39913d9b4deb455c859553b2cb291f23\n\n- grenade https://zuul.opendev.org/t/openstack/build/9ea1eff5e37b4f94a2fce94e76764c36 : SUCCESS in 1h 00m 00s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6c79b047ef8e49c4b2d592f7abe7910e : SUCCESS in 59m 42s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/07c74e5df9204d9e9e694a71e5495202 : SUCCESS in 1h 32m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c9e72b31067e47a9b65008a15f71ea85 : SUCCESS in 47m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c4e259c309b46f3af3df994adee69f3 : SUCCESS in 7m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/642d8c8fa3e04fdb99a9b5612cd4a332 : SUCCESS in 6m 12s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8a1b0d44d31b4a0aba50587e2567a2fd : SUCCESS in 7m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cfa1035bb1f34310a25e4a7f20af6ee7 : SUCCESS in 7m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1007ce75e594165badad7a5fb008ff1 : SUCCESS in 5m 43s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d08faaf13a3e4d39a40d1c2d3539b33a : SUCCESS in 6m 22s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/d3565efcc9c8416a886fb3c41f4f8afb : SUCCESS in 10m 07s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/29e89ac45b944380a4d0f903dd1c2224 : SUCCESS in 1h 59m 15s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/150f8dbeb9c4434ebbe0f5001374d913 : SUCCESS in 1h 44m 43s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/45ad965896a54389b6ff1f80d79d8ad7 : SUCCESS in 1h 52m 54s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/8c4972f048454ac292e7317de6762a9e : SUCCESS in 2h 15m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b9fa74a8684442ef8d5d293498741f32 : SUCCESS in 59m 31s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/6ba7970bedc14437a38bfc1ecbf65b82 : SUCCESS in 1h 45m 26s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/2ac763fff7b74bbfa72b4f2df64a1b09 : FAILURE in 36m 00s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/da493b5223874575bbc3b3415da18006 : SUCCESS in 1h 51m 14s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/78ba0ccf0de044e98c028b8915446912 : SUCCESS in 32m 40s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/dd87857da4f44d448632e819717f1e60 : SUCCESS in 1h 06m 49s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5aee4f89ee1e94b434e2147d8a51038799103ff5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-09 08:23:36.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"46959a3474e01bdc0658187471d83d709ad20805","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-09 10:35:49.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f17bd459020b49b9ac22e9c6f274714b\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/daa1248f1153480bad3888becfdb50fb : SUCCESS in 10m 31s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5bab11209cd44054af3ffedd771861b4 : SUCCESS in 14m 34s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c46e01bc7a65456e9c0e3354e2043e13 : SUCCESS in 17m 21s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"68054e30bf704bfdbaa2bcda7621b2ae9b08cb7f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-09 10:42:35.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a28ce5aefd25463b98d57ac62c120e5e\n\n- grenade https://zuul.opendev.org/t/openstack/build/933db884996149779f25b38489ae45c7 : SUCCESS in 1h 08m 50s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5c58015db8fb44888456fcceee22d399 : SUCCESS in 1h 11m 02s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/997943ba1be54a8db85fae5aff6b05ac : SUCCESS in 1h 42m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/27a2d351cc3a4aff94647a57fc426a6d : SUCCESS in 58m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a76974d59e14417e96e4bb5f945905e9 : FAILURE in 6m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a9f67b69b5934854b9d53246bc769e23 : SUCCESS in 6m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5e1c2ede7c2a4e8884690e9ca08678b1 : SUCCESS in 3m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/837537a3076040cdae4d95c2d1de80bf : SUCCESS in 8m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a144e00fecb34313ba4f9e428438ad14 : SUCCESS in 7m 51s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3c5f4ef32cde408180827905bd0feffd : SUCCESS in 6m 41s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/551b6c30803c4476a478e28dc98c3f8c : SUCCESS in 9m 58s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/8225cfc6e9534ee1b5f275aaa13d1b48 : SUCCESS in 1h 55m 41s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/f2233e3f9e42423cbc54467420ffea79 : SUCCESS in 1h 59m 00s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/47e9c57db45c4e5f80347655d458ce4b : SUCCESS in 1h 51m 38s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/58736441d8f346dabc5e764617d5a395 : SUCCESS in 2h 11m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fe0ecbfa5a7f4feda942383a1577dbf2 : SUCCESS in 54m 10s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c0a48ceff95f41da92b7bb0a2c9a514c : FAILURE in 1h 57m 51s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/4c8b6aea2dc74a9ea3eeb96810e55158 : FAILURE in 32m 02s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/bfae785304374db1b2e2b0a2ad155707 : SUCCESS in 1h 57m 24s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/a220e6c2c7c0407fb4e02356fc24bafb : SUCCESS in 29m 02s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/0311aebe9f5f475d89152429a57f7d6f : SUCCESS in 1h 04m 23s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2b61ea0e9e047fae4da54219cc7da9f49673a2fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-18 15:26:31.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"7b0e4bef0418ace1c377af81f9742bf88eec486b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-18 19:56:40.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/20c49f329e9c49efba8bf3a7e091ea46\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ef2be2f1221845cd975d9ca90474b244 : SUCCESS in 10m 41s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1046273ff775438eb28cfd3f91e703f0 : SUCCESS in 11m 11s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/290d43efe2c4412284c6e1f1998311b0 : SUCCESS in 13m 12s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"491875260fe43bcaeab07057723e674e9691ab47","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-18 21:21:23.000000000","message":"Patch Set 4: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/adeaab782ad5495cab87e665c1e770fe\n\n- grenade https://zuul.opendev.org/t/openstack/build/f02f458877524e9f9123a1be834bdfc2 : SUCCESS in 53m 06s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0f45fa45d22b4577b0a53b35c5337424 : SUCCESS in 35m 15s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/309203c823bb4b1b86179f0ee2b77ce7 : SUCCESS in 1h 18m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4f95a12e829845ee9caad2174e39d049 : SUCCESS in 1h 05m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6e5e5a863ac418692c81856c3c1d9ee : FAILURE in 5m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/42799ed83d4e44eb8264d3b420ef4d0b : SUCCESS in 4m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8e0d7669e92b403ba1b2a7117e1d6266 : SUCCESS in 2m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/77c15ae3691143dba11d41376aea84a9 : SUCCESS in 10m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b284a73d8724ffe9e5bf9fa39efae0a : SUCCESS in 3m 50s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f36b0f0c811c402c958a30100b4eaf4f : SUCCESS in 4m 59s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/984192d6ace84b7daf076b69791b1c84 : SUCCESS in 6m 13s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/536cfdc460444251ba64e46451a9c9d1 : SUCCESS in 1h 34m 29s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/25bdc8f1e739434eb5b0b7d788e374c9 : SUCCESS in 1h 53m 54s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/defa0a227061430b9c04ab752d2c685e : SUCCESS in 1h 01m 26s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/110bd11e6f444d33925b3ca0404b809b : SUCCESS in 1h 13m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6fa70309e93f48c5947704ad7a631771 : SUCCESS in 28m 57s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d8064c8ce2944e2b81a74c53902abcca : SUCCESS in 1h 35m 25s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/3fe4c4dd7d464ef6b2c0949b2465e4fc : FAILURE in 16m 07s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/31278e97f71f4c04b48e7cc10fed0f94 : SUCCESS in 1h 00m 01s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/5a077e881d1b429b9b63b18ef0710e5e : SUCCESS in 27m 37s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/27450d981aeb44c3808e1cee788d71a3 : SUCCESS in 34m 33s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"e245d846aae25928d7f8e507b113b2059b75acfa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-19 01:13:36.000000000","message":"Patch Set 5: Published edit on patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"f0880394da23d2ce721687db722950ced6ac75ef","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 01:31:01.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee935aaf5c4643d295ea09c2e07a6540\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5d840cac816e4e1f93e21c0721f981bd : SUCCESS in 9m 48s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9ef14f799fcc482f9b1d2b3ec4001936 : SUCCESS in 12m 15s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/1a438348df3642dcb22579c6214d7452 : SUCCESS in 15m 30s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"6474037e0c77ade3c2cd7f436b79582cb6e9fb18","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 03:19:52.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/27a076b78e064541a6ba0602b9e0aecf\n\n- grenade https://zuul.opendev.org/t/openstack/build/aa99d2b1737f483eb920a390c0a89a05 : SUCCESS in 44m 26s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4dea61e8ddd64504b19eb74cb0832806 : SUCCESS in 1h 02m 36s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/38587fe071a84eb9a10d85924bbcab2c : SUCCESS in 1h 16m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f9753f3b2c9646d18fb2426a2af42745 : SUCCESS in 1h 04m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cbc9c23e510c49f183b3c6f2c11e0b19 : SUCCESS in 3m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/910c16257c584c94a3a8bbc0a9a85d50 : SUCCESS in 2m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a66495d420004afa8180d9ca72de4b31 : SUCCESS in 4m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1bf1bd9a722649feab7c0c08c077520a : SUCCESS in 4m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b06a1b2e7a9f4422b281be3b04038761 : SUCCESS in 3m 57s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/fafcc4386c6240bd8b88243becb09862 : SUCCESS in 7m 41s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/f712bc8904de4ea18ac7b820579760eb : SUCCESS in 9m 16s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/d961f982fe14402c95495b4c62d0c399 : SUCCESS in 1h 53m 46s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/6cd73bd6f4224e32bc86faae74a865f9 : SUCCESS in 1h 06m 50s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/456cab94d21a44c1984001db2d30edc2 : SUCCESS in 1h 05m 17s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/00896101cef447488d6cb1de6c04ea53 : SUCCESS in 2h 04m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7d9057c0e7034e3ebaa299f43bf3010c : SUCCESS in 56m 07s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/29447eaaae3f4a7292bb4a3f8b72cf1c : SUCCESS in 1h 14m 14s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/7fbfa83212b44078af0b941e841f2d86 : FAILURE in 23m 52s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/b2aec25603e7487dba4a761e995caa51 : SUCCESS in 58m 59s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f9210cb0479549d698491a0b9c1f69f9 : SUCCESS in 19m 42s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/8f102575b8c54ec0900a780add493b49 : SUCCESS in 32m 59s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"52b4be98ca783305034f0152ba7cc69fd59ca566","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-08-19 10:20:37.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"67e9ee7594c3ec47e743a1983fac0222fa1739d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-20 02:23:02.000000000","message":"Patch Set 6: Patch Set 5 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"f3902d4bdc7b0846eb665c457fee59bc19b96da2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-20 03:25:55.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a38f2f15adce41e588e773d8c8837cd0\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5e51e108af5d46cab5c0b2b637d1b295 : SUCCESS in 10m 56s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/81843cd5ccc4417da27fc4aea624b294 : SUCCESS in 14m 03s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/7b22a1dd1d5b4227ae548b6a57f0adc0 : SUCCESS in 10m 27s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"f2df75a322d3cbda2d3410451dbbdf3b1a4c7ece","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-20 04:26:05.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/31814d61ffa846d085ada2a765e77a42\n\n- grenade https://zuul.opendev.org/t/openstack/build/6793938a3359447c9e22aa9c1d46b631 : SUCCESS in 53m 46s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9613370041be4e1c9b7f3e4de24be2e7 : SUCCESS in 1h 05m 09s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ccd0919fa99d41548eff36bccdcc7a64 : SUCCESS in 1h 37m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8516c18c77974be284c367fb41bb5b9e : SUCCESS in 53m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b1e4a61aded49bf969cd630ed173af2 : SUCCESS in 7m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6b6ddd9a149147bd8166c377f620faa7 : SUCCESS in 5m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2960a94b447c45f48e367c24ca573d0d : SUCCESS in 5m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2e51071d1c104c4f83dbc500c672cf56 : SUCCESS in 7m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85e4204c38a9453197c16a8db2c5aa1a : SUCCESS in 6m 46s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/73275bacdead44609570abfa6a6e566f : SUCCESS in 4m 59s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/6b8c0d59939a4d409d116901224f9319 : SUCCESS in 6m 52s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/b2cbf941383443ddb9be73749ff42344 : SUCCESS in 1h 40m 33s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/6041041313be49b795d5a130a432bc80 : SUCCESS in 1h 48m 09s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/557db23b4b654f3fb44f0ca4ffc03b0a : SUCCESS in 1h 20m 21s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/80de88c976994ddd83676351516323c4 : SUCCESS in 1h 51m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3237d8b8e58f4ad49a15ed0565834320 : SUCCESS in 28m 33s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b0e224218bef4fe0b52aaf7dd71c3773 : SUCCESS in 1h 58m 16s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/3ddca1c8396d4e9780b9eec23e4efae8 : FAILURE in 30m 13s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/4d07c937d92c417eb8bd779757761c25 : SUCCESS in 1h 55m 37s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/a493fd9d2eea4cd3b212b8246d46bc39 : SUCCESS in 18m 05s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/ce6e0fe0d0fe471c943b7831e44802b8 : SUCCESS in 54m 25s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"dfd03d2f237c9aa13cfe4366fd5297b47f2d1c26","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-09-12 15:55:51.000000000","message":"Patch Set 7: Patch Set 6 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"d1ef3fbd629aa2321d0e4123a2c44727c9dd12c9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-12 16:58:43.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0927375428064651a61e587c1f8c1cc8\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d2bbe0a251eb4240ae1aa961c397c49e : SUCCESS in 12m 25s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/bfe45bcbc12045949be1ad61681ad94c : SUCCESS in 9m 07s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ab7b6a0ede234ec28c70c05429fff48a : SUCCESS in 9m 10s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"52f8a1659586347523e58941cce156eaeb8386c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-12 18:12:30.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0ffd15ef61044f48b40a872fcfad0a20\n\n- grenade https://zuul.opendev.org/t/openstack/build/a6f15f613c6d4a32a80e2f87b16c7f95 : SUCCESS in 33m 27s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4f133d9ef85c4372b80b9eec12c7df33 : SUCCESS in 54m 10s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b6933354fd934d9bb86762c17c19968e : SUCCESS in 1h 17m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/61de6e0e34af48849dcde308fb1dce24 : SUCCESS in 51m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f950c8b6f8cf47068eec9421a683257f : SUCCESS in 5m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/29ed2e77f87948ff867c4a19a8837fe1 : SUCCESS in 5m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0d74f8a689ed4b11bf51ef822e8c18e3 : SUCCESS in 6m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e4bd2744bd7d48cb9a4ea9389f80e4b6 : SUCCESS in 6m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4139e0a30fe14d6e93ab4774811f45d9 : SUCCESS in 5m 39s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/08b0d58d523440809a455d54648efb51 : SUCCESS in 6m 07s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/5a6835f26a8a41aca62a60002ee11941 : SUCCESS in 9m 53s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/250f7fe15a7441e7ad2103233f5adf0b : SUCCESS in 1h 51m 48s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/e15a4f0125284b018dcda8ea3680c592 : SUCCESS in 1h 44m 55s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/3a14850028a9497eb382a9f4659ae4f5 : SUCCESS in 1h 02m 09s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/002cc68a967f4fc1bb7fa4df7634fe36 : SUCCESS in 2h 10m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/71897ef4b4ba4a658a3e8a9735cd156f : SUCCESS in 50m 17s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/56540c3a642647cc91826695a11a923d : SUCCESS in 1h 47m 51s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/46bcc54544274f60857fe453259f5976 : FAILURE in 29m 58s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/27462c3ab224456cbf5366f206e9efc0 : SUCCESS in 1h 08m 45s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/16ac34a097494c9488ccace8199b43da : SUCCESS in 19m 02s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/41f549f9dc6b4c9981299649d1795141 : SUCCESS in 32m 27s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"9c916bf00aa8bf4abfb776070bbc70e6a274ec17","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-09-15 13:58:17.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"84f50bf15ea4471f59021baf4445d3619f3a2918","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-01-06 15:06:07.000000000","message":"Patch Set 7: -Workflow","accounts_in_message":[],"_revision_number":7},{"id":"9b8878ceb0ee1a9beffb07ead7a2b4fc1a54e47e","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-01-06 15:06:11.000000000","message":"Patch Set 7: Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"f4087949cf71ef5170d95f329a30728617136653","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-11 20:39:38.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"9d58cfc230d9ff78f177da1cc4560f6d1543dfa7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-15 14:08:23.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"99dc84562277087c15fc67beea6a9335e855672b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-15 16:25:42.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/902be46ea4d346b7858324073c43c59a\n\n- grenade https://zuul.opendev.org/t/openstack/build/d568e2553a8047099f57f47ccbd07a66 : SUCCESS in 30m 08s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/df72fc2593bc4ffc878f4819c5b5adef : SUCCESS in 57m 47s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/296e929080234aef84cd5ce271d253ca : SUCCESS in 1h 30m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4a84a7acbf56448794b9aeb002d87343 : SUCCESS in 51m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f848687a308e49fca24fa76d37d199df : SUCCESS in 6m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/730749ace5574cc8b0fdddf957b2e103 : SUCCESS in 4m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5e72575a230d4024bcfb68c07e7a3edd : SUCCESS in 9m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83062d6c29804a0e9e9187a0d7a61a8c : SUCCESS in 6m 34s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c1cda338362d43c391fcab62ef0bb866 : SUCCESS in 6m 33s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/e4b01ed896fe4a37b4749f9f04c09d2c : SUCCESS in 5m 42s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/17ae5b6e26c5460c9297602cf67590c3 : SUCCESS in 1h 48m 43s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/6f1f67f399b640e3be1c978f15d42d63 : SUCCESS in 1h 51m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/af7d2dcc41df438f981b5cb3ef97ec11 : SUCCESS in 58m 13s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/cfc1bc96633349f293a7c9e28a60263f : SUCCESS in 1h 49m 53s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/3eb6cb52a8894b888353dc69802a554c : SUCCESS in 26m 00s","accounts_in_message":[],"_revision_number":7},{"id":"c8c45b65d8851c37c05f9b82d759e09fdb9fb049","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-15 16:25:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"bfcee43af0bf1f137686cbf500cc72b93fa9097a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-15 16:28:00.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/67786471658f4e5ea6e38f91dac739d7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/082b0b6d98da4354ac67f9a059485363 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"ad0c02b144f008bf6daa4becc9d6785974e13a89","revisions":{"988d9f761715d8d78fbd73d88ce4eae49de00b36":{"kind":"REWORK","_number":1,"created":"2025-08-02 18:36:56.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/34/956434/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/34/956434/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/34/956434/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/34/956434/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/34/956434/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/34/956434/1"}}},"commit":{"parents":[{"commit":"4f28a36ee6e9bfa0fd9bedc999e781c40738f071","subject":"Enable E711/E712","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4f28a36ee6e9bfa0fd9bedc999e781c40738f071"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-02 18:33:11.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-02 18:36:48.000000000","tz":540},"subject":"Simplify import of sqlalchemy","message":"Simplify import of sqlalchemy\n\nAvoid importing a same module using different names.\n\nChange-Id: Ic6888fb0b362ba46c869517b1c4030aca1e18f77\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/988d9f761715d8d78fbd73d88ce4eae49de00b36"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/988d9f761715d8d78fbd73d88ce4eae49de00b36"}]},"branch":"refs/heads/master"},"90adc685fe859c78b257ed16bf7586c4101c545b":{"kind":"REWORK","_number":2,"created":"2025-08-03 02:29:59.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/34/956434/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/34/956434/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/34/956434/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/34/956434/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/34/956434/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/34/956434/2"}}},"commit":{"parents":[{"commit":"4f28a36ee6e9bfa0fd9bedc999e781c40738f071","subject":"Enable E711/E712","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4f28a36ee6e9bfa0fd9bedc999e781c40738f071"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-02 18:33:11.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-03 02:27:19.000000000","tz":540},"subject":"Simplify import of sqlalchemy","message":"Simplify import of sqlalchemy\n\nAvoid importing a same module using different names.\n\nChange-Id: Ic6888fb0b362ba46c869517b1c4030aca1e18f77\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/90adc685fe859c78b257ed16bf7586c4101c545b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/90adc685fe859c78b257ed16bf7586c4101c545b"}]},"branch":"refs/heads/master"},"d6c73ce8e0d63d463a40ff179751ea1f5dba857a":{"kind":"REWORK","_number":3,"created":"2025-08-09 08:23:36.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/34/956434/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/34/956434/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/34/956434/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/34/956434/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/34/956434/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/34/956434/3"}}},"commit":{"parents":[{"commit":"e7ba90d96a41aaec8d29363adc9435112bb37af6","subject":"Enable E711/E712","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e7ba90d96a41aaec8d29363adc9435112bb37af6"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-02 18:33:11.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-09 08:22:56.000000000","tz":540},"subject":"Simplify import of sqlalchemy","message":"Simplify import of sqlalchemy\n\nAvoid importing a same module using different names.\n\nChange-Id: Ic6888fb0b362ba46c869517b1c4030aca1e18f77\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d6c73ce8e0d63d463a40ff179751ea1f5dba857a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d6c73ce8e0d63d463a40ff179751ea1f5dba857a"}]},"branch":"refs/heads/master"},"1d2c12e3ffadd7dc58b832c9fae098735b815696":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2025-08-18 15:26:31.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/34/956434/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/34/956434/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/34/956434/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/34/956434/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/34/956434/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/34/956434/4"}}},"commit":{"parents":[{"commit":"a1ed166dccd0409eeb6f4126f9a9d07d3a8bc2bd","subject":"Enable E711/E712","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a1ed166dccd0409eeb6f4126f9a9d07d3a8bc2bd"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-02 18:33:11.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-18 15:26:31.000000000","tz":0},"subject":"Simplify import of sqlalchemy","message":"Simplify import of sqlalchemy\n\nAvoid importing a same module using different names.\n\nChange-Id: Ic6888fb0b362ba46c869517b1c4030aca1e18f77\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1d2c12e3ffadd7dc58b832c9fae098735b815696"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1d2c12e3ffadd7dc58b832c9fae098735b815696"}]},"branch":"refs/heads/master","description":"Rebase"},"2a92686596171af0304600306b6783f2ca3352b6":{"kind":"REWORK","_number":5,"created":"2025-08-19 01:13:36.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/34/956434/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/34/956434/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/34/956434/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/34/956434/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/34/956434/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/34/956434/5"}}},"commit":{"parents":[{"commit":"a1ed166dccd0409eeb6f4126f9a9d07d3a8bc2bd","subject":"Enable E711/E712","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a1ed166dccd0409eeb6f4126f9a9d07d3a8bc2bd"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-02 18:33:11.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-19 01:13:36.000000000","tz":0},"subject":"Simplify import of sqlalchemy","message":"Simplify import of sqlalchemy\n\nAvoid importing a same module using different names.\n\nChange-Id: Ic6888fb0b362ba46c869517b1c4030aca1e18f77\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2a92686596171af0304600306b6783f2ca3352b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2a92686596171af0304600306b6783f2ca3352b6"}]},"branch":"refs/heads/master"},"02b0e3a226e45c4e53e9f5d21dec951a6df390aa":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2025-08-20 02:23:02.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/34/956434/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/34/956434/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/34/956434/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/34/956434/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/34/956434/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/34/956434/6"}}},"commit":{"parents":[{"commit":"c634a6af83894a7547136e8e2a9cccd1d06cb9db","subject":"Enable E711/E712","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c634a6af83894a7547136e8e2a9cccd1d06cb9db"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-02 18:33:11.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-20 02:23:02.000000000","tz":0},"subject":"Simplify import of sqlalchemy","message":"Simplify import of sqlalchemy\n\nAvoid importing a same module using different names.\n\nChange-Id: Ic6888fb0b362ba46c869517b1c4030aca1e18f77\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/02b0e3a226e45c4e53e9f5d21dec951a6df390aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/02b0e3a226e45c4e53e9f5d21dec951a6df390aa"}]},"branch":"refs/heads/master","description":"Rebase"},"ad0c02b144f008bf6daa4becc9d6785974e13a89":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2025-09-12 15:55:51.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/34/956434/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/34/956434/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/34/956434/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/34/956434/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/34/956434/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/34/956434/7"}}},"commit":{"parents":[{"commit":"05d4e3aac9be9e238015b5cc5aa83c6c76e84cc5","subject":"Enable E711/E712","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/05d4e3aac9be9e238015b5cc5aa83c6c76e84cc5"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-02 18:33:11.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-12 15:55:51.000000000","tz":0},"subject":"Simplify import of sqlalchemy","message":"Simplify import of sqlalchemy\n\nAvoid importing a same module using different names.\n\nChange-Id: Ic6888fb0b362ba46c869517b1c4030aca1e18f77\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ad0c02b144f008bf6daa4becc9d6785974e13a89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ad0c02b144f008bf6daa4becc9d6785974e13a89"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
