)]}'
{"id":"openstack%2Fglance~956628","triplet_id":"openstack%2Fglance~master~Ie733373c093bed4866719f163a75abaff4abe6a2","project":"openstack/glance","branch":"master","topic":"spec/set-size-on-upload","attention_set":{},"removed_from_attention_set":{"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2025-08-25 12:31:28.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie733373c093bed4866719f163a75abaff4abe6a2","subject":"Update API reference to mention x-openstack-image-size header","status":"MERGED","created":"2025-08-06 07:32:06.000000000","updated":"2025-08-25 12:34:50.000000000","submitted":"2025-08-25 12:31:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"956628-spec/set-size-on-upload","meta_rev_id":"73e4101b09fccb44551fa185a111d887e409171c","_number":956628,"virtual_id_number":956628,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-25 12:31:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":2,"date":"2025-08-25 12:23:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":1,"date":"2025-08-25 12:23:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-06 07:49:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-25 12:23:02.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"35240f0512d2643234cccf1b615062439a9a59ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-08-06 07:32:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"51c74df3273d69bb5b07a3a9bf36546107648356","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-06 07:49:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1bf1a1a2799245b4a4ed1a1af1d092c2\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/d71139016ae944b690e960d5e1d270a4 : SUCCESS in 4m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4fa12f576b14750a6a25d8f00a681e3 : SUCCESS in 8m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/239e1e4c7a4741d4ad2a3a80955263b4 : SUCCESS in 6m 13s","accounts_in_message":[],"_revision_number":1},{"id":"01ae462ffdd343af0e4686cd96ad12b2cac043ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-08-07 06:31:16.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"6dc3e32ceab2538e6587181cb62d7539f514b993","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-07 06:42:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aaf79cea8d0445918a25bf2fbcbc1c32\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/c228c70e08444bec9924144ae31ba5b8 : SUCCESS in 4m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36a5c4dc93444944b7eb780cebc21e12 : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1337224d32a944e3a777b9dc20472977 : SUCCESS in 3m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/95edb67cbba84a8999c13c1e05da49f5 : SUCCESS in 5m 29s","accounts_in_message":[],"_revision_number":2},{"id":"37556708fdc76ed442bafdbde33ab5eba9a3e8db","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-08-25 12:23:02.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"334f2358ce875c73544872c6c2bd0332d9c4210e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-25 12:23:14.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3e0da47889b7244202453b2fe79ad207a162dd68","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-25 12:31:27.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f0034b8ce1524467911b455904cea4ab\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/e53bfc2f9f374b4db6e33be644fd87bd : SUCCESS in 3m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb8d0400fe364ff686b94114ca692179 : SUCCESS in 6m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f9cd0adc0c5419d82c011e9e6355a28 : SUCCESS in 7m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8c0609990fd647afa61397e19a9fdb5a : SUCCESS in 6m 12s","accounts_in_message":[],"_revision_number":2},{"id":"acdf3e0f315eb8f5198708c1f058147ed4665a38","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-25 12:31:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"73e4101b09fccb44551fa185a111d887e409171c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-25 12:34:50.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5b32c1f71ac34a069d49047354af911e\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/2960a822943f40b08765136042bbe051 : SUCCESS in 57s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81b9809d97164b7d9762c3ad4866c775 : SUCCESS in 2m 54s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7db8bd748dfb4b67a4314748bec65f99 : SUCCESS in 2m 33s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"df0d0aad86b54a3554c5bef602d1049660223065","revisions":{"b6078be3f190c96b6c862061a3afff461abaf2c9":{"kind":"REWORK","_number":1,"created":"2025-08-06 07:32:06.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/28/956628/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/28/956628/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/28/956628/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/28/956628/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/28/956628/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/28/956628/1"}}},"commit":{"parents":[{"commit":"62a323aeddcc0d8895b99d111ffb89992687d340","subject":"Read image size from request header","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/62a323aeddcc0d8895b99d111ffb89992687d340"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-08-06 06:39:51.000000000","tz":330},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-08-06 07:05:46.000000000","tz":0},"subject":"Update API reference to mention x-openstack-image-size header","message":"Update API reference to mention x-openstack-image-size header\n\nAdd API reference documentation for the x-openstack-image-size header\nin PUT /v2/images/{image_id}/file and PUT /v2/images/{image_id}/stage\nendpoints.\n\nAssisted-By: Cursor (claude-3.5-sonnet)\nChange-Id: Ie733373c093bed4866719f163a75abaff4abe6a2\nSigned-off-by: Abhishek Kekane \u003cakekane@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b6078be3f190c96b6c862061a3afff461abaf2c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b6078be3f190c96b6c862061a3afff461abaf2c9"}]},"branch":"refs/heads/master"},"df0d0aad86b54a3554c5bef602d1049660223065":{"kind":"REWORK","_number":2,"created":"2025-08-07 06:31:16.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/28/956628/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/28/956628/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/28/956628/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/28/956628/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/28/956628/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/28/956628/2"}}},"commit":{"parents":[{"commit":"62a323aeddcc0d8895b99d111ffb89992687d340","subject":"Read image size from request header","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/62a323aeddcc0d8895b99d111ffb89992687d340"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-08-06 06:39:51.000000000","tz":330},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-08-07 06:05:13.000000000","tz":0},"subject":"Update API reference to mention x-openstack-image-size header","message":"Update API reference to mention x-openstack-image-size header\n\nAdd API reference documentation for the x-openstack-image-size header\nin PUT /v2/images/{image_id}/file and PUT /v2/images/{image_id}/stage\nendpoints.\n\nAssisted-By: Cursor (claude-3.5-sonnet)\nChange-Id: Ie733373c093bed4866719f163a75abaff4abe6a2\nSigned-off-by: Abhishek Kekane \u003cakekane@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/df0d0aad86b54a3554c5bef602d1049660223065"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/df0d0aad86b54a3554c5bef602d1049660223065"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
