)]}'
{"id":"openstack%2Fglance~957024","triplet_id":"openstack%2Fglance~stable%2F2025.1~Ic58233302001a2396e84400ef7081afa51748a2a","project":"openstack/glance","branch":"stable/2025.1","attention_set":{},"removed_from_attention_set":{"25402":{"account":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"last_update":"2025-08-11 19:06:37.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic58233302001a2396e84400ef7081afa51748a2a","subject":"Add task cancellation tracker utility","status":"MERGED","created":"2025-08-11 14:23:39.000000000","updated":"2025-08-11 19:19:14.000000000","submitted":"2025-08-11 19:06:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"957024","meta_rev_id":"d8bdacd9a8a9194d524143f8fa736dd009684fe9","_number":957024,"virtual_id_number":957024,"owner":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-11 19:06:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2025-08-11 17:06:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2025-08-11 17:06:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-11 14:23:39.000000000","updated_by":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"CC"},{"updated":"2025-08-11 16:45:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-11 17:06:28.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"cf0f872bfb01b6eeb00510d486f3531a99aa84c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2025-08-11 14:23:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"22405f632be50914107431e6c395c489fbffe918","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-11 16:45:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d2aeac94969f405fbc5ce0dd0c6c66d5\n\n- grenade https://zuul.opendev.org/t/openstack/build/a70b9b58f4a24282bbce55964182a714 : SUCCESS in 49m 28s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/759ea92b49c846fa9ef0e063a11394a9 : SUCCESS in 49m 00s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0a5b5f72a1124a76be8f08a9aedbf63e : SUCCESS in 1h 26m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6b0b9094dbf54d6bbe14d73dcb32f427 : SUCCESS in 56m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5eb46a696ae54aa4b14611abea082d0c : SUCCESS in 4m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3d82514fcde441b8933d74cc52b525ba : SUCCESS in 5m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/532736cc7d164200b24d6832b87a7409 : SUCCESS in 6m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d2d53d25b4545f6be866cad54e4fe73 : SUCCESS in 6m 01s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/941dc0f692c04e039edcd1e4f3f77b1a : SUCCESS in 9m 09s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9b8ad02bf03941748eb458f49d6debec : SUCCESS in 8m 27s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/9439ceee1cd544f2a66c74362a3c17e7 : SUCCESS in 12m 05s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/f79cd461416c4ab097bdb15a24f884f5 : SUCCESS in 1h 51m 01s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/b2bde49e7ab9473bb632e1773d59d53e : SUCCESS in 1h 39m 48s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/71ec3bcc2d184f809193cb494955d697 : SUCCESS in 1h 53m 43s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/3444032facb34170b49ad9ee026646a5 : SUCCESS in 1h 50m 24s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/839d8bd4cca84461bd0a59df4d4448ff : SUCCESS in 2h 16m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d09109886c84487b8e128a4a6f99ac1a : SUCCESS in 28m 34s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/dbcd18f8082e46a4a5d625573a003769 : SUCCESS in 1h 56m 17s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/8187999cc0f142c7ac9ec58587e8f6cd : SUCCESS in 2h 16m 43s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/5bfe6cc417c94cab81972ad674657ed3 : SUCCESS in 57m 57s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/66614ebd61964ee99677e2e7635bf38a : SUCCESS in 2h 00m 44s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/bfee9983cbb045f4bc8af7b1967a16b3 : SUCCESS in 32m 14s","accounts_in_message":[],"_revision_number":1},{"id":"91377d04bf215f7e8c2b703fcfe4f96fec9bc1ad","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-08-11 17:06:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7837041461b7179b77b069a283f063a4612c45ec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-11 17:06:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d0e77c6a59597b584d08e063f4855d3550c187c4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-11 19:06:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b29dfbdca7314196989e901c42161cf8\n\n- grenade https://zuul.opendev.org/t/openstack/build/c7c86f6b3588461dac50de11d8172b54 : SUCCESS in 32m 19s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d9d2f57dff044246af6581980b3e2323 : SUCCESS in 30m 01s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0b460898374c43a5aabb4b51b708755e : SUCCESS in 1h 28m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/21a0f5247e95414f81332a4ed722c6cb : SUCCESS in 1h 04m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8545e623afbf4173aef1a5ec27a35ee1 : SUCCESS in 5m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0e196c77ca95499d9e5a9d797fe740f4 : SUCCESS in 3m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/696f47b03cb14c64ad3dece9e68e523c : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e7c8a985e3b479fa924f7755db6e806 : SUCCESS in 5m 22s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/88425feaa624446aacb12a150142d8eb : SUCCESS in 7m 07s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c634fc094550456bb64075c6a436436a : SUCCESS in 7m 00s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/48f73b2d302340ce9007d744f035c4eb : SUCCESS in 14m 21s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/b2eaa475fc164b378e75ba86639f9197 : SUCCESS in 1h 05m 38s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/63e7e87abf4042278baee4f23e7fa551 : SUCCESS in 1h 03m 37s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/62a9954a647942a2b5fe1c727e8309e8 : SUCCESS in 1h 45m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/13d461c4722c4df0b0c1927960fa68ec : SUCCESS in 29m 17s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/0aae0f59c1bc4a85885e904b71e36948 : SUCCESS in 1h 29m 39s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f0fbed92122a41aca4efcb1f5631d7a7 : SUCCESS in 15m 13s","accounts_in_message":[],"_revision_number":1},{"id":"c42a1589c2c3c1af9b31016beb8c3672e9c56f01","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-11 19:06:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d90d8cb938fb71f66ab1b73abfe941d738665262","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-11 19:07:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/50d3653bc5d04435b4fcf5e287ccbd2a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ace0e5c21e84e25abddeb606c6c9402 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":1},{"id":"d8bdacd9a8a9194d524143f8fa736dd009684fe9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-11 19:19:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db34519a1f6b4cf08eb13ed605a2a13c\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f5f9500e69e5443f83eda3842f03dde1 : SUCCESS in 10m 57s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b4781ea61c354170b7183b81dac7ad20 : SUCCESS in 11m 30s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8e10a6511a2c52a3d0c0571f2ac7b915a15ccabc","revisions":{"8e10a6511a2c52a3d0c0571f2ac7b915a15ccabc":{"kind":"REWORK","_number":1,"created":"2025-08-11 14:23:39.000000000","uploader":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"ref":"refs/changes/24/957024/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/24/957024/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/24/957024/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/24/957024/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/24/957024/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/24/957024/1"}}},"commit":{"parents":[{"commit":"99b54964031408555917539b73eb5c18e463f89a","subject":"Merge \"Fix vTPM metadefs.\" into stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/99b54964031408555917539b73eb5c18e463f89a"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-05-27 20:00:14.000000000","tz":0},"committer":{"name":"Francesco Pantano","email":"fpantano@redhat.com","date":"2025-08-11 14:24:52.000000000","tz":120},"subject":"Add task cancellation tracker utility","message":"Add task cancellation tracker utility\n\nThis module provides functions to register, cancel, and signal cancellation\nof long-running tasks using lock files and external locking. It ensures\natomic operation registration, safe cancellation, and proper cleanup,\nsupporting concurrent task management.\n\nChange-Id: Ic58233302001a2396e84400ef7081afa51748a2a\n(cherry picked from commit 1311d5fe2f1e308e1abd930ab17a8fb368b670fe)\nSigned-off-by: Francesco Pantano \u003cfpantano@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8e10a6511a2c52a3d0c0571f2ac7b915a15ccabc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8e10a6511a2c52a3d0c0571f2ac7b915a15ccabc"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
