)]}'
{"id":"openstack%2Fglance~957572","triplet_id":"openstack%2Fglance~master~Iab40d4cf1d6fb836a370c8b3ebfbb84182265d7a","project":"openstack/glance","branch":"master","topic":"bug/2120737","attention_set":{"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2025-08-26 18:12:54.000000000","reason":"\u003cGERRIT_ACCOUNT_11604\u003e replied on the change","reason_account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Iab40d4cf1d6fb836a370c8b3ebfbb84182265d7a","subject":"Fix: Policy for new location APIs","status":"MERGED","created":"2025-08-15 21:10:54.000000000","updated":"2025-08-26 18:20:01.000000000","submitted":"2025-08-19 12:15:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":12,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"957572-bug/2120737","meta_rev_id":"6f0fe12f6881b0ca994487b1a9b116913dff89e4","_number":957572,"virtual_id_number":957572,"owner":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-19 12:15:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2025-08-18 05:53:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2025-08-18 05:53:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}],"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-15 21:27:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-08-15 23:54:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-18 05:53:01.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2025-08-19 01:14:49.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2025-08-26 18:12:54.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"CC"}],"messages":[{"id":"ed96376e50e5dcae5d57f7086bad06fc13727dac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2025-08-15 21:10:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a1f963679c21bce682d6919dcacf1f6ff8f39fae","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-15 21:27:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c764a60826184f58abc432264bb24db8\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c773fe33d3824d6ab62504c9416e81a6 : SUCCESS in 9m 15s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4fc971ff384e4f3aaad7722a73588fd4 : SUCCESS in 9m 52s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/9ef256ab66d2414191264e34d4ededec : SUCCESS in 14m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c0860e0ae1e9c3123ffd0b8e33689129c4c5a875","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-15 23:54:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b9360e0e1915444fa82c8b6fdb4700e7\n\n- grenade https://zuul.opendev.org/t/openstack/build/c1d37382efa4403696faed3882d87dbe : SUCCESS in 1h 07m 47s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/79d2f4032cd447289629214ad4d867b3 : SUCCESS in 1h 12m 42s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5e45df7b4384422ebc12c3a1b44fd028 : SUCCESS in 1h 42m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c089998e9c7c452aa70c7ba341e2603e : SUCCESS in 1h 10m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c5235c5ee9446c3a281967c71fd8444 : SUCCESS in 6m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/336f9c0146b7410394aa919a9138ac39 : SUCCESS in 6m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/622e15acae8a4f999dd3780331451d15 : SUCCESS in 3m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9974bcc24a7644fda380439148f028b1 : SUCCESS in 9m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77b5f31a45eb49e2ba783401a21ade4b : SUCCESS in 7m 26s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/12bb9ec09e2b41a39ccba43be00b2e51 : FAILURE in 9m 19s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/296b58c4fc064db099c647c1f97b35ec : FAILURE in 6m 29s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/9378d5f6d3524940937fc818cede7924 : SUCCESS in 1h 59m 28s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/353cf436d70646dea020b746bc78b39d : SUCCESS in 1h 12m 45s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/907bf5a0e78943d2a9d9e23ccc562e10 : SUCCESS in 2h 03m 15s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/54f7bdc5954443528ac46766db49647d : SUCCESS in 2h 39m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0272221e9f5b4bdbb4a89c4e811e22ea : SUCCESS in 29m 37s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/25493d5cdbc94676869bf259eb8f6239 : SUCCESS in 1h 49m 09s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/9c553eb4da4043448abc2ddf771aeb76 : FAILURE in 16m 52s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/15a62baddef743b29caed9769891ab56 : SUCCESS in 59m 30s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/68ae67652d2f4830b5595c345e9f2f31 : SUCCESS in 36m 43s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/5a18b30b1026431ea173a4831fb606e3 : SUCCESS in 26m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"58cacef261d63678bec8efd9c9193e960d2bff56","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2025-08-16 09:54:41.000000000","message":"Topic set to bug/2045769","accounts_in_message":[],"_revision_number":1},{"id":"98f69185c16e8b2de303c2b9a9e4bca25478d86d","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2025-08-16 09:55:06.000000000","message":"Topic bug/2045769 removed","accounts_in_message":[],"_revision_number":1},{"id":"e86a55e56f27413664749d5165940be499a5d7a5","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2025-08-16 09:55:10.000000000","message":"Topic set to bug/2120737","accounts_in_message":[],"_revision_number":1},{"id":"c650c24f29d0c3666af074a2885e4942b4eb451b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2025-08-16 18:55:12.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"5834129e239d7e32ca3e1bdd4ed935c1708bffc7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-16 19:12:11.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b4346c9fd619451e9973b34bda2ceed6\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ad644009989b4249916c34e95bd9b5dd : SUCCESS in 10m 13s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/79a0022324a942daa5889833b685acb0 : SUCCESS in 9m 42s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/3ccfa1a70a8f4b5e9660dbccd34382f4 : SUCCESS in 14m 53s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"01a78c0812c5fd8c6178d0c6354075e737d807ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-16 21:21:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d7e7623d85a74d3fa96e15d0625b5533\n\n- grenade https://zuul.opendev.org/t/openstack/build/1ebe764dbc95410e9c1ce3072961722d : SUCCESS in 1h 04m 16s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/bfad85690abd4928910122bca0f1b6e7 : SUCCESS in 32m 34s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2c2be49771bc460c8c52ed3f46b2d9e0 : SUCCESS in 1h 29m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/51d21166a0f1455e8a7e441bdf626f00 : SUCCESS in 1h 05m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f6afc304e704a04a62a63492bf8816b : SUCCESS in 2m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1ed2802414bc4f2b8630b5bd969350e4 : SUCCESS in 6m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/65fa27cbdd4748ad9638e40052ffa37a : SUCCESS in 8m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/32bff1c422584f40bfb552d063fa0c9e : SUCCESS in 9m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3199b93f197f4662bb2468b380b7ef74 : SUCCESS in 2m 53s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3d22a1d424ff447493a149b812771a81 : SUCCESS in 7m 28s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/0d923553bd7743849823f567fceec237 : SUCCESS in 11m 40s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/f3727febe8d6426cb8d95506b8360c9c : SUCCESS in 1h 49m 05s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/5aab8ffdbf694044a00d65188de117bc : SUCCESS in 2h 02m 48s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/fcb3526d9f5e412ca33a6b58b9eef7e5 : SUCCESS in 2h 00m 56s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/c1133f7fccfb4875a560cf726f40663e : SUCCESS in 2h 18m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2ca63dd3fbfc414683600011e24dc961 : SUCCESS in 58m 45s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ed87e5caf47c4aeebd4b98c26b045e86 : SUCCESS in 2h 08m 06s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/7d397a9593f0485595da9cbb117d8a60 : FAILURE in 15m 53s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/a25049647c534149a32a975c0227b700 : SUCCESS in 2h 05m 58s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/7219fcd7ab9e45279673dc9451bf55ba : SUCCESS in 34m 43s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/539edaa4e9c6420d962df6c254245a75 : SUCCESS in 1h 06m 20s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b75fb867783e5227a098cec5094df0f3be95110a","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-08-18 05:53:01.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0c680df02aba6296aebf46de06a57d012650e61a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-18 05:53:27.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ff31322e96fff36ab5e8258b0e35f0d568e48485","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-18 07:12:55.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c24ea7a62e1849f9b0732daacd1e852b\n\n- grenade https://zuul.opendev.org/t/openstack/build/1f6a4ae1f9494b1db4f7cc2a3839c75c : SUCCESS in 1h 00m 51s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5f95383e5f574cff949b845e2c303d2f : SUCCESS in 1h 02m 01s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1a82cdd20c414fa58ccf900ef43d4389 : SUCCESS in 1h 06m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bbb1ca272cff45eb918e0d4463d46e70 : SUCCESS in 32m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/246fe7e496894d2ab5fc8e3acec4327a : SUCCESS in 6m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c550f7e970e444c2868441dc85ffb493 : SUCCESS in 4m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/db7e0e746c6b47be9a4119367f6b08f1 : SUCCESS in 4m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51ffaa202bfa4b8c9ad76b9b9ceb4a63 : SUCCESS in 2m 58s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7aee6572b5bd493e80855553a5738771 : SUCCESS in 4m 44s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/bf62a88b15864a79b51edcb435454110 : SUCCESS in 7m 30s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/19c1eef1ba59415b8fc27d36665d8c28 : SUCCESS in 1h 14m 39s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/758c02eb220f447bab6f66ea6f732352 : SUCCESS in 1h 17m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/459baee2c9904c8da36480f1485b9293 : SUCCESS in 30m 44s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/91f9a3fbd9684530b2ab05f78c85138d : FAILURE in 1h 02m 51s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/40df1a7354e0437dae898f92f63d8639 : SUCCESS in 19m 57s","accounts_in_message":[],"_revision_number":2},{"id":"16f740765927e440dee753409ec09f75a4309b3f","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-08-18 07:18:09.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"661745bb8fa1df445d7aaf39685c2ee107036cd9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-18 07:39:06.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9f3e442e255b4ee5bd71dba91203be0d\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/deaf965394644cb395b7c0a09c145276 : SUCCESS in 7m 33s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/23ce1679438645b49eea0b4995796d33 : SUCCESS in 10m 32s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/5c9c74477c96421dbd29cb1e44c81b6f : SUCCESS in 15m 12s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"60952f1b171084b2a01f945e297a35dff41db7c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-18 09:18:04.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4c3fbe7e2a444c67b6b3c2712300b3e5\n\n- grenade https://zuul.opendev.org/t/openstack/build/ffa359eae7094463904cf57c28e8ddff : SUCCESS in 30m 00s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/30f20114c09d447dac87fead80e3508c : SUCCESS in 35m 53s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/919925f2be37461685d5391dd8d2b389 : SUCCESS in 1h 53m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8c22093ca0894e74b4fda09aa62221d1 : SUCCESS in 1h 01m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a483f4a3f23424dbd596947131e4a49 : SUCCESS in 3m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/01cd59e372664316b2f46b959e161ef8 : SUCCESS in 4m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c3560211496244e0aafdfd0bac3aa8d7 : SUCCESS in 3m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2a22c03d265e47c78909ae18fa2d3f68 : SUCCESS in 4m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c56b1e005e842b69e8fbcf355ab3d41 : SUCCESS in 2m 51s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d2efc8a205474040a74461f6e3588b23 : SUCCESS in 4m 21s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/428cd2bad8d04ddbb82087334898af4c : SUCCESS in 7m 41s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/ad23e8ee62a847728889a72f56258575 : SUCCESS in 1h 52m 51s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/1c7b60e300474949b5a883c8e41a232c : SUCCESS in 1h 03m 03s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/4f2be193c04c482ab1c09dd6f366371f : SUCCESS in 1h 15m 42s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/087bb23693704ceca3547874ef5bd6d2 : FAILURE in 1h 54m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0683f7d8f1404ceda5c98415425150e7 : SUCCESS in 1h 00m 21s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9fb62717a9e747d38cb6a8337436d193 : SUCCESS in 59m 30s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/aa3791597d8145a2bc148b53a2c12b95 : FAILURE in 31m 31s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/49996a990e2e48a58fdd5dd686864a84 : SUCCESS in 1h 15m 08s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f7fde27948f240fe9effe933168597d3 : SUCCESS in 19m 09s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/45c842afee8c4fac8534342350a6c92f : SUCCESS in 1h 04m 04s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"510fa9ee7b3b36dac260fde614ed12024f718c3e","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2025-08-18 11:30:49.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"62d73d7cdbccca90743fe79010c4ca6c6ed2d15e","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2025-08-18 11:30:57.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e4c1c08b219a4393ed432150e0ae31cfa44c38ec","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-18 12:52:40.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2a0a776f2cfc42229d3340c7ec37f05f\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b7f2c201688545018cc8d6551b55ee17 : SUCCESS in 10m 17s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ef9dc5a4ce414028b1114bb1584810fd : SUCCESS in 10m 49s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a9d84f5731224d108f6db48f0c144c64 : SUCCESS in 16m 17s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5d75046e113d08e56d6b72c5a0aaaccf118be7d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-18 13:29:11.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9f2bd8c042a946ad8b50d09b3e16f647\n\n- grenade https://zuul.opendev.org/t/openstack/build/aba4ea86ee67468d953ed4a8eac6056e : SUCCESS in 30m 11s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/268463220e21483c8ec06463094befb5 : SUCCESS in 1h 10m 31s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/84b7ed7d95d44cc7b7db126fad0f3865 : SUCCESS in 1h 35m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5e1b336f151443679e2602883d96c219 : SUCCESS in 32m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef71154b31e1404dab3fac0940b83620 : SUCCESS in 2m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e496a9a292e94c43a197fb4f5ccedaa8 : SUCCESS in 2m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c713d6da00e04733baa2201ed6854db4 : SUCCESS in 7m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4d4b8fc685e04c598159b0faccfeb09d : SUCCESS in 7m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e07119b4c664b37a1f528e14b7d98fc : SUCCESS in 7m 32s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e012b27e31454af38decb2cd80a4f6ff : SUCCESS in 9m 25s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/311499fcd6b344519ced0436fe2b8345 : SUCCESS in 7m 39s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/4c41e7d575724dc4a43011fb8a6ed7ba : SUCCESS in 1h 06m 37s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/290a3e4b100e4ea89787fd5230db1861 : FAILURE in 1h 00m 13s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/7d016be84c03412484a679a9d977c6ee : SUCCESS in 1h 54m 20s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/a63a0d70f4f04c20b27812aa9d005d82 : SUCCESS in 1h 22m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/aa65e4dde0e84da18cc2b4cfbc921cad : SUCCESS in 31m 04s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/0cca2db125a0442eb1927c13a789b156 : FAILURE in 1h 44m 56s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/411962d4c7f24a0892eb1f63efc9fe03 : FAILURE in 16m 40s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/09049793645f4337befecdee2dd96f94 : SUCCESS in 1h 39m 43s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/d7e5433872714bc2a267a05404d96150 : SUCCESS in 27m 15s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/fe3eb69a39c644d9a82aa5581c40e7c8 : SUCCESS in 1h 17m 49s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f8eae2efb6736eb66fd01eea7c031a165f14214d","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-08-18 13:31:27.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6b2d412d5fef9b3b671079e760e297165f025422","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-18 17:35:16.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/925fdc0b085f4871ae4f51fd34037fd0\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5a3ce58cbaf645cfa25fd23b0f99ea19 : SUCCESS in 6m 29s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9d96613e3bd548f9bee8edcb494466aa : SUCCESS in 10m 34s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/bb765836fddd4461a3cf7d3fbc70e365 : SUCCESS in 15m 02s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2bab669c617b44f7739b24691fe96f5d2290e3c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-18 18:16:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9610052f303f4155bdb211583da6c7b1\n\n- grenade https://zuul.opendev.org/t/openstack/build/6bb82d71b20641eaacf70d10df1dbacd : SUCCESS in 54m 03s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0749ab6621ef42379f95ccae85ee1525 : SUCCESS in 55m 06s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9a20f0db0b7d42fda46b1719ae8dbf76 : SUCCESS in 1h 35m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bb0276dded474f5d908c786c03327579 : SUCCESS in 1h 03m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4adc86433c03460bb9540e09a353edc3 : SUCCESS in 2m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/162d3ba95afb4774b9fd9490b5105ccd : SUCCESS in 3m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5af070cd38e24e55872e2296b74e2c7b : SUCCESS in 2m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7b6ae3a8ac9c43bc872d01c6d4ccbdb9 : SUCCESS in 10m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cecdc5e608104b10a132859b7d886efc : SUCCESS in 7m 23s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/808a29c8ca614f68855ab67a94c36c37 : SUCCESS in 7m 49s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/94dff0fcc6244e7988bc3d0e7fda4dae : SUCCESS in 11m 35s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/618610dd635845569d7c33b3490802c2 : SUCCESS in 1h 53m 00s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/dd6df765f17047d3b9d4fc2e45a8a2e8 : SUCCESS in 2h 10m 50s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/985c206656f34c12957a3d85fa6e8f78 : SUCCESS in 1h 56m 45s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/d5aa28ef436545c082db1215fd843cc4 : SUCCESS in 2h 07m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5e880b24bc6448f28f523b7fff88f757 : SUCCESS in 54m 14s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/917db5dcb2e94b3b8b51df0c74e7075d : SUCCESS in 1h 28m 07s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/4ca5c3eb0ec84127a439906647983710 : FAILURE in 29m 33s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/cd642f3baa6d466fa32f7105b805ee7a : SUCCESS in 1h 05m 55s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/992275e708eb4d079600069781c198fd : SUCCESS in 32m 11s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/5fbb7aa5c2a14c538e2c5bf6177f18a6 : SUCCESS in 53m 48s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"352d40d81b2c83e43560d328a451448fd5a181cd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-18 18:17:21.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"87ef59e9f43dd9f39faab9e394eeac0b26e74053","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-18 20:24:34.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6452a08a42424d50b2b2125141b08ffb\n\n- grenade https://zuul.opendev.org/t/openstack/build/77ab56bf587d4bbab9d5761239207227 : SUCCESS in 53m 31s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/76f2b78acdbb465482679d7ba29e686d : SUCCESS in 51m 43s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ef7f9938a6d84c14b99622761f786da7 : SUCCESS in 1h 36m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/da90fedece074c46bdc3cc9931576510 : SUCCESS in 51m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/46afaec7d6d6403b904431f8191e4ac8 : SUCCESS in 5m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cec48cf26b324783b406927936931763 : SUCCESS in 4m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6f1524c296064a738a03910b1d09a056 : SUCCESS in 6m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd32f6e8888b4e38a6dbb0b95905021c : SUCCESS in 8m 30s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/15798c28d8044feab72740c404dfabf2 : SUCCESS in 8m 07s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/9b79eb4de1d34f9a81ae269cb76bd430 : FAILURE in 9m 42s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/506d214c7676412abf4af189bf34542b : SUCCESS in 2h 04m 38s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/1a561914c2fd4099981c50b9bb1b4c4d : SUCCESS in 2h 05m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1dc5b3bf63b545a0a3e010621b080bf7 : SUCCESS in 1h 00m 29s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/1e7f49b0ba014d4e8cb78bd5832ce77c : SUCCESS in 1h 46m 16s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/395ff394d7994d55bdfdb25f704cee02 : SUCCESS in 25m 33s","accounts_in_message":[],"_revision_number":2},{"id":"340fcf7059f4921ad920ce9782c24d44ceb2fd0d","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-19 01:14:49.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"51434f17df4a173c7e447c9bb1ef42933079ba72","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 01:31:02.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5a79f93aac634330ad2747ca2b3aa29c\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3911151876944dcf85ec0d5dd1367058 : SUCCESS in 6m 56s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/fce44b4094c044e5a47892c839781556 : SUCCESS in 8m 59s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c264f399ec974c3182fd5b8439c61f10 : SUCCESS in 14m 20s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"afa0fda99aab5ba48bd6a44a5fdc0f615ac8eae9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 03:26:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/baf43145f8e54d5194905742a09009f5\n\n- grenade https://zuul.opendev.org/t/openstack/build/f833626b31724648bd1d129d8cd5daf3 : SUCCESS in 58m 18s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7f35963130cd476f81a30f3168b39762 : SUCCESS in 1h 02m 12s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b5f4eeb238244088bc58789f40ca3821 : SUCCESS in 1h 31m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/df1034c0409f44ed8969138a437fc8a5 : SUCCESS in 37m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d79fc9b81e94863ae30cb2305eb5abb : SUCCESS in 3m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e283b43d2d504f37b76db8d55223a2d5 : SUCCESS in 7m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d91de204db1744efa22038123b18c897 : SUCCESS in 5m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5e63048a74964d0083277f824733c7b4 : SUCCESS in 10m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fada016b25ae4440a152e84e8ea573a5 : SUCCESS in 5m 31s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/de828e5d92324e11a1e8aaf361474f3a : SUCCESS in 6m 57s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/dcb8641846564d429c014d6b7aeb29ed : SUCCESS in 12m 42s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/3248febdbf164068900118b0c0067b5d : SUCCESS in 1h 16m 27s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/1916e88a9557430d9984b96eb47ed59d : SUCCESS in 1h 04m 22s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/c4769c8ef16344dabe22189c98eedc7c : SUCCESS in 1h 04m 50s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/cd56b716a5ab4b399137d4e4d7f59088 : SUCCESS in 2h 08m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/16e0bbf74d694fa89619135217fa61e4 : SUCCESS in 36m 03s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f4e0255cbe6b48e3a6ac3900f5e67824 : SUCCESS in 1h 55m 25s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/5803f13d26e14394ac3928e008c4f14c : FAILURE in 22m 39s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/10113ce7388c4bcaa48b940a7c58b2ad : FAILURE in 1h 10m 56s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/220185085c364ac08996fd10631d957e : SUCCESS in 14m 59s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/ec20bc6616df4ee7a0d7c322e0f69f75 : SUCCESS in 26m 56s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8f391c51c54ff77f621feb1d711d7cc10c7d4c71","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 03:26:25.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"13270de592c10252da6c75800749a690cb4a2431","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 05:21:13.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b1454aa8753d4b21a5e8f3e452a1e76e\n\n- grenade https://zuul.opendev.org/t/openstack/build/e92a1cc1a9644462a1d3a18b108cdb3d : SUCCESS in 57m 25s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b80ffdd329af430b82dbdcd2abf6dd23 : SUCCESS in 1h 00m 23s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8b7353a7ac2441bcb314d951417e220c : SUCCESS in 1h 26m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8910fa6766f74b3dbc3ebddf2755b5f0 : SUCCESS in 51m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b93c729ff08848d1ab45c84435be93ed : SUCCESS in 3m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5280081deb504611846dc89b94c7b476 : SUCCESS in 2m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2c3bc0ce69d24fc591614bc540dc4902 : SUCCESS in 9m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/199618199cd14e1997d1da5a90d174ad : SUCCESS in 4m 32s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1d5d1649a399486381bb12afb3bd50a7 : RETRY_LIMIT in 1m 58s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/abaa3573f1d348d3a3698e6e3627e770 : SUCCESS in 10m 38s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/a65830ad85704961969dc3ba0e63bc39 : SUCCESS in 59m 15s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f8d45d601a404754a9962fda72cdc941 : SUCCESS in 1h 53m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/63e0426b666e491eba21eb212da63e1d : SUCCESS in 29m 58s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/2531cab12fd6428b9cff5a79ad9cabe7 : SUCCESS in 1h 44m 12s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/24f026c795c645bcb1de545f4a4e3b92 : SUCCESS in 20m 04s","accounts_in_message":[],"_revision_number":2},{"id":"4f62e5b7a5ab4a4a9666702706f739b2acdbdb67","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-08-19 05:40:31.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f964851341c0e295bf3745e97632348f68819683","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 05:57:36.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bf87012d9eae41319c617868ed0b7de0\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/394f22e43ad3496389c8e1dbc18d62ab : SUCCESS in 8m 46s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8c3bbdc259044881b71235bc1536532b : SUCCESS in 10m 03s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e438d65c9d7e4f36a17f3bcbceedd931 : SUCCESS in 14m 48s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"df2adbad18ca13b87b6c0806d4b3a6880f117167","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 07:42:28.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b3a344fc010a4830b1afbae0c67d01f4\n\n- grenade https://zuul.opendev.org/t/openstack/build/b86cb5b1006649c1b1c6525055b43cb7 : SUCCESS in 42m 27s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/db02fabf2c1d4bc0b6345219dd6ff9b3 : SUCCESS in 54m 15s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a045a2fbe9fb48e2b0824bfa17841b2d : SUCCESS in 1h 29m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4dcc205b08dc4539b3b7703a04b3cbb8 : SUCCESS in 37m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7483edfa458f4aae89db9b13afc2eaa7 : SUCCESS in 4m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/687fbe022f4f47d082cdb5ddcc0c142f : SUCCESS in 6m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/05c0a9ecc75e43328f6a2ee7e3daea12 : SUCCESS in 7m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4a47608fc5bc464db90e8b3b30c74829 : SUCCESS in 6m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86f2f27fd5444e3e9b0337293e60a96a : SUCCESS in 3m 58s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6e42eb557e3f4c5aa668dedbd7571b17 : SUCCESS in 6m 07s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/0693e4ce360e47338643af2942425429 : SUCCESS in 9m 56s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/ece1a1617cb2483cabac3e9cdf042ad1 : SUCCESS in 2h 01m 18s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/0a0d268848ca42a7be56fb7f9fdc7c92 : SUCCESS in 1h 03m 32s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/5c50e8ad622f4807b322f910658f4f5e : SUCCESS in 1h 41m 47s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/d0e6ebd22b344bab96b112ddddacef05 : SUCCESS in 1h 49m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b563179ec14d45c697d40d378549d329 : SUCCESS in 40m 16s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c51e7d9eedc845b8b1e4512915b9d448 : FAILURE in 59m 47s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/e98fe4a7b584406ebdf9acd5c22374fe : FAILURE in 30m 56s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/958757b71cde40949831d9c4d4812547 : SUCCESS in 1h 53m 14s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/1a6ebd7c9dcd43d3abf904d5b7e327b5 : FAILURE in 32m 33s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/d82c67628028401d839d19bb8841b2d7 : SUCCESS in 45m 51s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4e48260c8567679e319002c122e41a3f6c7fc05c","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-08-19 07:45:08.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"dc833429add78fe22b9c47e5badd0c3ca99bc0f7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 08:19:20.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2247779350a840afb31380b1eedf0c0e\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9ef91af6bf8d4af7be65e84a0fdbdea3 : SUCCESS in 13m 08s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c7ef72cac0b248c2ab6ae5dbf28ffd9d : SUCCESS in 12m 40s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d2dd8302ad6f4b49829d6b58d0dbef7e : SUCCESS in 16m 10s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"bf708e0e026bfb00f83c679bb785197dacb37c15","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2025-08-19 09:27:51.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5232b1de4c7afdf976c0c5acfdeb6bbebca397a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 10:02:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4271ca158d9844cfbbe8d04f790ae612\n\n- grenade https://zuul.opendev.org/t/openstack/build/3cfc848a69c84f32b96c0d85e40b3a2d : SUCCESS in 31m 27s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1c7b0df99ef54decb5a6415e898357e6 : SUCCESS in 59m 40s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6bfa1f672e8a4d4d9fca060578de8ef2 : SUCCESS in 1h 37m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/489dcd5bc33f4a7c897012187b877572 : SUCCESS in 39m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4ddb40e54d4456ba56096d1450ae5e9 : SUCCESS in 3m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/720502b8d40c4341a46566583e22bd0a : SUCCESS in 5m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/295ea65aee1c400bb63a7e13851db9a3 : SUCCESS in 7m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/32badda9a06148f19ee091cebb880776 : SUCCESS in 5m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/becacaa38d03476ab8de24d6b683ca02 : SUCCESS in 4m 14s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0d40ad2af7f5482d8af0c34e78750f23 : SUCCESS in 6m 42s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/5339b339a63f4adeb0413a6d9f9f7875 : SUCCESS in 8m 41s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/d0405fb2bd7d45bdb3d398deaf575f97 : SUCCESS in 1h 12m 15s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/0894d5ebc3b34b0099e4839b1b6a8b4d : SUCCESS in 2h 04m 23s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/a92f960a300b4c109fed9c3237d1ec2a : SUCCESS in 1h 39m 19s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/6de809bfcbc147c99ab870f20aba1123 : SUCCESS in 2h 11m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/88e0c5cb57474d03849e7b018ac2e6c2 : SUCCESS in 41m 50s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/67d5e0f597714dd6bad7b3cbf620ddc8 : SUCCESS in 1h 09m 57s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/ff6b20158d474dde9bf5668ce9ff20cc : FAILURE in 24m 19s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/09d7c97f570e43b9843ff78c0c61ef50 : SUCCESS in 1h 43m 11s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/17a22da1d301409b873dea5d48956b53 : SUCCESS in 16m 14s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/c4f56fd4c5cc48a19d5c4cb74a4f401c : SUCCESS in 36m 47s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c19e64d0b1affdcd9898cb68b1206f407342582c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 10:02:56.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"160689701d6ee3586ce50ca61282bea3aaf00c6c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 12:15:34.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b82f37c1bd6746d7b3654c218d1b70ce\n\n- grenade https://zuul.opendev.org/t/openstack/build/9fe5a52f74104a15883c12facd4be9e8 : SUCCESS in 52m 33s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/479300a2a9ae493a98abe81e8765e5f3 : SUCCESS in 32m 42s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/538f3c00096a425a919b096229c84d84 : SUCCESS in 1h 21m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7ca2f470f9f64caeb00003c9fc18d382 : SUCCESS in 52m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d7d73236fdd4556b565928fbfef9303 : SUCCESS in 5m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ebde7175569b492785d92d16eafeb62e : SUCCESS in 2m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5c6055c2e7c5440299bcbab999f0e2e1 : SUCCESS in 3m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a42513c81f724bef92ace50ee3d98bc0 : SUCCESS in 3m 36s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/82f7e2f3c1114c81818c79bfde783a83 : SUCCESS in 5m 36s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/983cde051b064e4992a84a24b8f28d0b : SUCCESS in 9m 43s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/8724d77c02444b2489007c640baa9613 : SUCCESS in 1h 52m 44s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/9d6ec3dd2ada455a9dfdd4c7595186e3 : SUCCESS in 1h 58m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2181175e1c514f199e34da0e0bb2496d : SUCCESS in 1h 00m 21s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a884ba9174de4fdf9d8d30afdf060c34 : SUCCESS in 56m 46s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/65403a2f8fe948af9aa0a698164f17fc : SUCCESS in 27m 53s","accounts_in_message":[],"_revision_number":2},{"id":"4a6b9283e3c0962089e5a54903ca085002a9337e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 12:15:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"ad9b8f0321dc96a01cd22c9f71ab544f363c1a9f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 12:16:41.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3e9525f7241a45969e35dff06ef2938a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d660eb444b97472e8550494e7439a464 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":2},{"id":"2f2884495c3ca98d1adaadac59808c42a8ebd067","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-08-26 18:12:54.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"6f0fe12f6881b0ca994487b1a9b116913dff89e4","tag":"autogenerated:gerrit:revert","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-08-26 18:20:01.000000000","message":"Created a revert of this change as Ifaf4b083218071f4706989f2931d0b40bc588afc","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e40759c15bb224671398f7142afe12fa0c3ac336","revisions":{"b354ed19a275e20dafee974426917b5b42b01f63":{"kind":"REWORK","_number":1,"created":"2025-08-15 21:10:54.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/72/957572/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/72/957572/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/957572/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/957572/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/957572/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/72/957572/1"}}},"commit":{"parents":[{"commit":"fda4fba5e9bf971691d1106aaf9ecf052455209a","subject":"Merge \"Drop SQLALCHEMY_WARN_20\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fda4fba5e9bf971691d1106aaf9ecf052455209a"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2025-08-15 20:27:10.000000000","tz":330},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2025-08-15 21:10:50.000000000","tz":330},"subject":"Fix: Policy for new location APIs","message":"Fix: Policy for new location APIs\n\nCurrently, in the new location APIs, we use ``role:service`` as\nthe check for the existence of service role or not.\nHowever, in a request with service token enabled, we get the\nfollowing context object:\n\n\u0027roles\u0027: [\u0027member\u0027, \u0027manager\u0027, \u0027admin\u0027, \u0027reader\u0027],\n\u0027service_user_name\u0027: \u0027cinder\u0027,\n\u0027service_project_name\u0027: \u0027service\u0027,\n\u0027service_roles\u0027: [\u0027creator\u0027, \u0027service\u0027]\n\nThis issue is not seen while adopting the add location API because\nof two reasons:\n1. The ``add_image_location`` policy validates SERVICE_OR_PROJECT_MEMBER\nwhere the project member check (({PROJECT_MEMBER} and project_id:%(owner)s))\npasses the policy check\n2. We include the ``service_roles`` from the context before the if check[1]\n\nFrom the above, we can see that there is a bug in the new location APIs\npolicy checks which is only visible when we start consuming the GET location\nAPI.\nThis patch fixes it to correct the check from ``role:service`` to\n``service_roles:service`` as already did in this patch[2].\n\n[1] https://opendev.org/openstack/glance/src/commit/fda4fba5e9bf971691d1106aaf9ecf052455209a/glance/api/v2/images.py#L1160-L1161\n[2] https://review.opendev.org/c/openstack/glance/+/942153\n\nCloses-Bug: #2120737\nChange-Id: Iab40d4cf1d6fb836a370c8b3ebfbb84182265d7a\nSigned-off-by: Rajat Dhasmana \u003crajatdhasmana@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b354ed19a275e20dafee974426917b5b42b01f63"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b354ed19a275e20dafee974426917b5b42b01f63"}]},"branch":"refs/heads/master"},"e40759c15bb224671398f7142afe12fa0c3ac336":{"kind":"REWORK","_number":2,"created":"2025-08-16 18:55:12.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/72/957572/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/72/957572/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/957572/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/957572/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/957572/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/72/957572/2"}}},"commit":{"parents":[{"commit":"fda4fba5e9bf971691d1106aaf9ecf052455209a","subject":"Merge \"Drop SQLALCHEMY_WARN_20\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fda4fba5e9bf971691d1106aaf9ecf052455209a"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2025-08-15 20:27:10.000000000","tz":330},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2025-08-16 18:54:57.000000000","tz":330},"subject":"Fix: Policy for new location APIs","message":"Fix: Policy for new location APIs\n\nCurrently, in the new location APIs, we use ``role:service`` as\nthe check for the existence of service role or not.\nHowever, in a request with service token enabled, we get the\nfollowing context object:\n\n\u0027roles\u0027: [\u0027member\u0027, \u0027manager\u0027, \u0027admin\u0027, \u0027reader\u0027],\n\u0027service_user_name\u0027: \u0027cinder\u0027,\n\u0027service_project_name\u0027: \u0027service\u0027,\n\u0027service_roles\u0027: [\u0027creator\u0027, \u0027service\u0027]\n\nThis issue is not seen while adopting the add location API because\nof two reasons:\n1. The ``add_image_location`` policy validates SERVICE_OR_PROJECT_MEMBER\nwhere the project member check (({PROJECT_MEMBER} and project_id:%(owner)s))\npasses the policy check\n2. We include the ``service_roles`` from the context before the if check[1]\n\nFrom the above, we can see that there is a bug in the new location APIs\npolicy checks which is only visible when we start consuming the GET location\nAPI.\nThis patch fixes it to correct the check from ``role:service`` to\n``service_roles:service`` as already did in this patch[2].\n\n[1] https://opendev.org/openstack/glance/src/commit/fda4fba5e9bf971691d1106aaf9ecf052455209a/glance/api/v2/images.py#L1160-L1161\n[2] https://review.opendev.org/c/openstack/glance/+/942153\n\nCloses-Bug: #2120737\nChange-Id: Iab40d4cf1d6fb836a370c8b3ebfbb84182265d7a\nSigned-off-by: Rajat Dhasmana \u003crajatdhasmana@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e40759c15bb224671398f7142afe12fa0c3ac336"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e40759c15bb224671398f7142afe12fa0c3ac336"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
