)]}'
{"id":"openstack%2Fglance~957670","triplet_id":"openstack%2Fglance~master~Ia38eb9b74b197dbcea0974ce04f6e5fefdf3017b","project":"openstack/glance","branch":"master","topic":"bug/2120834","attention_set":{},"removed_from_attention_set":{"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2025-08-19 03:06:16.000000000","reason":"Change was abandoned"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-08-19 03:06:16.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ia38eb9b74b197dbcea0974ce04f6e5fefdf3017b","subject":"Revert \"Remove LegacyEngineFacade\"","status":"ABANDONED","created":"2025-08-18 11:04:48.000000000","updated":"2025-08-19 03:06:16.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"4c88664bfe8178c9ec0b5217d9c3a05352b9d323","_number":957670,"virtual_id_number":957670,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-08-18 13:10:08.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":-1,"date":"2025-08-18 12:09:15.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":-1,"date":"2025-08-18 12:33:35.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-18 12:09:15.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2025-08-18 12:10:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-08-18 13:10:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5641263807fb026e065a16147c5b566aabfad85d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2025-08-18 11:04:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fb8c29db5c1fd82f7b9cb88fe64f634807d66b7d","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2025-08-18 11:06:51.000000000","message":"Topic set to bug/2120834","accounts_in_message":[],"_revision_number":1},{"id":"82f0d7ed95786dd9edc1af8771dc7044a0d08854","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-18 12:09:15.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"49251823702e6f8841b55c562ffcd276f92e74d7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-18 12:10:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/634385c6b8ae40b8a5e0802b3a65a40e\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/36746f45470549d992a336e5237d6fea : SUCCESS in 11m 46s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/08d0a6b6f0734e21bab7c2cf0be16f41 : SUCCESS in 10m 05s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b51238c8694d4fd7bba09a9883001b7b : SUCCESS in 14m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7618714e391d473585f1d1a78882cc5cf5c221c7","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2025-08-18 12:33:35.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fc41480caad869c3dc6da5e2c597ea7be635581c","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2025-08-18 12:34:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e10d328d934d48b067b98269653ea5ec45c70bd5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-18 13:10:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fd6ea865f3024d658a0ccf2410375274\n\n- grenade https://zuul.opendev.org/t/openstack/build/bcc4faa33d974dfd98193d24563da62b : SUCCESS in 59m 49s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7e9e33c0ba9f4171b9f4e016e66cc475 : SUCCESS in 1h 05m 49s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/132f36c2a47c4b379a67b6e2fc8658ca : SUCCESS in 1h 36m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7b5a7d47aa1541668a584a583b954369 : SUCCESS in 1h 01m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ff7e854ec6940a2896846e7e8626f91 : SUCCESS in 3m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/34ed3f0fe6ec4f1fa38ea8467ba64324 : SUCCESS in 6m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/28c1362baf3c44e4861a6f966e633dfb : SUCCESS in 3m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f7e133e6e8b94ed89d29918913b01b61 : SUCCESS in 9m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e5f29a8c86b4ef58d73b1464f65766d : SUCCESS in 4m 52s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b11d3dbf4de949d1accb76f9b709552e : SUCCESS in 6m 47s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/86f8a3a29abb49dabff0a424c4d8040e : SUCCESS in 11m 52s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/47f20c34c0ae4c1689e09e1d86aa18d3 : SUCCESS in 2h 00m 00s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/9a085bc96c10433f908d97b0f445fe9d : SUCCESS in 1h 39m 40s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/82e7620ada4c4a8d87fa427099c12866 : SUCCESS in 1h 23m 32s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/8564f9c44e9340f3b7190f27ea3259af : SUCCESS in 1h 16m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8b011c8ee7274430ba50813c3345058e : SUCCESS in 58m 14s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/1540ac55ae1740808c751d6c866ae217 : FAILURE in 1h 35m 43s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/bbbef498ff494f699eee92ac848447e2 : FAILURE in 32m 31s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/bf3ef5a0e2564c9da0bf211edf55a3a2 : SUCCESS in 1h 57m 51s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/d37c8675f40e4d33a832d59293511fc8 : SUCCESS in 24m 45s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/e46d5e8d6437478a8b0151c28431dae1 : SUCCESS in 1h 10m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4c88664bfe8178c9ec0b5217d9c3a05352b9d323","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2025-08-19 03:06:16.000000000","message":"Abandoned\n\nwas for confirmation, in favor of https://review.opendev.org/c/openstack/glance/+/957682","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ee81eb0947fb46d111fb095ac4c15c1a728117ee","revisions":{"ee81eb0947fb46d111fb095ac4c15c1a728117ee":{"kind":"REWORK","_number":1,"created":"2025-08-18 11:04:48.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/70/957670/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/70/957670/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/70/957670/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/70/957670/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/70/957670/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/70/957670/1"}}},"commit":{"parents":[{"commit":"fda4fba5e9bf971691d1106aaf9ecf052455209a","subject":"Merge \"Drop SQLALCHEMY_WARN_20\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fda4fba5e9bf971691d1106aaf9ecf052455209a"}]}],"author":{"name":"Yatin Karel","email":"ykarel@redhat.com","date":"2025-08-18 11:03:14.000000000","tz":330},"committer":{"name":"Yatin Karel","email":"ykarel@redhat.com","date":"2025-08-18 11:04:44.000000000","tz":330},"subject":"Revert \"Remove LegacyEngineFacade\"","message":"Revert \"Remove LegacyEngineFacade\"\n\nThis reverts commit 4653c3e338bb626cc5b50ed25e33a8968379cd0b.\n\nRelated-Bug: #2120834\nChange-Id: Ia38eb9b74b197dbcea0974ce04f6e5fefdf3017b\nSigned-off-by: Yatin Karel \u003cykarel@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ee81eb0947fb46d111fb095ac4c15c1a728117ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ee81eb0947fb46d111fb095ac4c15c1a728117ee"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{}}}]}
