)]}'
{"id":"openstack%2Fglance~957682","triplet_id":"openstack%2Fglance~master~I4297c3ddd73456a7b2458d608ad2511e7e797340","project":"openstack/glance","branch":"master","topic":"bug/2120834","attention_set":{},"removed_from_attention_set":{"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2025-08-18 14:23:25.000000000","reason":"\u003cGERRIT_ACCOUNT_9303\u003e replied on the change","reason_account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-08-19 12:27:20.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4297c3ddd73456a7b2458d608ad2511e7e797340","subject":"Fix broken osprofiler call for sqlalchemy","status":"MERGED","created":"2025-08-18 12:08:37.000000000","updated":"2025-08-19 12:28:16.000000000","submitted":"2025-08-19 12:27:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"957682-bug/2120834","meta_rev_id":"c5cbe369e1028cb81c9ef6f70fc0d95a99529ff2","_number":957682,"virtual_id_number":957682,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-19 12:27:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"recommended":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":1,"date":"2025-08-18 13:53:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2025-08-19 10:18:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":1,"date":"2025-08-19 10:18:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-18 12:54:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-08-18 13:53:59.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2025-08-18 14:14:37.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2025-08-18 14:20:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0710de449cbfed7d26051883ad1957fa718e55e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-18 12:08:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"404d65deb1703b6dc887e1b0d9660ccefeaf2188","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-18 12:54:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/04bb3497a8254a56a62f9acd2b610357\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0fe8d84f8767410c85811d9ef953f7b2 : SUCCESS in 6m 35s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f04af3388f864f75af28dbccda134713 : SUCCESS in 11m 16s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/5686541ffc4c4d5e9bbaccdb51db877f : SUCCESS in 10m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5e1ae995da349545c11f001ef0139dd30ce43181","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-18 13:07:33.000000000","message":"Topic set to bug/2120834","accounts_in_message":[],"_revision_number":1},{"id":"171e12a74ff7ac49b7d5de606afcd672ea8f6387","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2025-08-18 13:53:59.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c63ff02567998bec55d53f844dc4b15124c47184","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-18 14:20:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8cb286ff6e804b52a59f62500499c035\n\n- grenade https://zuul.opendev.org/t/openstack/build/2b8c6ff0fb7f488a982b255a8ac1eeeb : SUCCESS in 1h 00m 55s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a90273c725c94fe1aa392a60dde9204a : SUCCESS in 59m 19s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/39b8bd6fca0745fe92936161c3771d79 : SUCCESS in 1h 26m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b97a1bd2426246eba91510e096141b66 : SUCCESS in 1h 02m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d5d32a0b5c742419cb73dbdef56dcf6 : SUCCESS in 6m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7438afdfda844b0cbb2217bef3db344b : SUCCESS in 5m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2c0ca6085b4d433eb84de3bb57c82783 : SUCCESS in 7m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/17e279ce10594241bf8b97e92ac5eaa6 : SUCCESS in 10m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4954376716745c18e718c5c1ed21e6b : SUCCESS in 7m 05s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/af45c8b4aad349bd97c611ea1a5bdf19 : SUCCESS in 7m 51s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/fb246f6feea143d986b83470602e22c0 : SUCCESS in 11m 51s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/1f2fa786b2334372a96c14e0c788e546 : SUCCESS in 1h 53m 48s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/146f311742c142e7a7f50a1292947974 : SUCCESS in 1h 54m 37s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/44714dc07fd447eab869945eab003630 : SUCCESS in 1h 54m 37s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/f11c9d1368d84d1b8b3335f049115b0d : SUCCESS in 2h 09m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0cb1aa2c1b5a4ec396cefca90b10d6b3 : SUCCESS in 59m 56s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ea2739b1bd664bf79aec4390dfe377e0 : SUCCESS in 1h 51m 30s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/8309954e10ff47ce8672fc117e3b7f44 : FAILURE in 31m 27s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/8d7e9e8b33b2429696fd2bbab3920ea3 : SUCCESS in 1h 48m 50s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/5db322d72e92460ea97c057f74a7841d : FAILURE in 32m 27s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/7dd17ad5763543e0ad1481da3b045b87 : SUCCESS in 1h 03m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"06175530c95a616078fe2dc2a0e0ccc093f0406f","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-18 14:22:01.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"444f8f5acc4a1a6022513973c15884825a59d96e","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-08-18 14:23:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7812a780efe3b13022102012f88db63240b31e07","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-08-18 14:24:12.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d8c1d069cbb11ef0eaf3655488b3a226875808d3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-18 17:50:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e7ffaed7c7de453baccef1f1cde3cd4f\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7e136bd945fd4c72a3adfa01601a7e6c : SUCCESS in 10m 45s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/cc2a90fc80254c06ac929bdbb75897c0 : SUCCESS in 9m 41s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/94ef107986da49ec827424504539c995 : SUCCESS in 15m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0d68bb47f237970051742a73e59edf938877b744","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-18 19:06:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/562e88b7a55749f3a81d1b95237cf9f6\n\n- grenade https://zuul.opendev.org/t/openstack/build/94a48252317a4e01b7b453cc3efcd5c5 : SUCCESS in 1h 04m 17s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/cc831aa212794381814790a24cb16304 : SUCCESS in 32m 33s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d65c6b52ccb34971baa409d5e1014c7a : SUCCESS in 1h 31m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/15b60be89ba244a188ab83ecc3638b2d : SUCCESS in 39m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/311b0d0007be4d8b830b4bc4ddd90057 : SUCCESS in 2m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/63743619482e4b318f0caa431a089a48 : SUCCESS in 2m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/814f765255034469801e861ba33ae7be : SUCCESS in 7m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fee6639165654b11bfd37402af966f24 : SUCCESS in 5m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f49550c7de441e98c8181d3034ed004 : SUCCESS in 3m 40s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4bf8b6f81db149659bf0a625fe00008a : SUCCESS in 8m 13s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/d5507caf394349a890bdb2741a96a5c2 : SUCCESS in 9m 44s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/887c385a62944f05a62bfd38f149981d : SUCCESS in 1h 17m 49s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/9bd5c56181564e8a85d087783ecb1062 : SUCCESS in 1h 07m 47s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/62caf835706849ef95b8d015f97cd59c : SUCCESS in 1h 12m 56s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/f1532134e3da48dba92a8eaf44402e37 : SUCCESS in 2h 08m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b5bfdb8be1cb434da97b1c380cc59c36 : SUCCESS in 49m 44s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/eeb23ef57a0a4421a5edb377520f62a8 : SUCCESS in 1h 10m 21s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/85add2937b3547e8b753b821bdde77d9 : FAILURE in 17m 18s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/c0644d6fc44143b7a6ff0c770df9a24b : SUCCESS in 1h 59m 08s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/b6ebe2abe2f84f2382c9c8fd8fc8c029 : SUCCESS in 32m 28s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/c1e3731f4e9844cabd167ec51e2dff40 : SUCCESS in 28m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f9df48208da0760c08bd96cffa584dbff6d4b949","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-08-19 10:18:15.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4fba71e6d706306ed3e9817a63fa260a56cdb8bf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 10:18:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6535ee1d4e3747f1b17fe419cb041b0b3de8b989","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 12:27:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/97c2c5c280614f11a0f6df7bdf69801b\n\n- grenade https://zuul.opendev.org/t/openstack/build/f56dee85c84f4a2b9ce2def1cdafc326 : SUCCESS in 1h 04m 30s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/473f16cbcc13404fbe14eff1d5f2601b : SUCCESS in 53m 18s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/eccbb80e743643c5b934b41d315b3a8f : SUCCESS in 1h 20m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6e989ceab45a4e41ab93f29634aa8ebf : SUCCESS in 1h 05m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a3cc880416a4f7caf1869301d909ab6 : SUCCESS in 6m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c4196f3f79144d2799a22de32958e4cc : SUCCESS in 2m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/301e19e724784ca6bf52a1b554c8ffc5 : SUCCESS in 3m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9383356d91d248b9b5ac9d6a8f5e4d2e : SUCCESS in 8m 05s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c2f76bf223464c8d92e415ffdb91b65d : SUCCESS in 6m 48s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/9d31c9703de747bbbf25cc2be564a21c : SUCCESS in 6m 24s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/b89fa915f73b4520b60ee6cda88343ad : SUCCESS in 1h 49m 01s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/0969e797528b44e897d42b7b1893cf12 : SUCCESS in 1h 02m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f880c8432746471aa2208b3efc9a3110 : SUCCESS in 34m 36s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/3daa3b6c15e044dfa320db6fecd1352a : SUCCESS in 1h 56m 26s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/5d9db089ee824a0788af8538a269f982 : SUCCESS in 19m 34s","accounts_in_message":[],"_revision_number":1},{"id":"1196c22f8ee2b0d4bda270d3ad8ee6db7fa9f7c2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 12:27:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c5cbe369e1028cb81c9ef6f70fc0d95a99529ff2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 12:28:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/07db039dd8944cff9633045870749ec8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/edf9757c0aaf4af8b9dd70f49362e9ad : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b038f5728f4fe1221ab08343b6205659d9d5c72d","revisions":{"b038f5728f4fe1221ab08343b6205659d9d5c72d":{"kind":"REWORK","_number":1,"created":"2025-08-18 12:08:37.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/82/957682/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/82/957682/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/82/957682/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/82/957682/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/82/957682/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/82/957682/1"}}},"commit":{"parents":[{"commit":"fda4fba5e9bf971691d1106aaf9ecf052455209a","subject":"Merge \"Drop SQLALCHEMY_WARN_20\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fda4fba5e9bf971691d1106aaf9ecf052455209a"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-18 12:07:29.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-18 12:08:25.000000000","tz":540},"subject":"Fix broken osprofiler call for sqlalchemy","message":"Fix broken osprofiler call for sqlalchemy\n\nCloses-Bug: #2120834\nChange-Id: I4297c3ddd73456a7b2458d608ad2511e7e797340\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b038f5728f4fe1221ab08343b6205659d9d5c72d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b038f5728f4fe1221ab08343b6205659d9d5c72d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
