)]}'
{"id":"openstack%2Fglance~958196","triplet_id":"openstack%2Fglance~master~Ibf359a495c69249611b7861a94619a131621a234","project":"openstack/glance","branch":"master","topic":"bug/2121144","attention_set":{},"removed_from_attention_set":{"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2025-11-25 04:25:15.000000000","reason":"Change was abandoned"},"37432":{"account":{"_account_id":37432,"name":"Koya Watanabe","display_name":"Koya Watanabe","email":"koywatan@lycorp.co.jp","username":"koya-watanabe"},"last_update":"2025-09-17 13:17:40.000000000","reason":"\u003cGERRIT_ACCOUNT_37432\u003e replied on the change","reason_account":{"_account_id":37432,"name":"Koya Watanabe","display_name":"Koya Watanabe","email":"koywatan@lycorp.co.jp","username":"koya-watanabe"}},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2025-11-25 04:25:15.000000000","reason":"Change was abandoned"},"8878":{"account":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"last_update":"2025-11-25 04:25:15.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ibf359a495c69249611b7861a94619a131621a234","subject":"releasenotes: Epoxy known issue for S3-compatible storage checksum behavior","status":"ABANDONED","created":"2025-08-21 13:14:38.000000000","updated":"2025-11-25 04:25:15.000000000","total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2745781ea6feb3fde8cd80b7895f571647497bd8","_number":958196,"virtual_id_number":958196,"owner":{"_account_id":37432,"name":"Koya Watanabe","display_name":"Koya Watanabe","email":"koywatan@lycorp.co.jp","username":"koya-watanabe"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-09-18 07:22:52.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":-2,"date":"2025-09-18 07:11:22.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}],"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-21 13:29:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-26 07:45:30.000000000","updated_by":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"reviewer":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"state":"REVIEWER"},{"updated":"2025-09-01 16:44:55.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"CC"},{"updated":"2025-09-02 15:08:55.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"4f2f33fddf2edb9dce576328a5c72a0a9887d60e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37432,"name":"Koya Watanabe","display_name":"Koya Watanabe","email":"koywatan@lycorp.co.jp","username":"koya-watanabe"},"date":"2025-08-21 13:14:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fbc2ccca90e962f568dee512219d97f5f6157c43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-21 13:29:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b13dc6fccc5641548000e446feaabd7d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d907bb1d66c4c069fdaa6346a0b0aae : SUCCESS in 6m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56b8f625555242ada3547c0ff8ece258 : SUCCESS in 5m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/15997941e3824f32a97afdd77c69540c : SUCCESS in 7m 45s","accounts_in_message":[],"_revision_number":1},{"id":"d62e01a88ce8a4fdf8946a9efcbcad76f49f76ce","author":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"date":"2025-08-26 07:45:30.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3d9e4351a32aaf9b477b87bc84e9f702a0021e7e","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-09-01 16:44:55.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a14bc4419cbfe3866bb64caa1ea31a8fe05f1c64","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-09-01 16:50:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fb9d5cabe970ac03137f826b171eed7beffa9a11","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-09-02 15:08:55.000000000","message":"Patch Set 1: Code-Review-2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"40ef7f8c5bd822659fc895860723f41e41f82008","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37432,"name":"Koya Watanabe","display_name":"Koya Watanabe","email":"koywatan@lycorp.co.jp","username":"koya-watanabe"},"date":"2025-09-17 13:03:39.000000000","message":"Uploaded patch set 2.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"changekind:TRIVIAL_REBASE OR **is:MIN**\")\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"18edf58c6e5d09489a5dcc9bbad75c2dfda6191a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37432,"name":"Koya Watanabe","display_name":"Koya Watanabe","email":"koywatan@lycorp.co.jp","username":"koya-watanabe"},"date":"2025-09-17 13:05:50.000000000","message":"Uploaded patch set 3.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"changekind:TRIVIAL_REBASE OR **is:MIN**\")\n","accounts_in_message":[],"_revision_number":3},{"id":"2042214559bd1b1c4160138db214fb9092fb1f2c","author":{"_account_id":37432,"name":"Koya Watanabe","display_name":"Koya Watanabe","email":"koywatan@lycorp.co.jp","username":"koya-watanabe"},"date":"2025-09-17 13:17:40.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"94460c08745f08edd46ee90a74e31b3f49638e08","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37432,"name":"Koya Watanabe","display_name":"Koya Watanabe","email":"koywatan@lycorp.co.jp","username":"koya-watanabe"},"date":"2025-09-17 13:24:47.000000000","message":"Uploaded patch set 4.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"changekind:TRIVIAL_REBASE OR **is:MIN**\")\n","accounts_in_message":[],"_revision_number":4},{"id":"38e2771afac00ae64719252de5ffca6c32843699","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-17 13:43:17.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b77727d2795d49c3ac634c0f7d525e0d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/82cdcaf984be4b7aaa55f7945c0e9fd1 : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7526ad965b14036a30a16982fc2fbb8 : SUCCESS in 7m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7d02e78334734aa6b06e88abbce07381 : SUCCESS in 6m 18s","accounts_in_message":[],"_revision_number":4},{"id":"20b81857bbc1f57adca5f9fa953edf8709a1db4e","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":37432,"name":"Koya Watanabe","display_name":"Koya Watanabe","email":"koywatan@lycorp.co.jp","username":"koya-watanabe"},"date":"2025-09-17 14:31:54.000000000","message":"Topic set to bug/2121144","accounts_in_message":[],"_revision_number":4},{"id":"b90c2846d32f57d41454bc10d75fd54d15cf04d5","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":37432,"name":"Koya Watanabe","display_name":"Koya Watanabe","email":"koywatan@lycorp.co.jp","username":"koya-watanabe"},"date":"2025-09-17 14:34:29.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":4},{"id":"cb5c2c38ba767296b107582cef7c772f4f2f930c","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":37432,"name":"Koya Watanabe","display_name":"Koya Watanabe","email":"koywatan@lycorp.co.jp","username":"koya-watanabe"},"date":"2025-09-17 14:34:55.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":4},{"id":"03030ae8d6dec745186dce9babc77ea8849c28db","author":{"_account_id":37432,"name":"Koya Watanabe","display_name":"Koya Watanabe","email":"koywatan@lycorp.co.jp","username":"koya-watanabe"},"date":"2025-09-17 14:57:38.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"bcc898a078f11eeeaf8d5a3e23e97c2b4251774c","author":{"_account_id":37432,"name":"Koya Watanabe","display_name":"Koya Watanabe","email":"koywatan@lycorp.co.jp","username":"koya-watanabe"},"date":"2025-09-17 14:57:50.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"8a78629c066acacdd5c7aafef297695782dbbef9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37432,"name":"Koya Watanabe","display_name":"Koya Watanabe","email":"koywatan@lycorp.co.jp","username":"koya-watanabe"},"date":"2025-09-18 07:11:22.000000000","message":"Patch Set 5: Commit message was updated.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"changekind:TRIVIAL_REBASE OR **is:MIN**\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"8afbfa014b7b9bbb5ec3ddefbd275916f73daa17","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-18 07:22:52.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8acd98b9e79141d8b52ab881e88ff8a9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7be1a36730a643058414503f81f0b77d : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/018ea76b064f446e93751344cecc7ece : SUCCESS in 5m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3cec0d0a03b0448fa272bcc89237f736 : SUCCESS in 6m 46s","accounts_in_message":[],"_revision_number":5},{"id":"4fa9e8c328de687bfaead57464c14c5b0e09c5ae","author":{"_account_id":37432,"name":"Koya Watanabe","display_name":"Koya Watanabe","email":"koywatan@lycorp.co.jp","username":"koya-watanabe"},"date":"2025-09-19 07:15:02.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"2745781ea6feb3fde8cd80b7895f571647497bd8","tag":"autogenerated:gerrit:abandon","author":{"_account_id":37432,"name":"Koya Watanabe","display_name":"Koya Watanabe","email":"koywatan@lycorp.co.jp","username":"koya-watanabe"},"date":"2025-11-25 04:25:15.000000000","message":"Abandoned\n\nThis is a bug related to Epoxy, so I created a new Change on the stable/2025.1 branch:\nhttps://review.opendev.org/c/openstack/glance/+/961597","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"0de3cfbd7a41828fc8987451b0b3e10eb346d612","revisions":{"7427acedeff76e3e0df4b68636a1f79a28abef0a":{"kind":"REWORK","_number":1,"created":"2025-08-21 13:14:38.000000000","uploader":{"_account_id":37432,"name":"Koya Watanabe","display_name":"Koya Watanabe","email":"koywatan@lycorp.co.jp","username":"koya-watanabe"},"ref":"refs/changes/96/958196/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/96/958196/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/96/958196/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/96/958196/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/96/958196/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/96/958196/1"}}},"commit":{"parents":[{"commit":"d5b2e4f47d61c206c6d90f09005b3a93a92e06a0","subject":"Merge \"Fix broken osprofiler call for sqlalchemy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d5b2e4f47d61c206c6d90f09005b3a93a92e06a0"}]}],"author":{"name":"Koya Watanabe","email":"koywatan@lycorp.co.jp","date":"2025-08-21 12:52:43.000000000","tz":540},"committer":{"name":"Koya Watanabe","email":"koywatan@lycorp.co.jp","date":"2025-08-21 13:14:19.000000000","tz":540},"subject":"releasenotes: Epoxy known issue for S3-compatible storage checksum behavior","message":"releasenotes: Epoxy known issue for S3-compatible storage checksum behavior\n\nAdd an issues-section release note for the Epoxy series documenting\nthat image uploads to S3-compatible storage can fail when the backend\ndoes not support the S3 checksum headers/validation used by default\nin Epoxy.\n\nThe note also recommends setting the following environment variables\nfor the Glance API service as a workaround until the backend supports\nthe checksum behavior:\nAWS_REQUEST_CHECKSUM_CALCULATION\u003dWHEN_REQUIRED\nAWS_RESPONSE_CHECKSUM_VALIDATION\u003dWHEN_REQUIRED\n\nCloses-Bug: #2121144\nChange-Id: Ibf359a495c69249611b7861a94619a131621a234\nSigned-off-by: Koya Watanabe \u003ckoywatan@lycorp.co.jp\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7427acedeff76e3e0df4b68636a1f79a28abef0a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7427acedeff76e3e0df4b68636a1f79a28abef0a"}]},"branch":"refs/heads/master"},"43dfbc64def9c5886051ef9a67461c0d80f048af":{"kind":"REWORK","_number":2,"created":"2025-09-17 13:03:39.000000000","uploader":{"_account_id":37432,"name":"Koya Watanabe","display_name":"Koya Watanabe","email":"koywatan@lycorp.co.jp","username":"koya-watanabe"},"ref":"refs/changes/96/958196/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/96/958196/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/96/958196/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/96/958196/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/96/958196/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/96/958196/2"}}},"commit":{"parents":[{"commit":"d5b2e4f47d61c206c6d90f09005b3a93a92e06a0","subject":"Merge \"Fix broken osprofiler call for sqlalchemy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d5b2e4f47d61c206c6d90f09005b3a93a92e06a0"}]}],"author":{"name":"Koya Watanabe","email":"koywatan@lycorp.co.jp","date":"2025-08-21 12:52:43.000000000","tz":540},"committer":{"name":"Koya Watanabe","email":"koywatan@lycorp.co.jp","date":"2025-09-17 13:02:58.000000000","tz":540},"subject":"releasenotes: Epoxy known issue for S3-compatible storage checksum behavior","message":"releasenotes: Epoxy known issue for S3-compatible storage checksum behavior\n\nAdd an issues-section release note for the Epoxy series documenting\nthat image uploads to S3-compatible storage can fail when the backend\ndoes not support the S3 checksum headers/validation used by default\nin Epoxy.\n\nThe note also recommends setting the following environment variables\nfor the Glance API service as a workaround until the backend supports\nthe checksum behavior:\nAWS_REQUEST_CHECKSUM_CALCULATION\u003dWHEN_REQUIRED\nAWS_RESPONSE_CHECKSUM_VALIDATION\u003dWHEN_REQUIRED\n\nCloses-Bug: #2121144\nChange-Id: Ibf359a495c69249611b7861a94619a131621a234\nSigned-off-by: Koya Watanabe \u003ckoywatan@lycorp.co.jp\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/43dfbc64def9c5886051ef9a67461c0d80f048af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/43dfbc64def9c5886051ef9a67461c0d80f048af"}]},"branch":"refs/heads/master"},"80b88e8e2ca1ef6a25dd54157c9b81c57c9cba2c":{"kind":"REWORK","_number":3,"created":"2025-09-17 13:05:50.000000000","uploader":{"_account_id":37432,"name":"Koya Watanabe","display_name":"Koya Watanabe","email":"koywatan@lycorp.co.jp","username":"koya-watanabe"},"ref":"refs/changes/96/958196/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/96/958196/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/96/958196/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/96/958196/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/96/958196/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/96/958196/3"}}},"commit":{"parents":[{"commit":"d5b2e4f47d61c206c6d90f09005b3a93a92e06a0","subject":"Merge \"Fix broken osprofiler call for sqlalchemy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d5b2e4f47d61c206c6d90f09005b3a93a92e06a0"}]}],"author":{"name":"Koya Watanabe","email":"koywatan@lycorp.co.jp","date":"2025-08-21 12:52:43.000000000","tz":540},"committer":{"name":"Koya Watanabe","email":"koywatan@lycorp.co.jp","date":"2025-09-17 13:05:30.000000000","tz":540},"subject":"releasenotes: Epoxy known issue for S3-compatible storage checksum behavior","message":"releasenotes: Epoxy known issue for S3-compatible storage checksum behavior\n\nAdd an issues-section release note for the Epoxy series documenting\nthat image uploads to S3-compatible storage can fail when the backend\ndoes not support the S3 checksum headers/validation used by default\nin Epoxy.\n\nThe note also recommends setting the following environment variables\nfor the Glance API service as a workaround until the backend supports\nthe checksum behavior:\nAWS_REQUEST_CHECKSUM_CALCULATION\u003dWHEN_REQUIRED\nAWS_RESPONSE_CHECKSUM_VALIDATION\u003dWHEN_REQUIRED\n\nCloses-Bug: #2121144\nChange-Id: Ibf359a495c69249611b7861a94619a131621a234\nSigned-off-by: Koya Watanabe \u003ckoywatan@lycorp.co.jp\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/80b88e8e2ca1ef6a25dd54157c9b81c57c9cba2c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/80b88e8e2ca1ef6a25dd54157c9b81c57c9cba2c"}]},"branch":"refs/heads/master"},"0ef9307e2cff27eb2bfb2132adbb825315eb1b4e":{"kind":"REWORK","_number":4,"created":"2025-09-17 13:24:47.000000000","uploader":{"_account_id":37432,"name":"Koya Watanabe","display_name":"Koya Watanabe","email":"koywatan@lycorp.co.jp","username":"koya-watanabe"},"ref":"refs/changes/96/958196/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/96/958196/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/96/958196/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/96/958196/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/96/958196/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/96/958196/4"}}},"commit":{"parents":[{"commit":"d5b2e4f47d61c206c6d90f09005b3a93a92e06a0","subject":"Merge \"Fix broken osprofiler call for sqlalchemy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d5b2e4f47d61c206c6d90f09005b3a93a92e06a0"}]}],"author":{"name":"Koya Watanabe","email":"koywatan@lycorp.co.jp","date":"2025-08-21 12:52:43.000000000","tz":540},"committer":{"name":"Koya Watanabe","email":"koywatan@lycorp.co.jp","date":"2025-09-17 13:24:39.000000000","tz":540},"subject":"releasenotes: Epoxy known issue for S3-compatible storage checksum behavior","message":"releasenotes: Epoxy known issue for S3-compatible storage checksum behavior\n\nAdd an issues-section release note for the Epoxy series documenting\nthat image uploads to S3-compatible storage can fail when the backend\ndoes not support the S3 checksum headers/validation used by default\nin Epoxy.\n\nThe note also recommends setting the following environment variables\nfor the Glance API service as a workaround until the backend supports\nthe checksum behavior:\nAWS_REQUEST_CHECKSUM_CALCULATION\u003dWHEN_REQUIRED\nAWS_RESPONSE_CHECKSUM_VALIDATION\u003dWHEN_REQUIRED\n\nCloses-Bug: #2121144\nChange-Id: Ibf359a495c69249611b7861a94619a131621a234\nSigned-off-by: Koya Watanabe \u003ckoywatan@lycorp.co.jp\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0ef9307e2cff27eb2bfb2132adbb825315eb1b4e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0ef9307e2cff27eb2bfb2132adbb825315eb1b4e"}]},"branch":"refs/heads/master"},"0de3cfbd7a41828fc8987451b0b3e10eb346d612":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2025-09-18 07:11:22.000000000","uploader":{"_account_id":37432,"name":"Koya Watanabe","display_name":"Koya Watanabe","email":"koywatan@lycorp.co.jp","username":"koya-watanabe"},"ref":"refs/changes/96/958196/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/96/958196/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/96/958196/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/96/958196/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/96/958196/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/96/958196/5"}}},"commit":{"parents":[{"commit":"d5b2e4f47d61c206c6d90f09005b3a93a92e06a0","subject":"Merge \"Fix broken osprofiler call for sqlalchemy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d5b2e4f47d61c206c6d90f09005b3a93a92e06a0"}]}],"author":{"name":"Koya Watanabe","email":"koywatan@lycorp.co.jp","date":"2025-08-21 12:52:43.000000000","tz":540},"committer":{"name":"Koya Watanabe","email":"koywatan@lycorp.co.jp","date":"2025-09-18 07:11:22.000000000","tz":0},"subject":"releasenotes: Epoxy known issue for S3-compatible storage checksum behavior","message":"releasenotes: Epoxy known issue for S3-compatible storage checksum behavior\n\nAdd an issues-section release note for the Epoxy series documenting\nthat image uploads to S3-compatible storage can fail when the backend\ndoes not support the S3 checksum headers/validation used by default\nin Epoxy.\n\nThe note also recommends setting the following environment variables\nfor the Glance API service as a workaround until the backend supports\nthe checksum behavior:\nAWS_REQUEST_CHECKSUM_CALCULATION\u003dWHEN_REQUIRED\nAWS_RESPONSE_CHECKSUM_VALIDATION\u003dWHEN_REQUIRED\n\nRelated-Bug: #2121144\nChange-Id: Ibf359a495c69249611b7861a94619a131621a234\nSigned-off-by: Koya Watanabe \u003ckoywatan@lycorp.co.jp\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0de3cfbd7a41828fc8987451b0b3e10eb346d612"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0de3cfbd7a41828fc8987451b0b3e10eb346d612"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Code-Review\u003dMIN"],"failing_atoms":["label:Code-Review\u003dMAX"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
