)]}'
{"id":"openstack%2Fglance~958567","triplet_id":"openstack%2Fglance~master~Ifaf4b083218071f4706989f2931d0b40bc588afc","project":"openstack/glance","branch":"master","topic":"bug/2120737","attention_set":{},"removed_from_attention_set":{"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2025-08-28 20:13:09.000000000","reason":"Change was abandoned"},"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2025-08-28 20:13:09.000000000","reason":"Change was abandoned"},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2025-08-27 02:06:59.000000000","reason":"\u003cGERRIT_ACCOUNT_8556\u003e replied on the change","reason_account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2025-08-28 20:13:09.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ifaf4b083218071f4706989f2931d0b40bc588afc","subject":"Revert \"Fix: Policy for new location APIs\"","status":"ABANDONED","created":"2025-08-26 18:20:01.000000000","updated":"2025-08-28 20:24:22.000000000","total_comment_count":17,"unresolved_comment_count":2,"has_review_started":true,"revert_of":957572,"meta_rev_id":"ae31ec9461fd69b7f3c4478647cca53ab3879b69","_number":958567,"virtual_id_number":958567,"owner":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-08-26 22:24:36.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"disliked":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2025-08-27 02:06:59.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2025-08-27 05:00:23.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":-1,"date":"2025-08-27 17:02:27.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"CC":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-26 18:20:01.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2025-08-26 18:20:01.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2025-08-26 18:20:01.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2025-08-26 18:20:01.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-26 18:45:19.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"}],"messages":[{"id":"50df1937a6b1863e8adcc9de37f0045c0fbf01d9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-08-26 18:20:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eac106685a4635e3ba87b7969ac3592f0ad373f5","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-08-26 18:45:19.000000000","message":"Patch Set 1: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"6574213fe0cba3e229d793cbd7b70514487d1409","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-26 18:47:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e260af10fe9a4ee5945883b5d09b457c\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/792ad8ab64a64d459fdd7f76d7dbbe4c : SUCCESS in 7m 58s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2470bc4499ba406596233e7cbfb79276 : SUCCESS in 9m 07s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/8ba8b4c2ec3a403a98e53bd8b35b4508 : SUCCESS in 17m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"dce0e06216211afb7a9943df1d25fc4c2ddeb714","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-08-26 19:03:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3687b4888923861c84349765bbacdf9d2df16b28","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-08-26 19:05:12.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"43355bcffff5c39009f97a52038056a92e19706f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-08-26 20:31:13.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":2},{"id":"67365d560993465be3dfc5cf59504803030df6ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-26 22:24:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c28c48491f6c4a76835456be80a6590d\n\n- grenade https://zuul.opendev.org/t/openstack/build/ff7706d7d6f0469b8e9526b7ca2da558 : SUCCESS in 34m 30s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a14400f11ab047818bb5dbedbd712371 : SUCCESS in 38m 54s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/559d92b32d394373a97762899e3f2d09 : SUCCESS in 1h 24m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b7a9d1674127467b81d1b50cf86d57cb : SUCCESS in 34m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d90d5e62791427da9730771d1606b27 : SUCCESS in 6m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/86cb07c9faaa4c6e84c8f2c3834e0990 : SUCCESS in 6m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b6617f5b737a4805970378169edf87fd : SUCCESS in 5m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a32eee0e2f884855bf910219f0823f96 : SUCCESS in 7m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a78ab645d075496c932e0101d811c026 : SUCCESS in 6m 54s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2a7e362343db4d5cabdf3a90d69be439 : SUCCESS in 7m 02s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/8e7cd2304a5143c29e11edf86e87196a : SUCCESS in 7m 36s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/1a58406ced2a4549bf35b4dff2f0d35c : SUCCESS in 1h 47m 34s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/e7a7147d1baa4707bb52e302b71214f7 : SUCCESS in 1h 12m 53s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f7905902786b45149c042bf637caa29c : SUCCESS in 1h 12m 59s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/613d7659e94c44108dfd162b14700e19 : SUCCESS in 1h 13m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8fd502290859401fa91f8913f45070c4 : SUCCESS in 33m 10s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c42b5b9102924be5b978924d932fafd1 : SUCCESS in 1h 39m 26s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/a7fc8b0352f94d02ac1515ab26154d58 : FAILURE in 18m 20s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/417210c7032443548a7418d02e8c9636 : SUCCESS in 1h 47m 43s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/d83a28828eb540ddae6dfaf0c7374767 : SUCCESS in 17m 52s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/9a5976e5c77e4734a69d7f899ab7fd80 : SUCCESS in 52m 57s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6d7cb29878ace98425f0e4ce895d52d6e1d692b0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-26 22:31:24.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b1a90168bc3b4579b5657c345826ac97\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d9bda92fe9074c659bf4366b2e8da527 : SUCCESS in 9m 52s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6afdcf099cc84c93a7cfe40c524e0697 : SUCCESS in 11m 15s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/58acd0edaf4d4a858e105ae8f9250a7e : SUCCESS in 15m 51s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9937712c1769b13b91d758ffed38f54c088291b4","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-27 01:56:12.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e90623da32c5ecb4a1309e297e56adb0138b4f68","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-08-27 02:06:59.000000000","message":"Patch Set 2: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"7dc089e57392e277e96b40a0188fccd5a0467a79","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-27 02:12:12.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"d58ccd506fe8364a4351ae47c8d6f013ddc55f7f","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2025-08-27 05:00:23.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"06945b8e6b363ec9dc68c6e50c8698d7a4014f42","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-08-27 17:02:27.000000000","message":"Patch Set 2: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"356d40b066a6a02038484c1fc3502214f608921e","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2025-08-27 17:20:04.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b76f17e5de2c4c8c9b5250ddefa390815753cf9d","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-08-27 23:59:54.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"074e94ecedaa9e786173d855707813d1aa31643e","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-08-28 00:11:37.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0c0fbc81af667665a35141449279bf49e51be809","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-08-28 04:15:48.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"61391016ab247ccdcae61ca6c0323fef0bf31dc7","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-08-28 04:18:36.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0d28f1f9a8e30d1abe0e82b863216e6dae00168f","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-08-28 20:13:09.000000000","message":"Abandoned\n\ngmann is fixing this properly https://review.opendev.org/c/openstack/glance/+/958715","accounts_in_message":[],"_revision_number":2},{"id":"ae31ec9461fd69b7f3c4478647cca53ab3879b69","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-08-28 20:24:22.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"17e424d2f3f18e575cbb448404bf3d0b31a8e2cf","revisions":{"3ce5b2cba13062f4e4f4074a322c66719cc24666":{"kind":"REWORK","_number":1,"created":"2025-08-26 18:20:01.000000000","uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"ref":"refs/changes/67/958567/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/67/958567/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/67/958567/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/67/958567/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/67/958567/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/67/958567/1"}}},"commit":{"parents":[{"commit":"e40759c15bb224671398f7142afe12fa0c3ac336","subject":"Fix: Policy for new location APIs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e40759c15bb224671398f7142afe12fa0c3ac336"}]}],"author":{"name":"sean mooney","email":"smooney@redhat.com","date":"2025-08-26 18:20:01.000000000","tz":0},"committer":{"name":"Gerrit Code Review","email":"review@openstack.org","date":"2025-08-26 18:20:01.000000000","tz":0},"subject":"Revert \"Fix: Policy for new location APIs\"","message":"Revert \"Fix: Policy for new location APIs\"\n\nThis reverts commit e40759c15bb224671398f7142afe12fa0c3ac336.\n\nReason for revert: This is a bug in how cinder was calling glance not in the glance policy.\n\nThe service to service APIs in https://governance.openstack.org/tc/goals/selected/consistent-and-secure-rbac.html#isolate-service-to-service-apis-to-the-service-role do not use the service token for validation the should assert the existing of the service role on the standard token which is what glance did previously\n\nChange-Id: Ifaf4b083218071f4706989f2931d0b40bc588afc\nSigned-off-by: Sean Mooney \u003cwork@seanmooney.info\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3ce5b2cba13062f4e4f4074a322c66719cc24666"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3ce5b2cba13062f4e4f4074a322c66719cc24666"}]},"branch":"refs/heads/master"},"17e424d2f3f18e575cbb448404bf3d0b31a8e2cf":{"kind":"REWORK","_number":2,"created":"2025-08-26 20:31:13.000000000","uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"ref":"refs/changes/67/958567/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/67/958567/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/67/958567/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/67/958567/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/67/958567/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/67/958567/2"}}},"commit":{"parents":[{"commit":"e40759c15bb224671398f7142afe12fa0c3ac336","subject":"Fix: Policy for new location APIs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e40759c15bb224671398f7142afe12fa0c3ac336"}]}],"author":{"name":"sean mooney","email":"smooney@redhat.com","date":"2025-08-26 18:20:01.000000000","tz":0},"committer":{"name":"sean mooney","email":"smooney@redhat.com","date":"2025-08-26 20:31:13.000000000","tz":0},"subject":"Revert \"Fix: Policy for new location APIs\"","message":"Revert \"Fix: Policy for new location APIs\"\n\nThis reverts commit e40759c15bb224671398f7142afe12fa0c3ac336.\n\nReason for revert: This is a bug in how cinder was calling glance not in the glance policy.\n\nThe service to service APIs in https://governance.openstack.org/tc/goals/selected/consistent-and-secure-rbac.html#isolate-service-to-service-apis-to-the-service-role do not use the service token for validation the should assert the existing of the service role on the standard token which is what glance did previously\n\nChange-Id: Ifaf4b083218071f4706989f2931d0b40bc588afc\nSigned-off-by: Sean Mooney \u003cwork@seanmooney.info\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/17e424d2f3f18e575cbb448404bf3d0b31a8e2cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/17e424d2f3f18e575cbb448404bf3d0b31a8e2cf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{}}}]}
