)]}'
{"id":"openstack%2Fglance~958715","triplet_id":"openstack%2Fglance~master~I50909e6bdb3227ca99b7eba642546da791f9552a","project":"openstack/glance","branch":"master","topic":"glance-service-api","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2025-09-12 13:48:30.000000000","reason":"\u003cGERRIT_ACCOUNT_5314\u003e replied on the change","reason_account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2025-09-15 12:41:58.000000000","reason":"Change was submitted"},"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2025-09-15 12:41:58.000000000","reason":"Change was submitted"},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2025-09-12 14:13:14.000000000","reason":"\u003cGERRIT_ACCOUNT_8122\u003e replied on the change","reason_account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2025-09-15 12:41:58.000000000","reason":"Change was submitted"},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2025-09-15 08:33:50.000000000","reason":"\u003cGERRIT_ACCOUNT_27615\u003e replied on the change","reason_account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}}},"hashtags":[],"change_id":"I50909e6bdb3227ca99b7eba642546da791f9552a","subject":"Fix glance service policy rule","status":"MERGED","created":"2025-08-28 03:39:18.000000000","updated":"2025-09-15 12:42:51.000000000","submitted":"2025-09-15 12:41:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":28,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"958715-glance-service-api","meta_rev_id":"e884e7d89ff438cb4815c6fa9ee15e638184bcba","_number":958715,"virtual_id_number":958715,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-15 12:41:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2025-09-12 13:48:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2025-09-12 14:13:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-09-12 08:57:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2025-09-12 14:13:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-28 03:57:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-08-28 05:24:08.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"CC"},{"updated":"2025-08-28 06:39:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-28 20:18:36.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"CC"},{"updated":"2025-08-28 20:23:28.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2025-08-29 05:34:40.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2025-08-29 05:34:40.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2025-08-29 18:03:09.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"CC"},{"updated":"2025-09-05 07:53:50.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2025-09-11 22:01:01.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"}],"messages":[{"id":"84b48429696cdb5af65686872b802b0041ad3ac1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-08-28 03:39:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b30e4ab2231362b274e27e6d511b763e89d49cf3","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-08-28 03:40:36.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"1fe94c9c86a1d17bbc7560672874f6e8117455b3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-28 03:57:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b00e468103354615bcf65bddbe9ace10\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/86260b3b936e4b4db69595cabb9876c3 : SUCCESS in 8m 57s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0db7ed1d323d43a38cac6009e610e6b4 : SUCCESS in 11m 45s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c924f9e6eae34b1f92f2a86ecc3fc120 : SUCCESS in 15m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"12dab4237df334e4063ca3706722fd54b550b533","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-08-28 04:09:25.000000000","message":"Uploaded patch set 2: Commit message was updated.\n\nOutdated Votes:\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"85fc23c63b87fbef68f3fb0c1c8de8aa28d57fa3","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-08-28 04:11:38.000000000","message":"Patch Set 2: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ef0eefbf797fcbce1a6f3893fd3734eed3194ee3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-28 04:54:22.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b107651edcfa4f03b6cd0cc7eb3de39a\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/64b4f0fca5844ed7805a75ded64fb076 : SUCCESS in 9m 26s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5ac25ee8548849a88745a8b6b6831a4f : SUCCESS in 11m 56s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/af07ca9c136a4a96a7f5f03c6f1f5083 : SUCCESS in 16m 12s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e72d1e3d3aa3cb2e3092df33926281d4e5c0529b","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-08-28 05:24:08.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"dc871e5c967917d99ec3417849b2144c0b71197b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-28 06:39:41.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/49b234d24a2a4135a0ce291b70b1e4c5\n\n- grenade https://zuul.opendev.org/t/openstack/build/b92cba28470d41e69686f4fda29dffaa : SUCCESS in 1h 00m 33s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2d17aa80bc7b4fa7bc30658977cb0f52 : SUCCESS in 1h 00m 40s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6c4ad08205d34a2f9bb641b75abb43d8 : SUCCESS in 1h 25m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b07aaa1c4127457bbe0e69a85fc3a92b : SUCCESS in 49m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/206e3d495a8341a7a10651dc49151498 : FAILURE in 5m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a480b9b776ff47a6acd2bdfb84d7cef4 : SUCCESS in 4m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7eba8c44fda84b35a749be8fa40b11fd : SUCCESS in 8m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6cec2d47b2dd4f65aa360a3c23760df9 : SUCCESS in 9m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4092ccaeeb634568918839e094ce1c1b : SUCCESS in 6m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/36b78b3ff91f43549043394cd76a99cb : SUCCESS in 3m 53s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7e9122d2582c47c9bdd87ecd3d868117 : FAILURE in 8m 53s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/cae96100ee634151be1b294f2c8e09dc : FAILURE in 6m 41s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/2230fedc9b084b3ea9026e94f9e06427 : SUCCESS in 1h 35m 37s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/e387686812c84f2f81bad7ccc24fbb03 : SUCCESS in 1h 46m 12s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/4cf509de226d4ae6b540c25359c6c81b : FAILURE in 1h 50m 15s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/18c12a39fd8c4ca9bb552ba667088d42 : SUCCESS in 2h 20m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/124b1be92500468ca1392f43923cd573 : SUCCESS in 56m 14s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a20e5251f328489f8fd9b2edb88d1dd8 : SUCCESS in 1h 56m 37s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/bf52c23daaa54c8ba2d05f09083c6013 : FAILURE in 31m 18s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/1750b39dc1e842db97424784e1cf5d77 : SUCCESS in 1h 48m 19s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/73d71d68448c490abcc0b355ebb8aa70 : SUCCESS in 18m 50s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/78bba59d111a41a4a1a1b00f9d33e07e : SUCCESS in 1h 08m 57s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"85564fe6b218ea9d9cca8e5e212afa32e8b22cac","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-08-28 20:18:36.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"8e40ee4c5ebf8db824a58aba3d0bbd27ffc322e1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-08-28 20:23:28.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":3},{"id":"6ab2de8d201e41587fef9c05f220c79a1c52a489","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-08-28 20:25:53.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"aca722e302e19c630af57a34ff8e568212bac4be","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-08-28 21:09:24.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"6735ab61234793ab0b0b1f64038cce25efed9359","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-28 23:18:12.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c31bb24be794cd68aaf098de37e8a3c\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a8b7740102ed4405bf9b9220d23762ba : SUCCESS in 10m 43s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/82db1c21313f45c3aee45e36e46f4703 : SUCCESS in 13m 42s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/3d765a7c310047199b495fe9b294fdca : SUCCESS in 12m 50s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"653d19f4ba893724495fd270f453b362d2cf1607","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-28 23:38:58.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a2d0f523d8148589c19e7da108687b7\n\n- grenade https://zuul.opendev.org/t/openstack/build/a59229c086b54c308286b81b1972a924 : SUCCESS in 1h 03m 24s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/976c01c68ced47e2848db3cf9212d230 : SUCCESS in 58m 09s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f5b81cca775a4e33b27e8f3e5fe5813b : SUCCESS in 1h 10m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dba16cd2b3024ee1bab10b5eea7db09a : SUCCESS in 48m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b1cda0a13314530abbcd2dcd2f6c994 : SUCCESS in 6m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/04abc96a72414f29b23f45cd4a6e9359 : SUCCESS in 6m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3b26d2c333d1429bb4b2fdcd677e91cb : SUCCESS in 6m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d8e2f4ace53345baa846fc23e3a8c4be : SUCCESS in 13m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c01129076d64722b1e6d677ed64a80f : SUCCESS in 5m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3cb7cf24de7441d88b7426fbc66a68ca : SUCCESS in 6m 10s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5334b462db3b4669a5acc82f7af0e2da : SUCCESS in 7m 31s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/363e0fa93b2f40068d6d11f2ad3e77e7 : SUCCESS in 11m 16s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/c8dfd356101c4966b42df42fc155ec84 : FAILURE in 2h 17m 40s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/08ad740b3af4456fab33f5e83443ec6b : SUCCESS in 1h 57m 42s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/79401bfb566140f2a1c1b851a1cec20b : SUCCESS in 2h 24m 19s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/1efa57eed63643e4ae1397c2270c6917 : SUCCESS in 1h 07m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c4bc3119ad0c49a5b0279211497321d2 : SUCCESS in 36m 46s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/56d2d63e0fdd4db1b8505e586942b451 : SUCCESS in 1h 43m 15s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/411594a7d5eb4293add51c6d5c18a200 : FAILURE in 32m 29s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/ff39458aec64452aba88943354e134eb : SUCCESS in 1h 16m 20s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/71f10829f11e420bae2b9bf8e3f0779f : SUCCESS in 19m 02s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/76d2e04d8edf4d93889abc8346202893 : SUCCESS in 30m 48s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"ac520f93a5ba5fa8be6f2d09c0c8f90c2c475b37","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-08-29 05:34:40.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"78c3a6e1218f64c943d73eefe7c86749303377a9","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2025-08-29 18:03:09.000000000","message":"Patch Set 4:\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"1b97e5fb4a95856c2cdd65839c9ba3dadcc3835d","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-08-29 19:23:10.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"dcda69c0cdd3a65f5cd9db710b64f3fbf19ff715","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-08-29 19:26:55.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"9001f928837fc6ce97acdc69211aaf29a08f2a2f","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-08-29 23:22:07.000000000","message":"Patch Set 4: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"5c22ee3d4da9721213b86ca8f660479b17e430ea","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-08-29 23:49:59.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"7481b3fcfa6094109429b58c8e41ccac328d27ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-30 02:06:03.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c82cf0ad411341f3bd6571c8a07c68e1\n\n- grenade https://zuul.opendev.org/t/openstack/build/a0e631d46e3247349f31ec826ad8e984 : SUCCESS in 47m 12s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7abfbaa2233a45518f9562e832218f82 : SUCCESS in 39m 24s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3fc94e05909a4d53a77c338255eb9643 : SUCCESS in 1h 32m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a05bee9205684bc3b9bc95ad945e6056 : SUCCESS in 36m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7a47cb2ddb94ffd85d6e65835481c4f : SUCCESS in 4m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/decb742fb1bb44c9a6dda3eba63b81d5 : SUCCESS in 4m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5f35bc43fcba4a1e857467517a36fa7a : SUCCESS in 3m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/243b0b09f4334cd79f21aeb1c1937647 : SUCCESS in 7m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d72b0ed20754527b9f81ff9ff1588e1 : SUCCESS in 5m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e1ecdc82f75849008d6707d7cbb003ae : SUCCESS in 4m 32s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/22581a1792a146b7ae273882df7962f7 : SUCCESS in 6m 50s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/a3a6ec7b819a4b75850939aaef2191eb : SUCCESS in 10m 44s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/72e3d065c30840abb0fb308f54c5e076 : SUCCESS in 1h 25m 57s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/782ce60e51a6441d9378b98e0aa0157a : SUCCESS in 1h 29m 59s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/1c19b5889e13432999865679549ab209 : SUCCESS in 2h 00m 32s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/b1ba28b8465d4d61bc09e7bd0eb8e20c : SUCCESS in 2h 10m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/aba8ee62bc0c4fac9710414959f1db13 : FAILURE in 53m 20s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/2b4096c339604d9784db6699f70eb41a : SUCCESS in 1h 57m 03s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/de3e2cc90705498daaaaf7b10ccb96f1 : FAILURE in 28m 47s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/7fdd83e4c8ad470cb94f15581cb3b179 : SUCCESS in 1h 26m 39s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/8ca6f82a2f29459cb9fe935b1cfb6554 : SUCCESS in 31m 24s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/e74db39dcc0c46b78422dcd8451d5f51 : SUCCESS in 26m 58s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"e33523528c16fc08e63f9a7b834be96613db3471","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-08-30 03:19:49.000000000","message":"Patch Set 4: -Workflow\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"7b0c11f7ff2238f7d848d4414aa987bd45eb8407","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-30 05:35:25.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5a26dd867deb4e53be01e03e92e47dce\n\n- grenade https://zuul.opendev.org/t/openstack/build/00bc28b065914451ab46e9535b612863 : SUCCESS in 1h 03m 59s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c1fb14e409b0446ba8fff67482ff9ffd : SUCCESS in 58m 33s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e8fb3c32a0bb4e60a59073fdfc254331 : SUCCESS in 1h 17m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fa3a4394e33b41d2b563fd2ad8a0e1f2 : SUCCESS in 1h 02m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a5257c75e264b019a842103c6bcd046 : SUCCESS in 2m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d4a0cf20c2d143a7a791102bf5050a3a : SUCCESS in 3m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/295b7ccdcb3847f7bef20bfce73530d2 : SUCCESS in 5m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6262a0d4f044492291ae2a70541485a4 : SUCCESS in 4m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2c314b6e1fe4b9aa61a0e1ed251ae7f : SUCCESS in 6m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a84c5a9b792a4ee0a4e79654ad934ed7 : SUCCESS in 7m 25s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e1a78d3762e9407db9b3d7d00db453c3 : SUCCESS in 7m 00s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/161e7a56dc51436286d7653273712fc2 : SUCCESS in 10m 02s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/2b24fda5dc12434b9e0a63268d286f45 : SUCCESS in 1h 18m 36s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/63304e764ffd416c930ea11ec7d5fc04 : SUCCESS in 1h 39m 38s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/cc1ee26040b547ceadef69059fbb4861 : SUCCESS in 1h 36m 02s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/21d123553f0743d1b3fb9905a2631b10 : SUCCESS in 2h 13m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ed1efb6a286c47d29bb5c7c34a4bb840 : SUCCESS in 30m 14s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/cb1964387fea4b0fb5014030042783df : SUCCESS in 1h 28m 37s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/b20dafe2cf9044ceaf5c3e1c436fa0e6 : FAILURE in 21m 28s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/ef42504b771a4682b37e9e54d7e00a3f : SUCCESS in 1h 50m 14s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/44ada9b2700b440cb5b1c5e9efc274c2 : SUCCESS in 24m 09s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/994cf1ccf17d4e0bb69f96bc1a85abad : SUCCESS in 1h 02m 09s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"2b019e39fc73394b66c0199059e9c2a31ea2bdad","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-30 07:50:54.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2c071c5c704d4b21a36db911f063bf01\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4062f2427383447b8e90c30227053c5b : SUCCESS in 9m 52s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/139d60d53a364c5cab3161b9c7bbbcae : SUCCESS in 12m 50s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c12da7d5f17240189a6f7d96347f393c : SUCCESS in 10m 26s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"40af8ae3ae9b6df6ba7721ccab31b6d604d9982b","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-09-01 15:12:17.000000000","message":"Patch Set 4: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"220ced0e622694a27b90911ac62b1c6faf31de7b","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-09-03 21:31:20.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ab0c0e5cbc3fb83f298bf759d6eb89d6cc831a76","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-03 22:09:00.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0a2eddf9064340519be1369f39cc47e9\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/04ea8cac2d714583999462bee100487b : SUCCESS in 11m 09s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2ba74a1e593a4f58bc2659ab780127fc : SUCCESS in 9m 05s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/634cd3c80d6c4b64b5dd1af62fef700e : SUCCESS in 19m 42s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"bf6ed0c661870e2ec2159101d8474ff8d4c06ba5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-03 23:44:17.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e35c64abc3eb4952be83a05d1d101eb9\n\n- grenade https://zuul.opendev.org/t/openstack/build/6d87721978d94144b42a110cb79c0284 : SUCCESS in 43m 40s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a0f807a341054d2e9d461c14066be198 : SUCCESS in 46m 33s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c2bb38d20c604c8e94ef756d23931e4b : SUCCESS in 1h 59m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c5e6e00b5b0348baa9d09dceb282eb5d : SUCCESS in 51m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a450ff8f9e24fbab3f407cc591cfce4 : SUCCESS in 4m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a14a337ee1c1481282729578f065f53d : SUCCESS in 4m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bfb7ce1571fe43d585510f03a4b32683 : SUCCESS in 6m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3bf70f7a4624409bad297600a2781360 : SUCCESS in 6m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac2b208645aa41ed9f8c5786b848dfd5 : SUCCESS in 4m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0258becf748542769fec6150da65548a : SUCCESS in 6m 58s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/782ec67ee1f947a48614cd285c9dbcac : SUCCESS in 7m 23s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/9a8e282a8ef14245981f87647a8c36bb : SUCCESS in 10m 08s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/4e65faff2f9a450092640e0a5a4b883e : SUCCESS in 1h 51m 55s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/947b4f1f77364dc2bd634a816f43eda0 : SUCCESS in 1h 53m 59s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/843329f8414a43da86e2138caf92e54b : SUCCESS in 1h 54m 42s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/fed8555c971740e78e943d16c1df44d4 : SUCCESS in 1h 51m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9aabed0a00e149efa39990d35d8b4e32 : SUCCESS in 1h 00m 44s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/359247e2b9c84e91adf002343b501095 : SUCCESS in 1h 39m 52s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/19a54e30cdad471aaee14744a70999b4 : FAILURE in 32m 26s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/661f3a2b3edb43c59265e9fd4e35406c : SUCCESS in 2h 07m 36s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/3d3189ce5a77483ea3b2d7d667ee6514 : SUCCESS in 32m 53s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/ad627ff1b3644ba3b82b86bf8fcdf1cc : SUCCESS in 47m 22s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"0f4e6eff14b581aa8b3c3e504e296eb9bafdcbe5","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-09-05 07:52:22.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"82834b4cca901e8a8adef45fa168e51066c86c36","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2025-09-05 07:53:50.000000000","message":"Patch Set 4: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"98131e52bb174dec5dec78d948479be81314fdb0","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-09-05 17:02:18.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"534a41e7c75c46f2ce6921309e012ec6187a0033","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-09-11 22:01:01.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"f8f88006318d18cca0dbcfcfc8b0f78a655b46b9","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2025-09-12 08:56:25.000000000","message":"Patch Set 4: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"17a855d3ecf0c304dbd5459208cf8915c369dc1b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2025-09-12 08:57:02.000000000","message":"Patch Set 5: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"472eae9e9f0206e605fef5c74cc25b8ef9e77c70","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2025-09-12 08:57:24.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"0213c7fb429bfc0e15859baa6cb48734eec8f38f","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2025-09-12 08:57:34.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"fd5e1bef8238c5f3a174b1fa40fd141828795b7d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-12 09:48:21.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/529402a875664219aca895c414a34dd9\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5fdfb988e46f4585897399212fd02e02 : SUCCESS in 10m 14s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9081753b658e4203b4068ee258977ed4 : SUCCESS in 10m 03s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/8da8d21d5072418285de80db6bce79e4 : SUCCESS in 8m 21s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"88494ba710c14341cf68025da5a7332e2885c498","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-12 11:14:53.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d2bad4b2b184329ad54d9da3f104d68\n\n- grenade https://zuul.opendev.org/t/openstack/build/3ddf1e29fcb7436f9e4f54246c919468 : SUCCESS in 58m 53s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e0a05709a73e4618935586dece88f6cd : SUCCESS in 47m 05s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/38468c8533bf4ba5bb5da6911662ba1a : SUCCESS in 1h 24m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b60b825a41d940bc9c93aeca7f1d41ad : SUCCESS in 34m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f8550c91484413aa46c2b87bb8ed027 : SUCCESS in 5m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4278fbc531224e31825cacd340add061 : SUCCESS in 5m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4ce88d1702014eeab1beba4453fbe330 : SUCCESS in 6m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/125b864119bc4747a9b80c63752042bd : SUCCESS in 8m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b229637cdf34fa8b2e5e2ff7712cf8d : SUCCESS in 3m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b54474d104114175b28934d16db41a08 : SUCCESS in 3m 47s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c8f7945ad9594469a1cea0e54780ac4f : SUCCESS in 6m 19s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/475ed0be341649febe521d5bd6cecef1 : SUCCESS in 10m 50s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/b90f5c5aed0c4a9798354260c4b97182 : SUCCESS in 1h 58m 43s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/fac3bc245bc245ac8453c39fe8aabde0 : SUCCESS in 1h 49m 58s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/2fbcff7067c24b1b9080e853225b9bbf : SUCCESS in 1h 32m 03s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/03b6e4454a25444a823887326e0b2bee : SUCCESS in 2h 09m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/992b749b7bbc42bfaa7a0fe6cf968989 : SUCCESS in 49m 04s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/bc42bd255b304fb6921b2499ed4925c6 : SUCCESS in 1h 39m 20s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/7c06b06b52dc470387427c42271b6659 : FAILURE in 17m 34s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/31a6b3043ebf4267b756be4301052ad3 : SUCCESS in 1h 09m 20s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/88429f2d63eb42d9b202b5cd4269119c : SUCCESS in 33m 43s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/285f2de628ec46cbbea9f3c653f1cf0c : SUCCESS in 1h 02m 32s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"5f0ac6442c8be5d1c9ca8ba880f68a8c786075bb","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-09-12 13:48:30.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"8d5918dd472bd3270ae8bd415828c0cde2f705cb","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-09-12 14:13:14.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"f7e0fc159c63088bb2e6673071ccc4913ecda3ec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-12 14:13:57.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"037abd03f224f9f7a0046b1cc6ae3aae9dc09287","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-12 16:26:12.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6742843c05714fbabb632ab842b50c00\n\n- grenade https://zuul.opendev.org/t/openstack/build/0b441f0108da4b348b46ba976cbf4986 : SUCCESS in 1h 02m 23s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d56baa94494e4ac48ef1a0954d1582e7 : SUCCESS in 32m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/473acf8e738f421fa3d2c0b3838fb8d8 : SUCCESS in 1h 32m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dd349a8efb854f93826f67c3ebe20ecb : SUCCESS in 1h 00m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70852bdec14b48cfbb4692f3300cfc8b : SUCCESS in 4m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/74dd8179cb764bc8a2e895813a94fb17 : SUCCESS in 5m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/441c98a70a0c43f6ad98a780d3e05b72 : SUCCESS in 5m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd87387655814d1aa5416d04c21c4f44 : SUCCESS in 5m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0b9ae5c1de8642359ee6ebaf5e93b50c : SUCCESS in 7m 00s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3a463c48b64440c6ab56fb67382e5ceb : SUCCESS in 6m 21s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/3f4c66ad452448cdb2480949c0ea039e : SUCCESS in 10m 03s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/1b5ecd273fb14a36850bfbfa1bddfc53 : SUCCESS in 1h 40m 25s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/ef7da27649a44b27a46d6b3461c29d76 : FAILURE in 2h 06m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d2735d947bf04d1ea8531b81dd53ed57 : SUCCESS in 1h 07m 38s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ffdd46de20ec4800994b421cefd726eb : SUCCESS in 1h 46m 02s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/7b3755921c204f578a88e9de230ed873 : SUCCESS in 31m 25s","accounts_in_message":[],"_revision_number":5},{"id":"874c12d9a823460a26b913c7b7c9e3bbfbc14e74","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2025-09-12 19:01:51.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"0822ad8cde88cccd4f2aa885a91db03989bb693f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-12 19:19:03.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/abd33279099a4808bc7b4e652d444b7d\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4f4fad57f5144b0497919ba7774abf96 : SUCCESS in 10m 36s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e3b2010b8ed44d778263cd3044357d28 : SUCCESS in 11m 14s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/9612aa239bc44af2a2cc2df36ab33aca : SUCCESS in 15m 00s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"495a578d83f2f6c1aa77561dc170a10878d355cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-12 21:16:38.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0495ede450604e1d9b1e7b807502f454\n\n- grenade https://zuul.opendev.org/t/openstack/build/b3d7ade521ab44c4a5572a31f623e90c : SUCCESS in 28m 14s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6fd4f3527b5f4e86bc3d5db7c588b6c2 : SUCCESS in 35m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ac0b1678983a45ec97d79d5c5090e5e9 : SUCCESS in 1h 35m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2afac276a22344948bebdbf755e03ace : SUCCESS in 1h 06m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dfe5314511c740f9935641c0278cd35d : SUCCESS in 5m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/75254ccdf6b24883a47b778b0b14ed9d : SUCCESS in 5m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cd737cd9f4d64364b9fb3088ccdbfe89 : SUCCESS in 5m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2530b68f303f47f3ba8d0aa8ab49b247 : SUCCESS in 10m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/681779f800044d04a2f2d6fc84ed4548 : SUCCESS in 5m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/538282168b8b45838ff2d35b9ab45c7c : SUCCESS in 10m 07s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5998d6d54dac4958b8e2a950041f30f5 : SUCCESS in 6m 18s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/c1e2fb8298e645709173c61d7044790b : SUCCESS in 10m 19s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/4a8e428fdd8a4af1ade55c04082f14d8 : SUCCESS in 1h 52m 32s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/2f19fc74caa04ff39b15f731b87acb4e : SUCCESS in 2h 09m 52s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/5255fb2b12204b17b82c5a4e8c7783ba : SUCCESS in 1h 15m 46s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/6e6587aa509640f3a8f04fd7aa4d8b71 : FAILURE in 1h 14m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d24656183c2a4ed4858987bff63fa81f : SUCCESS in 53m 21s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/870b52a22ab149f1a6d2a9036925beee : SUCCESS in 2h 04m 48s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/79fd65614ac44043bacc85b821863871 : FAILURE in 24m 25s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/c897378c28de4a018fd15fab72aef0de : SUCCESS in 1h 47m 44s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/d36cbbf6820c419a9c43b75369db3529 : SUCCESS in 35m 33s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/998cb7d9a737486a9818bb72289c62a0 : SUCCESS in 1h 22m 40s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"d755ab3fdb67485c7d5a091b31137ea25db9fe50","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2025-09-15 08:33:50.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"93262fdc93082ea13f2df037f99a3d455fd629ee","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-15 10:21:48.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f4a96849292843b1a1653e88a8ecf329\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/33a7b6057c5240cfbd8cee2247b82904 : SUCCESS in 10m 59s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/135b74b5165e4659bf611f4578872df0 : SUCCESS in 13m 02s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/db1e3540cd4d4853ab7a700c387fd2cb : SUCCESS in 19m 29s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"dbd2eff63900e19241b565b2370812d0fa7c1517","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-15 10:43:21.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b6d86702701a4ff6840c6c181c11e02e\n\n- grenade https://zuul.opendev.org/t/openstack/build/64bc9b0bdcec4adc877020bf52c6b21e : SUCCESS in 54m 57s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d7319f37d3854600b11839a52fdb53f8 : SUCCESS in 57m 47s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/867c174fe6d245c69c052a2d55690bfc : SUCCESS in 1h 46m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d9d090022cd94bb6abb0cb5bc5a96936 : SUCCESS in 53m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e3f5bcb09524505ae27a7cd708370ee : SUCCESS in 4m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dbb4c51ba64e40aa82491639ce68b07a : SUCCESS in 6m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/54cf87091ddd49208d0751bb8c45ed43 : SUCCESS in 5m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2ea81ee763384e46b0fcf1e74a24bd55 : SUCCESS in 8m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19387b533ecc47308f69012c1ee0696c : SUCCESS in 5m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/de480a1620b74b2eb6917ca03b0e9630 : SUCCESS in 6m 57s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a37e1052a88245f3ac8e9f55c264bc74 : SUCCESS in 5m 58s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/ea06c7cbefde412e9b1c0971a1f99484 : SUCCESS in 10m 09s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/4c4d272425ba443a9c406d3c17253e8a : SUCCESS in 1h 44m 49s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/82135533ce12497d9f2289eba039134f : SUCCESS in 2h 04m 10s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/a4fe0a0866104169bcae42cfa88c3acb : SUCCESS in 1h 57m 01s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/a854f2d2bdaf4ba39d134f5dff214d05 : SUCCESS in 2h 07m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ff517746f99d448b8f9532f7a09fcfae : SUCCESS in 40m 23s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/253ffe8124744ddbbbe5fc09912ba5d2 : SUCCESS in 1h 49m 07s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/8bbbe1cd8803497bb5c4a6cf9a91d5d2 : FAILURE in 33m 39s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/f2d755b1e50a42a5901ce65db56e8fca : SUCCESS in 1h 50m 01s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/8a2606d2f02c4da3a5f0516130c228cd : SUCCESS in 32m 22s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/dd80dd6b20ff46089e7cec7532026be6 : SUCCESS in 1h 09m 38s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"eeed11cdbba8e55d23fa725afdcf2ccb8a9c4e3a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-15 10:43:54.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"2624a22862e54f6582070519d466a91d53a073d7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-15 12:41:57.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/925053e202294f42b04fdf43b1caa450\n\n- grenade https://zuul.opendev.org/t/openstack/build/2071775958324325b9b8f20e58651b11 : SUCCESS in 51m 59s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9a482a16b60d4d569fae604e95874bee : SUCCESS in 54m 29s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9a6b1e0828d54d2387a4151050331101 : SUCCESS in 1h 17m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1bc1e98487da4882bf5c50620f3d9c03 : SUCCESS in 48m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e465021527f44979df6b6ecbb1a3875 : SUCCESS in 5m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a8ceddd014fd48448c6765fc351b37a0 : SUCCESS in 4m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c09e971d9bdd49d9828d95144c57d0fa : SUCCESS in 6m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a18e8d511a134a59b50e3a63b24a1a9a : SUCCESS in 6m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cc1e775ea2954d91bd93a0ace27df8ad : SUCCESS in 8m 18s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c1a2b659ca40421d9782f8a145c767bc : SUCCESS in 7m 34s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/33d0fbf0e3504cb5a38bd30759d20364 : SUCCESS in 11m 57s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/e1621dda4a7b402bb081494bfb741ae9 : SUCCESS in 1h 38m 27s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/31e0023cff4847f2bc560f9ec74427ab : SUCCESS in 1h 35m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c3c1f985e5c3486ab122e1b33f21d9b7 : SUCCESS in 49m 14s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/004cbd03c58c4e6c9c7fe46c6129e6cd : SUCCESS in 1h 47m 57s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/670f0892045c42f2909e2cdd9c5c4e4d : SUCCESS in 35m 36s","accounts_in_message":[],"_revision_number":5},{"id":"ad63361c81a3f1d7ebd4aaedb51a2735571e73c1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-15 12:41:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"e884e7d89ff438cb4815c6fa9ee15e638184bcba","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-15 12:42:51.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b465392271f42998e57f059148d93d7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5614a112da1b49b1abfebed86ad4c0f6 : SUCCESS in 44s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/94d9927bdd11422d8fd2a916aa816a49 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"301d8f580f4ce6a4f6089f877c0f200ff4ff633b","revisions":{"467a70c3f269376455e62dcc150b1a1738495c5a":{"kind":"REWORK","_number":1,"created":"2025-08-28 03:39:18.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/15/958715/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/15/958715/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/15/958715/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/15/958715/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/15/958715/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/15/958715/1"}}},"commit":{"parents":[{"commit":"6c33a667a9f5ddce07b6131f4a5cb7460a4bdf17","subject":"Merge \"Replaced usage outdate egrep to grep\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6c33a667a9f5ddce07b6131f4a5cb7460a4bdf17"}]}],"author":{"name":"Ghanshyam Maan","email":"gmaan@ghanshyammann.com","date":"2025-08-28 03:19:27.000000000","tz":0},"committer":{"name":"Ghanshyam Maan","email":"gmaan@ghanshyammann.com","date":"2025-08-28 03:39:16.000000000","tz":0},"subject":"Fix glance service policy rule","message":"Fix glance service policy rule\n\nThe OpenStack services communicate with each other by\npassing the user token and service token wrapped in\nkeystoneauth\u0027s ServiceTokenAuthWrapper. The only purpose\nof passing the service token is for long-running\noperations and in case the user token gets expired. But\nglance use that service token for RBAC (check \u0027service_role\u0027\nwhich is role from service token) which is not correct.\n\nFor RBAC, we need to check if a user from other service\nhas the \u0027service\u0027 role or not. Service needs to load the\nconfigured user auth plugin (where the user should have\nthe \u0027service\u0027 role) from keystoneauth and pass that to\nthe other services (for example, cinder change[1]) and\nglance will use that user role to verify the policy permission.\nTo fix that, we need to make the service APIs default to\n``role:service`` and not `service_role`:`service`.\n\nThis commit does one more change. Cinder does not have the\nway to configure the glance service user, we are adding the\nnew config in this release. For backward compatibility,\nwe need to allow admin access in service policy rule. In\nfuture release (after one SLURP release), we cna remove\nthe admin access.\n\nChange-Id: I50909e6bdb3227ca99b7eba642546da791f9552a\nSigned-off-by: Ghanshyam Maan \u003cgmaan@ghanshyammann.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/467a70c3f269376455e62dcc150b1a1738495c5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/467a70c3f269376455e62dcc150b1a1738495c5a"}]},"branch":"refs/heads/master"},"26c1a01b8bddf08aa6880390ffb9dda35c68c90c":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-08-28 04:09:25.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/15/958715/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/15/958715/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/15/958715/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/15/958715/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/15/958715/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/15/958715/2"}}},"commit":{"parents":[{"commit":"6c33a667a9f5ddce07b6131f4a5cb7460a4bdf17","subject":"Merge \"Replaced usage outdate egrep to grep\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6c33a667a9f5ddce07b6131f4a5cb7460a4bdf17"}]}],"author":{"name":"Ghanshyam Maan","email":"gmaan@ghanshyammann.com","date":"2025-08-28 03:19:27.000000000","tz":0},"committer":{"name":"Ghanshyam Maan","email":"gmaan@ghanshyammann.com","date":"2025-08-28 04:07:51.000000000","tz":0},"subject":"Fix glance service policy rule","message":"Fix glance service policy rule\n\nThe OpenStack services communicate with each other by\npassing the user token and service token wrapped in\nkeystoneauth\u0027s ServiceTokenAuthWrapper. The only purpose\nof passing the service token is for long-running\noperations and in case the user token gets expired. But\nglance use that service token for RBAC (check \u0027service_role\u0027\nwhich is role from service token) which is not correct.\n\nFor RBAC, we need to check if a user from other service\nhas the \u0027service\u0027 role or not. Service needs to load the\nconfigured user auth plugin (where the user should have\nthe \u0027service\u0027 role) from keystoneauth and pass that to\nthe other services (for example, cinder change depends-on)\nand glance will use that user role to verify the policy\npermission. To fix that, we need to make the service APIs\ndefault to ``role:service`` and not `service_role`:`service`.\n\nThis commit does one more change. Cinder does not have the\nway to configure the glance service user, we are adding the\nnew config in this release. For backward compatibility,\nwe need to allow admin access in service policy rule. In\nfuture release (after one SLURP release), we cna remove\nthe admin access.\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/958718\nDepends-On: https://review.opendev.org/c/openstack/cinder/+/958716\n\nChange-Id: I50909e6bdb3227ca99b7eba642546da791f9552a\nSigned-off-by: Ghanshyam Maan \u003cgmaan@ghanshyammann.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/26c1a01b8bddf08aa6880390ffb9dda35c68c90c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/26c1a01b8bddf08aa6880390ffb9dda35c68c90c"}]},"branch":"refs/heads/master"},"2c9471da8a744742bff0e80fb702d36b5cb992a7":{"kind":"REWORK","_number":3,"created":"2025-08-28 20:23:28.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/15/958715/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/15/958715/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/15/958715/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/15/958715/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/15/958715/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/15/958715/3"}}},"commit":{"parents":[{"commit":"6c33a667a9f5ddce07b6131f4a5cb7460a4bdf17","subject":"Merge \"Replaced usage outdate egrep to grep\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6c33a667a9f5ddce07b6131f4a5cb7460a4bdf17"}]}],"author":{"name":"Ghanshyam Maan","email":"gmaan@ghanshyammann.com","date":"2025-08-28 03:19:27.000000000","tz":0},"committer":{"name":"Ghanshyam Maan","email":"gmaan@ghanshyammann.com","date":"2025-08-28 20:22:37.000000000","tz":0},"subject":"Fix glance service policy rule","message":"Fix glance service policy rule\n\nGlance service APIs are default to \u0027service_roles: service\u0027\n\n- https://github.com/openstack/glance/blob/6c33a667a9f5ddce07b6131f4a5cb7460a4bdf17/glance/policies/base.py#L116\n\nThe issue here is the service token, which is sent from the\nservice for the user token expiry case but glance uses that\nservice token (keystonemiddleware sets the service token roles\nin Requestcontext in \u0027service_roles\u0027 field) for RBAC, which\nis not correct.\n\nThe OpenStack services communicate with each other by\npassing the user token and service token wrapped in\nkeystoneauth\u0027s ServiceTokenAuthWrapper. The only purpose\nof passing the service token is for long-running\noperations and in case the user token gets expired.\n\nFor RBAC, we need to check if a user token has the \u0027service\u0027\nrole or not. Service needs to load the configured user auth\nplugin (where the user should have the \u0027service\u0027 role) from\nkeystoneauth and pass that to the other services (for example,\ncinder change depends-on) and glance will use that user role\nto verify the policy permission. To fix that, we need to make\nthe service APIs default to ``role:service`` and not\n`service_role`:`service`.\n\nThis commit does one more change. Cinder does not have the\nway to configure the glance service user, we are adding the\nnew config in this release. For backward compatibility,\nwe need to allow admin access in service policy rule. In\nfuture release (after one SLURP release), we cna remove\nthe admin access.\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/958718\nDepends-On: https://review.opendev.org/c/openstack/cinder/+/958716\n\nCloses-Bug: #2121622\n\nCo-Authored-By: : Sean Mooney \u003cwork@seanmooney.info\u003e\n\nChange-Id: I50909e6bdb3227ca99b7eba642546da791f9552a\nSigned-off-by: Sean Mooney \u003cwork@seanmooney.info\u003e\nSigned-off-by: Ghanshyam Maan \u003cgmaan@ghanshyammann.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2c9471da8a744742bff0e80fb702d36b5cb992a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2c9471da8a744742bff0e80fb702d36b5cb992a7"}]},"branch":"refs/heads/master"},"50581a0134b46c2ec6286f88112a9bf52afb59cb":{"kind":"REWORK","_number":4,"created":"2025-08-28 21:09:24.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/15/958715/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/15/958715/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/15/958715/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/15/958715/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/15/958715/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/15/958715/4"}}},"commit":{"parents":[{"commit":"6c33a667a9f5ddce07b6131f4a5cb7460a4bdf17","subject":"Merge \"Replaced usage outdate egrep to grep\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6c33a667a9f5ddce07b6131f4a5cb7460a4bdf17"}]}],"author":{"name":"Ghanshyam Maan","email":"gmaan@ghanshyammann.com","date":"2025-08-28 03:19:27.000000000","tz":0},"committer":{"name":"Ghanshyam Maan","email":"gmaan@ghanshyammann.com","date":"2025-08-28 21:09:18.000000000","tz":0},"subject":"Fix glance service policy rule","message":"Fix glance service policy rule\n\nGlance service APIs are default to \u0027service_roles: service\u0027\n\n- https://github.com/openstack/glance/blob/6c33a667a9f5ddce07b6131f4a5cb7460a4bdf17/glance/policies/base.py#L116\n\nThe issue here is the service token, which is sent from the\nservice for the user token expiry case but glance uses that\nservice token (keystonemiddleware sets the service token roles\nin Requestcontext in \u0027service_roles\u0027 field) for RBAC, which\nis not correct.\n\nThe OpenStack services communicate with each other by\npassing the user token and service token wrapped in\nkeystoneauth\u0027s ServiceTokenAuthWrapper. The only purpose\nof passing the service token is for long-running\noperations and in case the user token gets expired.\n\nFor RBAC, we need to check if a user token has the \u0027service\u0027\nrole or not. Service needs to load the configured user auth\nplugin (where the user should have the \u0027service\u0027 role) from\nkeystoneauth and pass that to the other services (for example,\ncinder change depends-on) and glance will use that user role\nto verify the policy permission. To fix that, we need to make\nthe service APIs default to ``role:service`` and not\n`service_role`:`service`.\n\nThis commit does one more change. Cinder does not have the\nway to configure the glance service user, we are adding the\nnew config in this release. For backward compatibility,\nwe need to allow admin access in service policy rule. In\nfuture release (after one SLURP release), we cna remove\nthe admin access.\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/958718\nDepends-On: https://review.opendev.org/c/openstack/cinder/+/958716\n\nCloses-Bug: #2121622\n\nCo-Authored-By: : Sean Mooney \u003cwork@seanmooney.info\u003e\n\nChange-Id: I50909e6bdb3227ca99b7eba642546da791f9552a\nSigned-off-by: Sean Mooney \u003cwork@seanmooney.info\u003e\nSigned-off-by: Ghanshyam Maan \u003cgmaan@ghanshyammann.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/50581a0134b46c2ec6286f88112a9bf52afb59cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/50581a0134b46c2ec6286f88112a9bf52afb59cb"}]},"branch":"refs/heads/master"},"301d8f580f4ce6a4f6089f877c0f200ff4ff633b":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2025-09-12 08:57:02.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/15/958715/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/15/958715/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/15/958715/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/15/958715/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/15/958715/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/15/958715/5"}}},"commit":{"parents":[{"commit":"6c33a667a9f5ddce07b6131f4a5cb7460a4bdf17","subject":"Merge \"Replaced usage outdate egrep to grep\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6c33a667a9f5ddce07b6131f4a5cb7460a4bdf17"}]}],"author":{"name":"Ghanshyam Maan","email":"gmaan@ghanshyammann.com","date":"2025-08-28 03:19:27.000000000","tz":0},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2025-09-12 08:57:01.000000000","tz":0},"subject":"Fix glance service policy rule","message":"Fix glance service policy rule\n\nGlance service APIs are default to \u0027service_roles: service\u0027\n\n- https://github.com/openstack/glance/blob/6c33a667a9f5ddce07b6131f4a5cb7460a4bdf17/glance/policies/base.py#L116\n\nThe issue here is the service token, which is sent from the\nservice for the user token expiry case but glance uses that\nservice token (keystonemiddleware sets the service token roles\nin Requestcontext in \u0027service_roles\u0027 field) for RBAC, which\nis not correct.\n\nThe OpenStack services communicate with each other by\npassing the user token and service token wrapped in\nkeystoneauth\u0027s ServiceTokenAuthWrapper. The only purpose\nof passing the service token is for long-running\noperations and in case the user token gets expired.\n\nFor RBAC, we need to check if a user token has the \u0027service\u0027\nrole or not. Service needs to load the configured user auth\nplugin (where the user should have the \u0027service\u0027 role) from\nkeystoneauth and pass that to the other services (for example,\ncinder change depends-on) and glance will use that user role\nto verify the policy permission. To fix that, we need to make\nthe service APIs default to ``role:service`` and not\n`service_role`:`service`.\n\nThis commit does one more change. Cinder does not have the\nway to configure the glance service user, we are adding the\nnew config in this release. For backward compatibility,\nwe need to allow admin access in service policy rule. In\nfuture release (after one SLURP release), we cna remove\nthe admin access.\n\nCloses-Bug: #2121622\n\nCo-Authored-By: : Sean Mooney \u003cwork@seanmooney.info\u003e\n\nChange-Id: I50909e6bdb3227ca99b7eba642546da791f9552a\nSigned-off-by: Sean Mooney \u003cwork@seanmooney.info\u003e\nSigned-off-by: Ghanshyam Maan \u003cgmaan@ghanshyammann.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/301d8f580f4ce6a4f6089f877c0f200ff4ff633b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/301d8f580f4ce6a4f6089f877c0f200ff4ff633b"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
