)]}'
{"id":"openstack%2Fglance~958886","triplet_id":"openstack%2Fglance~stable%2F2025.1~I491c486a0b967baba9bbdca4cd241ba691a94435","project":"openstack/glance","branch":"stable/2025.1","topic":"bug/2118756","attention_set":{},"removed_from_attention_set":{"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2025-09-03 16:45:11.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I491c486a0b967baba9bbdca4cd241ba691a94435","subject":"Fix inaccurate usage reporting when delayed_delete is enabled","status":"MERGED","created":"2025-08-29 13:40:04.000000000","updated":"2025-09-03 16:48:24.000000000","submitted":"2025-09-03 16:45:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"958886-bug/2118756","cherry_pick_of_change":956086,"cherry_pick_of_patch_set":2,"meta_rev_id":"23418637ed24203f820d0e2ffe4dfa844bd82790","_number":958886,"virtual_id_number":958886,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-03 16:45:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":2,"date":"2025-09-03 14:32:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":1,"date":"2025-09-03 14:32:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-29 15:51:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-03 14:32:34.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"512867246fdfae8cb40dd0407f37385a2fb9e56f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-08-29 13:40:04.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"e7efcfb27750ec031f077674b9a976a971b10aff","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-08-29 13:40:32.000000000","message":"Topic bug/2118756-stable/2025.1 removed","accounts_in_message":[],"_revision_number":1},{"id":"85ee7e48364d5a0c0e424e1915a4572de03c4e67","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-08-29 13:40:39.000000000","message":"Topic set to bug/2118756","accounts_in_message":[],"_revision_number":1},{"id":"14dc9d776e1a899bc3eda5dd6d492df2b724b612","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-29 15:51:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c74a67a8aa2e4376aea07c11611ca5f7\n\n- grenade https://zuul.opendev.org/t/openstack/build/a224dc0b55b241319e6459765972aee9 : SUCCESS in 56m 12s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/aa83f1bbed4c41a187cbfbedc6febb91 : SUCCESS in 1h 02m 46s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ef9e817a0ff145a590ff103d8a780d2a : SUCCESS in 1h 31m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a1ba0dab7382421aadc1b650d277ba5a : SUCCESS in 1h 03m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/456514f19ab940628491e9743f32c3b9 : SUCCESS in 3m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ac653fd7443d46998fb5072c44684ba9 : SUCCESS in 5m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/760ac0c908f249bd8ae4c2cecffbb307 : SUCCESS in 7m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0f43f7994c746acb28b453aaf2ab19f : SUCCESS in 4m 10s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/df300ae3df174022813b939d2617a671 : SUCCESS in 10m 26s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a2274e319cf54067bb4eaff07971bb96 : FAILURE in 9m 51s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/5beb6436b70149f69bb299ed7e319651 : SUCCESS in 9m 21s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/93b3f74c3815443a910550d1af74619c : SUCCESS in 1h 56m 41s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/e2df1d30ea754610a3bd4d18403e9136 : SUCCESS in 1h 30m 04s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/47f50ed5a4bc4943bdc26a34eca17ccf : SUCCESS in 1h 35m 23s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/79fba597b55748a98308da4b92716cc7 : SUCCESS in 1h 56m 10s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/b2050c391bac4c30a31bcdd89ba0bd5c : SUCCESS in 2h 08m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/53eabbd4a1f24707b8d971c0144ce309 : SUCCESS in 58m 14s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/4abc14f67cc14ff2bd2cf44aa9a3b730 : SUCCESS in 1h 38m 59s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/2ce137c8fee44a02b99574d980de074b : SUCCESS in 1h 30m 43s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/57998539eafc4c789ef250d4511a5287 : SUCCESS in 26m 00s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/eebc4e24469d464a979cf598af1347b5 : SUCCESS in 1h 49m 09s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/55ff0a4a154e4c8390cb8adf90a2ffb4 : SUCCESS in 31m 23s","accounts_in_message":[],"_revision_number":1},{"id":"29e20e1550e7589cb4b1b0948323ebfed47f01fe","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-30 05:21:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/30410a91c22f4d4094ef0dbc2cc16f4e\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/52a51ffb25e149f68c2a5363120843a1 : SUCCESS in 12m 02s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0052c620f5b245b49b10c054b574c2f1 : SUCCESS in 10m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"843df9e492794e140e81c7a784e33f000552b8ff","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-09-03 05:38:50.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ac0edde313a754a7775c2928dfb002feaa45bb8e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-03 05:49:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b3a4e292af6844c0af9c991e7df5affe\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b3f901958c65488a814a802172966423 : SUCCESS in 8m 48s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7ab059ea6a5d4475b56c61a6285ed9e9 : SUCCESS in 9m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"626a1518a1be633df7bd9d5036455b2c4817815d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-03 07:58:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d75a18ec7ee0403b84cca4e7064f8b35\n\n- grenade https://zuul.opendev.org/t/openstack/build/e44bde84c8294ad5b3d34a650241fee0 : SUCCESS in 1h 01m 34s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/cd58882070644677a8acc584eaef3f85 : SUCCESS in 49m 41s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/215242b3806c489a8047b3822ef62741 : SUCCESS in 1h 30m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ddaf08bfe0ab42f2ac6f5641e1881917 : SUCCESS in 33m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f716539fb424ec9b6b489e45f668377 : SUCCESS in 5m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6081b29289a84205a6f6ca2cf91f1829 : SUCCESS in 5m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5cd5060397584733b2865e54ac10209f : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a1c1c5208804990a6a004f02f8750da : SUCCESS in 5m 42s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/351f10aebd824198beab81d751ed526b : SUCCESS in 8m 30s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5a4847344b234f8e85be4ca45dd65a70 : SUCCESS in 8m 10s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/754360e3487047f09595453764471491 : SUCCESS in 12m 11s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/a6ec5c61d90349328d58cc5e9fe411fc : SUCCESS in 1h 49m 10s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/51c28e6d44bf4b07bd72beb1ad88a5f1 : SUCCESS in 1h 42m 43s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/a0e159427b23443587a6db03093367a7 : SUCCESS in 1h 51m 15s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/0a0f459feaeb44bbb292afcfc98d1953 : SUCCESS in 2h 14m 14s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/a729073741e9496cadfeef8e435d6218 : SUCCESS in 2h 14m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f1b9aae921484957ae40d769315a3f05 : SUCCESS in 56m 09s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/bc99e70257a04c0e884fadef3d1f4c93 : SUCCESS in 1h 43m 13s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/3fb82a8ed3d8462a865a15cd3bb8d3be : SUCCESS in 2h 08m 50s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/5ad5d0268a9446158beb827a494553aa : SUCCESS in 57m 56s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/5252a87485c74805af20883946b99199 : SUCCESS in 1h 45m 21s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/bef37f5ccefc4ac2b561846b05df54e2 : SUCCESS in 28m 06s","accounts_in_message":[],"_revision_number":1},{"id":"3b2635bb479354a925eae94ff7b3638cdb06a1a5","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-09-03 14:32:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3670ddc5bd5e732a920fc28a9e1580487f24db8e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-03 14:33:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e4c76e81e6dc612eadbfbe1ab837cc304cea8b04","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-03 16:45:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/466236fa2f5046258560d36a8bded15a\n\n- grenade https://zuul.opendev.org/t/openstack/build/2d05b775d17b4ceba771995544453eca : SUCCESS in 48m 17s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d8e00c1ee67245fba30e66b2b834f15a : SUCCESS in 28m 48s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c84362ae15e742dfb3398f3225a4facf : SUCCESS in 1h 42m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0779ffef9b0c456e9a8b6772240a4a0c : SUCCESS in 1h 01m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a18f45e665d242b28dbbb519bc33c7c6 : SUCCESS in 5m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9b8fb13d4a464d9eb6c68ab7df7c36ac : SUCCESS in 5m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4cc089181723424495c8ff2776c73aae : SUCCESS in 5m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e3f91f2b7b14ce8a9c8c34c9e388faf : SUCCESS in 5m 50s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/18defcf77e1c4daf9bed409af33b87d2 : SUCCESS in 9m 17s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c8bcee2c281f44268e05b3dabf06522f : SUCCESS in 8m 36s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/7b4dd60f7b214098a8cf5b492dc95c37 : SUCCESS in 10m 54s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/1c17bc5f52234273be278f4984e04dcd : SUCCESS in 1h 57m 07s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/ff3485ddbbf146c5a99e8f8eb18427f8 : SUCCESS in 2h 03m 05s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/28082f70040f484db6a6d5f9fbe66f65 : SUCCESS in 1h 59m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/17524f9719c24aacb1a1f20ea39cc0ee : SUCCESS in 42m 20s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/301b5bd7a141412d81aaa74f37b38add : SUCCESS in 2h 05m 04s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f433d707c92442569ca628c30b2d66b1 : SUCCESS in 35m 01s","accounts_in_message":[],"_revision_number":1},{"id":"68d6bb443128dcffc085fd786f98da50804f361d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-03 16:45:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"23418637ed24203f820d0e2ffe4dfa844bd82790","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-03 16:48:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9972aa12aaea4425be4664f2d2dce39d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6f10457f08e40cab6d016c9b228b75c : SUCCESS in 2m 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d70e40e4cc993dacb88336a7a06fe57d4c670a22","revisions":{"d70e40e4cc993dacb88336a7a06fe57d4c670a22":{"kind":"REWORK","_number":1,"created":"2025-08-29 13:40:04.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/86/958886/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/86/958886/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/86/958886/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/86/958886/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/86/958886/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/86/958886/1"}}},"commit":{"parents":[{"commit":"2cf0c070e0b4198804b1175612626e7f0b462af7","subject":"Fix: Policy for new location APIs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2cf0c070e0b4198804b1175612626e7f0b462af7"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-07-29 14:02:39.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-08-29 13:40:04.000000000","tz":0},"subject":"Fix inaccurate usage reporting when delayed_delete is enabled","message":"Fix inaccurate usage reporting when delayed_delete is enabled\n\nThis change addresses issues where usage reports include images in\npending_delete, leading to potential user blocking and limit\nenforcement errors.\n\nCloses-Bug: #2118756\nChange-Id: I491c486a0b967baba9bbdca4cd241ba691a94435\nSigned-off-by: Abhishek Kekane \u003cakekane@redhat.com\u003e\n(cherry picked from commit 13b3c7283fd085131a3dd73f2b7300b91ca97007)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d70e40e4cc993dacb88336a7a06fe57d4c670a22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d70e40e4cc993dacb88336a7a06fe57d4c670a22"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
