)]}'
{"id":"openstack%2Fglance~959984","triplet_id":"openstack%2Fglance~master~I9063663a3b14161a67da59a68f5f7c92fd5447f5","project":"openstack/glance","branch":"master","topic":"bp/amd-sev-es-libvirt-support","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-09-08 20:50:29.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9063663a3b14161a67da59a68f5f7c92fd5447f5","subject":"Add new metadata definition of memory encryption model","status":"MERGED","created":"2025-09-07 13:05:52.000000000","updated":"2025-09-08 20:53:35.000000000","submitted":"2025-09-08 20:50:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"959984-bp/amd-sev-es-libvirt-support","meta_rev_id":"6e484e6fc63ad0d2a62e688594d473e8163acfd5","_number":959984,"virtual_id_number":959984,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-08 20:50:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":2,"date":"2025-09-08 15:48:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":1,"date":"2025-09-08 15:48:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-07 13:25:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-09-07 15:07:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-08 15:48:12.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"4a8f9e44ea0f42e01040ed142cbb4a83c33a602e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-09-07 13:05:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d710b4c1725f90446ba9e42cf6124ac1e4afabb3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-09-07 13:11:25.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"69ff4334f084c23e82c7059596ddcdd715699832","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-07 13:25:31.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e3bda37230d496c99abe3ed2c0908e8\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c0f80ec30a7046c1ba23b5fbfb5a20f2 : SUCCESS in 8m 19s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f9fba6be255d4416ad17cd3d13389a35 : SUCCESS in 8m 28s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/0cfb4d7a5f854c4ab1309253c8af559a : SUCCESS in 12m 30s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8c40e7f6ef7b3c38523181407dbd181b8b585938","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-07 15:07:44.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5434a46b08c0414281beec5bd17ba5dd\n\n- grenade https://zuul.opendev.org/t/openstack/build/2e303989b1a947bf8eaa29e3f4fca597 : SUCCESS in 28m 58s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9292a20ee8da44808196243ce5c443fb : SUCCESS in 33m 39s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/fdf0d0e6585244d1bcdafae0c424d488 : SUCCESS in 1h 26m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f6fab2dd58ea44979db86fa01c0b72aa : SUCCESS in 31m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc3b812bdc7f4e9d875d7ca0c37176f9 : SUCCESS in 6m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4af85cad5b0443edb57688a1669699ed : SUCCESS in 2m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f5138d21b0a44ce89be79d88c9b5b53a : SUCCESS in 3m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a35e28d3ac494de3a5356616cbca41db : SUCCESS in 7m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08eb3970aae345eb9bfdf01189f4a515 : SUCCESS in 6m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/de79b7d2872b4b00b722607459e20741 : SUCCESS in 4m 47s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c9d22ecbc171498c841bb21c767f15c4 : FAILURE in 6m 15s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/eef79bc6da204641875a89297e052e22 : FAILURE in 10m 07s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/4408bfcad458452da5f4880f87065515 : FAILURE in 1h 09m 17s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/248ce3c95b784c6cb81ff8b12d959d9b : SUCCESS in 1h 09m 20s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/1847f8cbd01d43458d3794ecaf822f76 : SUCCESS in 1h 51m 20s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/4cbe339535ad43e68f3f7fc819c7372f : SUCCESS in 1h 40m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8e2fbf5abd4f49e98388aca897459312 : SUCCESS in 42m 11s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/8b55c95de76442c7856f6cef61cf28a3 : SUCCESS in 1h 21m 11s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/7f8cd9820e4d4b6aa8566338cb59c9db : FAILURE in 30m 04s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/6c39c01e48a545d19d753d81c3c1e881 : SUCCESS in 1h 19m 28s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/2da086e785fc4b33b5e3518c34cf5f48 : SUCCESS in 34m 59s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/f2bb48b2a12440e3ba29061964a42081 : SUCCESS in 1h 11m 31s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"361cf5adefeef97ebb0473069bd80e5dc828f9ee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-09-07 15:09:15.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"4fd230bd2a6d27d9b5e6bffdefd2244b6d4dc799","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-07 15:24:12.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d67272871abb4bb5900cff2031f59653\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/37697766c35d442d80a84e1d711d4771 : SUCCESS in 8m 02s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2a3bb1c3f52348609674c00b6a885701 : SUCCESS in 9m 34s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/3d625282bbcf487a9b320e1cae42f64a : SUCCESS in 13m 22s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7fd67b8c9c6214237c192077ec25a4e58442016e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-07 17:38:19.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4aeb80c8c5b94baa8413fefb082b30ba\n\n- grenade https://zuul.opendev.org/t/openstack/build/a252d55cea6a40a78b59786aed829edc : SUCCESS in 32m 35s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/84f629a764db4f8ea456df2ffe1a80ef : SUCCESS in 42m 27s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/47571712042e4ebea59c7503c5cf3bbc : SUCCESS in 1h 46m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3a8919afe4784a82902b83a58f48ea45 : SUCCESS in 1h 05m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c37bd65eac64443db946d712736b88c3 : SUCCESS in 4m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/13f37c376e9c4166b0c35ce23d542c64 : SUCCESS in 4m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bc90772a75064ecd91405a84d4599703 : SUCCESS in 5m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9a3c019e13524b59a83a2d0a792bccc3 : SUCCESS in 7m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68407bbf080845b0be92d0378a3b0b11 : SUCCESS in 3m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0b933c9cf8da4b29948fe0137b73d0a5 : SUCCESS in 4m 34s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/553f64e6bc514257a492ec93dcc7642f : SUCCESS in 5m 59s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/85c28be32b9e44d1a091623b3929cabb : SUCCESS in 7m 06s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/f1c18f0257e446fb94808a2466f1f8ca : SUCCESS in 1h 15m 24s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/b97a9e2188bb46a3a36db607233cfffd : SUCCESS in 1h 53m 24s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/6d7dbaded7214454a3babf0d2ec41b77 : SUCCESS in 2h 03m 23s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/3292723b140149c58cf69c8e0e2c18e2 : SUCCESS in 2h 23m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/79fd1ca5cb9a4ed0a7dd3665cb0f96d6 : SUCCESS in 58m 35s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/cddaf04eb4af4ec595873b9d5d65a8ac : SUCCESS in 57m 45s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/36d048fcbf8b47dcbb50375b16ebbc55 : FAILURE in 30m 31s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/de50756cffa44983895a33c5cfb24553 : SUCCESS in 1h 01m 56s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/44906c81acbf4464992feed0b920e453 : SUCCESS in 32m 49s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/4a91593bd7a74115af4f3b48278287e8 : SUCCESS in 48m 41s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"207124c16655a29071c94d6074866346af3e021c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-09-08 15:38:18.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"16dbdae8238534c9b1065a44e7d37fae5ffc4b6f","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-09-08 15:48:12.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"5bb36ca63849da6eb09b80a611029a78800de0f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-08 17:54:34.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f5a225d320c54f8193857ee4ba18c061\n\n- grenade https://zuul.opendev.org/t/openstack/build/dbdd08b6b6d64ef992327879a66507a1 : SUCCESS in 28m 10s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a7e3fd553e654dee973055e68b757519 : SUCCESS in 45m 59s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e2c04441eb2d4cff8bf187f80c4ec465 : SUCCESS in 1h 19m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/86876681e9734f0ea51a842d38d5e81d : SUCCESS in 32m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a1accffc29c42d5a0045e152a8259a5 : SUCCESS in 3m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/06e51beafaaf4d01844d415a4018f1c0 : SUCCESS in 4m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/da6dcfe83814425e98051c77ca6eab61 : SUCCESS in 5m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b6da6ff16e8c4ebc85fb149e2a4f82ef : SUCCESS in 9m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d361020e31fc40d69a67a2c861543ea4 : SUCCESS in 7m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/da3801f4e8fc4566b68c189ee33b61cc : SUCCESS in 7m 59s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/277bf3b0b3a0402e91219530d00d1f9c : SUCCESS in 7m 03s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/6622666d7aed46e891750e3cdafc5b77 : SUCCESS in 9m 27s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/9041c0d88d644c9ea62d232b4a529ce3 : SUCCESS in 2h 01m 25s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/2b10d9a27e744fafb074bc06f85dc3d0 : SUCCESS in 2h 05m 34s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/c4e0e19505f24032912df3fbbf129252 : SUCCESS in 2h 10m 31s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/f86c91a7ad194f2b8bbb25343c302827 : SUCCESS in 1h 50m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9f17005d8d1c47a9b458e819f034e319 : SUCCESS in 58m 42s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a994607c9f6b42ccb2571bc4bae3b5aa : SUCCESS in 1h 42m 14s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/4422e783faff48c5b329a04453cc541e : FAILURE in 19m 12s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/29b3710a5e73412dbdc00c0ab3115546 : SUCCESS in 1h 08m 19s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c539748ab9f8410ca94fc21786f12b07 : SUCCESS in 31m 59s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/66ad10814a5048a68f90109d8a569724 : SUCCESS in 43m 14s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"d00a83ff72bcbed8407026fa9fed66ffc45f043e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-08 17:55:18.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"83dcbda4b10efa502bf63b502ceabc2a9246b13e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-08 19:58:54.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/630a95a204bf4e70af506101a4ab2ec5\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1774af169dd64fa7babc0d56429e599b : SUCCESS in 11m 08s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0d93da8ab70a4b8fad4f3ce87b6f58e8 : SUCCESS in 13m 14s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/84b0ea675b7b4c0fbde1f9499d80c22a : SUCCESS in 21m 06s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"dccd3f98b7ddeebc149dc424a29945038e16c0fc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-08 20:50:29.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e46cbc20ea4247c78ffb1def47f0488a\n\n- grenade https://zuul.opendev.org/t/openstack/build/b69af55fa4584ebaa20d05811016fbdf : SUCCESS in 58m 33s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c2c2862f40ff4857b5fea6bce14f4d9b : SUCCESS in 57m 55s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/856e06ea8da142c8b6d6fd9fbc9950c8 : SUCCESS in 1h 46m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/87c8802a266a435295a3a53ebbc120b6 : SUCCESS in 32m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb24894f4f8c4ed392ab7f6f16c2f3b0 : SUCCESS in 5m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/91227d7c7ca24a7da2f192a4ce05dc59 : SUCCESS in 3m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/723ee41ec79d4772bb9c7aa929ea94cd : SUCCESS in 6m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc00e67891b8473380b4b4747fa93a6b : SUCCESS in 6m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9c26abbbc7064ee6ae3bc386921b31a1 : SUCCESS in 6m 01s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a5436fa587f5444f8fa4c20d1b50eed8 : SUCCESS in 4m 52s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/2bad3f1670ee48c29c644561fbd3f4e8 : SUCCESS in 9m 51s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/74d6e7b818f0483a9939ec761977cf7c : SUCCESS in 2h 07m 07s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/c56e4f8d84bf43ddb0daae5ac3c33e3f : SUCCESS in 2h 12m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ab4ec6d1182f4c5184e5811b8edf278b : SUCCESS in 59m 39s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/6d1ffa1f141c4433bbcaa1eefc5e6fcc : SUCCESS in 1h 38m 06s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/193590e89730406ba73c9e9adca83f38 : SUCCESS in 27m 32s","accounts_in_message":[],"_revision_number":4},{"id":"8934cfc6c2963f057ada183742c76b27633223ca","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-08 20:50:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"6e484e6fc63ad0d2a62e688594d473e8163acfd5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-08 20:53:35.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/65d30072b5164c3188496512d8070527\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c6ac77fa8ac4b9e9715f9b754bda966 : SUCCESS in 2m 55s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e6aacd086862465991dc62cc32f587a7 : SUCCESS in 2m 37s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"309e9356a67c5604bc7e5773bf11cd64d1c1379f","revisions":{"2c577ea2b2a21eec61932df9fd2061a7aa399324":{"kind":"REWORK","_number":1,"created":"2025-09-07 13:05:52.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/84/959984/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/84/959984/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/84/959984/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/84/959984/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/84/959984/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/84/959984/1"}}},"commit":{"parents":[{"commit":"1ff89ce2c9dc3191fa4e8aa867e56863bf635b7b","subject":"Merge \"glance-multistore-cinder-import-fips: move to CentOS 10\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1ff89ce2c9dc3191fa4e8aa867e56863bf635b7b"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-07 12:59:54.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-07 13:05:27.000000000","tz":540},"subject":"Add new metadata definition of memory encryption model","message":"Add new metadata definition of memory encryption model\n\nNova recently introduced support of SEV-ES, to extend the existing\nmemory encryption support using SEV.\n\nAdd the new field of flavor extra spec and image property to determine\nthe hardware/CPU feature for memory encryption.\n\nDepends-on: https://review.opendev.org/c/openstack/nova/+/926106\nChange-Id: I9063663a3b14161a67da59a68f5f7c92fd5447f5\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2c577ea2b2a21eec61932df9fd2061a7aa399324"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2c577ea2b2a21eec61932df9fd2061a7aa399324"}]},"branch":"refs/heads/master"},"8472614e56dc4f13435cd2cc985d77990a85f831":{"kind":"REWORK","_number":2,"created":"2025-09-07 13:11:25.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/84/959984/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/84/959984/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/84/959984/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/84/959984/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/84/959984/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/84/959984/2"}}},"commit":{"parents":[{"commit":"1ff89ce2c9dc3191fa4e8aa867e56863bf635b7b","subject":"Merge \"glance-multistore-cinder-import-fips: move to CentOS 10\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1ff89ce2c9dc3191fa4e8aa867e56863bf635b7b"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-07 12:59:54.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-07 13:10:36.000000000","tz":540},"subject":"Add new metadata definition of memory encryption model","message":"Add new metadata definition of memory encryption model\n\nNova recently introduced support of SEV-ES, to extend the existing\nmemory encryption support using SEV. Add the new metadef value to\nexplain the flavor extra spec and the image property to determine\nthe hardware/CPU feature for memory encryption.\n\nAlso fix the outdated link of nova documentation about AMD SEV support.\n\nDepends-on: https://review.opendev.org/c/openstack/nova/+/926106\nChange-Id: I9063663a3b14161a67da59a68f5f7c92fd5447f5\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8472614e56dc4f13435cd2cc985d77990a85f831"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8472614e56dc4f13435cd2cc985d77990a85f831"}]},"branch":"refs/heads/master"},"3f85903feb31f490d4bd5968300c79f683f1cb7e":{"kind":"REWORK","_number":3,"created":"2025-09-07 15:09:15.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/84/959984/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/84/959984/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/84/959984/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/84/959984/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/84/959984/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/84/959984/3"}}},"commit":{"parents":[{"commit":"1ff89ce2c9dc3191fa4e8aa867e56863bf635b7b","subject":"Merge \"glance-multistore-cinder-import-fips: move to CentOS 10\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1ff89ce2c9dc3191fa4e8aa867e56863bf635b7b"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-07 12:59:54.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-07 15:09:07.000000000","tz":540},"subject":"Add new metadata definition of memory encryption model","message":"Add new metadata definition of memory encryption model\n\nNova recently introduced support of SEV-ES, to extend the existing\nmemory encryption support using SEV. Add the new metadef value to\nexplain the flavor extra spec and the image property to determine\nthe hardware/CPU feature for memory encryption.\n\nAlso fix the outdated link of nova documentation about AMD SEV support.\n\nDepends-on: https://review.opendev.org/c/openstack/nova/+/926106\nChange-Id: I9063663a3b14161a67da59a68f5f7c92fd5447f5\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3f85903feb31f490d4bd5968300c79f683f1cb7e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3f85903feb31f490d4bd5968300c79f683f1cb7e"}]},"branch":"refs/heads/master"},"309e9356a67c5604bc7e5773bf11cd64d1c1379f":{"kind":"REWORK","_number":4,"created":"2025-09-08 15:38:18.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/84/959984/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/84/959984/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/84/959984/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/84/959984/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/84/959984/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/84/959984/4"}}},"commit":{"parents":[{"commit":"1ff89ce2c9dc3191fa4e8aa867e56863bf635b7b","subject":"Merge \"glance-multistore-cinder-import-fips: move to CentOS 10\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1ff89ce2c9dc3191fa4e8aa867e56863bf635b7b"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-07 12:59:54.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-08 15:38:08.000000000","tz":540},"subject":"Add new metadata definition of memory encryption model","message":"Add new metadata definition of memory encryption model\n\nNova recently introduced support of SEV-ES, to extend the existing\nmemory encryption support using SEV. Add the new metadef value to\nexplain the flavor extra spec and the image property to determine\nthe hardware/CPU feature for memory encryption.\n\nAlso fix the outdated link of nova documentation about AMD SEV support.\n\nDepends-on: https://review.opendev.org/c/openstack/nova/+/926106\nChange-Id: I9063663a3b14161a67da59a68f5f7c92fd5447f5\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/309e9356a67c5604bc7e5773bf11cd64d1c1379f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/309e9356a67c5604bc7e5773bf11cd64d1c1379f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
