)]}'
{"id":"openstack%2Fglance~961465","triplet_id":"openstack%2Fglance~master~I8e340403808287a594cb742c24bcf6343fa42caa","project":"openstack/glance","branch":"master","topic":"pyproject","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-10-10 15:01:09.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8e340403808287a594cb742c24bcf6343fa42caa","subject":"setup: Remove pbr\u0027s wsgi_scripts","status":"MERGED","created":"2025-09-17 10:53:36.000000000","updated":"2025-10-10 15:02:07.000000000","submitted":"2025-10-10 15:01:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"961465-pyproject","meta_rev_id":"a10c97eb7f109b0a241c3e11f003a73b1ed05934","_number":961465,"virtual_id_number":961465,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-10 15:01:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2025-10-10 12:52:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2025-10-08 15:53:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2025-10-10 12:52:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-17 11:34:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-09-17 13:39:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-25 11:46:14.000000000","updated_by":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"reviewer":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"state":"CC"},{"updated":"2025-10-08 15:53:04.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2025-10-10 12:52:53.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"e47622e1ff86919364e7ba06de495600a9a7229d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-09-17 10:53:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"966f33e8765ce8a2195fb2b93a2626e645460b2b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-17 11:34:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c5f62da5ecb6498889dc91433497b548\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4f39a7edc3cc4ecd8943ecd07a18d51e : SUCCESS in 8m 50s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b912e505fa25483bb245c60de8841773 : SUCCESS in 9m 33s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f39e3dff3e2948338910805f2166dc6b : SUCCESS in 13m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"146f00cb6d6bc363e60d51dfab1cd27b578de489","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-17 13:39:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c48fad1537014351b7e1112140046b44\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/fbbe6cc2badb4c8fb25df4f15aa42588 : SUCCESS in 2m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/7b2b75fbcfc64a1baa0367d289e2575b : SUCCESS in 1h 01m 47s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f19773cac5b347dea68194e7eca66443 : SUCCESS in 45m 16s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/95e971feae6f462b830c08df9889c057 : SUCCESS in 1h 23m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a2cfe89782f24cfea2fc12c32857cfa1 : SUCCESS in 57m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/abeb626855e543e6ab8ca6be2e76b427 : SUCCESS in 4m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2e50968b53be45a1886b8564d0633e78 : SUCCESS in 6m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/79cceb9dee1b413ba9dab61f4052be5e : SUCCESS in 5m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/832b5b4a7fab49b7a30b405b0abba8ff : SUCCESS in 7m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8915e70a97846e0bb4c1d6451364c6c : SUCCESS in 5m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e0a1f9868e9d4c22bd893a6eba8a2f77 : SUCCESS in 7m 51s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b38e7330e17f4ec2aef0cb84ae3f4893 : SUCCESS in 7m 03s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/0c84c8ab8dc34efd8fc999373e1b325b : SUCCESS in 9m 17s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/37b1ee63031c424eb52d749cdaf47e7e : SUCCESS in 1h 33m 19s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/86d2e7495aa540a59838f086e3b45f97 : SUCCESS in 2h 26m 29s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/b014cd5ec5524d56bd8d9778ba17c08b : FAILURE in 1h 44m 55s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/384bedf5e33e4f638b800bb3d78ead30 : SUCCESS in 2h 11m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/14683f47883648c69273efc35ae60f4f : SUCCESS in 1h 04m 52s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/505b5eb26b3e4be7912da10ac0eff725 : SUCCESS in 1h 41m 47s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/c725622b25164215a32626a97716ec54 : FAILURE in 41m 18s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/e15db87ceb98436e8195408d26ac5065 : SUCCESS in 1h 42m 17s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/53938346359e4141a4e7d0d2c24ee19d : SUCCESS in 25m 48s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/642d75c9e31644cb99f7c18d1babce96 : SUCCESS in 1h 02m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"480234dfb50fa70b6492f217a7b6cc61419470d2","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-09-17 14:59:14.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c4fed021965cb6c3774007988f55f60aeaad7bd2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-17 16:27:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc5ebe8231d6488980b56968ac21ea3e\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/aa7d0fba6c5045b2856cffbb2e8a0a78 : SUCCESS in 9m 48s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0e5aaffe4e554b03a3c1ea7b3c559fcc : SUCCESS in 11m 10s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/8b823bb99a7d4faa980c0f4e7a18d50d : SUCCESS in 20m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7fa113c7bcacf6438e4aec6d5c159be5942d9752","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-17 17:42:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/130027240d514de2a95342d37f93391c\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b41107ce0bf245bd965c1740ef2b882d : SUCCESS in 3m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/c53266afd98b427c9b4cf8501f72c600 : SUCCESS in 46m 03s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/411672fad9a745bf97cd4b36445d030b : SUCCESS in 1h 03m 46s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/93c0f2904b4d403f9409e017c171a375 : SUCCESS in 1h 36m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4fe2fad7b20e482b9e0bd0bda92a000c : SUCCESS in 57m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d14dde8cb78441a3801e83ddf55b89f0 : SUCCESS in 5m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cbcd0460cd1848b9a7f64bebff42e7fc : SUCCESS in 5m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/99ee5a1d02ee46d785e2497ceccb2bd9 : SUCCESS in 5m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0f02b89f874248de9245af93119137a5 : SUCCESS in 7m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67c160864eab4909991fafeae229d31e : SUCCESS in 6m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bae481dd6b8c4115a79186436b0d6e96 : SUCCESS in 7m 04s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2fd8aeec4c774d6894194d0226e517f1 : SUCCESS in 7m 32s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/fa04898c78824425ab528f21d467b535 : SUCCESS in 11m 46s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/6efe195dab1f469d8e3d8eafa2c6a181 : SUCCESS in 2h 00m 09s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/791f5d9c72384c54a2f211671dfb113e : SUCCESS in 1h 57m 06s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/5e808a225b7e47059e83ccb3615ed528 : SUCCESS in 1h 51m 41s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/c3e81764e17346a9b874ce531d1b9efd : SUCCESS in 2h 20m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9e27781cab964f3a9ea08ea1e82986a5 : SUCCESS in 54m 41s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/538eb6eb48b6403a8e22b2b5a2c52ca2 : SUCCESS in 1h 53m 47s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/05cb44444c01426782dd4f7ad90a4cd9 : FAILURE in 29m 38s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/94052c12f452474bbc16500de82a59c0 : SUCCESS in 1h 53m 15s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/610f53ecf20a431e9a5afc0f21b026cd : SUCCESS in 34m 14s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/680e427a4d244abbb8614cd4e65647ad : SUCCESS in 1h 00m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2c97d6b3536bdae3da1f34e45b87aa1c4e3ae982","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-10-08 15:53:04.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f9a0cf91b9fdd00e6b4f6700d062bd4935ddedc5","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-10-10 12:52:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"58cb533abcdfab4f8f89708e82a559941312441d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-10 12:53:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"74c86b25e41025e62b61305ba6a7c60fdc0c163a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-10 15:01:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/943d0b5379eb4aeb8f3a4450396f1822\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/79d7664b63f245219cf708641a0ec0a8 : SUCCESS in 2m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/ad60a7bda7cf4fa68e8720fde47b70f8 : SUCCESS in 29m 33s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/59a1c09a50a9479f8332dc92faebaa59 : SUCCESS in 46m 26s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c024e16629054a718f0341eaf7a4132b : SUCCESS in 1h 34m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7ddfec9973ef45ada48e00e74d5d46d6 : SUCCESS in 1h 02m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa056c9b99784803bf951a40fa140dcb : SUCCESS in 2m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0e7c667eafa84bc5abf1bf3fefb7e532 : SUCCESS in 5m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/83a6ae267ff547da9f6224f1a2bd5e85 : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c63fa173e2b44b7a5a722f190846191 : SUCCESS in 8m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0be542fc425641869ba1f6e06aea5bc4 : SUCCESS in 5m 28s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d1d1e93fbb3b435f81fd7937306dc0d7 : SUCCESS in 5m 59s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/cb28b0b0607d4674ad897caebb61da2e : SUCCESS in 9m 38s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/7f041e6644304e6796c04a2475f7a44c : SUCCESS in 2h 02m 23s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/329a5c7ee875418ea86b78fc985b276d : SUCCESS in 1h 55m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ac7984f11aa5421390699e1c11d3e7d4 : SUCCESS in 34m 03s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a735b809d07c45b98b199727c678e342 : SUCCESS in 1h 46m 55s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/240cf00624884caa9913d4d288342725 : SUCCESS in 19m 32s","accounts_in_message":[],"_revision_number":1},{"id":"bb0dcdf91841acd943a98afa21df110238602ea8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-10 15:01:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a10c97eb7f109b0a241c3e11f003a73b1ed05934","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-10 15:02:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc076329ca504ed68518cad1012d3779\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5e0a440ea2447ea9047bd00a398ad11 : SUCCESS in 43s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8f09e83b85f846ee926d43d6dc162864 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c1827a3f3334a6a686d85d1060ab49afe39a62bb","revisions":{"c1827a3f3334a6a686d85d1060ab49afe39a62bb":{"kind":"REWORK","_number":1,"created":"2025-09-17 10:53:36.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/65/961465/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/65/961465/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/65/961465/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/65/961465/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/65/961465/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/65/961465/1"}}},"commit":{"parents":[{"commit":"7a21722ecdd349610102ce4ea67705dd0c2c4fb0","subject":"Add pyproject.toml to support pip 23.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7a21722ecdd349610102ce4ea67705dd0c2c4fb0"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-09-16 16:47:58.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-09-17 10:53:27.000000000","tz":60},"subject":"setup: Remove pbr\u0027s wsgi_scripts","message":"setup: Remove pbr\u0027s wsgi_scripts\n\nThis is technical dead end and not something that\u0027s supported by default\nin the any versions of setuptools released in over 6 months (and\ntherefore, by extension, pbr). It\u0027s time to remove it.\n\nSome references to wsgi scripts and the eventlet server are removed.\n\nChange-Id: I8e340403808287a594cb742c24bcf6343fa42caa\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c1827a3f3334a6a686d85d1060ab49afe39a62bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c1827a3f3334a6a686d85d1060ab49afe39a62bb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
