)]}'
{"id":"openstack%2Fglance~961466","triplet_id":"openstack%2Fglance~master~I89c8a43069f60612eca9613d47047b8a5190aeae","project":"openstack/glance","branch":"master","topic":"pyproject","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-10-10 15:04:39.000000000","reason":"Change was submitted"},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2025-10-08 16:17:05.000000000","reason":"\u003cGERRIT_ACCOUNT_8122\u003e replied on the change","reason_account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}}},"hashtags":[],"change_id":"I89c8a43069f60612eca9613d47047b8a5190aeae","subject":"Migrate setup configuration to pyproject.toml","status":"MERGED","created":"2025-09-17 10:53:36.000000000","updated":"2025-10-10 15:05:35.000000000","submitted":"2025-10-10 15:04:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"961466-pyproject","meta_rev_id":"88e6dedb813c270e6d4b9e3801037c89c54402a4","_number":961466,"virtual_id_number":961466,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-10 15:04:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2025-10-10 12:53:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2025-10-08 16:17:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2025-10-10 12:53:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"date":"2025-10-08 16:17:05.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-17 11:35:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-09-17 11:54:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-18 13:31:44.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2025-09-18 13:52:16.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"CC"},{"updated":"2025-09-18 17:42:15.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2025-10-10 12:53:39.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"7100bf1fc6dd3512b62c5f3f391be9b42381864b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-09-17 10:53:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"262a9c3435585e8b99919dc50b1a4c2e1e936b2b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-17 11:35:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/696ae9bcef944a0ab27d3ba7d7bb74e4\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6d241313f2f343598a4c7d97b42b380e : FAILURE in 5m 16s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/64d0c6fc034d4449895e3c6faaa57aa1 : FAILURE in 4m 47s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ae8e9799cce04a4f9cf53edfa28a1362 : FAILURE in 13m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"968ef13235587924531382b5bb57335f99af277b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-17 11:54:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3e33d6384fdc40278f4e2f1d7e361ac5\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/0380a368ed09419e8397c1a514c37938 : FAILURE in 2m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/265af02a7388445292c37d0d24864252 : FAILURE in 39m 54s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3ab584fb5c8e46978f80a739ece19dda : FAILURE in 35m 14s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a87d5e56597a43b3a85d971f1cc49c91 : FAILURE in 8m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9c06040384c04afcaa4754dc06e4adb2 : POST_FAILURE in 9m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f34bfed762f0450d840a09094b848d98 : FAILURE in 3m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/69649bc484ab4021969e8aeaaa7b2143 : FAILURE in 3m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cb54a5f10a4642cfbe013aef7d77211b : FAILURE in 3m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f65a70c3b53f470da5a5265c68b0ef24 : FAILURE in 4m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4fe6c4abd8644d0fbf42a704c51817cf : FAILURE in 3m 05s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/86287b474e9a45c68d0355f077c8fde5 : FAILURE in 3m 01s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/0e0ce9c55aa149ceb6691313e870bf0f : FAILURE in 5m 19s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/8b1f462c61804afb8a80eb8dc9a8c84d : FAILURE in 14m 59s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/49bed9c8ee564e4ba4fc549a37516e8e : FAILURE in 8m 06s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/095846ff9b334020ac7ca93a5cea1bcd : FAILURE in 11m 36s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/d9aa5437e2be455dac3870a043a1dbdf : FAILURE in 8m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/67681b9618b4410f87a6e870d8727918 : FAILURE in 13m 04s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/80ce9364d3594d2ca2deaea936a71160 : FAILURE in 9m 16s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/99bf0cedff7c43c9a232e43f0244ca77 : FAILURE in 25m 24s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/e714a735a1514a25b1411f5bb7155266 : FAILURE in 8m 37s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/86c9a1a74bdc443c991a09e580bb966c : FAILURE in 16m 20s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/39be8468a7c54c7ca0c902aadf2bead2 : POST_FAILURE in 18m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"44f858257c21b9388384080a5bce95c84d738367","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-09-17 13:15:02.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"6972909bb36110a262fab013d7b6cf0ee60d3a0f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-17 13:29:24.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/429d6d22445149cba91625155b417a08\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8fd7ed17b57243e0b2aa3333eef61ef0 : SUCCESS in 8m 25s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1670e06e0e6743e49aa45f8700f66f90 : SUCCESS in 8m 32s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/6dfb664da00744469f3acca925ecf447 : SUCCESS in 12m 25s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d20b1d7f82e30a3160ff10d2c37c8180f9f879c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-17 15:41:56.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cb3c82475b6043ff93d9e2bb98fadc0d\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/aa18ea066b114822a22dea9543c66f8b : SUCCESS in 2m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/042ea303566b4691ae32340dfcea2247 : SUCCESS in 53m 28s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/82344436d98c484988b75ad99f75b218 : SUCCESS in 1h 00m 56s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0e4dc9de942344468469f223acfc1567 : SUCCESS in 1h 32m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/99714b72f85644e99bf50502e91936bf : SUCCESS in 49m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/efbced7105c640f9bb846c3d46d0a9d4 : SUCCESS in 4m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/63db30237c44467c99ce0810cba4dfee : SUCCESS in 5m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bd28258b6c5c4a4aad9a7b2243c07378 : SUCCESS in 5m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ae071108fb204bc8b54514a6b0fcadef : SUCCESS in 8m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb98692da0ff4ef9b50a93aa7baf5a5c : SUCCESS in 6m 07s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/79c8ad5f1ca74c1884e5462873080763 : SUCCESS in 7m 24s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/e097d2f1058d4871b39d9841ea0cb104 : SUCCESS in 10m 20s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/f5d6ee412b084388abc3cf5e1be4bd8a : SUCCESS in 1h 56m 48s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/7582293ce5434534a6d8c6fdbdd5c6c3 : SUCCESS in 2h 04m 18s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/38ca195e070147e09f157477643ae8d8 : SUCCESS in 2h 07m 14s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/30db1970b78e48d3ac7f69711f07f226 : SUCCESS in 2h 16m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e80176fdeb374bfba471e8dae17a8e75 : SUCCESS in 42m 59s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/55bc8c3ab8494790a3a827d68b3f1dc4 : SUCCESS in 1h 37m 29s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/19da7871ce0a446fa586c7be16b9a96b : FAILURE in 32m 28s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/386dcb0db3214fffa3beed3810ee35ac : SUCCESS in 1h 44m 49s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/d8edef6473264b4b96af7202a8b2c627 : FAILURE in 34m 56s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/837840b7587a4af1bdd5bb5bf710a725 : SUCCESS in 1h 10m 46s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"de601bfa09798073b2407d0f90d43ebe8db67272","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-09-18 13:31:44.000000000","message":"Patch Set 2: Code-Review-1 Workflow-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"f884dd28d3ad16452fbdc49eae32dd776bd5e413","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-09-18 13:52:16.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"92e5ce0323d4ab5403fb1c5dd7897abc97e5f6fd","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-09-18 17:42:15.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0ccf2fdd935f691d9b3c9bcf9e8afed2ae5dcef7","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-09-19 18:11:22.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2dd2812d5331556eb4b89ac33ac1563dba359646","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-10-08 16:17:05.000000000","message":"Patch Set 2: -Workflow Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"95f7382b0ce106023e435f95bde9348dfbd11293","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-10-08 16:18:42.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6d2277d26d6777ad9677122353660b37d2dca970","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-08 17:09:35.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f5cd3ed14b90474d9c510b824bbc098d\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/003f6f0d853d406fbadb74f58afb6dab : SUCCESS in 9m 48s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c70a3d3e476b4f37a0b438a21ccc6c6a : SUCCESS in 7m 48s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/0ffdc14ba50e41b6b0c2f9dce0c190f6 : SUCCESS in 14m 40s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7298dc8137063a1030eefdad17938afb2d83f654","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-08 18:54:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/841dcf9da24e4adf92f431c1c6d3c912\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/56b9417313cc4f589c3ce2883f78bb46 : SUCCESS in 3m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/5773cd2b0223473dbc9b7f4d41966a48 : SUCCESS in 59m 47s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ad7d364eabdc4576a4c2de987b3f16a7 : SUCCESS in 39m 03s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/43801a593e374e21b526ca04bc9fa081 : SUCCESS in 1h 47m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4620203df86e4c9db9e6d62f304148b5 : SUCCESS in 38m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d726379e84754f309eac8bc61e147c69 : SUCCESS in 5m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/58cd61ed4474447e8554896b5fa16bc0 : SUCCESS in 4m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/246b8ac0eddb449a924f3b3868a851ef : SUCCESS in 4m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1dd2a641b3e8434cb6774ab3a2d3eb0a : SUCCESS in 8m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f857adee3af4bd5990761c6ebbf71a6 : SUCCESS in 6m 13s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ade99af1d19c4cf59fba5c00b06983ca : SUCCESS in 7m 25s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/965a071cc4834a4ca78a537cc2bcf3a7 : SUCCESS in 8m 50s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/bce721cc6b1e4c8a848818e7dfe5e9cf : SUCCESS in 2h 02m 37s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/3fd21a8bfd49499ab247c8d5c3cfdfb5 : SUCCESS in 2h 03m 03s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/3171c9ca8d9f4bd0923d527b72ad187a : SUCCESS in 1h 17m 50s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/34e81960bd974afb932b179d86bc7565 : SUCCESS in 2h 30m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0a1382aaaf6d4632b2c0bcea1d5ce290 : SUCCESS in 57m 58s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/5f77dee4c4e440d78c07b9106b4d596f : SUCCESS in 1h 12m 34s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/90fbfcc2c9794f16b9b7208b63e5b5f2 : FAILURE in 22m 43s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/550dec5385a74ede8b685958f913b55f : SUCCESS in 1h 21m 07s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/d57710f8d9a34ca38bd6c15dd91a9dfd : SUCCESS in 32m 30s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/fb568ffa487a49ee97f1703a6a73e77d : SUCCESS in 34m 36s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ca75029cac77853b928fd61afb9874bd2b176e26","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-10-10 12:53:39.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"7b504e76aebc6e4f4e84f8e2fa168ed078401f6d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-10 12:54:21.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"714c71109620d809cfd57eb15fb59ecea5b1cee8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-10 15:04:38.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/df0a7dd0848f4d9cbcffdf5da77e072f\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/0633b4ff8431428aa4b35517d6a121d3 : SUCCESS in 2m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/605413297b214965bd8b63db9d3d733d : SUCCESS in 49m 28s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e616195c15f443c9bf8a5ea4a6e397c3 : SUCCESS in 43m 41s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/de008652cbaf4b5d968d3826a4c5bc21 : SUCCESS in 1h 26m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9c336b16cbec43cfa641d3f09714f453 : SUCCESS in 41m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dffa089df06249c499a4b7f93d3fd6ba : SUCCESS in 3m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b25c75a6240742529751fdf3f5143303 : SUCCESS in 5m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ad8883c34db342bda55dc5328fcc1816 : SUCCESS in 5m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8394f852dc24fe1ab2422a03b47eb88 : SUCCESS in 4m 57s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f0e096d712714caa8ad0b8a48e351080 : SUCCESS in 6m 37s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/b1b517a053474468a73a604b018acb07 : SUCCESS in 10m 15s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/5a844feed0554a77b6ca75602d7f0a22 : SUCCESS in 2h 05m 58s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/5a0d16aa99f44e6e8153916fd067d091 : SUCCESS in 2h 03m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b4e77493c7534e35869fe7d93ae17615 : SUCCESS in 58m 10s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b08687ed9fec4f2fb6683447c1592c9b : SUCCESS in 1h 49m 41s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/91250051cc2b4e58926ccf56541c35b4 : SUCCESS in 31m 08s","accounts_in_message":[],"_revision_number":2},{"id":"b4fbb638153a83916d975904b9093f95626c8be5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-10 15:04:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"88e6dedb813c270e6d4b9e3801037c89c54402a4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-10 15:05:35.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/720fca05d83645c1ad606c6e02216c16\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea787c7a29a44400b532e9a080100667 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c45485f8a02b7f4efea98bb0acab0d61f3d07bf1","revisions":{"73d5aadb0030d576d5659763c9c6167881ee3e07":{"kind":"REWORK","_number":1,"created":"2025-09-17 10:53:36.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/66/961466/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/66/961466/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/961466/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/961466/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/961466/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/66/961466/1"}}},"commit":{"parents":[{"commit":"c1827a3f3334a6a686d85d1060ab49afe39a62bb","subject":"setup: Remove pbr\u0027s wsgi_scripts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c1827a3f3334a6a686d85d1060ab49afe39a62bb"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-09-16 16:47:40.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-09-17 10:53:27.000000000","tz":60},"subject":"Migrate setup configuration to pyproject.toml","message":"Migrate setup configuration to pyproject.toml\n\nChange-Id: I89c8a43069f60612eca9613d47047b8a5190aeae\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/73d5aadb0030d576d5659763c9c6167881ee3e07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/73d5aadb0030d576d5659763c9c6167881ee3e07"}]},"branch":"refs/heads/master"},"c45485f8a02b7f4efea98bb0acab0d61f3d07bf1":{"kind":"REWORK","_number":2,"created":"2025-09-17 13:15:02.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/66/961466/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/66/961466/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/961466/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/961466/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/961466/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/66/961466/2"}}},"commit":{"parents":[{"commit":"c1827a3f3334a6a686d85d1060ab49afe39a62bb","subject":"setup: Remove pbr\u0027s wsgi_scripts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c1827a3f3334a6a686d85d1060ab49afe39a62bb"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-09-16 16:47:40.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-09-17 13:14:05.000000000","tz":60},"subject":"Migrate setup configuration to pyproject.toml","message":"Migrate setup configuration to pyproject.toml\n\nChange-Id: I89c8a43069f60612eca9613d47047b8a5190aeae\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c45485f8a02b7f4efea98bb0acab0d61f3d07bf1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c45485f8a02b7f4efea98bb0acab0d61f3d07bf1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
