)]}'
{"id":"openstack%2Fglance~964294","triplet_id":"openstack%2Fglance~master~I885d626991686e1ba19eedb1416cacb6af39ea20","project":"openstack/glance","branch":"master","topic":"bug/2128698","hashtags":[],"change_id":"I885d626991686e1ba19eedb1416cacb6af39ea20","subject":"Fix glance_store configuration section heading","status":"NEW","created":"2025-10-17 10:45:59.000000000","updated":"2025-10-17 11:07:42.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"90d41f8d72b77f0834d3ef1b1c17c5356e4e82c3","_number":964294,"virtual_id_number":964294,"owner":{"_account_id":38494,"name":"Matthieu Cote","display_name":"Kaddy","email":"cote.matthieu08@gmail.com","username":"Kaddy"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-10-17 11:07:42.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-17 11:07:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6a1339b2ae8c8dca5d975fdbed653a0b0c2266f2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38494,"name":"Matthieu Cote","display_name":"Kaddy","email":"cote.matthieu08@gmail.com","username":"Kaddy"},"date":"2025-10-17 10:45:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"90d41f8d72b77f0834d3ef1b1c17c5356e4e82c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-17 11:07:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/81e210c4ca384c5983962e123f782351\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57f1bf19259c440d9f3966f6fcc72ba2 : SUCCESS in 6m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8555c724e3d14e03bc65bc65fce1ba54 : SUCCESS in 4m 03s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d2950367d687e386f06b6333ca70ae5dc42c1867","revisions":{"d2950367d687e386f06b6333ca70ae5dc42c1867":{"kind":"REWORK","_number":1,"created":"2025-10-17 10:45:59.000000000","uploader":{"_account_id":38494,"name":"Matthieu Cote","display_name":"Kaddy","email":"cote.matthieu08@gmail.com","username":"Kaddy"},"ref":"refs/changes/94/964294/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/94/964294/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/94/964294/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/94/964294/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/94/964294/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/94/964294/1"}}},"commit":{"parents":[{"commit":"ab6114611fb0e5234d21ddc25fec8ee8902a13aa","subject":"Merge \"Replace remaining usage of healthcheck middleware filter\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ab6114611fb0e5234d21ddc25fec8ee8902a13aa"}]}],"author":{"name":"Matthieu Cote","email":"cote.matthieu08@gmail.com","date":"2025-10-17 10:25:18.000000000","tz":120},"committer":{"name":"Matthieu Cote","email":"cote.matthieu08@gmail.com","date":"2025-10-17 10:27:32.000000000","tz":120},"subject":"Fix glance_store configuration section heading","message":"Fix glance_store configuration section heading\n\nThe step describing [glance_store] configuration actually\nrequires modifications to three sections: [DEFAULT],\n[glance_store], and [fs]. Updated the section heading to\nexplicitly mention all three sections and clarify that\nthe [fs] section must be created.\n\nCloses-Bug: #2128698\nChange-Id: I885d626991686e1ba19eedb1416cacb6af39ea20\nSigned-off-by: Matthieu Cote \u003ccote.matthieu08@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d2950367d687e386f06b6333ca70ae5dc42c1867"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d2950367d687e386f06b6333ca70ae5dc42c1867"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
