)]}'
{"id":"openstack%2Fglance~964375","triplet_id":"openstack%2Fglance~master~I254945a34ac5b9b6cef8c478d1e9b3b5b53e6080","project":"openstack/glance","branch":"master","topic":"bug/2128971","attention_set":{"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2025-10-22 07:55:40.000000000","reason":"\u003cGERRIT_ACCOUNT_38494\u003e replied on the change","reason_account":{"_account_id":38494,"name":"Matthieu Cote","display_name":"Kaddy","email":"cote.matthieu08@gmail.com","username":"Kaddy"}}},"removed_from_attention_set":{"38494":{"account":{"_account_id":38494,"name":"Matthieu Cote","display_name":"Kaddy","email":"cote.matthieu08@gmail.com","username":"Kaddy"},"last_update":"2025-10-22 07:53:45.000000000","reason":"\u003cGERRIT_ACCOUNT_38494\u003e replied on the change","reason_account":{"_account_id":38494,"name":"Matthieu Cote","display_name":"Kaddy","email":"cote.matthieu08@gmail.com","username":"Kaddy"}}},"hashtags":[],"change_id":"I254945a34ac5b9b6cef8c478d1e9b3b5b53e6080","subject":"Clarify which endpoint ID to use in oslo_limit config","status":"NEW","created":"2025-10-19 16:29:46.000000000","updated":"2025-10-22 07:55:40.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"747460c0cbe1e7cad998939fa9bb96bd2cc625aa","_number":964375,"virtual_id_number":964375,"owner":{"_account_id":38494,"name":"Matthieu Cote","display_name":"Kaddy","email":"cote.matthieu08@gmail.com","username":"Kaddy"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-10-19 16:38:49.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-19 16:38:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-20 14:01:50.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"CC"}],"messages":[{"id":"66ce6672e29d74bed455d2434b51c0973670449b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38494,"name":"Matthieu Cote","display_name":"Kaddy","email":"cote.matthieu08@gmail.com","username":"Kaddy"},"date":"2025-10-19 16:29:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c90c007f49c4ddb973cb19148c72c607a36407f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-19 16:38:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a2fecbd1eeaf422f9b3df3d9d0d3a724\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8e2cadbae7844ad80b7a5063ac51a0e : SUCCESS in 4m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93bcff1afd9b40d195efbcecf3b6e5f6 : SUCCESS in 7m 33s","accounts_in_message":[],"_revision_number":1},{"id":"d09b2bf12d1df6bfbd8ecb4cdc474801848c4379","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2025-10-20 14:01:50.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"747460c0cbe1e7cad998939fa9bb96bd2cc625aa","author":{"_account_id":38494,"name":"Matthieu Cote","display_name":"Kaddy","email":"cote.matthieu08@gmail.com","username":"Kaddy"},"date":"2025-10-22 07:55:40.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4a96ffc9c233470801ab0d0d55d73b27a95e05a1","revisions":{"4a96ffc9c233470801ab0d0d55d73b27a95e05a1":{"kind":"REWORK","_number":1,"created":"2025-10-19 16:29:46.000000000","uploader":{"_account_id":38494,"name":"Matthieu Cote","display_name":"Kaddy","email":"cote.matthieu08@gmail.com","username":"Kaddy"},"ref":"refs/changes/75/964375/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/75/964375/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/75/964375/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/75/964375/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/75/964375/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/75/964375/1"}}},"commit":{"parents":[{"commit":"ab6114611fb0e5234d21ddc25fec8ee8902a13aa","subject":"Merge \"Replace remaining usage of healthcheck middleware filter\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ab6114611fb0e5234d21ddc25fec8ee8902a13aa"}]}],"author":{"name":"Matthieu Cote","email":"cote.matthieu08@gmail.com","date":"2025-10-19 07:58:26.000000000","tz":120},"committer":{"name":"Matthieu Cote","email":"cote.matthieu08@gmail.com","date":"2025-10-19 07:59:50.000000000","tz":120},"subject":"Clarify which endpoint ID to use in oslo_limit config","message":"Clarify which endpoint ID to use in oslo_limit config\n\nThe oslo_limit configuration step instructs users to replace\nENDPOINT_ID but does not explicitly state which of the three\nendpoint interfaces (public, admin, internal) should be used.\nUpdated the text to specify the public interface endpoint.\n\nCloses-Bug: #2128971\n\nChange-Id: I254945a34ac5b9b6cef8c478d1e9b3b5b53e6080\nSigned-off-by: Matthieu Cote \u003ccote.matthieu08@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4a96ffc9c233470801ab0d0d55d73b27a95e05a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4a96ffc9c233470801ab0d0d55d73b27a95e05a1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
