)]}'
{"id":"openstack%2Fglance~964508","triplet_id":"openstack%2Fglance~master~Iea2fffa75b973b5a34da6fc4145841888ea19034","project":"openstack/glance","branch":"master","topic":"bug/2085321","attention_set":{},"removed_from_attention_set":{"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2025-11-24 15:40:49.000000000","reason":"Change was submitted"},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2025-11-24 13:58:16.000000000","reason":"\u003cGERRIT_ACCOUNT_8122\u003e replied on the change","reason_account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}}},"hashtags":[],"change_id":"Iea2fffa75b973b5a34da6fc4145841888ea19034","subject":"Inject os_glance.* section in the property protection file","status":"MERGED","created":"2025-10-21 23:52:18.000000000","updated":"2025-11-24 15:41:50.000000000","submitted":"2025-11-24 15:40:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"964508-bug/2085321","meta_rev_id":"d97ed6fb26b623da8d4ad9e93aa88b12d325c7ca","_number":964508,"virtual_id_number":964508,"owner":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-24 15:40:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2025-11-24 14:01:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2025-11-24 14:01:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-22 00:10:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-10-22 02:15:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-28 09:14:57.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"CC"},{"updated":"2025-11-06 12:15:32.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"975d93652bbb08582c6b0251beb7192c5f3158cf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-10-21 23:52:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bf882aed2a81eb08514f62df2a4c88b6cd7a0dd5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-22 00:10:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e1f1e9a8e4714296aa090708e7f3b470\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/62489f7034ed49fd9f633ceb36e6a575 : SUCCESS in 9m 52s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/aaf8a256af0b47ccb12cb72195e4e73e : SUCCESS in 8m 35s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/aa4b5c2acf104695925d44b223018c7b : SUCCESS in 15m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a3d9c6b53a3a1c35eeceaa7df915610296cb3715","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-22 02:15:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a5383a3aa9c413ab2e6a278e99eb94c\n\n- grenade https://zuul.opendev.org/t/openstack/build/cd8a300299b840c3b1033f692b1458fd : SUCCESS in 56m 18s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d203a11512cc4f0db52c35d030de8a67 : SUCCESS in 1h 03m 24s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/13a24505ae9f4e46bc6f50cb8d65cc8e : SUCCESS in 1h 27m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/18b1665f7d7a4ccfa1ccb34f11ec2bd8 : SUCCESS in 1h 00m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7692d1624ff04c9b9b77228cbb06309d : SUCCESS in 6m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/402d2fe53b974d158d2c0c1464227629 : SUCCESS in 5m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b864d29dd90b447cb395a88a39d7536b : SUCCESS in 10m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad8d103cfd0c47bf830b6258bc912a1d : SUCCESS in 5m 55s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c057a4bd85254c3d9d38355fc0c94368 : SUCCESS in 6m 49s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/6587bcbeb24d4588acca9ebff63eae01 : SUCCESS in 12m 58s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/b3a1f40dbe4b4c99bad1291417fc9ab0 : SUCCESS in 1h 48m 31s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/7bf46b4a683e45f29ca2b89f87bf67b0 : SUCCESS in 1h 16m 30s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/ae76af532b074f90825ebbde90088247 : SUCCESS in 2h 17m 27s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/2fc52ce677e146cfa81b3b13f1c322b2 : SUCCESS in 2h 03m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/85c260c7a53442fba1235ace25447e08 : SUCCESS in 56m 57s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/3d543cc4b2ad4efb9c53c41103c2a588 : SUCCESS in 1h 14m 14s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/7456c0ed0851499ea28a328549276bb2 : FAILURE in 14m 06s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/a70cfb3147404965a147aa76529cef0a : SUCCESS in 1h 46m 40s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/28efe777e2804039af4da4fbeb4bb93c : SUCCESS in 16m 02s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/412a9f6a34e9452886fd27ce5cb8de5a : SUCCESS in 1h 05m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8c80d2caf7a8c06d3f54e39421198a8d568eae84","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-10-28 09:14:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"30a793995a173f45874ae9154417be8c7b86d48b","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-11-05 17:55:07.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4dfd5c34cc566186b5cf4be17135c10610da7c62","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-11-05 19:07:36.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"a51c6f0cab79f85595d591a6e994b17472b172d0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-05 21:22:30.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd89b68fd786422a86ace051c3c0839a\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/49a40639155c4d738bd24176c3495165 : SUCCESS in 6m 02s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/58b2b4687d2f4a5f9938304647fff9f7 : SUCCESS in 9m 49s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ee8c80c0d49649dd94f998b5ad9db5b4 : SUCCESS in 14m 14s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a8175844ba309dc19562886804e15d64d9f74483","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-05 21:26:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/145c843cf9654644b835611d24b1e977\n\n- grenade https://zuul.opendev.org/t/openstack/build/4392f4453ae14a6c89400ebf584328be : SUCCESS in 1h 17m 21s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f368f755908e45e2a2d552febf97cb13 : SUCCESS in 1h 10m 33s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1ed6af77560c4cc489f618b1f069213b : SUCCESS in 1h 28m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ceca93f9d8dd46eb93c916009820527f : SUCCESS in 1h 01m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43771c66989545c889d725414684b5cd : SUCCESS in 4m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/321c1f0cef654f01b69fb585bf38154c : SUCCESS in 6m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0d38b387f9514c0d910020d462ada9cb : SUCCESS in 8m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60527bb3dea64bbe88b0947b69ec4111 : SUCCESS in 6m 16s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3cde3f00961f44cb959eed7b188b27ec : SUCCESS in 6m 35s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/80f2a8e6152c41c5991a63cfea0483f6 : SUCCESS in 9m 53s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/901aa5ebc5ad414dbb5da95d08e17c57 : SUCCESS in 1h 43m 46s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/e18216eead994d7484e1b5b5fbeed088 : SUCCESS in 1h 10m 38s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/51845ebe11184de5b17722d905d4dcf0 : SUCCESS in 2h 07m 44s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/52df3b3fa74c426c821ced8664573dbf : SUCCESS in 2h 12m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8e8edc2475a943daaeaa0afbba579960 : SUCCESS in 41m 59s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/16e3e025d0c14b5d938e4937367fb5f9 : SUCCESS in 1h 58m 03s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/95b179af2e8c4ee5958574e4aa715530 : FAILURE in 20m 06s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/30904ca4ad8b4e0397ddfdfa8f95a4b3 : FAILURE in 26m 15s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/7b23cf770a92402c8d2cca464ea7adaa : SUCCESS in 28m 26s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/e40bed34964d463d99cd7e4d80178433 : SUCCESS in 1h 06m 17s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0a4b651a8b05a69d979e74e14df865c73a128043","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-11-06 12:15:32.000000000","message":"Patch Set 2: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"7d06fbf9040446f1ffb4190dc9df26f2115aedd0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-11-17 12:05:58.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"f942120fde3dae59afd83f9f5245ca8cc8f1fe10","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-17 12:36:16.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/997d7e3a583247bdae2aa2f407092cfa\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/24145d5f8f9a4e82b325e445e8e9f9af : SUCCESS in 6m 22s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d3e629aeb26d4a37a16a6b995c8dd220 : SUCCESS in 6m 14s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4a02f98545924bc6a9180ca138928142 : SUCCESS in 14m 06s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"283ca05463cda1004c8522fe14b8bd0c41701ba4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-17 15:09:05.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e3ed701612b542aa9cc479c1223303b5\n\n- grenade https://zuul.opendev.org/t/openstack/build/b3b375479f9442ad8a167dcd2ac7f2b7 : SUCCESS in 58m 41s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/71eaf8a933e742e2812dc59257a2275a : SUCCESS in 1h 17m 03s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6e8bea6559f04003b1b731c8f9a8eefa : SUCCESS in 1h 33m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a89ec68083ec4eee87e3a2f6fa63fcde : SUCCESS in 33m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f883c493bd2e4d3694bba77d4d4befd7 : SUCCESS in 4m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/13d7c0bdda7442de87dac54c22898689 : SUCCESS in 4m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8e7d505dff58480f9829113d03092e73 : SUCCESS in 9m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f51d372a24c046f28520327583faa5b4 : SUCCESS in 6m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4e79ee3009f04afe9ef0ab6dc0af9ca2 : SUCCESS in 6m 16s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/74368780b2f7480daa91b91d3fded7ca : SUCCESS in 5m 50s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/ae88503d5c24447da5d0ddce6ffd46ec : SUCCESS in 9m 59s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/21f7b0bf0ce84051b9dde5ae05a2df3f : SUCCESS in 2h 14m 46s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/8534ed91dab542a192b1857bbf45663c : SUCCESS in 1h 55m 05s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/df5274b33f524159bbbe6a87188dab29 : SUCCESS in 1h 48m 27s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/41e4462e4b1440d2b701cf575bed7afa : SUCCESS in 2h 57m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b30536ca841d48cda4197f72b6177885 : SUCCESS in 1h 19m 13s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/672fa5e453714c40a3cb811c935e79b5 : SUCCESS in 1h 53m 30s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/25b3ca9becd34d5ba6bd43b47288b35e : FAILURE in 14m 56s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/c81939be112b46d68775addc3e975cc4 : FAILURE in 34m 09s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/74ee0fe156ae432a8a8765f25fa90f01 : SUCCESS in 33m 48s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/83a4b738ea2842faa9770ca33f5ced11 : SUCCESS in 1h 06m 14s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"486328574099aa3a3587605fc54096eba69e7ced","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-11-24 13:26:33.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"731e93184b6b6dbe70f7137e19c6e3a5f88d9312","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-11-24 13:41:45.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"83e86049367d8a485d6379f9974770a73be0db74","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-11-24 13:56:25.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ae9e928c6e40606e7f7987a34dcdd92d3ee2f2a1","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-11-24 13:58:16.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"67d6ef5b73504fb1d9f695a5da1f0bf250c18eb9","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-11-24 14:01:57.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"fe6ab5e8b35d762f752629bc6b59558ef41802cc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-24 14:02:35.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"ec8e470028c59e371dc3ed8f3978ed3324a4f843","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-24 15:40:49.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a6887ab3682a4ff087c0f535ff2ce1c4\n\n- grenade https://zuul.opendev.org/t/openstack/build/950102b4ab104663978b2689d8bbe237 : SUCCESS in 1h 00m 36s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/78eb282aa54445f29e2ecb20002bb55b : SUCCESS in 43m 52s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9efe19646ae2441589e2f72e5cdd4543 : SUCCESS in 1h 33m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6887d97d540e4583a89777f665644d05 : SUCCESS in 42m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab2d70aa7ec9491e84cd6b928d297aac : SUCCESS in 5m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cbddef1e91724b2aa84e670117ae750f : SUCCESS in 4m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/04277e0f0da347b883a9320eaa785117 : SUCCESS in 7m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f47a85f9c2c48a79d8ca45494af8ddc : SUCCESS in 6m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/13e145a9cb664a75a90c033770bed690 : SUCCESS in 6m 15s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c1e089d3bb8344a88852a4d2609e314b : SUCCESS in 7m 27s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/bf2ad580052743deb8df12e62d7cd18d : SUCCESS in 9m 23s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/5d7b96e25ea048a0bcf3d052ce990f01 : SUCCESS in 1h 35m 52s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/833705c1c2d7452c8266802572c08acd : SUCCESS in 1h 05m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0acba0fd8ee04cc1badcf03d4dcd2782 : SUCCESS in 59m 50s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/4c8350ed70c44c82bce7932a5191cdd4 : SUCCESS in 1h 20m 31s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/452a74417d2446b1a8eec2e52a9d74d2 : SUCCESS in 23m 41s","accounts_in_message":[],"_revision_number":3},{"id":"218c6ed7d29685fc227d6b6a7d93d6901d2f15d8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-24 15:40:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"d97ed6fb26b623da8d4ad9e93aa88b12d325c7ca","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-24 15:41:50.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/73dad1ced9504b20acc602eff57e25d4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b11fbc90922414ab04da53f09a7864f : SUCCESS in 48s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1e4ea880916b435690429ec1c60489b5 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"613cc42a390d06ed0d4107246cc79722bda107a0","revisions":{"1d6810df03c07c7ed0b8bb2c640aca9a329b0f08":{"kind":"REWORK","_number":1,"created":"2025-10-21 23:52:18.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/08/964508/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/08/964508/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/08/964508/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/08/964508/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/08/964508/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/08/964508/1"}}},"commit":{"parents":[{"commit":"1a89edc86b10cbe14e1877d119de409b7b75863f","subject":"Merge \"Migrate setup configuration to pyproject.toml\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1a89edc86b10cbe14e1877d119de409b7b75863f"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2025-10-21 23:41:51.000000000","tz":120},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2025-10-21 23:50:25.000000000","tz":120},"subject":"Inject os_glance.* section in the property protection file","message":"Inject os_glance.* section in the property protection file\n\nUsers who use a property protection file and use either the import\nworkflow or the metadata injection plugin will get a\nglance.common.exception.ReservedProperty exception because we try to add\nos_glance_* properties to their images.\n\nTo work around this issue, users could add the following lines to their\nproperty protection file:\n\n[os_glance.*]\ncreate \u003d @\nread \u003d @\nupdate \u003d @\ndelete \u003d @\n\nHowever, it is not something users should have to do. This patch\nautomatically injects this section into the config we read from the\nproperty protection file.\n\nChange-Id: Iea2fffa75b973b5a34da6fc4145841888ea19034\nCloses-Bug: #2085321\nSigned-off-by: Cyril Roelandt \u003ccyril@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1d6810df03c07c7ed0b8bb2c640aca9a329b0f08"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1d6810df03c07c7ed0b8bb2c640aca9a329b0f08"}]},"branch":"refs/heads/master"},"6ce7e09c38c3bcb2c1245289f36e1748952f02e4":{"kind":"REWORK","_number":2,"created":"2025-11-05 19:07:36.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/08/964508/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/08/964508/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/08/964508/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/08/964508/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/08/964508/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/08/964508/2"}}},"commit":{"parents":[{"commit":"ffd134553f24fc4c900f8f36ed033eb41f8cd805","subject":"Merge \"Lazy update S3 URL on credential rotation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ffd134553f24fc4c900f8f36ed033eb41f8cd805"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2025-10-21 23:41:51.000000000","tz":120},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2025-11-05 19:06:29.000000000","tz":60},"subject":"Inject os_glance.* section in the property protection file","message":"Inject os_glance.* section in the property protection file\n\nUsers who use a property protection file and use either the import\nworkflow or the metadata injection plugin will get a\nglance.common.exception.ReservedProperty exception because we try to add\nos_glance_* properties to their images.\n\nTo work around this issue, users could add the following lines to their\nproperty protection file:\n\n[os_glance.*]\ncreate \u003d @\nread \u003d @\nupdate \u003d @\ndelete \u003d @\n\nHowever, it is not something users should have to do. This patch\nautomatically injects this section into the config we read from the\nproperty protection file.\n\nChange-Id: Iea2fffa75b973b5a34da6fc4145841888ea19034\nCloses-Bug: #2085321\nSigned-off-by: Cyril Roelandt \u003ccyril@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6ce7e09c38c3bcb2c1245289f36e1748952f02e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6ce7e09c38c3bcb2c1245289f36e1748952f02e4"}]},"branch":"refs/heads/master"},"613cc42a390d06ed0d4107246cc79722bda107a0":{"kind":"REWORK","_number":3,"created":"2025-11-17 12:05:58.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/08/964508/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/08/964508/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/08/964508/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/08/964508/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/08/964508/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/08/964508/3"}}},"commit":{"parents":[{"commit":"ffd134553f24fc4c900f8f36ed033eb41f8cd805","subject":"Merge \"Lazy update S3 URL on credential rotation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ffd134553f24fc4c900f8f36ed033eb41f8cd805"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2025-10-21 23:41:51.000000000","tz":120},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2025-11-17 11:53:35.000000000","tz":60},"subject":"Inject os_glance.* section in the property protection file","message":"Inject os_glance.* section in the property protection file\n\nUsers who use a property protection file and use either the import\nworkflow or the metadata injection plugin will get a\nglance.common.exception.ReservedProperty exception because we try to add\nos_glance_* properties to their images.\n\nTo work around this issue, users could add the following lines to their\nproperty protection file:\n\n[os_glance.*]\ncreate \u003d @\nread \u003d @\nupdate \u003d @\ndelete \u003d @\n\nHowever, it is not something users should have to do. This patch\nautomatically injects this section into the config we read from the\nproperty protection file.\n\nChange-Id: Iea2fffa75b973b5a34da6fc4145841888ea19034\nCloses-Bug: #2085321\nSigned-off-by: Cyril Roelandt \u003ccyril@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/613cc42a390d06ed0d4107246cc79722bda107a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/613cc42a390d06ed0d4107246cc79722bda107a0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
