)]}'
{"id":"openstack%2Fglance~964796","triplet_id":"openstack%2Fglance~master~I0feeddbf32ce9870ed7f4e515c03d3ae5b82139d","project":"openstack/glance","branch":"master","topic":"redundant-downloads","attention_set":{"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2025-10-24 18:00:56.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I0feeddbf32ce9870ed7f4e515c03d3ae5b82139d","subject":"WIP: Avoid multiple downloads to cache","status":"NEW","created":"2025-10-24 15:37:56.000000000","updated":"2026-04-21 13:53:46.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b68d5232e599d6f1347b430fe8e765c07e0e05a9","_number":964796,"virtual_id_number":964796,"owner":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-10-24 18:00:56.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-24 17:54:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-10-24 18:00:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"144edeb67a267d7fe9f68358d98cc8e8b9f282d0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2025-10-24 15:37:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6e74f913160147c5cc7d653aaaf57d060db46b6f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-24 17:54:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aa240175bc2a41c1be96a219087da9c1\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/23e585c27e75404690b5b4f0f43df43a : FAILURE in 7m 31s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e4c455b0b9e347d49b453f4305591c1c : FAILURE in 11m 49s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/13b5dd8e0e3945a4a639f6860715b873 : FAILURE in 8m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ec26ebe86f6b5f33966f7f8ca384d433fe068a06","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-24 18:00:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c644dbd6621a476da982cd70c3c8b8f3\n\n- grenade https://zuul.opendev.org/t/openstack/build/e5fbbb6fd1f24b449b62328befec8130 : FAILURE in 1h 09m 22s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6d5162998fda4ad197de187e30700929 : FAILURE in 51m 36s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6e9ced66fd9346018525a8b4c963366d : FAILURE in 1h 24m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5233aa682de04e1a9ae543e2361e6d95 : FAILURE in 53m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85f8c11899cc48e8bf4f105baab5c331 : SUCCESS in 5m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3c2e8cfbd7d048e2b5c4e7e79f8ffb05 : FAILURE in 5m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/393456cea4dc4dddb74f90b0ebf587a5 : FAILURE in 8m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9dc4fac196604f809df9afe0ac558b1c : SUCCESS in 5m 35s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/388d1bc5132d4424b12142d56fd13ab9 : FAILURE in 7m 13s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/ad677c2d29ee4e4483f98bad137eaa98 : FAILURE in 9m 42s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/e834d90794604ac5bc60e2ea3f3f8e24 : FAILURE in 1h 03m 08s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/b4bfc96ca4624f868951c9851e6267fe : FAILURE in 55m 36s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/a08155e941694b32a6208dbfa4861192 : FAILURE in 1h 24m 17s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/118a762ff76a474f86d9470ad5665a0a : FAILURE in 1h 40m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a62fdbb242dd481b9f6623c3c6c27d2d : FAILURE in 49m 46s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/1835064aa951453b8feecda82e39e6fc : FAILURE in 1h 41m 54s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/604f459ddd094b63afb8d9417a9b54c1 : FAILURE in 13m 22s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/403827e0dc8f4561a4e3ee896d86c502 : FAILURE in 2h 02m 52s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/6e7562fbc5f74fcb9774bbf8a293af9c : FAILURE in 26m 55s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/d419e05b798040e0b65a24fab704cbc4 : SUCCESS in 56m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b68d5232e599d6f1347b430fe8e765c07e0e05a9","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-04-21 13:53:46.000000000","message":"Topic set to redundant-downloads","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2b77f65e5768af831e74b82b55e12336e959d4b4","revisions":{"2b77f65e5768af831e74b82b55e12336e959d4b4":{"kind":"REWORK","_number":1,"created":"2025-10-24 15:37:56.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/96/964796/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/96/964796/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/96/964796/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/96/964796/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/96/964796/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/96/964796/1"}}},"commit":{"parents":[{"commit":"ab6114611fb0e5234d21ddc25fec8ee8902a13aa","subject":"Merge \"Replace remaining usage of healthcheck middleware filter\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ab6114611fb0e5234d21ddc25fec8ee8902a13aa"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2025-10-24 15:36:05.000000000","tz":0},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2025-10-24 15:36:53.000000000","tz":0},"subject":"WIP: Avoid multiple downloads to cache","message":"WIP: Avoid multiple downloads to cache\n\nSigned-off-by: Rajat Dhasmana \u003crajatdhasmana@gmail.com\u003e\nChange-Id: I0feeddbf32ce9870ed7f4e515c03d3ae5b82139d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2b77f65e5768af831e74b82b55e12336e959d4b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2b77f65e5768af831e74b82b55e12336e959d4b4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
