)]}'
{"id":"openstack%2Fglance~965056","triplet_id":"openstack%2Fglance~stable%2F2025.2~Ib281d01ecf5187fc1a02917cc1016c1572afb071","project":"openstack/glance","branch":"stable/2025.2","attention_set":{},"removed_from_attention_set":{"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2025-11-11 13:56:21.000000000","reason":"\u003cGERRIT_ACCOUNT_9303\u003e replied on the change","reason_account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2025-11-11 19:29:31.000000000","reason":"Change was submitted"},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2025-11-06 18:49:19.000000000","reason":"\u003cGERRIT_ACCOUNT_8122\u003e replied on the change","reason_account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}}},"hashtags":[],"change_id":"Ib281d01ecf5187fc1a02917cc1016c1572afb071","subject":"Lazy update S3 URL on credential rotation","status":"MERGED","created":"2025-10-28 05:51:40.000000000","updated":"2025-11-11 19:30:36.000000000","submitted":"2025-11-11 19:29:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"965056","cherry_pick_of_change":963875,"cherry_pick_of_patch_set":9,"meta_rev_id":"c67205d448dea5c8327b553e3b9aac482ef39cc2","_number":965056,"virtual_id_number":965056,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-11 19:29:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2025-11-06 18:49:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2025-11-06 18:49:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2025-10-28 05:52:11.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-28 05:52:11.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2025-10-28 05:52:11.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2025-10-28 08:01:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6d6e7c3a96f3b63741a9b855e6027150b2ece1c0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-10-28 05:51:40.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"f5a6e3cdff6057d3fd6b9c317aaac1787567a3dd","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-10-28 05:52:11.000000000","message":"Patch Set 1: Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"3776c832cc8387e18cc0fd57f65c26af805e66c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-28 08:01:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5405dfdfe11643c6a5a347fa94682ab3\n\n- grenade https://zuul.opendev.org/t/openstack/build/9f58f8ccc6394855987a0fd35ee12f47 : SUCCESS in 1h 02m 13s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d281a2b32f814222a2c643d93cf0952e : FAILURE in 19m 22s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4b187c4f666846c2b818635df20bbabc : SUCCESS in 1h 15m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7d43e06257ce4728a01cd82b7409a686 : SUCCESS in 32m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/448c707a4b194cdaad6f1986fef53cf1 : SUCCESS in 4m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6e014470f5aa4baf8b1457a92d45c91e : SUCCESS in 4m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/67c370d050284439bb2df6fb23af53da : SUCCESS in 5m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fe75399aa45e4bf0bdd8fa1fa4fc6073 : SUCCESS in 4m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/937ce9de180849fd99c4aa77811ff907 : SUCCESS in 5m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5b48236a475b450da18b50b9b2f1195f : SUCCESS in 6m 47s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1d39bfa0112c4291a92a7a1abef6b912 : SUCCESS in 6m 01s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/4e600a5f88074678b0344f0fc9c9266c : SUCCESS in 9m 09s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/26100eb008fd401999c17e06158a84d8 : SUCCESS in 1h 46m 02s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/321874224a72416ea6833cf86ecfea4e : SUCCESS in 2h 02m 30s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/14012a0bb3c045d5bf1d122560d16323 : SUCCESS in 1h 52m 00s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/7d2a1961076147f9969f65a672d0fa69 : SUCCESS in 2h 08m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/af918d2aeb1a41c19a74c60cbafd05d9 : SUCCESS in 29m 52s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/699f17c893f4438496f55b45f91923a3 : SUCCESS in 1h 45m 17s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/00c2a00cbfbc43e28c9c24bb4cf73b50 : FAILURE in 13m 22s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/31ec50ec7b544833b8393b30c598a118 : SUCCESS in 1h 56m 24s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/2b552e45c941403e8a106d13ced68219 : SUCCESS in 35m 18s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/a1cf8aedcb7045a8b38b7b82667fb792 : SUCCESS in 1h 09m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f4a490cd7a6b4cc008f90306a3c39d24a80d2e67","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-10-28 08:08:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9d65d5791ec08b49db62a8bc7f628d7c579a1de1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-28 10:17:18.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/dd92e45a324947309a0cb87a8dfef8ad\n\n- grenade https://zuul.opendev.org/t/openstack/build/25d444f4015d40d3ac37f4edf1b39748 : SUCCESS in 29m 25s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/bb7764c5f99341f8925ec1914c618bfc : FAILURE in 11m 34s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7cd59e66cdcc471b9b79af005aac62b0 : SUCCESS in 1h 34m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e64a7889b38645a08417f431f6c48957 : SUCCESS in 1h 04m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5fbd918453904eebbd87810ee1f93677 : SUCCESS in 6m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f0e893f7b1b2430aa3c0d3dc7436b18e : SUCCESS in 6m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d5bc6f493370419a978855b636893fd2 : SUCCESS in 3m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/91815bcc255444dbac19ccee099936a9 : SUCCESS in 7m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8b4d1a52f1f4f09b1544f5783a6044f : SUCCESS in 5m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/636f01dda8b44a9fa9dca1a8a99ed07f : SUCCESS in 5m 51s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a048353442474f3586aef50a60d07e3f : SUCCESS in 6m 24s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/b31cd38b52bb43a88e9e358800901699 : SUCCESS in 10m 00s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/55828f9e255841b291992a64dfafeca9 : SUCCESS in 1h 28m 49s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/ee8182dd757c4f358eec635cf7ca491d : SUCCESS in 1h 57m 23s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/496e3d261a5b4da1aa1741640b0b8a9a : SUCCESS in 1h 11m 02s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/7b7a426005284f62972137f52efa6079 : SUCCESS in 2h 03m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/76532e534f8e484ead00fc8795b48ee8 : SUCCESS in 28m 13s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/83f474bf991545568c2b333aeadcf0f8 : SUCCESS in 1h 57m 36s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/39c15812c9184e79aaa92dc8af7c81c2 : FAILURE in 27m 45s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/d318ac85f9d44148b2a4b1a76f4678e9 : SUCCESS in 2h 03m 33s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/cd595098a94f404da1b9f598583e59d8 : SUCCESS in 33m 24s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/c1109cf883c34ad28c79b36190bc1204 : SUCCESS in 1h 06m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"efa585b58fc1ed85e740e2abb8756313eff7180a","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-11-05 16:22:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fe5c3810c95874941db47f6c1b0c1e481c503c29","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-05 18:53:39.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/980a9c1c97d649cf8ae368ebe12f4fb7\n\n- grenade https://zuul.opendev.org/t/openstack/build/74267fc0e14648c7962551a2efda740c : SUCCESS in 58m 48s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f8949427f6d64d79949dbe3d63957b28 : FAILURE in 45m 20s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/82649def355d40eb9a7903bc3e80a2d9 : SUCCESS in 1h 48m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1bb19487b53c408db0e9ede59ac54e7e : SUCCESS in 58m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f7aa68f1c174d12919067cecb19b7b7 : SUCCESS in 4m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/917c9bbf63df403eae48ed8dd01d995b : SUCCESS in 5m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5e3bc03574c34610b22ce6f41ed6e886 : SUCCESS in 5m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3fac27ff89ce4fd5a95edb5477b28fc8 : SUCCESS in 8m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2dec804c6f1a4d4087194418a398210b : SUCCESS in 7m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a66412e416624b6c9e79c33ca4fe1cd5 : SUCCESS in 7m 39s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/732bdd4b3d9a493c8f10fb56f4be92e9 : SUCCESS in 7m 30s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/bc0f855c78f44554a939c5ca4f4b07e1 : SUCCESS in 12m 09s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/317e34bea5564fb3ab0de09e17025a69 : SUCCESS in 2h 03m 50s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/297e4c94ecdf465993fd74a591d0f7d0 : SUCCESS in 1h 08m 15s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/ee040b3ae33c4b77a82d1b7e02d92746 : SUCCESS in 2h 10m 24s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/843ee8bc61f0421eb25dc5536d633b00 : SUCCESS in 1h 45m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c8b1f6e76bb344079540c6cd807b9f3e : SUCCESS in 55m 47s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/98967bae1df94bb0874879adfef622cf : SUCCESS in 1h 54m 06s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/38df13d182684629b20d33f05b2bb989 : FAILURE in 30m 57s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/735b69491de1486c8347490060736b0e : FAILURE in 27m 28s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c11411f53c6f4b12bc439c41635a2947 : SUCCESS in 37m 39s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/16deb09257614c47b2f6397851c8c04c : SUCCESS in 1h 09m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3565ef2b514b5f309d13b375ee7be53d8ceb9baa","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-11-06 05:38:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"950905bdf300605c54a5a90a006bd86bfdd6bc46","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-06 08:27:00.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/65565c3593324584b8e694260210a9c7\n\n- grenade https://zuul.opendev.org/t/openstack/build/f728ec280d8a42c3b5146396d69551ef : SUCCESS in 51m 05s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e73f4632d07b44768e9fe3b17d068f8a : SUCCESS in 59m 26s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ff3e024261ca4d6ca4e84c27c42cf031 : SUCCESS in 1h 35m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5570a25b7e4a40bdb8d248d5f5e22d14 : SUCCESS in 1h 01m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e4baae11f1d4cb3a2d1a67e7b62fbf3 : SUCCESS in 5m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/99455e8a1e1540a49d81f552a7170bd2 : SUCCESS in 4m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3348a06e1b89489db52a18aa8dd15ede : SUCCESS in 5m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/492b895c21e940128622a8784f3d69e8 : SUCCESS in 8m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/289aa08e6b884574bd7d3579ec1e754d : SUCCESS in 6m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d473d45b52be499dbe42e58891e9bebb : SUCCESS in 7m 00s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0aae0a83a6c2459d9c1fa4ba6c33f27f : SUCCESS in 7m 17s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/74e19bf429d843cbb7d0421bc3de8cc8 : SUCCESS in 10m 30s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/efb65d1a7e6e4aa19f70dfd1bc05f26c : SUCCESS in 1h 53m 00s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/2776cd18be234532bc45e12df7711504 : SUCCESS in 1h 42m 17s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/570f435e351c4cdc88a0304c64f551be : FAILURE in 1h 55m 47s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/a8936d7a6333409aa011fc63bc5a0332 : SUCCESS in 2h 33m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/df0cb7ad80ab4d75a71ecec2fafe7ccc : SUCCESS in 59m 28s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/4a36e42460b941349dfe9ac0097e85f9 : SUCCESS in 1h 48m 01s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/e9d58eff51b54d1aa8f61ffd43895634 : FAILURE in 26m 50s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/93243ccf40284fe7973a273257c52273 : FAILURE in 12m 52s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c09cc2d803834f979f9ea6bde031524b : SUCCESS in 22m 51s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/4f62e2545a5c4ea2af3e7ac2c9e85b76 : SUCCESS in 58m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"abff57f83721e2835add9743564c6df63c247cc4","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-11-06 08:28:12.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a6f55bd566115be05c11ec9f6ded7aa2d707be3a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-06 10:32:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/15b1630018424f4091ca315e07310d45\n\n- grenade https://zuul.opendev.org/t/openstack/build/b29fabae0994459f84622ae38c04bc1d : SUCCESS in 37m 25s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e78e865b4330432aa024b75e1be405fc : SUCCESS in 56m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7a71e8b49a6345ffbe4ad84cec7471c0 : SUCCESS in 1h 31m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dbd5f8c05b074c94869059b46ed1c597 : SUCCESS in 1h 00m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1feca0f2e3024572b6fc2c872b9a9c49 : SUCCESS in 5m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dba7188b28a34c69999f259ca77fd8e5 : SUCCESS in 5m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4daffb7c295e4893a5654806e70f2529 : SUCCESS in 5m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/584cd510487c4f3b9e9c36450487e3ed : SUCCESS in 8m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/675af89a201a46fd875344ecc837566a : SUCCESS in 6m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2d86be56710f4d0ba14d8244efa074a4 : SUCCESS in 6m 01s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8c546edc1a4d48608d4ab87e0ed04d43 : SUCCESS in 6m 27s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/9c4d0acc54a94a23b2debfc0ed21bc03 : SUCCESS in 9m 38s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/90b9efe7a1d449f8b9b88e9826a8253b : SUCCESS in 1h 59m 24s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/16835ca9c70e4a26bae81ef4b775a875 : SUCCESS in 1h 28m 11s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/ad8cdb752a4b48708cc0e06924bd361e : SUCCESS in 1h 46m 05s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/57f8e44fc26e403789c39a5114355263 : SUCCESS in 1h 24m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6345d074dc4e43e2bfd581a99c8fd53b : SUCCESS in 57m 20s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/1bf213a943b34ff8892f620d950d0343 : SUCCESS in 1h 46m 33s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/8a8f1bcb2caf4eca8780ac950a0a0f73 : FAILURE in 28m 02s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/4c81bbb98258496fbc2df4e65bca80cb : FAILURE in 27m 56s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/d4a052bc73da4ba6993328dce34c16c4 : SUCCESS in 22m 11s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/867ec955ae3a45599bf281f22f70a011 : SUCCESS in 31m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"02dbd0edfc26d528fd47374635744ddadb182346","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-11-06 18:49:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6dc347689bb085daa50d001225b8a8b845bde202","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-06 18:49:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ce3755ebce8a754286cb5fb75c0bf2a20761a675","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-06 21:01:54.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f72070c13aad4a21ba6b956624dfbd0a\n\n- grenade https://zuul.opendev.org/t/openstack/build/633c0ff33f4f4f30b17e18362560676c : SUCCESS in 37m 42s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b608c53a1c0e4f918a0d6a323f0c58d1 : SUCCESS in 1h 07m 31s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/75257dc3b6354f23a2c18f3ad0f71d97 : SUCCESS in 1h 34m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/61fc011e606e47bf902439e58b3844f1 : FAILURE in 59m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29f7525fd6d44105872408fc8a5d9c1c : SUCCESS in 5m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8281bfa0a0474852aab36020bb07f4fc : SUCCESS in 5m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e6c0a9fa3a004b6dba3078dff8b7a02a : SUCCESS in 4m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/221f07501c6748729804a0a0cc42db50 : SUCCESS in 5m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8d69cc63c6c64807a6e52b82859bfc5a : SUCCESS in 7m 44s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0be19234c4074b7aafbf122b656d9885 : SUCCESS in 6m 41s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/d2207597c89e4cd584eae6bb6b38f11d : SUCCESS in 9m 37s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/5113ee8b88194bcb871dcee1a6d1f9fd : SUCCESS in 1h 19m 58s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/e931cacc62984a8eb54cd9e0f4e77212 : SUCCESS in 2h 08m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/11b14d7dcb3d4f7abc9b2a7d4fc9b70d : SUCCESS in 1h 00m 05s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/cfffae14be694291a65165e4d6858207 : SUCCESS in 2h 07m 29s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/dac9c5c7af614d9b917804ddb3502263 : SUCCESS in 36m 37s","accounts_in_message":[],"_revision_number":1},{"id":"2200e67ad327c5e8019b580109f8b072d5f41376","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-11-07 06:35:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"12265c2180edaeab19c0765d2b1f8a82e422aa3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-07 08:28:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/02f379f8314b498b9833cb1eaf7cfceb\n\n- grenade https://zuul.opendev.org/t/openstack/build/8c9be75f53fa4499bdd24282d064c9bf : SUCCESS in 59m 19s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9773f042ff884fd49ccd5f167296a7f1 : SUCCESS in 46m 36s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/387b7885a0034c2ea5de079934963152 : SUCCESS in 1h 31m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/67fd30d471e74d9a814ca3fe6d6bce16 : SUCCESS in 1h 00m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d07505a971d49f095edeaff37dcc086 : SUCCESS in 4m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2ba69863d38442d8bef1d0588a8e6a2b : SUCCESS in 4m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a3befcd0663a4a0badfa46684f4b2891 : SUCCESS in 6m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9bc7eaa25194422a9f9f2e7e22c6f83a : SUCCESS in 7m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07d6124db40b4a6bb03600fa1a4ecd2d : SUCCESS in 4m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/11f84557851f4253a2bf138205a12a81 : SUCCESS in 6m 58s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7f017188be0e47dd84aca883597a858d : SUCCESS in 7m 50s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/264a38a805b04d13a03f5a0e78c8c5d7 : SUCCESS in 11m 38s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/f5664cf4106f440a9ec5f2592e7aa36b : SUCCESS in 1h 20m 16s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/fb569687d71544e9b2ec82e819f5f89b : SUCCESS in 1h 46m 33s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/0b126001543a4c579a1f8a280c78e9a8 : SUCCESS in 1h 37m 15s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/282f58661a4c4cda8c2b9b9af3ad5841 : SUCCESS in 1h 41m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8134cbb60ba0415f967a1b8edc371bce : SUCCESS in 54m 35s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/506d573fe0a44fbcb597a8a764c5b44b : SUCCESS in 1h 43m 34s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/e424f822d4464d3086841d78d0bf0238 : FAILURE in 27m 51s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/d3e6364772dc47a5863f686d855a69a8 : FAILURE in 17m 19s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/06a5b77a1595431982ed2df678acb547 : SUCCESS in 31m 05s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/9d64fb9df8524fafa900f3b3ca7a14ec : SUCCESS in 1h 01m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4aa9f058ec4680f8745f2bb25949f319742349db","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-07 08:28:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2236e496ddd575f54b366ca463ac305949340d4f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-07 10:19:28.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b42534ab59fe49a199c5099cd07acde9\n\n- grenade https://zuul.opendev.org/t/openstack/build/289eac8067d14ac1ba9816b38ffec1e7 : SUCCESS in 58m 23s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/560a94faede345efa602e6752e81a60f : SUCCESS in 1h 02m 44s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3d396c4145ce421c989f8412a26094ee : SUCCESS in 1h 37m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c28510cff762466d855c412589980470 : SUCCESS in 44m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f93c51b5f76a470081ffb2054300a1f4 : SUCCESS in 5m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dfc07c3685044a0b886785ef6b588ad0 : SUCCESS in 4m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d69685ca00264accbe285ef0a14f5bb2 : SUCCESS in 5m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3b95c20df0e4581b8ee5cd100504d38 : SUCCESS in 5m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1b052f1b56bd46278b6b825566b4fd7b : SUCCESS in 5m 45s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c95dd8b5cb7c4daa833d9ea57f7d9c52 : SUCCESS in 6m 52s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/093446d880c64996a378cbf7f7efcbcd : SUCCESS in 9m 57s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/b82d7f74a4bd44c2aeb43ca8e78c67c6 : SUCCESS in 1h 45m 19s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/62eb9e4b923b4d08b6e56f7dd557aae4 : SUCCESS in 1h 45m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/19e697c7a66c428abf21c0a89e006ad3 : SUCCESS in 1h 03m 14s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/8fe30029bcf8428c9d22d6cccca703df : FAILURE in 1h 12m 44s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/97ad1e73095346c2aea551f53adf9cfb : SUCCESS in 25m 02s","accounts_in_message":[],"_revision_number":1},{"id":"8f9742d5645eebf87d69ef1db93f9687bd4b92a5","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-11-11 13:56:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3a9fcef145430a19e5153fe3d56fc7daed5d192e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-11 16:23:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d4f007a614334fc9ba9c06c2327a396b\n\n- grenade https://zuul.opendev.org/t/openstack/build/2455129e9ac949088ccaf0ef2f8fed8f : SUCCESS in 32m 09s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b037918e799e4723b01cff0e900cddac : SUCCESS in 1h 08m 16s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3664c0a0d54a4227aef4a2e559026631 : SUCCESS in 1h 21m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8e2e106072744f5f8e58fdd6586d2622 : SUCCESS in 1h 01m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6805980130d545b284ea0f3fd94b057a : SUCCESS in 3m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1f914c0d2aad4439b27d5220091bc149 : SUCCESS in 5m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cdbf61e61e464c12a24550617ed9638e : SUCCESS in 5m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6fb0569bcced4e30a1266484017b942c : SUCCESS in 7m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/acc5a77400e1435a8b7bd2a49734a2eb : SUCCESS in 4m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e662880d7bc04b8291bdd711f0315c43 : SUCCESS in 7m 22s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/cf04b218f82c459e83ef1c9e36b92f79 : SUCCESS in 7m 30s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/aad11f4f88744635bd3ffab0d7a1eb82 : SUCCESS in 11m 18s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/c16162f2f6e64563b5ae79db37bc9ec0 : SUCCESS in 1h 49m 55s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/14a6b7762b8f454cb9d6aea2a2dec69a : SUCCESS in 2h 06m 04s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/d88bef7e2ffe42a4b513044a8eda13e0 : SUCCESS in 1h 25m 25s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/452ee2f8381f4c04ac6fc2e143f3c5c8 : SUCCESS in 2h 21m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/da255acc1c8745e68cabc98ae82c0c2c : SUCCESS in 49m 38s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/37bda604e97b48d9a8f079e3e4e71f4e : SUCCESS in 2h 00m 52s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/f07483a81de2454295c6a2852287776a : FAILURE in 28m 19s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/cdecdffa75b04b36be0ec739b846436d : FAILURE in 24m 31s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/38c5c3a154b742b8b6dd9f2cdbf5c840 : SUCCESS in 32m 02s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/2cd5702d78c540399b4fce685e7eea3d : SUCCESS in 32m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"63de342933112de4c5676462723bc718b6523655","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-11 16:24:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c497ef3835387aeb9b2bfa366f9002e30aa304b3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-11 19:29:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/07d0704054ab4bdd98c27fb7dbf15042\n\n- grenade https://zuul.opendev.org/t/openstack/build/a13c8fb7bba4446e875a1007fbd84ae1 : SUCCESS in 33m 12s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f8be287b9c8d4c06a214d3fb7c96bb7d : SUCCESS in 45m 20s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7790604376514de6a970a34c51809236 : SUCCESS in 1h 22m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9c640259f0ea4b7ba683148d045abfb0 : SUCCESS in 1h 02m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45b733215df24b059bbfabb5250ffbf9 : SUCCESS in 3m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b2b37217a9464345a7d64f5845c04de3 : SUCCESS in 4m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2c616316a6cf4e749651669d47630cd6 : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41586566961246eb9f00c40079e395d7 : SUCCESS in 7m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4d43521e53f047f4bee127b3fc11ca3a : SUCCESS in 7m 13s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/10b51f484dff4eba947ef7c2046faf7e : SUCCESS in 7m 24s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/f0e5b01b5649418181428ca2bfab527f : SUCCESS in 10m 02s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/b6012a148dda45c5b37cfbff5ab74fa8 : SUCCESS in 1h 39m 08s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/2b5113d2b30d417181b2c38f0ba077d3 : SUCCESS in 1h 35m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b7b9830679094793ba5d834940541421 : SUCCESS in 1h 00m 49s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/8247dd7cc04a40c98dc4d699e991c7cd : SUCCESS in 1h 27m 17s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/4689a845621f4e37a231b88ef2c95b76 : SUCCESS in 23m 19s","accounts_in_message":[],"_revision_number":1},{"id":"3143a3be125ad7415bfcdd53c3315f48c92e1d3e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-11 19:29:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c67205d448dea5c8327b553e3b9aac482ef39cc2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-11 19:30:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b7fcf8d27b34759a4540fdf1e1e5b93\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31177bb4274a4505915f677b901d8d6c : SUCCESS in 50s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5af29ebb259541b49be894443750dd8e : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ac5505179f51cb88d83ec40113d5c6801ef397b3","revisions":{"ac5505179f51cb88d83ec40113d5c6801ef397b3":{"kind":"REWORK","_number":1,"created":"2025-10-28 05:51:40.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/56/965056/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/56/965056/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/56/965056/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/56/965056/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/56/965056/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/56/965056/1"}}},"commit":{"parents":[{"commit":"2ebeab5fca074cdd5ab194c1a0cdd079fd83d52a","subject":"Update TOX_CONSTRAINTS_FILE for stable/2025.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2ebeab5fca074cdd5ab194c1a0cdd079fd83d52a"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-10-14 18:37:35.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-10-28 05:51:40.000000000","tz":0},"subject":"Lazy update S3 URL on credential rotation","message":"Lazy update S3 URL on credential rotation\n\nEnables automatic updating of S3 credentials in image location URIs\nwhen store configurations change, ensuring continued access to stored\nimages.\n\nThis change covers below scenarios:\n- Upgrade from single store to multi-store, update store name in\n  location metadata\n- Single store credential rotation\n- Multistore credential roataion for all s3 stores\n\nCloses-Bug: #2127798\nChange-Id: Ib281d01ecf5187fc1a02917cc1016c1572afb071\nSigned-off-by: Abhishek Kekane \u003cakekane@redhat.com\u003e\n(cherry picked from commit 9bb7ffd4eda728fe74de2baf7767c5f655916ad4)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ac5505179f51cb88d83ec40113d5c6801ef397b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ac5505179f51cb88d83ec40113d5c6801ef397b3"}]},"branch":"refs/heads/stable/2025.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
