)]}'
{"id":"openstack%2Fglance~965172","triplet_id":"openstack%2Fglance~master~Ia3a1ceaff3964bf62e19ebf46a53b885750ac4a7","project":"openstack/glance","branch":"master","topic":"db-migration","attention_set":{},"removed_from_attention_set":{"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2025-11-11 17:42:57.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia3a1ceaff3964bf62e19ebf46a53b885750ac4a7","subject":"Change DB migration constant to 2026_1","status":"MERGED","created":"2025-10-28 19:01:01.000000000","updated":"2025-11-11 17:44:03.000000000","submitted":"2025-11-11 17:42:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"965172-db-migration","meta_rev_id":"05d79f52f1526faf715008995ef0d49516c1e42a","_number":965172,"virtual_id_number":965172,"owner":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-11 17:42:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2025-11-06 12:17:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2025-11-06 12:17:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-28 19:13:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-10-28 21:25:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-06 12:17:14.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"1a9487aae568866da24c822b12b4a7c900594c60","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-10-28 19:01:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a9d10db71a7e3923949923fe3060ebd70f982a25","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-28 19:13:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5955ded9ab9d4b0d8d9b61ed0b267024\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/aeb5f6c375c64932a4e78f47a5f15489 : SUCCESS in 9m 21s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/aa2e702bf7af448a985413602b8da028 : SUCCESS in 10m 14s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d5dc4cc21c2e40668b575a67bf33bc2b : SUCCESS in 10m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1d82be98d5af93fd76ffa244bec9083e2b5c53b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-28 21:25:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b867e66db6ad4f9a84bb41a5f25ffc2f\n\n- grenade https://zuul.opendev.org/t/openstack/build/9892b7f0b1484689bb06fda366a26d12 : SUCCESS in 44m 54s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1dd4bf2b0ece44aaa665450d98c5291e : SUCCESS in 59m 56s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/87273ff8d0c142cc92caea1490522362 : SUCCESS in 1h 40m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/034241834c654d9eae9838143d18dd41 : SUCCESS in 49m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c8bb689840b490b974571a08431444a : SUCCESS in 5m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/76f481412416471ea6c7d5a1fa996f9d : SUCCESS in 5m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/03dbae7449ab44eebfebdb9fbd159853 : SUCCESS in 4m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/863fc414858d47818b0c07337f44fac9 : SUCCESS in 8m 25s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0ad35f03946c4fe280b762191ca6d666 : SUCCESS in 7m 13s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/a99d90850f8b463caffb98a405b1ac5e : SUCCESS in 10m 17s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/a5e2d95905cd471d9325842e3e6c9bd0 : SUCCESS in 1h 03m 32s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/f6a951d269e147399076a3d89cc34b9c : SUCCESS in 1h 59m 14s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/d176734e1ba04e9ca65d2436651cfe83 : SUCCESS in 2h 10m 04s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/bab31733498142cda6eebf5dcd4c8216 : SUCCESS in 2h 21m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/674f3753c76f4b33967771f95c630761 : SUCCESS in 33m 54s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/cf6e823b676b4159bbf5acfd216821d6 : SUCCESS in 2h 00m 40s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/657c9d5feb7b41368d220efea2fb4129 : FAILURE in 22m 06s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/d3dd7b897ca24064a555119951e36a66 : SUCCESS in 1h 25m 15s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/84a714d20cab45e89277ecc8226943cc : SUCCESS in 31m 59s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/91960ca3f21942c98dfc8a820f671b37 : SUCCESS in 29m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6b9e9a3ec9a8f590137179898181d34e0ce722f0","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-11-06 12:17:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3d8960ff5c814e5a8398e2dab11c95df3a33a314","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-06 12:17:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"88ecc0fed6daa1c78284d55d17eae2de502550c8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-06 14:10:38.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fc3e5ce906ea43f6ba7954a7adedde4e\n\n- grenade https://zuul.opendev.org/t/openstack/build/10f1ea29aa6d4508929d4a88d244575a : SUCCESS in 34m 52s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/bf6dd9c00c1a499a8b776bcf5c954a9b : SUCCESS in 35m 37s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bea68896465f4071a83b4e077590d1e1 : SUCCESS in 1h 39m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b20b6409c24e49e0a0763eaf3a45fe76 : SUCCESS in 1h 00m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/872bad6dc388469a8d7f35d98e496ae7 : SUCCESS in 4m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b801b1b3e77643f1955838d119423c1d : SUCCESS in 4m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/73366e233d0b44a8864d83b2c76453b2 : SUCCESS in 6m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce638fa8a20644648dfba731830ae66e : SUCCESS in 5m 36s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/060e0772431f47dcbb22c7b3ea4eda1b : SUCCESS in 6m 27s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/69007a2eeb2043a3bb4898940905ea6c : SUCCESS in 9m 56s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/e67fa4802e4744fc94e148cdee719bc2 : SUCCESS in 1h 46m 02s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/e81fdae991224a38af26f205b21420df : FAILURE in 1h 48m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/31788a514cf34c33a04a14720682edb4 : SUCCESS in 1h 02m 03s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/499a46992b2f471ea3d505256935d72c : SUCCESS in 1h 44m 38s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/188a8ca11c21494ca7574cf6c89ceeac : SUCCESS in 21m 08s","accounts_in_message":[],"_revision_number":1},{"id":"e40053cb2660ca458b25768dbd310af0f9db3bcd","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-11-11 13:06:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4c08bf5a97b5083065b08d7a8272e18564854847","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-11 13:20:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/92a294cec9124875b49489dc66e4b071\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ee6b844e590a4cf68574326693b27000 : SUCCESS in 8m 24s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e84cf447aad44172874443f32fee97ba : SUCCESS in 8m 39s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/438f460cedce4eeca0f46e06595b9136 : SUCCESS in 12m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d6898297319a6d42105e456bf2a49c39df1d667c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-11 15:30:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8ac1b74d006747b198e485392c73cdf8\n\n- grenade https://zuul.opendev.org/t/openstack/build/76a5ce6774ab4a5db51eb032d9919775 : SUCCESS in 34m 00s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/44a7bdfb3be441b0b2e47647623718e0 : SUCCESS in 51m 14s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a8f7a875b6d649e19a73b7652354f184 : SUCCESS in 1h 16m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/379dddff893245ba8e47aad5ff8468a6 : SUCCESS in 1h 00m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3514bfb3285d4c2cb1f03ba7faa2ae1f : SUCCESS in 5m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2284a672b06d415489dbeacd27d256db : SUCCESS in 4m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ea5d0cc797e44dce988c8a3a445acc20 : SUCCESS in 11m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/acde3f43889748c1890992faf4c5a100 : SUCCESS in 5m 00s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e8f56dc51f464ca1a97f95cbd03ad88d : SUCCESS in 6m 53s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/04f52f4073304ed795e416d352e4a244 : SUCCESS in 8m 57s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/dad5b5b696464f86b94c9163bb254d27 : SUCCESS in 2h 19m 08s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/6f13dad5635143a8a1bb08b19a313573 : SUCCESS in 2h 13m 34s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/6d055354a1984a258c907fb9999a1653 : SUCCESS in 1h 44m 57s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/4da5a066ee56459ea0c437aa40fdb4ee : SUCCESS in 1h 51m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/59574b9f005847a880e1026c7b55323c : SUCCESS in 36m 43s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/cc39f4dad8db47f59c5f8e69a042cf57 : SUCCESS in 1h 51m 47s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/723d5f29f206416c93ea4f284af88685 : FAILURE in 24m 30s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/6aa50b5267c84eaebb957e6fbfda6813 : FAILURE in 24m 33s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/95311c3fdcae4671be9bed37717dbd82 : SUCCESS in 37m 10s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/946e8632ff244e06b5ecfc8be27f05a4 : SUCCESS in 1h 03m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a5553cddc2d644aecb0c49c57a4fa69ad1c0a9b3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-11 15:30:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f5464fdff1ec03572dac2537fc3f9ce07035ad00","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-11 17:42:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b33d6f5a357e495eadc640eee0ee26d8\n\n- grenade https://zuul.opendev.org/t/openstack/build/c405dad57cfd42588ddefd9ff50727ac : SUCCESS in 33m 30s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4ca9998e253b415399523778f10442f0 : SUCCESS in 1h 00m 10s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5b95632da117460295eea6422018ee64 : SUCCESS in 1h 45m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2af01f65a74c4f7a83115f6dd228a5fe : SUCCESS in 36m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5cbbd9398cb4eb58f5f8fcc2e6c0922 : SUCCESS in 6m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/73341c6d5e344f98990a75ec87e2c6f8 : SUCCESS in 4m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e1569adad69b4528b5e61a374e2bd694 : SUCCESS in 9m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a2997045a2b4ff0be1a608c1873e230 : SUCCESS in 5m 41s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6be037b52bf34572a0f81e2f20fe47f8 : SUCCESS in 6m 56s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/c28b157e39f04a4fa6222245c31e5686 : SUCCESS in 12m 52s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/219d8cf430a847438936ad912ab9c633 : SUCCESS in 1h 57m 44s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/b2f6332a53d643c782bb4149c70bcc0d : SUCCESS in 1h 54m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b1334c00af3c4a0ba296c78abf1c6cb1 : SUCCESS in 54m 45s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f1b9ba633826411495d310ffa8244076 : SUCCESS in 2h 06m 38s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/d35d96f8454b44da89fab79ad8d6719d : SUCCESS in 34m 28s","accounts_in_message":[],"_revision_number":1},{"id":"d5520015355bdd0758d35f06bc4fd3a2ab619290","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-11 17:42:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"05d79f52f1526faf715008995ef0d49516c1e42a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-11 17:44:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b5a4457ec1c14ceea88f1ff7287bcdd0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44569807ec224c7885ce413d57f35116 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ad612406e7eddfc34f5dbc94973772c252b24eb7","revisions":{"ad612406e7eddfc34f5dbc94973772c252b24eb7":{"kind":"REWORK","_number":1,"created":"2025-10-28 19:01:01.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/72/965172/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/72/965172/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/965172/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/965172/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/965172/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/72/965172/1"}}},"commit":{"parents":[{"commit":"ffd134553f24fc4c900f8f36ed033eb41f8cd805","subject":"Merge \"Lazy update S3 URL on credential rotation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ffd134553f24fc4c900f8f36ed033eb41f8cd805"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2025-10-28 18:49:23.000000000","tz":60},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2025-10-28 18:49:43.000000000","tz":60},"subject":"Change DB migration constant to 2026_1","message":"Change DB migration constant to 2026_1\n\nChange-Id: Ia3a1ceaff3964bf62e19ebf46a53b885750ac4a7\nSigned-off-by: Cyril Roelandt \u003ccyril@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ad612406e7eddfc34f5dbc94973772c252b24eb7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ad612406e7eddfc34f5dbc94973772c252b24eb7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
