)]}'
{"id":"openstack%2Fglance~967162","triplet_id":"openstack%2Fglance~master~I1e7fe0720581dfa815b5e0cd2b01379cab2a5f33","project":"openstack/glance","branch":"master","attention_set":{},"removed_from_attention_set":{"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2025-11-14 16:43:37.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1e7fe0720581dfa815b5e0cd2b01379cab2a5f33","subject":"s3: Do not log access keys","status":"MERGED","created":"2025-11-14 07:05:27.000000000","updated":"2025-11-14 16:44:27.000000000","submitted":"2025-11-14 16:43:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"967162","meta_rev_id":"9741a7bb6b64904de29e47cd3ee98ecc83d8d22b","_number":967162,"virtual_id_number":967162,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-14 16:43:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":2,"date":"2025-11-14 14:35:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-11-14 16:35:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":1,"date":"2025-11-14 14:35:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-14 07:21:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-11-14 09:30:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-14 14:35:57.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2025-11-14 16:35:51.000000000","updated_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"reviewer":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"state":"REVIEWER"}],"messages":[{"id":"a22244270bd494edd26598a33787d97912fc5b5a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-11-14 07:05:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1ffc6cd34b54970e0e03e17545baa9e440739fb6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-14 07:21:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/806cf53e46bc44e7b9ab2cef58d2b7b3\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9b9d65de6bc945f49cf9c289ae52e7f7 : SUCCESS in 9m 15s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/fc82337615354e5e8b9db0857df25380 : SUCCESS in 9m 57s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/124747b58ff44ddf8e25b4147d4e7a17 : SUCCESS in 13m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3c966b62f8033ee618a573a704ec519d9d2c29fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-14 09:30:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/10d11099c5164a2db87986defa588497\n\n- grenade https://zuul.opendev.org/t/openstack/build/b309f63c52e046af8b61f0a6926e83c3 : SUCCESS in 36m 39s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/38f86b12c84e43f1883b3a56695d6837 : SUCCESS in 33m 23s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/43a9d8c3500b43e2b908c3deb6ebcdaf : SUCCESS in 1h 16m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/aea9cbf158f448e6abcfbe569fa7a3d5 : SUCCESS in 39m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f97908d677b24f5f88f436b0fa69c394 : SUCCESS in 2m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/70e611a7521540f59a0936e49e204c80 : SUCCESS in 2m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9a8dc1263be244b5ac49ea2fcd3c92ef : SUCCESS in 12m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a9357923ab5420fa0cfd6c9d7bd9dd8 : SUCCESS in 5m 43s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e15b3ec9f8e1470c8dc8241a5d51e094 : SUCCESS in 5m 51s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/b414678af66544e9b5b242256ffeac32 : SUCCESS in 8m 35s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/35195cd1aa28468aa389d32faaa0e718 : SUCCESS in 1h 05m 24s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/4e335e3749a943e090ad9a76e3df4166 : SUCCESS in 1h 44m 57s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/c122a0d963f54cd1a38e701f99fb277a : SUCCESS in 2h 03m 36s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/e250ad5bb11348908d31aa79d2a71d16 : SUCCESS in 2h 20m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8107a575a1f543d292b7ffb59f2e1c63 : SUCCESS in 28m 27s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/239e9568ab52466e90e17aad9c73b307 : SUCCESS in 1h 40m 27s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/79ffc6451807487385550b2f896b9c53 : FAILURE in 11m 45s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/d5d4ab8109fb4ac3bbbeab75434dc2b4 : FAILURE in 28m 12s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/e8c3817aecfd40b1a626dc7afd94d767 : SUCCESS in 15m 55s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/35f90e64205a4b98a33fa580cd7b87fd : SUCCESS in 56m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"988cac16f4a661fe09208fd5a9671e54ff3a9837","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-11-14 14:35:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7dd87dbea17ece5bb18db6a7d4357af40eb4c0ea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-14 14:36:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"83391fda3600f77dfb56c881c29b0f6eb41ce5bd","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-11-14 16:35:51.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"766ef674905d2e83b28066f25dc7813f4e9d5b2e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-14 16:43:37.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/17aa6ee344fe4ac6bd31231366a8f87b\n\n- grenade https://zuul.opendev.org/t/openstack/build/99e074010e124058accd5a5934d91509 : SUCCESS in 30m 32s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d390ee6c0aeb407590d7eafb1f41fac8 : SUCCESS in 45m 35s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/44e50b61bde840f1bd3d6fd0eae579c8 : SUCCESS in 1h 37m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/09af83efb7bd491c8b6b30b5f4c3e10d : SUCCESS in 51m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0197ff368f54485e99b1c82ae62475f4 : SUCCESS in 3m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/56080a66ec714204bf8dab9e114d2811 : SUCCESS in 3m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6977a500a5fe4d98953cea40c60d416e : SUCCESS in 8m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a2fd492fe37420b851cb06b72d52dd5 : SUCCESS in 3m 49s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6043c5a73aff4f0aa1d1aa3be1ff1c52 : SUCCESS in 6m 23s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/d1594f18c64f4b98b833d8935063003b : SUCCESS in 6m 22s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/d2be9b780c9149118e52cd0d228db6ed : SUCCESS in 1h 56m 14s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/691361bac357472cae591894af715043 : SUCCESS in 1h 09m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/10ce9e0da6934a9fba27dc3192cc4807 : SUCCESS in 33m 23s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b2c847ec8fca4b2e94882299148c8495 : SUCCESS in 2h 02m 54s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/e3d75addf2c84a5697ef09f2d359b9b0 : SUCCESS in 32m 39s","accounts_in_message":[],"_revision_number":1},{"id":"bd3825ce0f0b37053b7156e429c8bda8acc9e971","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-14 16:43:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"9741a7bb6b64904de29e47cd3ee98ecc83d8d22b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-14 16:44:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1fc3cc86f4c74b61b707ccef46432eb0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b25a317f77e24510b029fa98e426b19f : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"81a838cc24da73e86e2cc85e06b21265fb960875","revisions":{"81a838cc24da73e86e2cc85e06b21265fb960875":{"kind":"REWORK","_number":1,"created":"2025-11-14 07:05:27.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/62/967162/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/62/967162/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/62/967162/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/62/967162/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/62/967162/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/62/967162/1"}}},"commit":{"parents":[{"commit":"ffd134553f24fc4c900f8f36ed033eb41f8cd805","subject":"Merge \"Lazy update S3 URL on credential rotation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ffd134553f24fc4c900f8f36ed033eb41f8cd805"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-11-11 17:10:52.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-11-14 07:01:57.000000000","tz":0},"subject":"s3: Do not log access keys","message":"s3: Do not log access keys\n\nRecent credential change for s3 has introduced one debug log\nwhich can leak access information of s3 store. This change avoids\nlogging URI in the log.\n\nCloses-Bug: #2131085\n\nChange-Id: I1e7fe0720581dfa815b5e0cd2b01379cab2a5f33\nSigned-off-by: Abhishek Kekane \u003cakekane@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/81a838cc24da73e86e2cc85e06b21265fb960875"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/81a838cc24da73e86e2cc85e06b21265fb960875"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
