)]}'
{"id":"openstack%2Fglance~967278","triplet_id":"openstack%2Fglance~stable%2F2025.2~I1e7fe0720581dfa815b5e0cd2b01379cab2a5f33","project":"openstack/glance","branch":"stable/2025.2","attention_set":{},"removed_from_attention_set":{"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2025-11-21 10:49:51.000000000","reason":"\u003cGERRIT_ACCOUNT_9303\u003e replied on the change","reason_account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}}},"hashtags":[],"change_id":"I1e7fe0720581dfa815b5e0cd2b01379cab2a5f33","subject":"s3: Do not log access keys","status":"MERGED","created":"2025-11-17 06:19:37.000000000","updated":"2025-11-21 15:25:15.000000000","submitted":"2025-11-21 15:23:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"967278","cherry_pick_of_change":967162,"cherry_pick_of_patch_set":1,"meta_rev_id":"da8a5aabcc69430056a7356b5166718386282472","_number":967278,"virtual_id_number":967278,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-21 15:23:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"recommended":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"all":[{"value":1,"date":"2025-11-17 06:43:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":2,"date":"2025-11-20 12:32:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":1,"date":"2025-11-20 12:32:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-17 06:43:00.000000000","updated_by":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"reviewer":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"state":"REVIEWER"},{"updated":"2025-11-17 08:22:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-20 12:32:31.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"669b8e73302de4eb404cd56439e4dd23b8240dc5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-11-17 06:19:37.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"9e67353896461a6e205bad91c8aa59d10ab4279d","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2025-11-17 06:43:00.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5e3f6ffb32d87c5f240f4f44790076a0c7e13e3e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-17 08:22:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/916a9e8324b44bd79a2f4f0073cd8d7c\n\n- grenade https://zuul.opendev.org/t/openstack/build/bf5cdead947841eca567c7e3ccdc989c : SUCCESS in 1h 02m 47s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b75a151157fa43c3b09000393fc7b781 : SUCCESS in 1h 05m 07s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/27401ca0340d48cab9cd9185875d4930 : SUCCESS in 1h 19m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ab01260e476c47b0a55b9b7f3fd4754e : SUCCESS in 42m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8fd91ebed7b1468a90ea730eb8f30326 : SUCCESS in 4m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a0d70213a4e84be99d7825e97b8fe9dd : SUCCESS in 5m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6725a7eecf2c4465bc27e3b16bc9472c : SUCCESS in 4m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/61b06303bb8f4878a4c2f5f9e3e097d3 : SUCCESS in 9m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9bf8cf0bd864d83b9517cf5ae0b1ffe : SUCCESS in 5m 40s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5566aa281da34e6c99654f67b6434714 : SUCCESS in 6m 53s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/322729b4493c4c80ae7998c61eb428da : SUCCESS in 11m 46s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/e9e7b07eb36b4159b11df81c459b00eb : SUCCESS in 1h 55m 05s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/b9f718581fcd43058ea431c43393daee : SUCCESS in 1h 50m 55s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/9da3e27777194f868f31645fc98daebe : SUCCESS in 1h 49m 52s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/275d10cf8d3d40879f01968d8a39ea28 : SUCCESS in 1h 56m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c1f1b15e4ca84c4cacdf4f43c873fddc : SUCCESS in 32m 27s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f6bf4e864c724c60ad9c37dce6f5c9ed : SUCCESS in 1h 39m 24s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/9e7967d2733d45c6b4d2d7e6151a85fb : FAILURE in 25m 57s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/7e966deeb18b488bb378c7ae7a4ef06f : FAILURE in 12m 40s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/07e2110612a6414e969af2c0cf630996 : SUCCESS in 17m 43s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/576ad5d64de241278eac92ddf8ae55d5 : SUCCESS in 48m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"19545a9cd29e946218bcd9ed52f2badda9ca67cf","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-11-20 12:32:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"91523ec2a9c5ffb53da7a96f00c91056a4033bde","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-20 12:33:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bbd29b5e867a75f6813b1a104c7e416cb2e93ac2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-20 14:32:01.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4717d7291c5a48f68363d9afcedaf153\n\n- grenade https://zuul.opendev.org/t/openstack/build/d8c75e88a66740dcb024a151d0ea21a0 : SUCCESS in 39m 46s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f504a23dfc93490a8f9c6add85111a29 : SUCCESS in 1h 00m 01s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/003e79e792d7469db1527ff10e435eff : SUCCESS in 1h 49m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3337631748ba480eb0c1c5615ecd261d : SUCCESS in 37m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/883ab954c51b40b485217acc259c36d6 : SUCCESS in 6m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2874c548f89545589bf26a365d4b604c : SUCCESS in 8m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7e5463fa76e24184a953ee0b0dda6221 : SUCCESS in 5m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9626fa8f4ad4711bb3b34692e2cda74 : SUCCESS in 7m 18s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e550e3fd083242f2b97ad9b0c13a52bb : SUCCESS in 4m 47s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/ace04b86114748fa9a8a8d4f7435cf92 : SUCCESS in 12m 28s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/d55a9c2dd1d549f68d818ffe8de7b9d3 : FAILURE in 1h 13m 32s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/d3780a1ebddb424fadd316ad4f90626d : SUCCESS in 1h 54m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/791d5776e21f4da18a328c97cb799150 : SUCCESS in 37m 37s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e78e822a81324ef1b67f79d8c48c2c98 : SUCCESS in 1h 17m 48s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c5b436b5487b46a2af4040089b977d9c : SUCCESS in 21m 08s","accounts_in_message":[],"_revision_number":1},{"id":"564880605ef7f2bdc7456459da3ab539c8f56dce","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-11-20 14:57:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"765773fd412fb37810b19d6d94e16c93bfcb2d67","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-20 17:12:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0f405b75e7d04e62b24d8130deee0e70\n\n- grenade https://zuul.opendev.org/t/openstack/build/3fd4549d582a42bba60c5a49a4f44222 : SUCCESS in 1h 02m 23s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/60e3e3132d27436c841108a73873b2e7 : SUCCESS in 34m 14s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8b27c2a47a184fddb6543c334e1ab70a : SUCCESS in 1h 28m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/41749cb757714d66b390eddaa668a312 : SUCCESS in 1h 06m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21a2663a0d3a449eab27e615a97a117d : SUCCESS in 6m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f2c75b584bf54d40b24cf955a436a6ef : SUCCESS in 5m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6c80e14cfa2b4a7e92f10929710cb775 : SUCCESS in 6m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/44479a6df8eb43958c4fc6c646152937 : SUCCESS in 8m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b04eef08a2384da49cff17a6ceb4674e : SUCCESS in 5m 59s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c0c544d221db4afdb4d23b02224a9821 : SUCCESS in 7m 53s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/129eb1cfb11943f484a576bf6c6bd494 : SUCCESS in 10m 42s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/506f0173b7f5446c84606dec7347cb46 : SUCCESS in 1h 57m 26s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/a53b98e0124a443daeae8fb091720d6b : SUCCESS in 1h 17m 04s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/ac4608d82aba481d9adfc53cd9f41760 : SUCCESS in 2h 02m 47s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/d9b27a5eb3204d8c91acdb4e216d9de5 : SUCCESS in 1h 46m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/567fc37fc47f4f64b3d09fa6fe4d27d8 : SUCCESS in 55m 53s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/93b25159017243c0a0e7fa054948be75 : SUCCESS in 1h 59m 15s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/6563ff7ea5324a1db10ce078e3040bc1 : FAILURE in 17m 08s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/2da5802fc68044129810a5c4d39f8a0b : FAILURE in 26m 16s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/afc8e763968f418bb15f15f9e82442f0 : SUCCESS in 41m 18s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/43eb63bb44c94178a3137da60a859a9b : SUCCESS in 1h 17m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0a101acc03dc0d7890e03359aa7ecd7edda31b5e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-20 17:12:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e128a4f77f3b28bacef4173d0b0d9b558e395fd7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-20 18:43:43.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ef3e45d8701749079f627352d5f42b78\n\n- grenade https://zuul.opendev.org/t/openstack/build/90a740e1907648f09ce34e17aad74af6 : FAILURE in 15m 05s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/cc39fadb6a2b47238d1baff1a067c5e5 : FAILURE in 13m 56s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8ced630219fe4532898d0fd65ed66acf : FAILURE in 7m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/20318ea6a2f04000bacfc46b2f44b8ca : SUCCESS in 32m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9be460cb50604b1898c512a3138fb6fd : SUCCESS in 5m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/198aa355e5af41d8800ab79d474dac2e : SUCCESS in 4m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c6972a464f6c4478b64e8abf05c5b4a4 : SUCCESS in 5m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09985011db69435fbf9584a8ca8dd536 : SUCCESS in 3m 21s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f3b6f20f8d0b46189245f6e7711600fc : SUCCESS in 7m 29s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/ba18bc5c8c8147ff95981b6db86c8056 : SUCCESS in 7m 45s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/42068f67302f41cc80b0545b69d658d6 : SUCCESS in 1h 21m 56s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/c5cbc1b36e294f0e9003183e6efb9331 : SUCCESS in 1h 29m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/08ca20ffe7604147b188889937bced80 : FAILURE in 6m 07s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/2887555ec36b40f9bb25cce55be50e27 : FAILURE in 6m 20s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/d2c1f72df51541428b1f2cdb1261091e : SUCCESS in 20m 29s","accounts_in_message":[],"_revision_number":1},{"id":"2f61fdd3a5218bb77eaf7c3d4ba4cca93ceb9986","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-11-20 19:24:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bd7d360715e8f6315bec77797e1faa45821bb9d5","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-11-20 20:35:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2df36b558ac18f8d11f751586deb4577afcef87b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-20 20:59:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e8fc60c39dd54f5a9b50e0d16930fd38\n\n- grenade https://zuul.opendev.org/t/openstack/build/c5f4b9f799d84b9ab492e604a37e8d65 : SUCCESS in 33m 50s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/379e709cd53142f3a37f34a8573d8e87 : SUCCESS in 32m 01s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e5c52bdd29894889b29fb085dd624b25 : SUCCESS in 1h 20m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6ce5d5a78eaf4ad48b2f2678fc303192 : SUCCESS in 36m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c5f4ad923e247b28be00adaf0f8b59e : SUCCESS in 5m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c2bcbed34c1a44a78af59c09238a99a0 : SUCCESS in 4m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/52ecbe27650c4991b3cbd5613840344b : SUCCESS in 5m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6af28c4994994ef9850ffb029b7f80eb : SUCCESS in 5m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b42ca1b293f5480ab1c699b55c614c46 : SUCCESS in 4m 28s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/379079fe1aa5479fa17e71e863b6cf57 : SUCCESS in 5m 27s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/4ae8545dbebe4f088e4846a4a5e47ade : SUCCESS in 8m 56s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/4fd8af06bc6a41b58e90c7e20d8a00b9 : SUCCESS in 1h 14m 07s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/f0c768f0578c4b24b241140eda3552d3 : SUCCESS in 1h 19m 47s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/b94a5b1a9bb044aa8912a121ee70c220 : SUCCESS in 1h 30m 31s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/a1068085aef445c8b70b2d362c8656c4 : SUCCESS in 1h 17m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/610a2e5bbdbb46f59175c0348f1cbc42 : SUCCESS in 32m 25s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/3b0718b1ba404c78a0522459dbb97e72 : FAILURE in 13m 16s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/5caed965cb464593a4036597f1ba5468 : FAILURE in 16m 34s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/a0dec6b1a1ea496ca8879cacd67a91b3 : FAILURE in 25m 03s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/98e354c311c847878a6301e41f7f6665 : SUCCESS in 33m 39s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/014478f0f4144878be69f44770ff70b2 : SUCCESS in 1h 09m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"58838470b3807b5fcd12dccc139281e7a6c92e45","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-11-21 10:49:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9f804901de457d5b87a197797af00cdbc0bbbd9d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-21 13:22:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/441826742e2348fc932ae240cad96db6\n\n- grenade https://zuul.opendev.org/t/openstack/build/1ca53495302046859d79153b692717e0 : SUCCESS in 41m 56s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e4cc89c0e2d84c1f833d625de6494110 : SUCCESS in 40m 52s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7c2d75fdaeb947dfbccb21f8110f18a5 : SUCCESS in 1h 18m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2c92297ec57a4c658380769cdeb2903e : SUCCESS in 1h 50m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc003dcdf9c744a8b739c96e430947cc : SUCCESS in 7m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ca1e6cef7c4248f4b48056e917aa8184 : SUCCESS in 5m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/65037f44466a4c59b678e3b0fff92627 : SUCCESS in 5m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/55ea87599c7c4f2dbd972dfa307f168b : SUCCESS in 10m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53af765da65a449ba7ca5897968f92ac : SUCCESS in 6m 22s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d5c5be5cb93b4eaea203d1a8e7081a31 : SUCCESS in 6m 40s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/ed4a20b0367d4b1fa0ae561492218829 : SUCCESS in 9m 53s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/f0566483249648558f1d89e531b56fa2 : SUCCESS in 1h 55m 42s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/681e1c893e4a4eed8b536c24aff22869 : SUCCESS in 2h 08m 00s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/a911a90652514339860eaca6119a5cff : SUCCESS in 1h 17m 16s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/9b2da8c4be8f48f8a426f27435d43e4c : SUCCESS in 2h 14m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0ad1566b6fdc461890a9e68a52df8d7c : SUCCESS in 56m 32s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/1f6cc9af3d4e4dd49c191b49e049f17b : SUCCESS in 1h 44m 24s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/c88b32d60c2d4322acdc217221cdcea8 : FAILURE in 30m 43s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/25220e4b06bd4cf4b0d34464d515f609 : FAILURE in 25m 53s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/63cd90b1ab1c46b695d5d5e7ca88899e : SUCCESS in 31m 40s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/12f768da2ca64fa687a16a5aa0a6d656 : SUCCESS in 1h 03m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4202af707da844af41f327182395641c9f411dc5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-21 13:23:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"47b16d1667538a6431a6534c213c03cd1db07013","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-21 15:23:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b80c54303cc94f60989dc3f48d4de62a\n\n- grenade https://zuul.opendev.org/t/openstack/build/ce12ca4c3f674fabacc965c77879866c : SUCCESS in 31m 01s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7eb6bc191c9a44008d703ba56c3eaa90 : SUCCESS in 38m 55s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/be5b59100a4244f4953a7814295dfeae : SUCCESS in 1h 24m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d0588590abb1432fae22e8a53af729f5 : SUCCESS in 33m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/feebb68af27b4f13918abdcf9737e400 : SUCCESS in 3m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/641101a84b714147bfcd1159278caa63 : SUCCESS in 4m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3c40fbc2bcfb49d29f1130dc5593187d : SUCCESS in 5m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52f74a45bc7c48e98fc4457dbc6aff19 : SUCCESS in 6m 25s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5865607ce1a14eceb902fe80edee8f76 : SUCCESS in 7m 34s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/a54f476100cc4992bf88be531a07159b : SUCCESS in 7m 14s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/5c2575b91b2d4e5c9c5954626702de3e : SUCCESS in 1h 52m 21s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/9fffc758bbaf4d7198c77cde700e2a05 : SUCCESS in 1h 47m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/002f6fc21eba4629ba2c4e634ecfe419 : SUCCESS in 32m 37s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/7d19a04475dc49e9a7a89c0caa3f464d : SUCCESS in 1h 39m 21s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/78a5695097a04ac5bac334b62992edd9 : SUCCESS in 17m 16s","accounts_in_message":[],"_revision_number":1},{"id":"c68c4ad351ae0dfa9c5035c938da72ee7368990e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-21 15:23:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"da8a5aabcc69430056a7356b5166718386282472","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-21 15:25:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d80e5d965e049f5989a93fa28d692c3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/067dc15831ba484bbdcca58053db15ac : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8ea9ba973f9564cdec1ca9717d90c1301d912edb","revisions":{"8ea9ba973f9564cdec1ca9717d90c1301d912edb":{"kind":"REWORK","_number":1,"created":"2025-11-17 06:19:37.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/78/967278/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/78/967278/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/78/967278/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/78/967278/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/78/967278/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/78/967278/1"}}},"commit":{"parents":[{"commit":"ac5505179f51cb88d83ec40113d5c6801ef397b3","subject":"Lazy update S3 URL on credential rotation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ac5505179f51cb88d83ec40113d5c6801ef397b3"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-11-11 17:10:52.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-11-17 06:19:37.000000000","tz":0},"subject":"s3: Do not log access keys","message":"s3: Do not log access keys\n\nRecent credential change for s3 has introduced one debug log\nwhich can leak access information of s3 store. This change avoids\nlogging URI in the log.\n\nCloses-Bug: #2131085\n\nChange-Id: I1e7fe0720581dfa815b5e0cd2b01379cab2a5f33\nSigned-off-by: Abhishek Kekane \u003cakekane@redhat.com\u003e\n(cherry picked from commit 81a838cc24da73e86e2cc85e06b21265fb960875)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8ea9ba973f9564cdec1ca9717d90c1301d912edb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8ea9ba973f9564cdec1ca9717d90c1301d912edb"}]},"branch":"refs/heads/stable/2025.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
