)]}'
{"id":"openstack%2Fglance~968203","triplet_id":"openstack%2Fglance~master~I96730704c241d94d02eac668aad2e5ab65f63d1e","project":"openstack/glance","branch":"master","topic":"eventlet-removal","attention_set":{},"removed_from_attention_set":{"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2026-01-08 17:51:08.000000000","reason":"Change was submitted"},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2026-01-08 15:45:38.000000000","reason":"\u003cGERRIT_ACCOUNT_8122\u003e replied on the change","reason_account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}}},"hashtags":[],"change_id":"I96730704c241d94d02eac668aad2e5ab65f63d1e","subject":"Functional tests migration - test_gzip_middleware.py","status":"MERGED","created":"2025-11-24 14:56:31.000000000","updated":"2026-01-08 17:52:23.000000000","submitted":"2026-01-08 17:51:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"968203-eventlet-removal","meta_rev_id":"37d28ea3ff549cfaeb4723b484e157bf21f7211a","_number":968203,"virtual_id_number":968203,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-08 17:51:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":2,"date":"2026-01-08 15:45:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":1,"date":"2026-01-08 15:45:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-24 14:58:49.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2025-11-24 15:43:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-11-24 16:13:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"16ed3b92389186926d96b09215d7477bf7a664ce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-11-24 14:56:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"532182d05903f93bde4db86c6fa544288eecfb27","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-11-24 14:57:07.000000000","message":"Topic set to eventlet-removal","accounts_in_message":[],"_revision_number":1},{"id":"70d27e89ada494f9cf4928fe52c1498286629e44","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-11-24 14:58:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"32568b3510d0ef58a079630b6cada27a2566b60d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-24 15:43:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ae32713da116441f845ea0058d107de2\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/fbb7dc2d15554683a88c874047c65714 : SUCCESS in 6m 11s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/95ba18e1d4914e8d9e7939c4aeea0222 : SUCCESS in 6m 18s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f4c877a12030450e90d947f14741ea70 : SUCCESS in 9m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"daee1d02896520913231b73c5ca3497f7450a820","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-24 16:13:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/24b62c8202884a19beeabdf479db11f2\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/57ad2021afea433ba5f1ffad7ef1eead : SUCCESS in 1h 10m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7f4f926447c34e71afe207a1096f19e5 : SUCCESS in 52m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c52ecb153f294919aa30b3674c03f0cf : SUCCESS in 5m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/821bf1933d164ff3a1172e4edcedf0f1 : SUCCESS in 4m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4d74d32987794622b49b48fd234a0107 : SUCCESS in 6m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/335ad19572e2404f800e3fb4635731e2 : SUCCESS in 3m 19s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/577705a9346444058640f88b9c88cab7 : SUCCESS in 6m 37s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/bc506bc85216439b8a61412423aed13d : SUCCESS in 10m 49s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/fb7d8b49d2e74c3097f3d8fea3c5243a : SUCCESS in 28m 52s","accounts_in_message":[],"_revision_number":1},{"id":"12a7c5a4fd3b56e8c29dd4e92110170afa035be5","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-01-08 15:45:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7e6274064d3520fb966100647257402c73aec914","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-08 15:46:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"28d0c976d47b29e96dd7483591263a284f1b4e07","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-08 17:51:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8e93c3df38cc48c693238ddece557d25\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/261a7dff4e0644fa8f213f7bc96a7e2d : SUCCESS in 45m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9e48f75ab27d440795c0fcb0137c7855 : SUCCESS in 35m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ddd4c28fc7674adea0aae727a9ac014f : SUCCESS in 5m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/77992a5f00d7428184f6f9f9334413c4 : SUCCESS in 3m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/518407eb769a462a81c03819392f55d1 : SUCCESS in 6m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09f3062183014763801ebf8ff146c46f : SUCCESS in 3m 40s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5f29ddf5975a43ffbbc017ef31bb9d7b : SUCCESS in 8m 10s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/adde5343f0374c1f876c4b48a213421e : SUCCESS in 6m 46s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/5cb74d0b81a1411ab6b20a7deb43667f : SUCCESS in 22m 00s","accounts_in_message":[],"_revision_number":1},{"id":"9d01bee1c4c0e768b7909fb42f37b3b4600ed0db","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-08 17:51:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"37d28ea3ff549cfaeb4723b484e157bf21f7211a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-08 17:52:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/58c100b393ec487fba0049540dc21a33\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9dbd927e8c314a2f99f4f34ff3ca94fe : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d573598598dbdfcf7b141842077e773c280e8919","revisions":{"d573598598dbdfcf7b141842077e773c280e8919":{"kind":"REWORK","_number":1,"created":"2025-11-24 14:56:31.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/03/968203/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/03/968203/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/03/968203/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/03/968203/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/03/968203/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/03/968203/1"}}},"commit":{"parents":[{"commit":"f418d3f00d7c1e2bd4d1458b0c500716eb0ba653","subject":"Merge \"[docs] Add s3 upgrade scenarios\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f418d3f00d7c1e2bd4d1458b0c500716eb0ba653"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-11-24 14:51:42.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-11-24 14:56:06.000000000","tz":0},"subject":"Functional tests migration - test_gzip_middleware.py","message":"Functional tests migration - test_gzip_middleware.py\n\nMigrated single test to use SynchronousAPIBase class.\n\nChange-Id: I96730704c241d94d02eac668aad2e5ab65f63d1e\nSigned-off-by: Abhishek Kekane \u003cakekane@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d573598598dbdfcf7b141842077e773c280e8919"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d573598598dbdfcf7b141842077e773c280e8919"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
