)]}'
{"id":"openstack%2Fglance~968432","triplet_id":"openstack%2Fglance~master~I24cacde5a9b2209671e9283cd38548ac05c75ac3","project":"openstack/glance","branch":"master","topic":"image_decompression","attention_set":{},"removed_from_attention_set":{"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2026-01-07 03:26:16.000000000","reason":"Change was submitted"},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2026-01-07 01:09:49.000000000","reason":"\u003cGERRIT_ACCOUNT_8122\u003e replied on the change","reason_account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}}},"hashtags":[],"change_id":"I24cacde5a9b2209671e9283cd38548ac05c75ac3","subject":"Fix ZIP file extraction in image_decompression plugin","status":"MERGED","created":"2025-11-26 08:07:33.000000000","updated":"2026-01-07 03:27:11.000000000","submitted":"2026-01-07 03:26:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"968432-image_decompression","meta_rev_id":"61116208e9f07f5228d73b4c17eb547d97d4e9cb","_number":968432,"virtual_id_number":968432,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-07 03:26:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2025-12-17 04:05:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2026-01-07 01:09:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2026-01-07 01:09:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-26 08:28:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-11-26 10:16:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-12-08 15:18:59.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2025-12-17 04:05:37.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"}],"messages":[{"id":"ba8917a38b315cbebbf68fc9833a81ef8bcbcc69","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-11-26 08:07:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"64e76e2a62e19ea2bd5c5a3aedfb9e620e727123","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-11-26 08:09:03.000000000","message":"Topic set to image_decompression","accounts_in_message":[],"_revision_number":1},{"id":"3e284d0e45b78c394ec30007643945d2df9ba7b4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-26 08:28:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cbe9b7337b50455ba35de7df304813bb\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/db91ff8fdf324b6d81e7b4ff18391df8 : SUCCESS in 10m 54s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/22678a5ec17b46c3a0eb1c6bae44222e : SUCCESS in 12m 12s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d5a13af3269e45ea85f2c25408dcc07b : SUCCESS in 18m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d795a404d81d558207be6e19cea2a04eb67f7431","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-26 10:16:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e8b06c5e51e54f63b4e559d40e16bb80\n\n- grenade https://zuul.opendev.org/t/openstack/build/c8c32561d0b847d8bca54fa169926246 : SUCCESS in 59m 38s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/08a00092b39142bcac4e220ffe9f92e8 : SUCCESS in 56m 41s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3929fdd938df484b984ff76576cb0e46 : SUCCESS in 1h 07m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3ac49e18d5334edc92076ae4876ced0b : SUCCESS in 1h 02m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3549c883c8d345b19c098c576336b2a7 : SUCCESS in 5m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/177b82a548b54a34b2918919261b6616 : SUCCESS in 3m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/17502ee59aab4ae380541b702ebbd64d : SUCCESS in 8m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae695655d8594eca86e9be9442563fa5 : SUCCESS in 5m 47s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3db8ef42d0ab486ebb54f76a65b3f667 : SUCCESS in 8m 32s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/77158927f51e4344860812994df77990 : SUCCESS in 8m 56s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/1a3e9e39b0f74c7284f180a51fd124d3 : SUCCESS in 2h 06m 57s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/ae9603b81cef4cb497846c73f1acffb4 : SUCCESS in 1h 48m 52s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f9bcf231e1be48c387fc879745fa04e7 : SUCCESS in 1h 49m 36s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/69c6636bc00f41d79e9a0a356277d771 : SUCCESS in 2h 06m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/98d1291e2a274f24817ede6ec1bcceb4 : SUCCESS in 56m 45s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/4d5eaabb01524f49a63b15e66a8ae97f : SUCCESS in 1h 51m 36s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/304e956965bf4dc6a3daaab266e656af : FAILURE in 23m 56s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/043c386467ab44868697a49616ddd41d : FAILURE in 12m 43s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/1f041d99509f48eeb32dc89ca70cbb55 : SUCCESS in 35m 24s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/76d697c7089f46eab2f442bf258e6e48 : SUCCESS in 48m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8f80c8aa3b798c45efe52728f8b042a8485c6cab","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-12-08 15:18:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"19235539d6691a8c1295adee0e5b7d864697b3d4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-12-08 15:46:02.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":2},{"id":"13a623bf114fdfe7c586570b46724f392733420b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-08 15:57:26.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/28e1443e9a2d433cb4008eda27b31a7e\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/81f6d2e070ee4b00ae40b72af69ecd3f : SUCCESS in 6m 21s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/16c6b3f82e5144b79a7a93e6793300a1 : SUCCESS in 6m 37s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/5b8348af4b434df29966ed2b714a3374 : SUCCESS in 8m 34s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2b3d406c7e3cf48a3256e55bd799e6fb802de8d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-08 18:44:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d1770ed2fea44feb9aa826792d9f2266\n\n- grenade https://zuul.opendev.org/t/openstack/build/7e7bf5124525486ca88a3871efd4ea00 : SUCCESS in 59m 47s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/89d2849fa0b84c8fb5fa4fc43459261d : SUCCESS in 58m 43s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e834cf9417bc428e8d3c404b4753233c : SUCCESS in 1h 33m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/00d8d89252e9441693bf89bc496767d5 : SUCCESS in 1h 03m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8575e7350f544a996e3642c9c8ff1fa : SUCCESS in 2m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/53942739138a430ea4e978af664ed0ef : SUCCESS in 4m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/98a2beed38a24b0d87238617f8794286 : SUCCESS in 8m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a067c80c3f484276a0dab2c723c6a5d3 : SUCCESS in 5m 14s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/418115dec4fa4924ba9c572f8b99b9d7 : SUCCESS in 6m 40s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/eb62e85bcc3c49f6a88dec458211e35c : SUCCESS in 8m 54s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/3acfaf4b5fb1495eb00e8d6a36835964 : SUCCESS in 1h 49m 54s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/26d44a9a387d4921a82d62b3f3d3c766 : SUCCESS in 1h 04m 19s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/79ad134f6dac4753bdaf433b35ab8769 : SUCCESS in 1h 33m 37s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/e05119fea5634e88bc73ea320d6f5951 : SUCCESS in 2h 45m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d43a050d4bd54258ae68b7c30a10b77a : SUCCESS in 58m 53s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/127863693f89478380ea179904831280 : SUCCESS in 1h 59m 57s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/2cd5f492500c488f9feb0d7f9217310f : FAILURE in 20m 17s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/a28e9019b57c4b9197813ff8cb583c9e : FAILURE in 26m 04s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/642531b184db43129c23dee12924bcb7 : SUCCESS in 16m 56s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/a9eb08bdc3e4494c92b5239f5fed6cc9 : SUCCESS in 33m 37s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c21d44c2d6069e741e5c81b60c2eb1c599053a52","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-12-17 04:05:37.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b883419525a88b0de7cf665f79f8fe4800e43edc","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-01-07 01:09:49.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"0ded1e13f1b16c998d08c5d7d77ec1e7432e713f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-07 01:10:15.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"277ee4b686e67f3d8335aea69760199b778156fa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-07 03:26:15.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d51b23b9ea054f48846b5c477b4a0278\n\n- grenade https://zuul.opendev.org/t/openstack/build/45fd6ad5addc49cd892e0001d259edcf : SUCCESS in 56m 42s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9f860b853c7c46d6aa76a6aa4de1238f : SUCCESS in 57m 00s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/53bbadbda07e49e39fe89279f1ad9fd5 : SUCCESS in 1h 39m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/034370f0b70447efaf622ac1ccdb4177 : SUCCESS in 1h 06m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a9790d3eeb8410a8521f457175ba9e8 : SUCCESS in 5m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/89d013c762a9454ca06a7a64b07163db : SUCCESS in 2m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/eb79c4478669496f8d5402323cd9632c : SUCCESS in 4m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c51e7a6a1fa4ccf951758850fcf0ee1 : SUCCESS in 2m 59s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e90abfe54b2a470dac3650d7973cd2e9 : SUCCESS in 6m 48s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/90db16a779d246cf96b7373916290dcb : SUCCESS in 9m 08s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/0925074d68a143fd868d4e7f01e2188f : SUCCESS in 1h 50m 18s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/e70748d2cd0f4df7b0ad15ef88486e44 : SUCCESS in 2h 11m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/66d953c9794b48a1984424bb6d328d18 : SUCCESS in 55m 46s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/2756586549274d93b2603d31c70bdfea : SUCCESS in 1h 42m 07s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/348ed5ba693648d3a301495d36d81dae : SUCCESS in 32m 32s","accounts_in_message":[],"_revision_number":2},{"id":"0e6268dc151b288262e1f7d2bb62d9650c31ede3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-07 03:26:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"61116208e9f07f5228d73b4c17eb547d97d4e9cb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-07 03:27:11.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0186a8857d50494295ec8393f67eb43e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc4a65ddf57a4e359ba0852cba15152a : SUCCESS in 42s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"76e7b27050dae4636208bd437550659e2b8305a3","revisions":{"caff3460c9a3ce78dfe8494871d9c2ae79f4e4d2":{"kind":"REWORK","_number":1,"created":"2025-11-26 08:07:33.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/32/968432/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/32/968432/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/968432/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/968432/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/968432/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/32/968432/1"}}},"commit":{"parents":[{"commit":"ec16dfb96fcf09834f3fa028c9e57c32dbc3242e","subject":"Fix LHA file decompression context manager issue","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ec16dfb96fcf09834f3fa028c9e57c32dbc3242e"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-11-26 08:00:43.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-11-26 08:00:43.000000000","tz":0},"subject":"Fix ZIP file extraction in image_decompression plugin","message":"Fix ZIP file extraction in image_decompression plugin\n\nThe zipfile.ZipFile.extract() method expects a directory path, not a\nfile path. This fix extracts to the parent directory and then moves\nthe file to the destination path.\n\nAlso updated documentation about the single-file ZIP limitation.\n\nCloses-Bug: #2132963\nChange-Id: I24cacde5a9b2209671e9283cd38548ac05c75ac3\nSigned-off-by: Abhishek Kekane \u003cakekane@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/caff3460c9a3ce78dfe8494871d9c2ae79f4e4d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/caff3460c9a3ce78dfe8494871d9c2ae79f4e4d2"}]},"branch":"refs/heads/master"},"76e7b27050dae4636208bd437550659e2b8305a3":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-12-08 15:46:02.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/32/968432/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/32/968432/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/968432/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/968432/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/968432/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/32/968432/2"}}},"commit":{"parents":[{"commit":"796b244ae4d148b9a1d090f853eec8f9f84152ac","subject":"Fix LHA file decompression context manager issue","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/796b244ae4d148b9a1d090f853eec8f9f84152ac"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-11-26 08:00:43.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-12-08 15:42:42.000000000","tz":0},"subject":"Fix ZIP file extraction in image_decompression plugin","message":"Fix ZIP file extraction in image_decompression plugin\n\nThe zipfile.ZipFile.extract() method expects a directory path, not a\nfile path. This fix extracts to the parent directory and then moves\nthe file to the destination path.\n\nAlso updated documentation about the single-file ZIP limitation.\n\nCloses-Bug: #2132963\nChange-Id: I24cacde5a9b2209671e9283cd38548ac05c75ac3\nSigned-off-by: Abhishek Kekane \u003cakekane@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/76e7b27050dae4636208bd437550659e2b8305a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/76e7b27050dae4636208bd437550659e2b8305a3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
